)]}'
{"id":"openstack%2Ftrove~914231","triplet_id":"openstack%2Ftrove~master~Ifbe862020f7c2b1f6684164d2a26bb5ceb37bec3","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2024-04-03 01:52:19.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-04-02 03:04:30.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Ifbe862020f7c2b1f6684164d2a26bb5ceb37bec3","subject":"Enable network_isolation by default","status":"MERGED","created":"2024-04-02 01:58:59.000000000","updated":"2024-04-03 01:54:15.000000000","submitted":"2024-04-03 01:52:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914231","meta_rev_id":"e9ff1322156810551bfa14b88f874ec3c33acaa5","_number":914231,"virtual_id_number":914231,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-03 01:52:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"all":[{"value":2,"date":"2024-04-02 15:19:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-03 01:06:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-03 01:06:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-02 02:10:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-02 02:43:28.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2024-04-02 03:20:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c3739f250dce129bab034adc3e5c5de5d8f588de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 01:58:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"966a9ba722af0cb04874f1e2c70c3d3b97be85b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 02:10:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01a66dabf7b34fd2a07f07b0c68daa72\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85f54c423d2a41fd8504246e6d27a83e : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/277a9271a2e749e58ad15b9ec0bc7710 : SUCCESS in 9m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e09a1a8b758092c3ac6f67ec27e51d2b5161851","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 02:25:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e5e9a986dcdf0972d821f2c3bb76da38cf9c6bb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 02:27:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ff63d26df7e0fe3e8984cae117fbc29b2050b9bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 02:38:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/537cecc3d9a7420cb2c58ba47cf0cbae\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60195330527e404ab652c3430c70c38a : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/001fb2d0213048a8b0e1c70a86161d8a : SUCCESS in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"940e281eab4df1be1e08c5025319373bfcf7134a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 02:43:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8609985e77cbbc610cf9465190a7e8fdfcd26043","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-04-02 02:53:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"96a8275ffa42fab214261cb65760b0a95ffa2dce","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-04-02 02:58:49.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"34805d9225d71e1615ef822638683a6e9a63c4a0","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 03:04:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"81abd76ad1c38ce7edc7e03f4cb8c63c2c1d7340","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 03:20:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b2a1480f1dc40419c6d92c157fda61a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11b295f3bacd4188a41c1a170c0cd760 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64ab561528514de6bda206e1924f8e7d : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3dafa0983e484b7980837a3adc2d4c5e : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b388b92eabb452eb238cef6023b6f50 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d7450f8d2c1f4378ba9ea15ba991e7ac : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef5c4d85249940edba33783bcc9e2ee8 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/378cfe3fd5474545aebca664abb6c7a7 : SUCCESS in 5m 53s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/954df47264dd4c17988fd7bebed0e288 : SUCCESS in 41m 47s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/2c02f24fba0b452cb5b89b127fb69cc8 : FAILURE in 35m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/73adef0eccdc48b0a0a1447e614f5438 : FAILURE in 30m 05s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a217258347ee4546812d8f9ea109404f : SUCCESS in 39m 23s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2ffe1203c96c4309845d334d50d82d2e : FAILURE in 34m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6fee7eb6e5b64bc9ae452210bec42dc7 : FAILURE in 35m 36s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/8ddcb5456824445ab331cdebecff4b09 : SUCCESS in 45m 08s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/271b7541d0d341c1ad3ad1823283d579 : FAILURE in 33m 50s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/deef89c9fb9148e79b400c60bb5fa57c : FAILURE in 32m 22s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b8e005a2d4584e128045a852fa6ed4c2 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":3},{"id":"cac84b462c359ade7a789abba58feaafc4a368a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 15:19:29.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"74e8c1e2fc2db863a61112fcd5c86834462ad359","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 15:31:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b9736aff63f44d185d6451dcf17ba53\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d07521627cde4924acae56b08d423005 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/af918467cd3c40e49937905d55bad741 : SUCCESS in 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9ab247d63b257ed481d3530d07d71cc4872c1e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 17:39:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08d440b2310640cd84d3f349dc88b497\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa79bd6fd5484dfc80739f22f453eeed : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/904a8ab49d2a4a328c3d35f8e74c3ecc : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e04449c6fa8a46df960a8896d8d8f406 : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d49287b88914c7b8c106b13e1c20f85 : SUCCESS in 6m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/270eb9ba07254780bf44f58350e68235 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fb6cbb02ad5450c9195891c72ace726 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/382e85cd0aa84388816fc20952f5a013 : SUCCESS in 6m 00s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/34ac53573c3b47d88db64907a891ded4 : SUCCESS in 39m 58s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/119fc23bc18d4620b94fea271c7b702d : SUCCESS in 1h 36m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/73184ea67d9446db871c2ca71d082d68 : SUCCESS in 1h 57m 27s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4e66ddc6f0c74403bcb02c53c5bbd166 : SUCCESS in 41m 24s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a562afcebbc04fc2b262e6403c17f04e : SUCCESS in 2h 18m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/056758d5a53f4120bb738262919c8042 : SUCCESS in 1h 38m 08s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/434005fe45ff4ca1b5d77c70ac503d47 : SUCCESS in 43m 20s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/b152b5f4c61e452fb4c969f5a6f794ad : FAILURE in 1h 12m 01s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/efe01ccab52146a5a99a8f7d54b86041 : SUCCESS in 1h 49m 26s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/697ca666be1b46bc8e87e56aa0114c7c : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":4},{"id":"1bb063f48b79875b6d35d066986b3e047cc07d01","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-03 01:06:10.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"03387e3b154b3c462355a3e8a9d2e9632d428842","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 01:06:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"116a78135cb5d738fe4eee055bb7744a848f84c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 01:52:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ba20cba18ec4a148e9eb445bc7cd7dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba6a3a0bef114046a4c579b2dbda0934 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acce9eb3503e44ad908865fe1d5bc22a : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/87b0df95ca904df59d93e4f2145581ad : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fac907201ceb481e838c808955ab38f6 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b36e199d37e401a8b771572bdd926c1 : SUCCESS in 5m 26s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/4c483b283b41445dad5b547bb1523393 : SUCCESS in 43m 56s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0b5106aed07b40d08002a8b2a5970845 : SUCCESS in 40m 28s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/1325e5472bb742c68bd69b4dca2fa896 : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":4},{"id":"326fc40e0405c0282dec0f9c0c519742a455fb2e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 01:52:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e9ff1322156810551bfa14b88f874ec3c33acaa5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 01:54:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/273511d2f69947f68f520af28e8aadb7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba95722092684ebc9e4aaaaae57fb7ee : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a3d0070ce5c4148a9974fb219a639be : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"421721a3e1c766b3e9f4aeabd705d2dad32a339b","revisions":{"130df4eb03d0d3e6341cada379b457712c8cead6":{"kind":"REWORK","_number":1,"created":"2024-04-02 01:58:59.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/31/914231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/914231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/914231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/914231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/914231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/914231/1"}}},"commit":{"parents":[{"commit":"e19a7d4e50edc399b83d57e905a34f96fafcad1d","subject":"Merge \"Add Datastore Version Registry Extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e19a7d4e50edc399b83d57e905a34f96fafcad1d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 01:56:35.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 01:56:35.000000000","tz":480},"subject":"Enable network_isolation by default","message":"Enable network_isolation by default\n\nWe already tested this feature on our CI for a whole cycle.\nit\u0027s ready to enable feature as the default value.\n\nChange-Id: Ifbe862020f7c2b1f6684164d2a26bb5ceb37bec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/130df4eb03d0d3e6341cada379b457712c8cead6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/130df4eb03d0d3e6341cada379b457712c8cead6"}]},"branch":"refs/heads/master"},"ebb30d629105202c49ea5e635514d358003687b6":{"kind":"REWORK","_number":2,"created":"2024-04-02 02:25:54.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/31/914231/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/914231/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/914231/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/914231/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/914231/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/914231/2"}}},"commit":{"parents":[{"commit":"e19a7d4e50edc399b83d57e905a34f96fafcad1d","subject":"Merge \"Add Datastore Version Registry Extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e19a7d4e50edc399b83d57e905a34f96fafcad1d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 01:56:35.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 02:24:09.000000000","tz":480},"subject":"Enable network_isolation by default","message":"Enable network_isolation by default\n\nWe have already tested this feature on our CI for an entire cycle.\nIt\u0027s ready to be enabled as the default value.\n\nChange-Id: Ifbe862020f7c2b1f6684164d2a26bb5ceb37bec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ebb30d629105202c49ea5e635514d358003687b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ebb30d629105202c49ea5e635514d358003687b6"}]},"branch":"refs/heads/master"},"ce25a41d3527bb54df7310ed83cba4e04a4cecd4":{"kind":"REWORK","_number":3,"created":"2024-04-02 02:27:31.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/31/914231/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/914231/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/914231/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/914231/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/914231/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/914231/3"}}},"commit":{"parents":[{"commit":"e19a7d4e50edc399b83d57e905a34f96fafcad1d","subject":"Merge \"Add Datastore Version Registry Extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e19a7d4e50edc399b83d57e905a34f96fafcad1d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 01:56:35.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 02:27:25.000000000","tz":480},"subject":"Enable network_isolation by default","message":"Enable network_isolation by default\n\nWe have already tested this feature on our CI for an entire cycle.\nIt\u0027s ready to be enabled as the default value.\n\nChange-Id: Ifbe862020f7c2b1f6684164d2a26bb5ceb37bec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ce25a41d3527bb54df7310ed83cba4e04a4cecd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ce25a41d3527bb54df7310ed83cba4e04a4cecd4"}]},"branch":"refs/heads/master"},"421721a3e1c766b3e9f4aeabd705d2dad32a339b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-04-02 15:19:29.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/31/914231/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/914231/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/914231/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/914231/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/914231/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/914231/4"}}},"commit":{"parents":[{"commit":"c0cd2a081539445619fa462670f58ae409ff4957","subject":"Merge \"fix error when build mariadb backup docker image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c0cd2a081539445619fa462670f58ae409ff4957"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 01:56:35.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-04-02 15:19:29.000000000","tz":0},"subject":"Enable network_isolation by default","message":"Enable network_isolation by default\n\nWe have already tested this feature on our CI for an entire cycle.\nIt\u0027s ready to be enabled as the default value.\n\nChange-Id: Ifbe862020f7c2b1f6684164d2a26bb5ceb37bec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/421721a3e1c766b3e9f4aeabd705d2dad32a339b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/421721a3e1c766b3e9f4aeabd705d2dad32a339b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
