)]}'
{"id":"openstack%2Ftrove~914232","triplet_id":"openstack%2Ftrove~stable%2F2024.1~If4bd6f0bd950331d1424bd2078d22f7a86f5b628","project":"openstack/trove","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-04-03 01:05:43.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"If4bd6f0bd950331d1424bd2078d22f7a86f5b628","subject":"Change settings for stable/2024.1","status":"MERGED","created":"2024-04-02 02:56:35.000000000","updated":"2024-04-03 01:54:08.000000000","submitted":"2024-04-03 01:52:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914232","meta_rev_id":"18408dde4783cc773461a7529766bb2dc78a7e04","_number":914232,"virtual_id_number":914232,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-03 01:52:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-03 01:05:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-03 01:05:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-02 03:10:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-02 03:30:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1acff48f5ba59092115ae21ffa27c62cdaf2b72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 02:56:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da57167bdb170bf1866c65c41eb11eaea3188e2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 03:10:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdbedf7dcddf4327b671ce0d903f7a13\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/635eed5965b74b609d60ce22abb9d122 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2fa07ac7d38f48d69021a3429e6d4ff5 : SUCCESS in 9m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"073a353ac7770707cfa035a0f80705ea0c84d748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 03:30:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/438f15ad5825468bb4dac89fdb48acdd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8602ba7c77a4526b87189ef039420a8 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5abba3c6fd63491ab0d168e4a65007d1 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fbc8c0d0cad4154a372dcd12cbe3c80 : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1b1ead705e945eebd7b2eac6af03a01 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/310a4c6f901f49a78b146a251553bb34 : SUCCESS in 5m 14s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/cf0f8add28ea4edf94d5c00893f13480 : FAILURE in 19m 28s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/c534cd1facbf4a9dbe53856d9fe912cd : FAILURE in 30m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/5e2e48c09591439493873e97a2a82882 : FAILURE in 20m 32s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f084e30ee21640079a06d4a08ddd175a : FAILURE in 28m 17s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/741005ded6554ecdb58ce08b16cffcf3 : FAILURE in 30m 03s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/eb58d4e79e4b456297124d6e7031e267 : FAILURE in 28m 17s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/0b0b96ad90ea44b4948e53c4b16482d9 : FAILURE in 25m 58s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/6bad5d6d1b624eefafebb5da093ff546 : FAILURE in 28m 19s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/7def21211f0c4768b12842de3ada675a : FAILURE in 29m 17s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff58986626834f6e82a063c10de75a16 : SUCCESS in 4m 53s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/ca33dfcfe5564719bf03be12a67ed4fa : FAILURE in 4m 19s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/813feda1bb054e07968fa44677ab1bd0 : FAILURE in 8m 21s","accounts_in_message":[],"_revision_number":1},{"id":"371a8248c2a38b8a9489cc6192cd435182ae7f81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 05:57:57.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1453c758b919791ef95f3f4722935cae8c8954ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 06:05:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21df3a7d8eef4dee8125b228b54fdc16\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2407c0f42cae40169d085dcc700b60da : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4afc31fe027348da8f70530b4d441ecc : SUCCESS in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a71c83e6b92abcc947e63c6fb37ff0d65bad6801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 06:41:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/178f19074bec4b6c86234ea728a2af68\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d067831a5374e1f9320bfc14e58a091 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a17831e1c6a4fbf80e5858996b3d7e7 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c27a0aa144f469b941c67f288a55146 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e527f54eab84bedae80a073d55d0145 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/487a412215a4481792d05c1b7c963747 : SUCCESS in 4m 36s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c9c420a3e0fe45febfcfe256f31e0e6c : FAILURE in 22m 20s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/555e6f49b73546399e0b9857d784b9bd : FAILURE in 20m 08s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/3339fb1aaed947e6871bcc71a81ee1af : FAILURE in 21m 55s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c6336d6b4acf464d9b18ec4166d0d686 : FAILURE in 20m 40s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c2e33452663143f8a158687d6d6eb5c0 : FAILURE in 21m 03s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d2e2f500ebf448a79e69619ea10ea91a : FAILURE in 20m 52s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6c30354f877343078fedd82d42e0bd76 : FAILURE in 19m 50s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/77c495b8695b42b781c4e75ba252178f : FAILURE in 22m 03s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/4d2658b45dfd4f10a46fe694b27fdaf6 : FAILURE in 22m 53s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4f6addabeb8a4111ad939a17dedcb02d : SUCCESS in 5m 09s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/d8a6f98b4c6043728f62c7a597aa9e88 : FAILURE in 4m 00s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/faded7a5117a41928800e539af81369d : FAILURE in 6m 47s","accounts_in_message":[],"_revision_number":2},{"id":"947121ebd9681a7a853c72aa15d448a5d07b4d46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 14:16:05.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a67128c5c887dae071c4a570974ce905e772e526","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 14:31:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea372960c7f64274ab15424c8306209c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2da65b523b54c9b9ca0add5aa04005d : SUCCESS in 13m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fc3ec4f61e54437b89dc3e5219464d20 : SUCCESS in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8050d5a122f606aeedc565228b7348a214104c98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 15:17:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"730a3d60ffface4a2965503acc48cd504b5b8674","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 15:18:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0b9540cbbaf2b92188c1fab205c9625e2467c783","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 15:31:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7acab218c634203aa721c96536fe4db\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/66c8dbbf49c6455ca7dd2bda6f496a0a : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8f2dc03ce1c84f4b921a701d8ebd28f1 : SUCCESS in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5836b9e55f07693db57ce2eb3f05e4f4a4ae3170","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 17:18:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e3c755befe94d1faa1ddf22fd4fc4dc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f76f429934641acbc502a64a53702d8 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c42a3e72a3a847a8b24b6555c98f0683 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4528f6d2b924e7c828ae4fa9e97ee6a : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a21d2fe48784727b3b219a1f3ba72dd : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cf94ec4f6fa4f4c98e30d9358749718 : SUCCESS in 6m 49s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/fbf88f745503446ca45b373c3a2b02d0 : SUCCESS in 47m 48s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/0cba2accc9d342b38ef95ace13d344d9 : SUCCESS in 1h 58m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/0b796aa7f11d4d859e4b1b523f2deb89 : SUCCESS in 1h 19m 35s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7ebb09403f65489d8e0423382bbb37a8 : SUCCESS in 41m 20s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/67a74a4b24ba473392954aa3acf31e78 : SUCCESS in 1h 28m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ba5cae826b224b08b0ad9324e70ba185 : SUCCESS in 1h 15m 20s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/7d4fa9300d5b459283bfaa30b4da008d : SUCCESS in 47m 32s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/00ce2eb571fa4f05ad1c392010579eed : SUCCESS in 1h 32m 22s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b683d415225c4b89afbe05b8a59caff8 : SUCCESS in 1h 45m 48s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/14df5ce1ab7349b49f64490dc611a23c : SUCCESS in 5m 13s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/791a1fd6a7b34f109f38a28ceb19095c : SUCCESS in 8m 33s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/cf9cce13b50741529700515efb524d25 : SUCCESS in 19m 13s","accounts_in_message":[],"_revision_number":5},{"id":"a4f6d91ce7ad9d13ca9a1988c4344236adab1664","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-03 01:05:43.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"41531043f8786f8c2d84ddcc85fa7569e69e95dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 01:06:18.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4eb9dfb138cc3aba89c5cb647ae28d6904f3bd49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 01:52:17.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49841e5603c74ac699b47e85d5c9774e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/294b875eea464b86ace51f9180c25f99 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2be04f4da6de4ff6aba21e9afb514a5c : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c32ccb646ac44ddcbcfb8e01652c81fc : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91c58cb72efd472797ba259540478b04 : SUCCESS in 5m 20s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/d69aa43835ee4ce4a317daee219f0bfc : SUCCESS in 45m 25s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/76463d234e304349a7f3dbe283f6aa91 : SUCCESS in 43m 22s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/11cd0181e4114a3aa12b12f7ccf5ed64 : SUCCESS in 41m 29s","accounts_in_message":[],"_revision_number":5},{"id":"7cec705de26f9e9e3d8691803a16b4cd8e8e8c30","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 01:52:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"18408dde4783cc773461a7529766bb2dc78a7e04","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 01:54:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7d9961796bb4033ae7767c7b9b67e82\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bfa4a2b44224e0994c00f1c62b9af76 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"87d60a1b5f3408e9ea3a9d1aca9a22af6a4cdabf","revisions":{"f5d970aca42d4e32e10beaface13753eebcb3e97":{"kind":"REWORK","_number":1,"created":"2024-04-02 02:56:35.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/32/914232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/914232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/914232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/914232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/914232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/914232/1"}}},"commit":{"parents":[{"commit":"5cb5a331a414efdc23c60a307da31945f3dec7c1","subject":"db: add commit action after querying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5cb5a331a414efdc23c60a307da31945f3dec7c1"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 02:55:34.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 02:55:34.000000000","tz":480},"subject":"Change settings for stable/2024.1","message":"Change settings for stable/2024.1\n\nChange-Id: If4bd6f0bd950331d1424bd2078d22f7a86f5b628\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f5d970aca42d4e32e10beaface13753eebcb3e97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f5d970aca42d4e32e10beaface13753eebcb3e97"}]},"branch":"refs/heads/stable/2024.1"},"8fec3f38529869503184cdf2e88a2257fd49a255":{"kind":"REWORK","_number":2,"created":"2024-04-02 05:57:57.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/32/914232/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/914232/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/914232/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/914232/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/914232/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/914232/2"}}},"commit":{"parents":[{"commit":"5cb5a331a414efdc23c60a307da31945f3dec7c1","subject":"db: add commit action after querying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5cb5a331a414efdc23c60a307da31945f3dec7c1"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 02:55:34.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-04-02 05:57:57.000000000","tz":0},"subject":"Change settings for stable/2024.1","message":"Change settings for stable/2024.1\n\nChange-Id: If4bd6f0bd950331d1424bd2078d22f7a86f5b628\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8fec3f38529869503184cdf2e88a2257fd49a255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8fec3f38529869503184cdf2e88a2257fd49a255"}]},"branch":"refs/heads/stable/2024.1"},"2f38a470973d5722946a4268cf2ea9fe660754c9":{"kind":"REWORK","_number":3,"created":"2024-04-02 14:16:05.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/32/914232/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/914232/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/914232/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/914232/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/914232/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/914232/3"}}},"commit":{"parents":[{"commit":"5cb5a331a414efdc23c60a307da31945f3dec7c1","subject":"db: add commit action after querying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5cb5a331a414efdc23c60a307da31945f3dec7c1"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 02:55:34.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 14:15:57.000000000","tz":480},"subject":"Change settings for stable/2024.1","message":"Change settings for stable/2024.1\n\nChange-Id: If4bd6f0bd950331d1424bd2078d22f7a86f5b628\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2f38a470973d5722946a4268cf2ea9fe660754c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2f38a470973d5722946a4268cf2ea9fe660754c9"}]},"branch":"refs/heads/stable/2024.1"},"5b898d1bc8cf7a818287a8db701d7b47cf641413":{"kind":"REWORK","_number":4,"created":"2024-04-02 15:17:02.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/32/914232/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/914232/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/914232/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/914232/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/914232/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/914232/4"}}},"commit":{"parents":[{"commit":"5cb5a331a414efdc23c60a307da31945f3dec7c1","subject":"db: add commit action after querying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5cb5a331a414efdc23c60a307da31945f3dec7c1"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 02:55:34.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 15:16:52.000000000","tz":480},"subject":"Change settings for stable/2024.1","message":"Change settings for stable/2024.1\n\nChange-Id: If4bd6f0bd950331d1424bd2078d22f7a86f5b628\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5b898d1bc8cf7a818287a8db701d7b47cf641413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5b898d1bc8cf7a818287a8db701d7b47cf641413"}]},"branch":"refs/heads/stable/2024.1"},"87d60a1b5f3408e9ea3a9d1aca9a22af6a4cdabf":{"kind":"REWORK","_number":5,"created":"2024-04-02 15:18:14.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/32/914232/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/914232/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/914232/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/914232/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/914232/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/914232/5"}}},"commit":{"parents":[{"commit":"5cb5a331a414efdc23c60a307da31945f3dec7c1","subject":"db: add commit action after querying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5cb5a331a414efdc23c60a307da31945f3dec7c1"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 02:55:34.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 15:18:07.000000000","tz":480},"subject":"Change settings for stable/2024.1","message":"Change settings for stable/2024.1\n\nChange-Id: If4bd6f0bd950331d1424bd2078d22f7a86f5b628\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87d60a1b5f3408e9ea3a9d1aca9a22af6a4cdabf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87d60a1b5f3408e9ea3a9d1aca9a22af6a4cdabf"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
