)]}'
{"id":"openstack%2Ftrove~914236","triplet_id":"openstack%2Ftrove~stable%2F2023.1~I43896ac8a5a85171f55216dc4f9c7354aa9957d2","project":"openstack/trove","branch":"stable/2023.1","hashtags":[],"change_id":"I43896ac8a5a85171f55216dc4f9c7354aa9957d2","subject":"remove .coverage file from repository","status":"MERGED","created":"2024-04-03 01:22:44.000000000","updated":"2024-04-03 08:16:15.000000000","submitted":"2024-04-03 08:12:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914236","meta_rev_id":"3f99da8c14b08eac7ee33ce7471ce5fa3136a394","_number":914236,"virtual_id_number":914236,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-03 08:12:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-03 04:51:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-03 04:51:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-03 01:29:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-03 04:05:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2661fbf3dacad83a5f6cda7fe55fdd6ac617a0c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-03 01:22:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce6e057bd927baf5d865237bd64bc662b1ebc44a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 01:29:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/782b5d421cf44537a20924ad1ba8aedd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad057b5730e54536b148adea36b80e1f : SUCCESS in 6m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3ce00bbea0220e52fc05d232a68f491cf7aab0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-03 01:33:53.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6250934036d80daebd4eef8758f9abae4eaf2b81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 01:39:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5341fc1b094d4b73ba2280d5129d9070\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5076487b226b4cf5a1bae286b2845445 : SUCCESS in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e084391babbb533a30cf3b0ac1cf0b900e7cf5d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:05:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43a9e36b819d46caad1b419e4f6b62c1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/414e731fd55146d88db19ce41aed0386 : FAILURE in 4m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0cc8addd0394a1f9d0898c7b0213044 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5690411c1204fca85ff593a199032c4 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c25da76842843b5a884478227b87695 : SUCCESS in 4m 30s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b8f306f19ea4966983f785f449e3a50 : SUCCESS in 4m 38s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f4affa2a94e34c5580db37c3bb52da64 : SUCCESS in 2h 26m 53s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/b06f4d6cb41b44079c3d867c9b97d03b : SUCCESS in 2h 30m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"08f0d46d26875b23ee86c214b2fc1db9c4a396ec","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-03 04:51:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"be16292503e96946636bcff359c6f88b85682b35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:52:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6d3408f00c0fb7c95d88fc3034bac8f509a0d86e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 08:12:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3a025abbe4a4d8193bf57793390e86b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a4e881e646a43cc8fbf71fa106a0468 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6c65a35df1e428ab22d83b1b1101b78 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c3594aee3342c59d1106cfd754ba6c : SUCCESS in 5m 35s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/342059f4c26646fe82f6cc30bccf331d : SUCCESS in 2h 27m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"605b61c6a5996227c725b8b641e4442dc291cf61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 08:12:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3f99da8c14b08eac7ee33ce7471ce5fa3136a394","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 08:16:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61533789f7114b7d97454bf88420c70b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef7d89c16a2146b092a1cabd08be9466 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3f08f37402f90f38d66e2c786b9118e781fe5d1f","revisions":{"0b3b7aaf33da07416189c3296e00d44243c97bfa":{"kind":"REWORK","_number":1,"created":"2024-04-03 01:22:44.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/36/914236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/36/914236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/36/914236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/36/914236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/36/914236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/36/914236/1"}}},"commit":{"parents":[{"commit":"2b6ea5e2f1f5b4cd90c888d4cd8c30acae6320bc","subject":"Retire TripleO: remove the TripleO jobs/ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2b6ea5e2f1f5b4cd90c888d4cd8c30acae6320bc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-03 01:22:15.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-03 01:22:15.000000000","tz":480},"subject":"remove .conver file from repository","message":"remove .conver file from repository\n\nChange-Id: I43896ac8a5a85171f55216dc4f9c7354aa9957d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b3b7aaf33da07416189c3296e00d44243c97bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b3b7aaf33da07416189c3296e00d44243c97bfa"}]},"branch":"refs/heads/stable/2023.1"},"3f08f37402f90f38d66e2c786b9118e781fe5d1f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-04-03 01:33:53.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/36/914236/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/36/914236/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/36/914236/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/36/914236/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/36/914236/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/36/914236/2"}}},"commit":{"parents":[{"commit":"2b6ea5e2f1f5b4cd90c888d4cd8c30acae6320bc","subject":"Retire TripleO: remove the TripleO jobs/ref","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2b6ea5e2f1f5b4cd90c888d4cd8c30acae6320bc"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-03 01:22:15.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-04-03 01:33:53.000000000","tz":0},"subject":"remove .coverage file from repository","message":"remove .coverage file from repository\n\nChange-Id: I43896ac8a5a85171f55216dc4f9c7354aa9957d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f08f37402f90f38d66e2c786b9118e781fe5d1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f08f37402f90f38d66e2c786b9118e781fe5d1f"}]},"branch":"refs/heads/stable/2023.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
