)]}'
{"id":"openstack%2Ftrove~914670","triplet_id":"openstack%2Ftrove~stable%2F2023.1~Id6fc86b570627bc6863dde330877966da3126fe6","project":"openstack/trove","branch":"stable/2023.1","topic":"retire-tripleo-stable/2023.2-stable/2023.1","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-04-02 05:55:02.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Id6fc86b570627bc6863dde330877966da3126fe6","subject":"Retire TripleO: remove the TripleO jobs/ref","status":"MERGED","created":"2024-03-29 05:41:21.000000000","updated":"2024-04-02 16:01:46.000000000","submitted":"2024-04-02 15:59:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914670-retire-tripleo-stable/2023.2-stable/2023.1","cherry_pick_of_change":914584,"cherry_pick_of_patch_set":3,"meta_rev_id":"e636cf93a4fc175ef4747ee1f6eebe655eaf621a","_number":914670,"virtual_id_number":914670,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-02 15:59:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-02 13:20:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-02 13:20:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-29 05:41:21.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2024-03-29 05:53:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-29 08:58:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e80feecf8e4e60f9db887f80a8705e2a0bce6012","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-29 05:41:21.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.\n\nThe following files contain Git conflicts:\n* devstack/settings","accounts_in_message":[],"_revision_number":1},{"id":"bc6c00fa47c30b80a9728a7273f864f02e1fbfc8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-03-29 05:53:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"661cd2ab4bbecb9b09f05a511664557d18649b05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 05:53:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd3498bd1c114df1bcb84643b447089d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8c17c218e4445a1bf64baa12b364e4a : SUCCESS in 10m 15s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/trove on branch stable/2023.1.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"branches\" attribute\nof the following project stanza:\n\n  ...\n          - trove-scenario-mariadb-multi\n      periodic:\n        jobs:\n          - publish-trove-guest-image-ubuntu-focal:\n              branches: ^(?!stable/.*).*$\n          - publish-trove-guest-image-ubuntu-focal-dev:\n              branches: ^(?!stable/.*).*$\n          - publish-trove-datastore-images:\n              branches: ^(?!stable/.*).*$\n\n  in \"openstack/trove/zuul.d/projects.yaml@stable/2023.1\", line 42","accounts_in_message":[],"_revision_number":1},{"id":"a59183b159f9b2130766f5cb7d279584606295c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 06:04:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/476b774a411049a082c5e3e4516ec27c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f53893fb7a734b62bb82b4718930a07d : SUCCESS in 8m 48s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/trove on branch stable/2023.1.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"branches\" attribute\nof the following project stanza:\n\n  ...\n          - trove-scenario-mariadb-multi\n      periodic:\n        jobs:\n          - publish-trove-guest-image-ubuntu-focal:\n              branches: ^(?!stable/.*).*$\n          - publish-trove-guest-image-ubuntu-focal-dev:\n              branches: ^(?!stable/.*).*$\n          - publish-trove-datastore-images:\n              branches: ^(?!stable/.*).*$\n\n  in \"openstack/trove/zuul.d/projects.yaml@stable/2023.1\", line 42","accounts_in_message":[],"_revision_number":2},{"id":"94d95dade3e11cbf9749f30bc812da95933c17bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 08:58:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9266e4019ca2438397afb8ea093ab615\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a574d35ac2e47b780bb9558235a8c28 : FAILURE in 4m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7decb0b7700b44f18fc4ab692b636662 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b87dd87a101e4703b7b02de3c25dd3d3 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b880f612477740359ed174c34f5a3c22 : SUCCESS in 4m 52s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c966df892e7b4a468b7ecbc335b8f54e : SUCCESS in 4m 21s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8e7adcfd1e6e44b280d3e017beceeef9 : SUCCESS in 2h 22m 07s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/44286a3791ab4f9da7fa2f022656e6b5 : SUCCESS in 7m 20s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/0c4efb27260243e29ce488299db04b61 : SUCCESS in 13m 09s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/f0c04e5df0944839875c4c7d79fb3d6e : TIMED_OUT in 3h 02m 34s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/227b7a68e68843eeb44898f9acdb414e : FAILURE in 1h 45m 52s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/trove on branch stable/2023.1.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"branches\" attribute\nof the following project stanza:\n\n  ...\n          - trove-scenario-mariadb-multi\n      periodic:\n        jobs:\n          - publish-trove-guest-image-ubuntu-focal:\n              branches: ^(?!stable/.*).*$\n          - publish-trove-guest-image-ubuntu-focal-dev:\n              branches: ^(?!stable/.*).*$\n          - publish-trove-datastore-images:\n              branches: ^(?!stable/.*).*$\n\n  in \"openstack/trove/zuul.d/projects.yaml@stable/2023.1\", line 42","accounts_in_message":[],"_revision_number":2},{"id":"163457be38abe069eaf70d8fcf60664d9bf041ca","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 02:01:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3240dac9fab0a64832faf4090c33157a0267e3e7","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 02:02:10.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"3fdf1a8e55445e04b917b2839b210d44547e1780","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 02:27:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1cf7a95776743de86c44b55560d54b7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67a7b9576337454f9112e7cc810ccdab : SUCCESS in 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"05fdbd0315b614765479e1bd0fb3a7204a66621a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 05:41:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa679954ae284885a78942434260636e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15c5fb2ee6de47519aad08b2e7ff773e : FAILURE in 5m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69b33cb3f9694ab0b844c7835e7f0769 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5e0442644a5475bacedc219cbbb0d82 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e96e97d65d4022b8e5d049971773c4 : SUCCESS in 6m 09s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a37ab6924fd64026a02cd15bd5c674ce : SUCCESS in 6m 23s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/aa87479cb3e04bcb9d5107fe184e163f : TIMED_OUT in 3h 02m 28s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/d867619cf0724326a0b2d9e5a63f51ed : SUCCESS in 11m 03s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/42d8e490fc5e4ce4ae9289b51c4e87ac : TIMED_OUT in 30m 53s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/bae7b0877277424c9057d2d1ee1187be : SUCCESS in 2h 17m 59s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0b7f9c05cec40a28ac0b5bf0cd83195 : FAILURE in 1h 45m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eb7e252b0342761327a4b239ecb4079b48d07318","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 05:55:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f261b47665d1afd9b36771147d2deb7bd610e576","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 06:01:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/368b22c05b87441cb7f5c096a0503f46\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fc550063e3b42a0915be568839428fd : SUCCESS in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b9c577d37fa9964070ce3302978af533a937564b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 09:04:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56e727ad418a49988da3fb23d90b11e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb31638a63ce4c9f821ff0bf2ddcfb69 : FAILURE in 6m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e86985f1c2554806bcac33e6ea1df1c5 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/750492bb12fa44b5b0ef3d3f535b688b : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce8318361e746cea38bdd0776eeb2a7 : SUCCESS in 5m 22s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c0244b436e554a3bb156bb1b54c6cab7 : SUCCESS in 5m 45s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/866043acdcb842e88ce92df649aedf99 : TIMED_OUT in 3h 03m 37s (non-voting)\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/5c0fcd0ddbf7483c99f071f06c82017c : SUCCESS in 9m 04s\n- trove-centos8s-guest-image-build https://zuul.opendev.org/t/openstack/build/99539f4303294fdca1da5b184a75928e : SUCCESS in 13m 15s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/affd8dfb44b346a9bac039e11f71ffc7 : TIMED_OUT in 3h 03m 25s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b812a75d6ae4b139ec1576e10de2f38 : FAILURE in 1h 51m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cce04ffb0465303da785f2074252730f4212bb82","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 13:20:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4d8ed61a184dd64f644e895ae59cdc4ffa4ba143","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 13:21:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"43d0cca3f27cd4d460bced1919c345cfe5427c16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 15:59:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eebbfc87d4724e1ebfae1ad1c92eb378\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bed93a9caba4757ba3ccabfc850b482 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfbf414bec9647d484165e496b3a0076 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de990996f51a490b9868e164bcb6b485 : SUCCESS in 4m 43s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/32fda54c58424b75817b92e5129b3059 : SUCCESS in 2h 36m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"55a119f238bb5f7c170ffa219ff1525b4d0ae743","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 15:59:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e636cf93a4fc175ef4747ee1f6eebe655eaf621a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 16:01:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cd90ab621bb48a2845f01f84aa5d54d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1412c89c8b0f4168b0fd7ce91f4f244b : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2b6ea5e2f1f5b4cd90c888d4cd8c30acae6320bc","revisions":{"024e695833eb73420233646946c2fb4ef6fa59e2":{"kind":"REWORK","_number":1,"created":"2024-03-29 05:41:21.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/70/914670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/70/914670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/914670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/914670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/914670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/70/914670/1"}}},"commit":{"parents":[{"commit":"363c1df7c94defb21e59eb6796d353b88987a2d6","subject":"stable-only: Pin bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/363c1df7c94defb21e59eb6796d353b88987a2d6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-29 21:48:52.000000000","tz":-480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-03-29 05:41:21.000000000","tz":0},"subject":"Retire TripleO: remove the TripleO jobs/ref","message":"Retire TripleO: remove the TripleO jobs/ref\n\nTripleO is retired so removing its jobs/ref\n\n- https://review.opendev.org/q/topic:%22retire-tripleo%22\n\nChange-Id: Id6fc86b570627bc6863dde330877966da3126fe6\n(cherry picked from commit a2447137d941d9c234f66cf7f89bfb6f47d0f238)\n(cherry picked from commit bc0ee00338ff1bfecf70dadd26e9b7b192dd052d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/024e695833eb73420233646946c2fb4ef6fa59e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/024e695833eb73420233646946c2fb4ef6fa59e2"}]},"branch":"refs/heads/stable/2023.1"},"9f9e17d00b4d03cc9bbadc256b994b01d3fbcb55":{"kind":"REWORK","_number":2,"created":"2024-03-29 05:53:37.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/70/914670/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/70/914670/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/914670/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/914670/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/914670/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/70/914670/2"}}},"commit":{"parents":[{"commit":"363c1df7c94defb21e59eb6796d353b88987a2d6","subject":"stable-only: Pin bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/363c1df7c94defb21e59eb6796d353b88987a2d6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-29 21:48:52.000000000","tz":-480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-03-29 05:53:31.000000000","tz":480},"subject":"Retire TripleO: remove the TripleO jobs/ref","message":"Retire TripleO: remove the TripleO jobs/ref\n\nTripleO is retired so removing its jobs/ref\n\n- https://review.opendev.org/q/topic:%22retire-tripleo%22\n\nChange-Id: Id6fc86b570627bc6863dde330877966da3126fe6\n(cherry picked from commit a2447137d941d9c234f66cf7f89bfb6f47d0f238)\n(cherry picked from commit bc0ee00338ff1bfecf70dadd26e9b7b192dd052d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9f9e17d00b4d03cc9bbadc256b994b01d3fbcb55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9f9e17d00b4d03cc9bbadc256b994b01d3fbcb55"}]},"branch":"refs/heads/stable/2023.1"},"2b6ea5e2f1f5b4cd90c888d4cd8c30acae6320bc":{"kind":"REWORK","_number":3,"created":"2024-04-02 02:01:53.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/70/914670/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/70/914670/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/914670/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/914670/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/914670/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/70/914670/3"}}},"commit":{"parents":[{"commit":"363c1df7c94defb21e59eb6796d353b88987a2d6","subject":"stable-only: Pin bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/363c1df7c94defb21e59eb6796d353b88987a2d6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-29 21:48:52.000000000","tz":-480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 02:01:45.000000000","tz":480},"subject":"Retire TripleO: remove the TripleO jobs/ref","message":"Retire TripleO: remove the TripleO jobs/ref\n\nTripleO is retired so removing its jobs/ref\n\n- https://review.opendev.org/q/topic:%22retire-tripleo%22\n\nChange-Id: Id6fc86b570627bc6863dde330877966da3126fe6\n(cherry picked from commit a2447137d941d9c234f66cf7f89bfb6f47d0f238)\n(cherry picked from commit bc0ee00338ff1bfecf70dadd26e9b7b192dd052d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2b6ea5e2f1f5b4cd90c888d4cd8c30acae6320bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2b6ea5e2f1f5b4cd90c888d4cd8c30acae6320bc"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
