)]}'
{"id":"openstack%2Ftrove~914679","triplet_id":"openstack%2Ftrove~stable%2F2024.1~I275bd77736020f02678fc34332edd277f5e2a6b6","project":"openstack/trove","branch":"stable/2024.1","topic":"fix_error_detach_cfg-stable/2024.1","hashtags":[],"change_id":"I275bd77736020f02678fc34332edd277f5e2a6b6","subject":"Fix error can\u0027t detach the configuration group binding","status":"MERGED","created":"2024-04-02 06:03:18.000000000","updated":"2024-04-02 16:02:31.000000000","submitted":"2024-04-02 15:59:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914679-fix_error_detach_cfg-stable/2024.1","cherry_pick_of_change":912342,"cherry_pick_of_patch_set":5,"meta_rev_id":"bf3e081921227036ee8514c25e22b7ed338a3fe0","_number":914679,"virtual_id_number":914679,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-02 15:59:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-02 14:46:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-02 14:46:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"CC":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-02 06:03:18.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"CC"},{"updated":"2024-04-02 06:15:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-02 09:04:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25282660e9214a82e49aa1dcb638427bfe61acdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 06:03:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"094ed10c66c82aef765d98f0f2c4e22eca791763","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 06:15:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/710f55e0be0a49f8acbaa2c414fb97ff\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c80e7da546624cf2a66e8d3bd1d602d5 : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a3e72069c6b442d88efc3b54ed56e41e : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc57f7d975c8113cdaa4a62bebd0ba51ea65e075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 09:04:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee537af935ec4523a87e3c91ba4c6acd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf33d1d438c64f6aaa9520013d5a9160 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b002065e99a64ef4a49962de5a6e3917 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ce6c1a092e14f7c81213dc3184a69c1 : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35437c218b5741eb99edc2b06820df5a : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57f174f8289c494690a91ec03525162f : SUCCESS in 4m 17s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/22e189b303fa43699d1365ce518b3221 : SUCCESS in 47m 17s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/bf37c77240124913ad80aed6f560bfa4 : FAILURE in 1h 44m 15s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/0c74017062284a4386d6ced1d3d402e5 : SUCCESS in 1h 43m 30s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/63bd2b98ffb94866ae6cfb41f0292cff : SUCCESS in 44m 38s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/16011ace8b674f59bbee91e28fb8107b : SUCCESS in 2h 01m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d648c04ac47e47e7950e2f40470e5f9f : POST_FAILURE in 1h 21m 49s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/4e97f23d7d6245c8928165fa0db97340 : SUCCESS in 45m 00s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/89d47e19dda54905acc8e69afe193ab4 : FAILURE in 1h 11m 29s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/16a0e20d31074f0f83d617f14f71a9f3 : POST_FAILURE in 1h 17m 58s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/21c31db30eff4be99466ce2608df0ded : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9f88e03b06a5a2ba426970ed33c1f6c68110af0b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 14:46:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0846bb7c9a6465987161788826bccc128dd928d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 14:47:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1bb41bc8f95c6bb28f60d54670bdf7c445f6872d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 15:59:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10f09f37c78b41aba40ef591948fc6ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f1764b0736e417dae2b3e6b1fdb8e41 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49a9413d224d424ba8454b4c70ea5d14 : SUCCESS in 7m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aef5c800fde94702a799d50dc23ca39b : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e00d6a273bb48e092d8bb6a195a58be : SUCCESS in 5m 44s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/d439c3ee49004c8f9ca7dce283049d0d : SUCCESS in 53m 01s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7c6886fe25fe4be7bd15560fcceceea3 : SUCCESS in 53m 08s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6b6c4a34a2f84f779f917c4b7651ff14 : SUCCESS in 43m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3978c04568b736a9fd416065cca4b9b9d8da7ebf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 15:59:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bf3e081921227036ee8514c25e22b7ed338a3fe0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 16:02:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daed065bff934d6a8bcccba5b3dbeddc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f03e8f12937476eb1037a6aa8793f12 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"452a9c27cfc6e6089dc72c307cc1e17434ba13da","revisions":{"452a9c27cfc6e6089dc72c307cc1e17434ba13da":{"kind":"REWORK","_number":1,"created":"2024-04-02 06:03:18.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/79/914679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/79/914679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/79/914679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/79/914679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/79/914679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/79/914679/1"}}},"commit":{"parents":[{"commit":"5cb5a331a414efdc23c60a307da31945f3dec7c1","subject":"db: add commit action after querying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5cb5a331a414efdc23c60a307da31945f3dec7c1"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-03-10 13:06:07.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-04-02 06:03:18.000000000","tz":0},"subject":"Fix error can\u0027t detach the configuration group binding","message":"Fix error can\u0027t detach the configuration group binding\n\nStory: #2011063\nTask: #49697\nChange-Id: I275bd77736020f02678fc34332edd277f5e2a6b6\n(cherry picked from commit dbb18bea1022e6c1f6bb5da721c3e791aa1f69b8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/452a9c27cfc6e6089dc72c307cc1e17434ba13da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/452a9c27cfc6e6089dc72c307cc1e17434ba13da"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
