)]}'
{"id":"openstack%2Ftrove~914862","triplet_id":"openstack%2Ftrove~stable%2F2023.2~I275bd77736020f02678fc34332edd277f5e2a6b6","project":"openstack/trove","branch":"stable/2023.2","topic":"fix_error_detach_cfg-stable/2023.2","hashtags":[],"change_id":"I275bd77736020f02678fc34332edd277f5e2a6b6","subject":"Fix error can\u0027t detach the configuration group binding","status":"MERGED","created":"2024-04-02 14:47:02.000000000","updated":"2024-04-03 04:03:48.000000000","submitted":"2024-04-03 04:02:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914862-fix_error_detach_cfg-stable/2023.2","cherry_pick_of_change":912342,"cherry_pick_of_patch_set":5,"meta_rev_id":"aaf9174ed22d8bb108d6f49f0f89fa28934bb364","_number":914862,"virtual_id_number":914862,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-03 04:02:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-03 01:10:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-03 01:10:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-02 14:47:02.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"CC"},{"updated":"2024-04-02 15:01:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-02 17:39:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44a708e7cf557ee8fcafb7b29faa086c693c77a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 14:47:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5241d03a89362e660fc3a37af157ec1a72775be4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 15:01:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3f4298f9c9a415cb7919c6d8f5054b5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d47eedf127b4ff395e04bfda7f78e76 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0278ce47b3e643df910a43f7429c5795 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4db85683f1be4459b1c9d4e2d2a7291a : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4feaa0b1e3b3438197e00140e8c810c0 : RETRY_LIMIT in 2m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6eaca2ba5f265836dc54aad6f99a19ff9b51f6eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 17:39:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f2cfb8c2861443a921db89b43cf10fa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f9eb83c5c2b4747b3eca5af6452ada3 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9723985787d049a7a7c52ad005f40d18 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9403779aafc74b6c9513b9d403301562 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8789a7964a1e41f8a438404f56e4c881 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba40e069871d4b939ec9c59675809181 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac5496e4ee63464d842a3e87d60764df : RETRY_LIMIT in 1m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c312a33ef8cd4c15993ff0f930fa97f0 : SUCCESS in 6m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7dac5ca41f764e4eb197e8545f964f56 : SUCCESS in 4m 57s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8f7c06de502148248543c4e0a57783de : FAILURE in 2h 45m 00s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/0d0718985af94fa1a5386cf0169ec23b : SUCCESS in 1h 30m 08s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/434cde01ac074c048a3dd3a6f9185a6a : FAILURE in 2h 43m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d897af57e99e272cef235b6cad93792f3db18907","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-03 01:10:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8a28ed5670da9a553021b587a15228733589532","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 01:10:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8c928d27b4e5afc3289b7e0c545efaeff0576845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:02:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7eaa865f3f924bfe97b67860aa0e7043\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f25ba32192be4923a85275be2131f3a7 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be20d9d9f1b54184abd043c0b581f002 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee267abb5c79471d9f52d74dd746189b : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea9088e3655f4bdf9368095bc89c159b : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3884afa83cfe4e09955b835cbc2f1619 : SUCCESS in 6m 01s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/66896ffaa7b745febba6e037dfdd1de5 : FAILURE in 2h 50m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02adfa0d7e2f2a517956680e5be094509e0c58dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:02:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aaf9174ed22d8bb108d6f49f0f89fa28934bb364","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:03:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2363157407044442b6947535b2307576\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf10b6e80dc4907aa23f6addfc9de5a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16e0f953cd52d182b67e8bdaabc8901541118ef0","revisions":{"16e0f953cd52d182b67e8bdaabc8901541118ef0":{"kind":"REWORK","_number":1,"created":"2024-04-02 14:47:02.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/62/914862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/62/914862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/914862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/914862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/914862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/62/914862/1"}}},"commit":{"parents":[{"commit":"700c7d8ecb848507b5d94434e788fa8cdeb12971","subject":"Merge \"Fixes the way to check db instance status\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/700c7d8ecb848507b5d94434e788fa8cdeb12971"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-03-10 13:06:07.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-04-02 14:47:02.000000000","tz":0},"subject":"Fix error can\u0027t detach the configuration group binding","message":"Fix error can\u0027t detach the configuration group binding\n\nStory: #2011063\nTask: #49697\nChange-Id: I275bd77736020f02678fc34332edd277f5e2a6b6\n(cherry picked from commit dbb18bea1022e6c1f6bb5da721c3e791aa1f69b8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/16e0f953cd52d182b67e8bdaabc8901541118ef0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/16e0f953cd52d182b67e8bdaabc8901541118ef0"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
