)]}'
{"id":"openstack%2Ftrove~914864","triplet_id":"openstack%2Ftrove~stable%2Fzed~I275bd77736020f02678fc34332edd277f5e2a6b6","project":"openstack/trove","branch":"stable/zed","topic":"fix_error_detach_cfg-stable/zed","hashtags":[],"change_id":"I275bd77736020f02678fc34332edd277f5e2a6b6","subject":"Fix error can\u0027t detach the configuration group binding","status":"MERGED","created":"2024-04-02 14:48:34.000000000","updated":"2024-04-03 04:16:55.000000000","submitted":"2024-04-03 04:15:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914864-fix_error_detach_cfg-stable/zed","cherry_pick_of_change":912342,"cherry_pick_of_patch_set":5,"meta_rev_id":"5760f1ba2b627b2b57380ad6f5996a428299121f","_number":914864,"virtual_id_number":914864,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-03 04:15:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-03 01:10:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-03 01:10:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-02 14:48:34.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"CC"},{"updated":"2024-04-02 16:56:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1149a0121ee8439c54aed6a3da431524716b346","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-02 14:48:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"62dff855956e12dfaa018219b22c821c1de9a515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 16:56:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca72d9ca3e294d1ab0f4e94ed60d69ae\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f5f5dcf204f40b4afb21dbecbbea381 : FAILURE in 6m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b5e150af1b9427ab7c1daef7a5c0588 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39b864238ecb4b9192ba9e5c4998e81c : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68a68f2824a94a529531df1610a629b7 : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64da1ea1e12b424882384f976d2c416d : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a109c167bc7247beb4279aaf7efdb35c : SUCCESS in 5m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/45c7ca386dab4ba08bbd1850e14133e4 : SUCCESS in 7m 09s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0307fe9b4cdb426b9d020eee9608a4eb : FAILURE in 1h 38m 13s (non-voting)\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/afdaee784b82413889ef9aae80b6f28c : FAILURE in 1h 34m 49s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dde221b6c0be44a69b907b40dc1d051e : FAILURE in 53m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e195814b45be1434112f2ee436b6c69c390adb35","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-03 01:10:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"450a33f6a74ae17588f48760b60bc5987df65b42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 01:10:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4889c2c02d79732f9efb450678796ce4ed2b904b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:15:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c11564636144f218279d58b74f9de4b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8353768e538849a495c7c92a22637d6c : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec1bc325e79b4e969896d62de1b575ac : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a02ff0625e44d1f81bbc63860f49e2d : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7bb0e1a5f8b402cb1169523e1addcdb : SUCCESS in 5m 18s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/754d8427458f4226a3e82d1723f6b8c2 : FAILURE in 35m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"520b9b87fc0da0ade2c6e5ea43d389625af9bb46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:15:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5760f1ba2b627b2b57380ad6f5996a428299121f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:16:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d1031518974483ab50933c836e6558d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1b18d8ad40043a1800d8c0f08008bd3 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8225f4463e96b8ce7ea10d2b05deb7379f239e2","revisions":{"c8225f4463e96b8ce7ea10d2b05deb7379f239e2":{"kind":"REWORK","_number":1,"created":"2024-04-02 14:48:34.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/64/914864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/64/914864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/914864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/914864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/64/914864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/64/914864/1"}}},"commit":{"parents":[{"commit":"99c7a9666738bb897e2bd5009c92b5c8ba2746cd","subject":"stable-only: Pin bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/99c7a9666738bb897e2bd5009c92b5c8ba2746cd"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-03-10 13:06:07.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-04-02 14:48:34.000000000","tz":0},"subject":"Fix error can\u0027t detach the configuration group binding","message":"Fix error can\u0027t detach the configuration group binding\n\nStory: #2011063\nTask: #49697\nChange-Id: I275bd77736020f02678fc34332edd277f5e2a6b6\n(cherry picked from commit dbb18bea1022e6c1f6bb5da721c3e791aa1f69b8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c8225f4463e96b8ce7ea10d2b05deb7379f239e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c8225f4463e96b8ce7ea10d2b05deb7379f239e2"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
