)]}'
{"id":"openstack%2Ftrove~914868","triplet_id":"openstack%2Ftrove~stable%2F2024.1~Ifbe862020f7c2b1f6684164d2a26bb5ceb37bec3","project":"openstack/trove","branch":"stable/2024.1","hashtags":[],"change_id":"Ifbe862020f7c2b1f6684164d2a26bb5ceb37bec3","subject":"Enable network_isolation by default","status":"MERGED","created":"2024-04-03 01:08:20.000000000","updated":"2024-04-03 04:18:11.000000000","submitted":"2024-04-03 04:15:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914868","cherry_pick_of_change":914231,"cherry_pick_of_patch_set":4,"meta_rev_id":"16ae9502b19b0012c29ba1eea8ab75f89fdb693c","_number":914868,"virtual_id_number":914868,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-03 04:15:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-03 03:05:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-03 03:05:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-03 01:16:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-03 02:59:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb9696bee739fc4662d5e14d237d073aba926296","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-03 01:08:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f6479c4dea8b821d8c7c7c2b9870ddba7822daa8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 01:16:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67aabec588cc48eb8d1e03de5fb28369\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb4c728ab347437fa662b9027d17d183 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/11a41bc3dd8749309ad0d13cd8ebd1d4 : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4f07d1f5b093d95533f4cb533d098b45d08d8d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 02:59:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bf254c5208c4ab2893d602150a0b051\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/677d321c285f4286bc6ebce2df4ab944 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bce0d9333dbe423f916b3b28bbb27581 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd2983f99cf5460ab75667de207ebc8c : SUCCESS in 8m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a84a5b215ad34db6bd15bcd7cc788806 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c7d2fe0aaa5447aadd132e770776a81 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b46b0907571440119e99762911212f73 : SUCCESS in 5m 38s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/f9bd881c049c45d29cd340dbbc81fbb3 : SUCCESS in 59m 21s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/10c751d86d404e7fadd3e5f73614fa0f : FAILURE in 1h 42m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/1596f71393f04092878f2bad0fd34402 : SUCCESS in 1h 49m 37s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5c89d936b0fa42629619449869d1cd8f : SUCCESS in 49m 49s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5f320a629c6741199a1430d29225e271 : SUCCESS in 1h 42m 41s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/cc146b982e1446e18b665ee6ef199a2b : SUCCESS in 1h 27m 37s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/2bf701d6239f414184b2cc15954253f3 : SUCCESS in 46m 51s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/7300cb17d7404469aa9d5b1cb2b70f02 : SUCCESS in 1h 41m 38s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/4739a0416ca946848c6b383b42d72f55 : SUCCESS in 1h 26m 36s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e2e776d1a17e42ad97b83ace5b90882e : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"26df5402ad59dede47622c7c681a0c106909291a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-03 03:05:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"21eae9f7a63b43be21a0d1942a967b3bc424f66f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 03:06:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a61ebeb64d2f734bd00802e30a5a5dc0c0fa1dbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:15:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10a50aa6ed744300a99e87b6a1c5fea7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9cb25d98ede40daab5ed47f9fbec24f : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a17c7a27062940e08fd64fe50c0ca48f : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/051b0ff771d147eaa3cfce7e0e466f37 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/144fdc979f7043c68d5552a54e667ed9 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c8d2fbc64d44a6d8d962ad19d8505e0 : SUCCESS in 5m 46s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/89bf2f819f824fc5a8da79f68e3bf769 : SUCCESS in 1h 04m 13s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/96238b6928ce4df583ab76b3ac22a4ae : SUCCESS in 1h 04m 06s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/04e3a61be8b04c93b86ec1b402577d50 : SUCCESS in 1h 00m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b3239acdc702ed8fa980ebd0a90050e421a124c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:15:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"16ae9502b19b0012c29ba1eea8ab75f89fdb693c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:18:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0bc2f800a6c4a8aa9e828cb40e645c0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eefd3072e1340e7a2732d45bd3a2bdc : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddd3aa18738540808040b4ce482d64fc : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7363a27135a05dc328e9a3ff9a9437b575eff9d","revisions":{"b7363a27135a05dc328e9a3ff9a9437b575eff9d":{"kind":"REWORK","_number":1,"created":"2024-04-03 01:08:20.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/68/914868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/914868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/914868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/914868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/914868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/914868/1"}}},"commit":{"parents":[{"commit":"0b99827ae285a53f43d734d87b16e7239f2891f3","subject":"Merge \"Fix error can\u0027t detach the configuration group binding\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b99827ae285a53f43d734d87b16e7239f2891f3"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 01:56:35.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-04-03 01:08:20.000000000","tz":0},"subject":"Enable network_isolation by default","message":"Enable network_isolation by default\n\nWe have already tested this feature on our CI for an entire cycle.\nIt\u0027s ready to be enabled as the default value.\n\nChange-Id: Ifbe862020f7c2b1f6684164d2a26bb5ceb37bec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b7363a27135a05dc328e9a3ff9a9437b575eff9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b7363a27135a05dc328e9a3ff9a9437b575eff9d"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
