)]}'
{"id":"openstack%2Ftrove~914869","triplet_id":"openstack%2Ftrove~stable%2F2023.2~Ifbe862020f7c2b1f6684164d2a26bb5ceb37bec3","project":"openstack/trove","branch":"stable/2023.2","hashtags":[],"change_id":"Ifbe862020f7c2b1f6684164d2a26bb5ceb37bec3","subject":"Enable network_isolation by default","status":"MERGED","created":"2024-04-03 01:08:48.000000000","updated":"2024-04-03 08:17:17.000000000","submitted":"2024-04-03 08:12:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914869","cherry_pick_of_change":914231,"cherry_pick_of_patch_set":4,"meta_rev_id":"ab7d16740174026e3d33cff2aca7d586ea413240","_number":914869,"virtual_id_number":914869,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-03 08:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-03 04:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-03 04:51:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-03 01:24:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-03 04:31:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fad073c768232951aea2651911e6f8ae612300ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-03 01:08:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c1b8e44bcd9722de6cef437b82e1e0a349a747d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 01:24:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4fcc7df21c742d5b06cd8f2464d1cd7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/339e4e03dbce478c9f45178d58c57d75 : SUCCESS in 13m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da506c3539834806ae2d14e3243ca955 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6d5f56861525490ba4f4e706a14e25e8 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/40d7143c3c6f459fb8502623ab808936 : RETRY_LIMIT in 1m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a184fd49afbec62f860786cda608fdd76e065fa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:31:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/579f6e5943414bdfa2baac42eeb4610b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dce75ace52b44339b962610154115623 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5554be8175d448f0aa30cf86b732e793 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/160ca4ca6f484a8b9764101e0c5f487b : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6b5fc146f8d479f8efb238c916d00da : SUCCESS in 7m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83d2a49d23fa4d2498a72c7d9bbde8cb : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/461277c4d9994ecd98bc69b7c1601747 : RETRY_LIMIT in 1m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/841a7c063e344d3bbe9d75f5d6e89dc3 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29c9f9a12d8d48bebd2fd444c0b3d070 : SUCCESS in 5m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1ef75d2dc5bc442f90b1064992e7f726 : SUCCESS in 5m 06s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/712e40ff8c6346069c28c0a9e9f01b01 : FAILURE in 2h 55m 56s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/4a28e3e05c4a4edb89f72166faa0c965 : SUCCESS in 2h 19m 42s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/235a9728cd3b48098bfce6849ee76526 : FAILURE in 2h 56m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f98e37fb6e923a7d6ec8e62b6d77373e35d33a96","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-04-03 04:51:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20e8948eb8a1f49316af30b0d8d8ad71613c3ebf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 04:51:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c001839c2aae40d155a5128cca8f0b12c785a4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 08:12:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4e94a56e7e24b0a8e9a524258d55252\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d6e30a69dd64959a8e304779604498c : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f38d886a4fe47adbc87feb2c74fc4f9 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30edcaf536054953a560def6c2c1dc5b : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/010e03d8ec5f46c89c37f74c3db5a326 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e89aaa7163d42ce9904debd143ecfab : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9195aa9e320f48178b0ad939aa9de7f4 : SUCCESS in 4m 27s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f9ef7f47048545feaae540bf4e63ad38 : FAILURE in 2h 54m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44bff5b424e2eb05487de6bc939bd979deed12bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 08:12:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ab7d16740174026e3d33cff2aca7d586ea413240","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 08:17:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d267e848e6c14483b2deb57687664f53\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3827b339bc240be91e6b756e4197737 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2173404083a54834adbeed6d18aa117c : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e9f8b70e2bc1965cd31e369d10d8fca7a40a630","revisions":{"9e9f8b70e2bc1965cd31e369d10d8fca7a40a630":{"kind":"REWORK","_number":1,"created":"2024-04-03 01:08:48.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/69/914869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/69/914869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/914869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/914869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/69/914869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/69/914869/1"}}},"commit":{"parents":[{"commit":"700c7d8ecb848507b5d94434e788fa8cdeb12971","subject":"Merge \"Fixes the way to check db instance status\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/700c7d8ecb848507b5d94434e788fa8cdeb12971"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-04-02 01:56:35.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-04-03 01:08:48.000000000","tz":0},"subject":"Enable network_isolation by default","message":"Enable network_isolation by default\n\nWe have already tested this feature on our CI for an entire cycle.\nIt\u0027s ready to be enabled as the default value.\n\nChange-Id: Ifbe862020f7c2b1f6684164d2a26bb5ceb37bec3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9e9f8b70e2bc1965cd31e369d10d8fca7a40a630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9e9f8b70e2bc1965cd31e369d10d8fca7a40a630"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
