)]}'
{"id":"openstack%2Ftrove~918661","triplet_id":"openstack%2Ftrove~master~Idd63f470a2b941720314b6356fe28cd8e394427e","project":"openstack/trove","branch":"master","topic":"story/2010922","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2024-05-10 08:23:16.000000000","reason":"\u003cGERRIT_ACCOUNT_28691\u003e replied on the change","reason_account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2024-06-14 07:44:59.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-06-14 06:31:15.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Idd63f470a2b941720314b6356fe28cd8e394427e","subject":"Introduces Alembic database migration tool","status":"MERGED","created":"2024-05-08 14:04:09.000000000","updated":"2024-06-14 07:47:26.000000000","submitted":"2024-06-14 07:44:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":60,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918661-story/2010922","meta_rev_id":"8395cd365d408a79b10b83289da330a882fc18e3","_number":918661,"virtual_id_number":918661,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-14 07:44:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-14 06:31:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-14 06:31:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-08 14:10:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-08 16:36:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-10 07:51:58.000000000","updated_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2024-05-10 07:51:58.000000000","updated_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"6afb3f8f281e831ddbbc225f5a006ffa7754ce6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-08 14:04:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"017fdc20ec6a2a4b12ff7e55e22d1b89b02e0261","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 14:10:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e415bb5060384e1aa4ca697d21e0f6fd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a22f77688e2466182f13ed6e4a5a562 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b04c132beb8d4b21b4fadd70638b9188 : SUCCESS in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f674e5614d75907143076ffc4048e6c493e7d5b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 16:36:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c9b940f988b452c978fc344b8c4b181\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f7cd159701d474daab42707c1c9c03b : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32d24ad9accd451cb7c9c637076faf77 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f044d2d39d0e4d6db7049c56ea51ebb8 : SUCCESS in 5m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33d243c7304449b79c2a9c5e23c2f03d : SUCCESS in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c48cbf062ccd4521afe5cc8965d70cd9 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed402c7b2c06439193282f78a57d01d9 : SUCCESS in 4m 51s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/ff27147788fd4eff96f0b93f7e98c664 : SUCCESS in 53m 56s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/62641dca4c8242d381ad24fdbd59caa3 : FAILURE in 2h 22m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/522ce610c3344ddc94b0c3430bb0c6f9 : SUCCESS in 1h 21m 45s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f86d2ae5bc1e4dbb9db12f958082234d : SUCCESS in 42m 23s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/faf46a3066f244328ca6ca778b08f1f9 : SUCCESS in 1h 51m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b0157167ff3f43f591378d4dd40ab4b4 : SUCCESS in 1h 13m 35s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/42263d9507d04e5baf556e98441d74c4 : SUCCESS in 1h 03m 26s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/4114bc7737a24fdb91efc8f6013373c9 : SUCCESS in 1h 22m 07s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/7bd83a4bb1354090b4fb7fac7c870823 : SUCCESS in 2h 00m 49s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/98b04076eaef471094e61b3c5c367749 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4899184866953d3915b880a20cdc1eaa86462082","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-09 02:21:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8dc1e41333e98216ba7a30d1c7af86f07910f15f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 02:39:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/452c494dee8448908db36e1333077e9d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f91e0dc4f8f94448866062d9733f1b51 : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fd6deefbb93b408c814b318f9500d2ee : SUCCESS in 14m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3bb080f969fc1ea3e06c823aacb3e18b736892f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 04:54:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eb9d57764df47c2bc0b3672d5be697a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e01657b0bd714f8da58f609d1db4029c : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d76dcc4766e4d648628989a71c60393 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8aa6f78fa7934358ab1edb43811e3538 : SUCCESS in 6m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64cbed1363cb46ac99454f8e21e7e2e1 : SUCCESS in 6m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/016ce799faf54f5db592184fd8867f08 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5173d8e2f89e4350866807453e74c26b : SUCCESS in 6m 25s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/6cf44e6415d642cdaf4cddf2d31e3adb : SUCCESS in 42m 41s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/7a1e08d80b2541c4bffef931a0ed840c : SUCCESS in 2h 17m 55s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/56d24ae7ee1e4942a5b8f963f27601f7 : SUCCESS in 1h 36m 25s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/46782212b21c402c8535a8f02c61f46a : SUCCESS in 1h 06m 55s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ba6019cf6e9b43d7b813140f7b1a09e5 : SUCCESS in 2h 08m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/aa3a92fef3794d97a9a241aab2fc14d1 : SUCCESS in 1h 43m 03s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/16cb81a41a194db7a890a107bb009108 : SUCCESS in 1h 05m 58s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/13ddec9acf9f4046ba8477ab6c6dbf36 : SUCCESS in 2h 01m 45s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/1814dbd5993c4fbba855716d53a04e23 : SUCCESS in 1h 44m 48s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3955d0a1c18c4a6ea07689c3205ba7dd : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"b967edb8f2c3ce73d1e3387beeb5974f9f500083","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-09 06:16:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c1a4e0868740a5b77ee35aa58e1746c0ca24cd20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 06:23:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37526ceb18e548cf82a0e7445e2fb5c8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c002ef9570bc4098b5fa8d32e2a18a3d : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8fc71ad05f7e438489673f9c211f9f97 : SUCCESS in 5m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"144554b04b490b7b85eeba0370bfc2c2eea4c14e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-09 07:00:10.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"cdbf96169c6f28008084dce41ce1f1b2ab1634b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-09 07:10:35.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"0172185e8323ee4db3048f5e96943f38373c64ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 07:10:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3507d3c5eeac4c4f91de3cf3159db150\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7359e31ce4f245e683653386efc3c090 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b7d0aeb284d94bed8c75ea23b580beac : SUCCESS in 8m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"518464f463953604c2c9e5dbd54eca8723e929cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 07:22:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd72b7e92f1f48c789d1d102c06d4ca5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2464b88634924beda3e0ac011187bcdd : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a1b7b9100cea40d4848ecf2ad683a461 : SUCCESS in 9m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1a0441de3f1ad0bd088be8bab9f09c62caf316e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 09:47:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc6ce02bb90543bb9a916f89c5c87df3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/765d33fb4fad4076afa10f0e79f7f962 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/924cabeb58e94e41a736dcb2cb2dcc2a : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f13f7f1914f241c48bd3c8927290a719 : SUCCESS in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6bb013313f5c42d4b78e960fe1e60feb : SUCCESS in 6m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c02ea44ef3024d6aa631981253ec053e : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c66e2dc0bf04290a94d76b47f2e1fe6 : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b6b9acbf0084c5694588a978a4e57e9 : SUCCESS in 5m 43s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/0b8ef81f2d3e44ee9192c1bc3da5ba57 : SUCCESS in 1h 01m 41s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/8baca90e7e75409db5d45fb38ecd354e : SUCCESS in 2h 34m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/5596c1a6d95a45c7b3e530481c0c398b : SUCCESS in 1h 55m 42s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/90d38dc8e93949cfbe259eef272f4744 : SUCCESS in 40m 17s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/43bc65a9914843f38cd73a39c7716f18 : SUCCESS in 2h 13m 58s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/68432799047c41489475afe5313ac899 : SUCCESS in 2h 00m 55s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6c1ae105681543bbbca1681daead4d94 : SUCCESS in 59m 27s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/72960a04c3164584b90a6290060a7ede : FAILURE in 1h 24m 10s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/f2491f190e034ef1a23ffac1fc8d2718 : SUCCESS in 1h 41m 04s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/40cc272936354c6ca447a9a8ba0ca7dc : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":5},{"id":"f73c088064dcc465fe26fa772885616c1e404b44","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-10 07:29:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1a60380e7fb6145214d9882e683b9847e081aae1","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-10 07:51:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"eb78f07d3ead77d0dcd3c70ee3ecd0325c0c03af","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-05-10 08:23:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"abec2900a32db46798c1d7d28a0eaaec51b0a9af","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-05-10 14:05:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"367ae2ea504c7ab56b54e2cb34ac597cbe7c6270","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-11 06:38:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c5bdea8c90edf20d33e4600f174ec92d2a90d8f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-13 16:41:00.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"7072d3874879d370535b5ed0ee1df8454dd1a576","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 16:53:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/182c8ccd9a08402d8c0b515e13a69ee6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b32f2cc021e45079b316602c4598088 : FAILURE in 6m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eed68533a8b94e66a5af135b6dbf3907 : FAILURE in 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"717ba4e5cb2217994685ed65686e73840a06d194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 17:08:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e280ee455724b978043eeb87a43c746\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a27839434d664b1a8915217bc812af8d : FAILURE in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d8866f85b98415bbc3b4aaac5823521 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7d4709c9ec04de7a90ba919a66b5d02 : FAILURE in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e504427c7984188ac1fd2107d25c12e : FAILURE in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee93f743a07049d88dfa09e41670150b : FAILURE in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7075ef6d2654fcba3b6c2d1859f2319 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f9b163f2515433fb3231e8d4439799e : SUCCESS in 5m 15s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/6f73325c370e465bb4e92ed324e08e9f : FAILURE in 22m 47s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/5ce40cd990164143a40b38627628f1a8 : FAILURE in 17m 17s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/e37f38265a764fd7a0f7b1fcddd9b9b9 : FAILURE in 24m 32s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b34df8448078476dbc0cdfdc2486dd8d : FAILURE in 16m 49s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/422ba0b8ccf74a8382ee5c3f28686288 : FAILURE in 18m 15s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4e847ae0092546c68d7498e8d92d1d17 : FAILURE in 24m 56s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/fbd715cf00fd45ea9a58d63d0a8d6d5f : FAILURE in 17m 26s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/4bbe22a164054e8389fdd7470224f401 : FAILURE in 17m 17s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b8b271f9951d4f098f2023618ca684cd : FAILURE in 25m 06s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c6ec8e64b403411aa23dffb539c8b65f : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":6},{"id":"ee343396d178868b81846a45939dd432a5ed5578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-13 22:59:59.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"ee9687c0d5ffda30135211d00a90de0531b3a6d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 23:07:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b66e9fd74d6440e5989dd4e61fcaf1bf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09653926c28340dda82173ce032e7b9b : FAILURE in 6m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8f7a6dd1cf304175ba2c0b151cc19137 : FAILURE in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7625e85aee44efe6b99d80e9f6c465ab9c361958","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 23:32:49.000000000","message":"Patch Set 7: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a756445a98354bfc9314f249800a1f05\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bbfb99302cb48a8900e29a9bb94a377 : FAILURE in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea446c27766c4ed6877114530eff4d13 : FAILURE in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2c23c63cafc4c7eb93e82764487cbd2 : FAILURE in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eccf203592524d4ba11dd8af30fe8c84 : FAILURE in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41becfb01d324c82a87c48b71514d7f9 : FAILURE in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab2172a4516c48e2981db163cb1adcec : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0eb71b4688e4549960fae7282042be9 : SUCCESS in 5m 48s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/992782a468634c24b75b96aca945f606 : FAILURE in 15m 48s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/383f56f5f6e24d4fa8c6d388cda5a42b : FAILURE in 22m 30s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/bbf3caa685c6405e8def82a20e661341 : FAILURE in 18m 41s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/75b058703c094069b69ac1437c7ae32b : FAILURE in 21m 07s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/226c06219793456885ca3cea89dab4ab : FAILURE in 16m 08s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/10ea6825cc784754b4bb777a382727c2 : FAILURE in 22m 16s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/627a4077e8b74297935aaab142734611 : FAILURE in 17m 55s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a907f299ee4941b4987e8f41e9526560 : FAILURE in 23m 14s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/702248cf8a154e32b72464a555371c64 : FAILURE in 17m 20s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bc2d88886e094dbcb499bc5ea96eaa2d : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":7},{"id":"9b8dc51fd34a891c39fc9f6676091e10f6aff0fe","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-18 08:33:02.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0c050a17db43f7deab080b80174e53db35d1ff68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-19 15:16:26.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"c03c96dd2ae071f066af263c2f0224976b451c81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-19 15:21:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99cb396513454a649e54f6a166c8c5f3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7dc945b5e2434671a1a9390305cf7e42 : FAILURE in 4m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d50d628c1eb4499e8e36999d76e22f41 : FAILURE in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ab2bd3d33a0d238fa4007cda788d7b05b073699f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-19 16:04:40.000000000","message":"Patch Set 8: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d48e7e49cde6443394ff87d64edf66d3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05674bb2b1594bf9b4ecaf75b57d4a9c : FAILURE in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a945321b3904ba6a7ceaba855934790 : FAILURE in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9419327503544bea1367cbe476978b7 : FAILURE in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b01a7a613f1431a87db7169cbfa400a : FAILURE in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a46043151e54c7686edbfb01af39fb2 : FAILURE in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c7b1bcd91da45b3b048ff50ddfbc261 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f5bfd6699fc431b9ad64000be58397a : SUCCESS in 5m 32s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/99f02b1e100148f3b56964068e60c352 : FAILURE in 22m 15s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/a3d0eb13f70e4def8d3b2a772ff519dd : FAILURE in 22m 50s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/0b08e6a23486486fa8ffbe1c3831b091 : FAILURE in 23m 05s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f932ad2c027742fab6b34c98a2fe3cc8 : FAILURE in 21m 58s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/09ec13bdc29d4e7c8f6fe762b7c72435 : FAILURE in 22m 48s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5b84bd6084be4bbc9434135dcf096368 : FAILURE in 22m 45s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6b208897b5db43f3aefca4d62d4283e8 : FAILURE in 15m 32s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/b2fec369aab24fd0a7f19c9455207e7d : FAILURE in 16m 01s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/7ef3fd43400d431dbb298dcceb5c1ce4 : FAILURE in 16m 06s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0235623ab1ed4c1b9da26646856b1258 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":8},{"id":"85e5d1f8e032723227007e88b9a63ef325e42e81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-19 22:28:56.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"a866cf18c69e287514e70e44e7798eb676578551","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-19 22:35:59.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8275da8e7ac04835b1052258aacf9f7d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0693b36f7c446b0a71c089f72ac5fc5 : FAILURE in 6m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1b7ff078b08f4b9f99598ec873a37982 : FAILURE in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"51dc54ad466dc2ad2e22711333dbbaad32b1611b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-19 22:52:38.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c15f05fbe7d44160b4cbc999e2dd9292\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33526e032d3e44e19ebffe968b8f67cd : FAILURE in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da2c4b7ab76e49b6b5795fc6ce516961 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9c504df30d24f8bb6ff3424fa143d5b : FAILURE in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d23825baee414229b5c327763d8418f8 : FAILURE in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d37cff7e5504fabb26bd21ccd68f8d8 : FAILURE in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57930aae9ef94d8c83c0ba73ddefaa6e : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40ebbd93ad414247be6b17674c6a6af4 : SUCCESS in 4m 39s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c7992735ffe346dea4f096aa3bf6323c : FAILURE in 21m 58s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/cc17ca7e2e99403db1f92b55410f34f5 : FAILURE in 18m 37s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/fb35e08faba14307b2bf32959fe63468 : FAILURE in 22m 33s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b337dbedcb0446a69a6f4003ccf019ea : FAILURE in 21m 54s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d2067c134adf4970bfed244f678901c8 : FAILURE in 18m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/43021c616ad54eb399432128efca78c6 : FAILURE in 22m 35s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/fa943aebeac2450c8587d4c9b8ad51d0 : FAILURE in 21m 28s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/1539b592dc3247a49cadbeeb673be989 : FAILURE in 22m 33s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/f06bd8e579c043859520ebf68a1ac3f8 : FAILURE in 18m 38s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f931057e271f405e93761c5f5e7db2ad : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":9},{"id":"fca0d11633fcd63a95b9731ca953e841c8823d18","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-05-20 03:22:51.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8f330b4f694f4d4240cddca582708996b583c98b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-21 21:44:51.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"1efae4cf02981e34dd754204ff9cbf485858d3c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 21:50:23.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff921056495b472dbc7cb8a66c1f21c0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d6d4a0efe58423ea78fb2c82b80c7a2 : FAILURE in 4m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2da9660a86564fb6b43359d30c926323 : FAILURE in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5eae246850bae2725c19db3abf50803d7870156b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 22:23:51.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d0b4afa4f1b49a8a42f68e94c02a6b9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/947acdfe451740018d6b7f68fcfd3d3c : FAILURE in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a26e2bad26c4b1a924b63a718e707d7 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e307b462a86430086526520df62c343 : FAILURE in 6m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/421194768fb7422f9d9825015db8784c : FAILURE in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f211e78e17324a7b82583a43b28d8db3 : FAILURE in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d43e663edc624c19acc36338d24536c0 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9fe801534e04cb5966cbe3b9a5d5b88 : SUCCESS in 4m 27s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/7a4e5e0106c342a48cdbcfb69c6d8711 : FAILURE in 19m 25s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/cf6d04221d8e43debcc322a9f510b2fa : FAILURE in 17m 13s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/b1053736b24d47ffb74f6cfce58286ca : FAILURE in 19m 14s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1d970dffd62449b397cb995e4f8c21ad : FAILURE in 15m 25s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/225fe21eff7a4b98a2dce7149239b8c7 : FAILURE in 18m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/010cf71651bb42d795e5ec7a39de40fe : FAILURE in 16m 48s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/b367674c1ab641c48f59dac6687e6939 : FAILURE in 17m 59s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/40382f0c2f354eb5af27355159bd4d8d : FAILURE in 18m 26s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/067fea944be94ceaad23f12512ab52e5 : FAILURE in 16m 57s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/febf3fc65e3549b893446629678c8278 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":10},{"id":"415729387fdc637e08d1b214581309c0d4404276","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-21 23:32:23.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"76380d2cb8de75e954f6f9a6f5dd5fbb8ad3a7d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 23:38:41.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c376f743b528490698fb836adbf0438b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e84a368d5de74074820c3be9a503057b : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/21e5325c72094d038760513e96a7b2b6 : SUCCESS in 3m 58s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"07476447d83ceb4f486a7d988bff357892638eb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 23:55:56.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37cb21052f64442ea16db49e4312bf6b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49d6edbf29864123b03f99923154243a : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7f88e3fb6b24e95bcc75d6c3efb1597 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fa695b2f5a34c6e9225bd9194d1b161 : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b522a2a403354467a7e5dee346837b8a : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22addd1a436a4141aeefcb434c40419e : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64dfbdad4c76466597f63a623de628d4 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e659929759e4791b6c0a9f7a0a77158 : SUCCESS in 6m 45s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/2e787d6ac7bc46b5a638c791f7484be3 : FAILURE in 20m 35s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/d0869d62b73f4408bd1e12682b081efb : FAILURE in 22m 19s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/13bcf6364f8e4d07b8612d37718e141a : FAILURE in 22m 20s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c4e54c5385964bb5a4b8937e3ec1252d : FAILURE in 21m 06s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/86588c563d834978864e9c2d738a4003 : FAILURE in 21m 39s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/cc6421a54a0c465199fc641089b94795 : FAILURE in 15m 54s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/411edcd58050413eb950bb27e2662d34 : FAILURE in 21m 22s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/21c39b1eb67145e89793a07364648225 : FAILURE in 22m 04s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/6e22738b7e7b42aca0920067274e1660 : FAILURE in 22m 22s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/98c0e103177f472993e75b324141c836 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":11},{"id":"096a3f53d27180d3d90cc6e90ac21e81585b8093","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-22 05:45:12.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"00d491acd048389831ec5ad302e809c3dec7384b","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-24 10:27:21.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fb60a0b3cba59e4e0c043e0f89b7e2004def56bd","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-05-29 05:16:18.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"dd57d6995c4c0859642700bf656a53117925c30d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-05-29 09:24:42.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"00ad29e443834e077adf5671871488b29eccc9d4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-05-29 16:43:14.000000000","message":"Patch Set 11:\n\n(16 comments)","accounts_in_message":[],"_revision_number":11},{"id":"dcb8a9dbd8c05e4125cb7cc5e48f8d0a156337d1","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-31 10:12:57.000000000","message":"Patch Set 11:\n\n(9 comments)","accounts_in_message":[],"_revision_number":11},{"id":"86bf2985c2daf11b959271f0a6f22d638bcd4ed7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-05-31 15:56:45.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"668421db48a61a16f5fbd765412e2f31a6dbd971","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 16:11:20.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea008bcc391d412fb8680f1561742f84\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b72202aec61e4c268f438752dec3a7f6 : RETRY_LIMIT in 1m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1f7956c6eb3a4923b244a3174affcebe : SUCCESS in 12m 03s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2168e671bedbe6885f73863461defc7d44fbcc1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 16:11:45.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea008bcc391d412fb8680f1561742f84\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b72202aec61e4c268f438752dec3a7f6 : RETRY_LIMIT in 1m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1f7956c6eb3a4923b244a3174affcebe : SUCCESS in 12m 03s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"01068cc7baf96c5b85f08fe371fced7b0798d457","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 16:24:48.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5a66216e23e4a56887071097c9a3664\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae692987f4bb439faedfcd12d4ea9eb5 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce759b36923c40879fcf4be5df64b350 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78266c13dd6547f6b881b4b19e056cee : SUCCESS in 6m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c3ce36e2ac3d4c15a1b27bcfd4dc3eed : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/58941ca5237b4cdd9bdba7f3814d272e : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0f8074480ad402a8652757e52d6df28 : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bedb6aaf19be43e6bc63333a517fb263 : SUCCESS in 5m 10s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/558b4be8d96b4eefb13f1be478b5bee1 : FAILURE in 23m 46s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/5e61010a1efd462f9400e36a20525d5e : FAILURE in 19m 01s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/b4bdb9d73757494fa32f7ab775d80b95 : FAILURE in 25m 59s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/89cd5deee95846388f9302d6a820dfc1 : FAILURE in 19m 50s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/77251116f7354bf2b710b62b10b5ed3d : FAILURE in 19m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7ccce4c00ca0410abb07b5da7b2f8311 : FAILURE in 20m 11s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/ab7845065376457a8b84cdb101eb0b89 : FAILURE in 17m 03s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/06e769cf9d6a4170b62d7afba2f2cd84 : FAILURE in 25m 10s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b93399e87fd247e48f0feacce868f4a9 : FAILURE in 18m 07s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c7d81c7049304fb18fe73a4bf5f5b733 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":12},{"id":"2b73f4bfa22d59b38028a38f0864cb7205d8141b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-06-04 06:25:43.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a24677fd19f1f9e5bde1f081eebed871343e3c81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-06-05 03:21:07.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"ac9ba3d31027ff15ec11af2014f69337698c19a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 03:29:36.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f4cc0f6f2bc46a59b1f78a3e4bb283e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9b48c5ec8e549bf8ed7ba169b3589ce : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9a617d96c3fa42609b90825057988923 : SUCCESS in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b92f6f45657d297770f75aaecc773d44992eef96","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-06-05 04:22:04.000000000","message":"Patch Set 13:\n\n(7 comments)","accounts_in_message":[],"_revision_number":13},{"id":"f2833eb1c91ce33c4586022e43a6fdc597279d13","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-06-05 04:43:08.000000000","message":"Patch Set 13:\n\n(4 comments)","accounts_in_message":[],"_revision_number":13},{"id":"58700b1f6c0b57e316df3c5a5eca48b2adfba942","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 05:15:36.000000000","message":"Patch Set 13: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/171bc1dd29cd4e928be123b40c08adaf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13ab630453af41a09911bfa4070b224d : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e29f4b1467e431a8f645ab228ec7bad : FAILURE in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/070468443c0945029a05d42250ab83b1 : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/960af15aabbe4ff5b19bbc1478ff2996 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f87dbddc2f5c4839b55781b7a3fc7cb8 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01f0b38b31374ae69631b0711343db94 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c17a90d50d56456abd507bf273ce9658 : SUCCESS in 6m 20s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/b1c0a9a2afd24119bfda6b97ba288a05 : FAILURE in 17m 45s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/3bf54870041d4a909f8d81c646cba54c : FAILURE in 16m 08s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/b98c1c5b669147308092682561aeb6d2 : FAILURE in 19m 32s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d597da48b8d346fb8eb694d90fe70ce7 : FAILURE in 18m 59s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7e6688ee8ee240baa0de5c5ca1dfdd26 : FAILURE in 19m 19s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fdba2c51443e488e89ff8791d0027515 : FAILURE in 22m 07s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/360fb2078a2f47fa9dd8f8c093376f1c : FAILURE in 18m 41s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/0585280b89a641488694e27678a13019 : FAILURE in 19m 01s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/3cd2161d3b5b407daacc001926f296a8 : FAILURE in 19m 33s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6132eabdbd4c43af966e79d41d01debc : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":13},{"id":"0acaa6a681d2c8f9b64a2e775a1378aa775e03b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-06-05 05:47:00.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"062186cb45663e76c9dfe3bbcb8ee8cc6ff56cd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 05:52:57.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2287026017104ec7be7f666483decadf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1d8cef5f9a7d47a8b774cfeb454012ad : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3266170131df40fc9b87ce891beb25bb : SUCCESS in 5m 09s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"076557da7846a386af04d02e926cdc35503029d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 06:07:44.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bc8ca69c6344938bb66cc9e2c29ffcb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6717ef95cb954354827200071445f814 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98ea206487354927947352eb56fccfa9 : FAILURE in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/058f5d41aa98452fbf34aca5f2aac595 : SUCCESS in 6m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf0e55be592e43e49195eb2e0635eb16 : SUCCESS in 5m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ad30b4920b243a785404ed028549a71 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1420fff83fa4fbbb21e080363888146 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63acd9d12d764386ba4da0b3bcd33f18 : SUCCESS in 5m 37s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/5511436f9fc04fbb95d24f3e9e57281f : FAILURE in 16m 41s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/5954a76bbbc04be696f493e6eb62aa0f : FAILURE in 18m 27s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/2c27922eda5d4a0fa56963cd10db72fe : FAILURE in 18m 45s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/01cabd5b191c42a3832686973a8374e7 : FAILURE in 17m 24s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3ca66c87952f4c77863348465925cce7 : FAILURE in 17m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/332d59f62e37402aab34d5937e8ad90d : FAILURE in 18m 28s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/12d2fb0bf1384fa9aae78f254743f861 : FAILURE in 16m 34s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/d310f06f4cb34755a30b2304871e6865 : FAILURE in 18m 49s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/01faf6a8e0234cfe9ca7a184ee0f7028 : FAILURE in 17m 34s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/57e5f006c16d4487b08c5a2084d092fe : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":14},{"id":"c18dd476bde21b23c617c8ae0c9a3275823e8ada","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-06-05 06:10:39.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"c87d0274119b2a4bbf84aaaa4f6ef803216a691e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-06-05 13:49:40.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0d006759703f514e0cea0f5f6efdbaab89d13eb2","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-06-05 22:46:28.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"19f561428a7bce01827f34d1fdd630f54bcf2e8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-06-05 23:18:19.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"49b0d7d9034fd9cb09b9e7a0b3a9d178f0961aeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 23:24:30.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83169275489346b087e25e93e059aba6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d71660773ec44e5a89cba7ada165a82 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c1a11e2dd1214fcd9e81760ed3834b8e : SUCCESS in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"2ebed0054413ef717bab954ebce3b8ea6e1a1018","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-06-06 00:06:24.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"f5a55ebbedbd693bf40d0792a5b6190d544581c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 00:32:04.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ede75f4e8a44e9e86b6b3529b6d6081\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9311af7ea2954c7bbce05a5ac01403a5 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8594591cf97145b0918dee53f7020be8 : POST_FAILURE in 7m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b2ee1e13d7e44a9ab7c8cec906b7c6c : SUCCESS in 9m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59fcf14b623241afa656f121dc6dd21c : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c805916b046f4842a1a148ef338cfbcb : SUCCESS in 11m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95bd720929064a718182170a64c975b5 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ba3cff28d2141d09755fae8014ddcc2 : POST_FAILURE in 34m 41s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/85b0c744710846d9ab971e841eac709f : FAILURE in 9m 36s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/d71fb561387942b7a0d9268685e0adfd : FAILURE in 11m 45s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/c4c3d653d0d74142ba7aa9b72fa23c69 : FAILURE in 1h 10m 32s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/be9c75a3f1364a07b93249e279a1f737 : FAILURE in 56m 58s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/33f9a7e8dd914857b3ad6822c48c9999 : FAILURE in 49m 07s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c77c10fe7b814ef09f7f2caf47589e92 : FAILURE in 1h 08m 52s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/edb67495e735473fbb45981066f3a80f : FAILURE in 57m 26s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/18ed36e227cb46178d175c7641738879 : FAILURE in 44m 00s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/d67db50947f44e53b4fc04632174912a : FAILURE in 47m 44s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/25b03add7ba9457086e0222ec8b93497 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":15},{"id":"9f769deffb1f1d421132e4c0791922248220a6e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-06-06 01:58:55.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"892e04a6bd13956addb2de85d7db273e08cc629c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 02:05:11.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68d3c44110ee4a8f9a714c60d65b9e8f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f42ed9d323a54f8aa53babd1deb356d0 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f1b50b6b55a949b9a04dc0d77e06a56f : SUCCESS in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"23953901007f8c455b5379f615fdad30717b5b4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 03:23:07.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f656a8e3bd6f4d2db1f8b2f2a0b27276\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c14dbe4b64ce41f0ba0eb708d8173238 : SUCCESS in 3m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6caa226eed414c7f98020597474f1e56 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532dab8617d6452d8d41e447dee369ec : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c706b3aa52e4e9ebf1830820a172437 : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3cb02cc71a07459c8eaa0cd10ac8ad9d : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cefe4d1d995e46b1b4ffa4d71cccd507 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/084e588757904d1093edfa02aa0af864 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bbedfe358ce41028248d784e5a788f9 : SUCCESS in 7m 03s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/7de7b2a22b144f97ad9ffa5070a4789c : FAILURE in 45m 23s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/6747e0425a864857b3e7510cd9c90d4a : FAILURE in 1h 06m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/250098a191e34b8384cbfde98372f396 : FAILURE in 39m 24s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fc01c6252e374dd5ac6506b4e8df8c19 : FAILURE in 54m 09s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3d7f728df91e466b8ef76ed1ebf1ac13 : FAILURE in 49m 38s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e1ee6d56a80342828451162cd99eb52d : FAILURE in 38m 46s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/63361647c1bf4c47a1143fcd36af118f : FAILURE in 41m 47s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/8320e50eba0c4a85ac14f007d59c13c2 : FAILURE in 36m 33s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/e6f10f8d21394acab949a309d81a42a3 : FAILURE in 53m 02s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/db1895be5f394022b24c39587781dfbd : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":16},{"id":"ed4c334210d71a7284a28005139e20959b1d02a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-06-08 06:18:44.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"9d9644b40fb7a64e5f176bb3d39a117b9decf628","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-08 06:26:08.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e34c01176f947e69d015a8ee909d191\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4904b9c97ca64457b6bb7a84ad45d01d : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e2fa2c7279734bb08fc0efd1572c8c3a : SUCCESS in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"b385f9e9a1caf020e08e78507bc68dfbca7c3cc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-08 08:01:06.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3fba662057841a4b36b2fa774e38d17\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/28c96ffddd2a4b57a0727048931eb360 : SUCCESS in 3m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45782e4e038241e4a2accb7764fe4c7f : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aec1f1c1dd8c4f52b9eab0fe8d6ef093 : FAILURE in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9af65dbd5921442c94a6ae9d54c752d4 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1250b696d79f4e6cb8c2b9ff39828cc6 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6a21817477804b2e81c3444a953fc3c9 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4885e60e4b4fecab14d2f2425c0742 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c07f06c77fd4bb8bea816563e47f961 : SUCCESS in 4m 53s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/119aa6763ded485cb333ce60885081a7 : SUCCESS in 51m 08s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/d030e96f2bdd4c92a85dfabfb7953c4d : FAILURE in 46m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/c5b9934b75ab49068cdda412b83e242a : FAILURE in 44m 50s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/094ab52dcecf432891454e4249196a7f : SUCCESS in 1h 00m 23s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/939ac4ed5cda4718ab1db46e1e5832ea : FAILURE in 1h 33m 02s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/54e8918e69fe47aea3805e5d39499bf9 : FAILURE in 1h 23m 50s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/e12f5204d8d04b8394eda1afab2b5a1a : SUCCESS in 44m 36s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/12dcde568f3e489aa7952156bd4d75d6 : SUCCESS in 1h 41m 19s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/f4e0d32b030c45e981336b0e973c86a1 : SUCCESS in 1h 16m 40s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f92143e828ed4eb59ee45ffd1b61c4ec : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":17},{"id":"b7b6e28b0b45e755afeb52af17e16a3443831014","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-06-08 14:42:30.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"bca93a368582a776d7034aa6bd323f50b75b5c2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-08 14:48:32.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5762e9ea019746a2b429cb52152345cc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/49896601ee614968bdf2ab2352c0de99 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8f4e8710ad7e40429ad149d53bd94466 : SUCCESS in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"2797a5b34a6dd8221732b72e55f19aa493f10fde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-08 16:54:56.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d262c2396194392bafda138317b9fc3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c6bc26bb42464e34ac1d336a19fea826 : SUCCESS in 3m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d333ee93d0fd4f2a8f903db2e701d960 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/424b609b62d447ea9ddf5f372aff9f6a : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/076012eff7da444cbc717ffbaa10a3d3 : SUCCESS in 6m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf4c66f736304f63a4846915cddd8ad8 : SUCCESS in 5m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/deefe9f208b94e2c8949b4731b6f1099 : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37ee188a160e4dfc89b96a3bad46490d : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/182429a85d0445b782d7761e846fab6b : SUCCESS in 5m 34s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/9892324c55d848b69e93d70b8b6e1c68 : SUCCESS in 50m 27s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/6ee1a23f22b149ab9e8f9a9609696da8 : FAILURE in 2h 10m 43s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/840bd1236c084eeaa8864b355d436e42 : FAILURE in 33m 11s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/216a899f6c564160877b4764f6fa41b4 : SUCCESS in 48m 43s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7b4a14172a3c4b1da142da93e55c0d30 : FAILURE in 33m 48s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1fd95377a3c64a1daeb679e4af25eb32 : FAILURE in 1h 21m 12s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/b6d364da8e6f4649a480a7beedbdca49 : SUCCESS in 51m 15s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/249f693f63be478caf30a20ba6cd6e9f : SUCCESS in 1h 30m 33s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/129d8f6d11d0494794a1845445fb3760 : SUCCESS in 1h 20m 09s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2941bae67d8845e68342f4c1be9dfaec : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":18},{"id":"cf121a6f3207f0fba0755f8042c4683d2500ccec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-06-09 15:07:00.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"dfc87a6a084bbb44a0c032b01688dbeafb97cce6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-09 15:14:51.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/743c3135d44a437a861fae297b49aad1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/498b3c27d05641cc9cbfacd9f70e040f : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/35e8021002824b4480a1feff6a29df99 : SUCCESS in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"98169acff7db764e71ebd3eceb2e8bca79c052f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-09 15:57:46.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c34891e5a164208a2b3ac245f7deab9\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d57656e791bc449e939ab7b8cbe81fdd : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dad606eada69409eb1fb42567acbbfed : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f2b967b81d24a0e9e33dc69553d6397 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7e0a8eb30014040af2c594cf344bafc : SUCCESS in 8m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03850a08b7ad4d83871d993ce7b88d85 : SUCCESS in 7m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/15dd1c8e3c114304a677bac1d4ae1565 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/510226d5c9c64d5dad2ee69336131640 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5e6a344bd4b4ac2868f5ed700e43b44 : SUCCESS in 5m 32s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/0574a07dbc024d258f3b31ec86ceda35 : SUCCESS in 44m 54s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/982e960b7d5b416b89452fd952f4f03f : FAILURE in 33m 38s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/873ba1d85d054cbda1539597f83d6995 : FAILURE in 30m 39s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3888bb45eb77428494bb9267090d1f7f : SUCCESS in 42m 37s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8df8f995b51341389539f2a09141d5f2 : FAILURE in 31m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3a5cee18cd564e67aad288748e00f41d : FAILURE in 31m 49s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/bf8c42172334478f8f57b86e67025e22 : SUCCESS in 44m 28s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/24071ef6bc7c4d6c830bc66e9a36d9db : FAILURE in 33m 43s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/65fd45a7a91442ed89455395489aa459 : FAILURE in 33m 26s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1c5143c9b6b649dcac72aec3c23bcc37 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":19},{"id":"2c902f0fdda7cdee512aef5d3c75984587688cdc","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-06-10 00:52:15.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"507398a69c5ad03e81a7d725257bd6826a8f4942","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 01:00:03.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f1b7caf2e114a209fd739c4fb3ae156\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7820cda474e64c658f3a6ddc996fe71e : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/af1000c85e664d5393fe58ef4e2f2539 : SUCCESS in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"f612c64545696e0903619aab36e8e92685847cb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 01:57:08.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbb473b2679f42b3be05475176873ac9\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2bc4beadc21b4f5b98f9dd2e0d7293cf : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fd9f58f809a496b909036b47ad9b98d : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaa52def955f40a598f80c0311d4b64e : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0b97a8360e5426aa88925e59d1c240a : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2dadc9d0c8bf4fc39391ce7d779be0b6 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a16f45a90064e34bbd45497598f61c4 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f78c6dc3904ba798f0a5aa2185999a : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1ad64523ea0479d8e4dfe7f1c9a11f6 : SUCCESS in 5m 38s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/69e3293735ee4ba49f1a4e0048c6056c : SUCCESS in 45m 20s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/b0a1fef67bb94707a8555d7c1ef03725 : FAILURE in 37m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/53e375aa0dd4472490906f40bef1ed29 : FAILURE in 43m 20s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/328a7647d1e84996aba8370d09363d78 : SUCCESS in 43m 20s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d65584d7055949bdb72d897abc80e27f : FAILURE in 37m 04s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f9e5969036fb4de4bc571a4b2f41ee20 : FAILURE in 35m 24s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/72cab49f70b942b493ace1d3693b91d5 : SUCCESS in 57m 49s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/80a1b342d2794beaa7e75de1c344e7b3 : FAILURE in 36m 35s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/32312e32578f468295883950ddb7b72f : FAILURE in 53m 49s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a9011765f1774a1c96f7415cca946201 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":19},{"id":"685d46c14a71d1d0186601f4232addbe5e088594","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-06-11 12:29:16.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"71a748a801aee065903e189caae60baf7ae95f21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 12:37:04.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5719eed4649443a8a31e8515be9bbfaa\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57762f06b75e4b25841f6c4d67722161 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a09fd2b4a9824cf589f4712feaa21cf7 : SUCCESS in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"a85d9ad34703e61dfdb36e0d5ce1848413fe3c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 14:11:03.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93a975f67d4e42debe59da3d60fbc4db\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6f8b21f0c8104d308ec61a8a6e3030bd : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60750dbd581f4368ab8c4b759c5434f0 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/973969bb086d4895b4063d5412c9a1c6 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9164b2cd1c184158a70e79863ce7ea54 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/49b3901b4d5e449597e7e6d3ab4ee168 : SUCCESS in 6m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3317f6a250c7469f8453d020cdf83f7e : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce5f7fb60344ba78b239c3d3ed4f37c : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d7553f0ffa84a158487ef0862a016ed : SUCCESS in 5m 34s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/55d130e90ff84506ac175cb1de80bad5 : SUCCESS in 43m 18s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/09a05af0e98c44cea93fc8734571d879 : SUCCESS in 1h 40m 29s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/b2cb4a09d75b46fdaef2b694c1b366c4 : SUCCESS in 1h 24m 10s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1bd0e55488dc4123a7d26409f3852e7f : SUCCESS in 41m 17s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/deb2218d4f0a4d9b9218191e3c41e32c : SUCCESS in 1h 29m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ba72a4977edb4cf4af89d727d2f31905 : FAILURE in 40m 11s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/1d710e18172f4bdfa97d1b69b4e95e5b : SUCCESS in 44m 06s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/4bc59242859e429b940b791a935ff9ed : FAILURE in 57m 03s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/04cb73682fd84de5a691ce51022c06d7 : SUCCESS in 1h 17m 06s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dca74e2a116e480299deae83e8c865c9 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":19},{"id":"9920d598b63e43214ed371d5ebb0191a5a641e5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-06-12 08:13:22.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"46e887c36a54abc5b2b15919427922fb32bb4521","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 08:23:25.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2282f9101b24be1b1a91e8244ea50aa\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/16a57a96346647eca1b09e946653de2a : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b3cd73d4668d4292b827e34a9dfe5352 : SUCCESS in 7m 56s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"14b7bc2766f95606836782d1f88d78e136742bb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 11:03:22.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9844e398ca6844318fcdfdf49314a45a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b23b001c06ef434eb4386c3caf710266 : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ab629bd10d54b20a2028e95d27e94f5 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f839501d3ee84f8bb5da99caf62ef331 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8be8376d72e94ca0a4244f13a81bed1a : SUCCESS in 6m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb3a3ab5b8d2498d91285f6d59cf3cee : SUCCESS in 6m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/33944144fa35451eb4e0ff67ce642b75 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ab6de7fa2dc4e81bca75b3486236c0c : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/611d01c5442a43779508b6526ad3785e : SUCCESS in 5m 45s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/752e0abb860241bb9d04d639210d0fe6 : SUCCESS in 49m 22s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/7248a48ea20540c5aa20c9a7e2567f91 : SUCCESS in 2h 48m 07s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/991bca537f9a48f088137b096ab6d6ef : SUCCESS in 2h 10m 50s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f0264138e7af4907b06104e0e5b00254 : SUCCESS in 1h 00m 05s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b4157d35db6b442fb3a172c4f681904e : SUCCESS in 1h 37m 50s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9c3176b8a3134a88a5a8690947b283a4 : SUCCESS in 2h 07m 47s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/91b9e4f69ed349d7bc20ac717cd0a23c : SUCCESS in 46m 18s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/11ffbf9823de4d9f9362f2f78820db90 : FAILURE in 1h 08m 22s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/5803e6c136f144ce99be283e4282c538 : SUCCESS in 1h 33m 58s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9ba067d3781a40c8a59bebe6b2f94af1 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":20},{"id":"f9d24ea586aaa94ac0d1c139540e0360e6adb243","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-06-12 12:27:12.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"a6843e53a64a11090fb211da396c3b3c5256efc6","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-06-12 13:03:32.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"c4e055ffd1aa4d6cec2aa36e713d27957b756545","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-06-12 15:14:01.000000000","message":"Patch Set 20: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":20},{"id":"34de565edc7b69ecd0f0f2b451de6a0c002865c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-06-13 08:21:04.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"63711a8b34a8c8b4f5e6d679a7c1e5c3cc1ddfe7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 08:27:52.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f489c0dad0e840449965aad81e14eac3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b299e4770e624dc1ae4c1488f5d39c40 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5f7aa68cfbb6492c84b472dd8349f4fd : SUCCESS in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"94e7a3a91943e8f2a51af70e037ede7337fff48d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 10:56:23.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/216fc8baa68047f0ab551c7813a1485e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/30450b9ef04b43fe94ec1581e6fa9527 : SUCCESS in 3m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad5f72cfd2384dc3ae3ee3a713ffd472 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f045680e677436eb25a00a82b4f1e8c : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd13a641b0bb4380b1336de629d5b896 : SUCCESS in 7m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ad29e24cd784c2a95aaf8d5712f7535 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/498a721534614814a08fee85d076fc78 : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/354c304271f84f1ebc501e60609d34ab : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c7815f4835249ec8a891b5ba2ff2892 : SUCCESS in 5m 25s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/d707968251124e4596b391dfb84d78f7 : SUCCESS in 1h 06m 55s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/c0ef2497850948748d6fc2d54ae9695b : FAILURE in 1h 33m 16s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/c9bdb26804ff458d96625e233f93288a : SUCCESS in 1h 22m 16s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fed083543ba34beb9bf9b36c1c66af2f : SUCCESS in 43m 43s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e0cf7b2a04b04a739e428e1c83a0a388 : SUCCESS in 2h 24m 37s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/67474a06ebf04f4b9e1c9641fd891a36 : SUCCESS in 1h 34m 22s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/469331f2191b4770bf4d1fa97688a0e1 : SUCCESS in 44m 24s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/86ddcffc1ae640b5a385429a5b7686a7 : FAILURE in 1h 23m 52s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/ae19bd8423ff477890355538aa2a94a6 : SUCCESS in 1h 16m 03s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fe6e3e8fcb5b441585b6cd50ae9a4d91 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":21},{"id":"1f2bc2e19c13cbfcc7c417c20db460728dc08699","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-06-14 06:31:15.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":21},{"id":"72e8343f1b6aa1552f6178a667dc8ab3d67ca53b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 06:31:50.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"d024a87c22edfba846f35bf40f8c0b67535690c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 07:44:59.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7af403fb91ef4a35b269d48d9a418562\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fdd986c5a85b4974a2a815a7df300ffd : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/270ce2c66d9949d0b60582407a171b90 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf84cf04ccff415fab214c22df9e4d00 : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/528ba2c991644433b1c9c31d7215e4cd : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35fe5fe774ca40dcb1aade03a9590aaf : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30a6cd39364c403091938e5e9a0dc25c : SUCCESS in 5m 54s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/f8a53382cb30415cbe7c6c376181caf4 : SUCCESS in 39m 46s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b8d9deb9aa1b423fa853df94ef803845 : SUCCESS in 1h 09m 05s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/8cb5dceb2ad8497f9808304886967adb : SUCCESS in 1h 11m 09s","accounts_in_message":[],"_revision_number":21},{"id":"cb49db554d1bbb4c683cc57539a61155b12fcaff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 07:44:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":21},{"id":"8395cd365d408a79b10b83289da330a882fc18e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 07:47:26.000000000","message":"Patch Set 21:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fac2937d62040988e17dacc9467fe59\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c3f1b5ff1c944388ba36ea8f3d8b932 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f23b113010449d4b41d5101fc50b113 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"114601edba9fc736af43a3f2705552d8c746115c","revisions":{"f0fdcf9a89fb40a3f44de78f476a846f5facbd64":{"kind":"REWORK","_number":1,"created":"2024-05-08 14:04:09.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/61/918661/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/1"}}},"commit":{"parents":[{"commit":"78088c57170a452f885133b454cf16797a8ab44e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78088c57170a452f885133b454cf16797a8ab44e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 14:03:53.000000000","tz":0},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work.\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f0fdcf9a89fb40a3f44de78f476a846f5facbd64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f0fdcf9a89fb40a3f44de78f476a846f5facbd64"}]},"branch":"refs/heads/master"},"033798e29a4b9619cdaa4734d1fff39b5a8f4aee":{"kind":"REWORK","_number":2,"created":"2024-05-09 02:21:44.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/61/918661/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/2"}}},"commit":{"parents":[{"commit":"78088c57170a452f885133b454cf16797a8ab44e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78088c57170a452f885133b454cf16797a8ab44e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-09 02:21:15.000000000","tz":0},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work.\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/033798e29a4b9619cdaa4734d1fff39b5a8f4aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/033798e29a4b9619cdaa4734d1fff39b5a8f4aee"}]},"branch":"refs/heads/master"},"5051ca472e622560eb5837ddcdc8bb5e184aab1b":{"kind":"REWORK","_number":3,"created":"2024-05-09 06:16:58.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/61/918661/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/3"}}},"commit":{"parents":[{"commit":"78088c57170a452f885133b454cf16797a8ab44e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78088c57170a452f885133b454cf16797a8ab44e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-09 06:16:24.000000000","tz":0},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work.\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5051ca472e622560eb5837ddcdc8bb5e184aab1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5051ca472e622560eb5837ddcdc8bb5e184aab1b"}]},"branch":"refs/heads/master"},"3f35b016e7130db8c6510b7e21207278e263ebe6":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-05-09 07:00:10.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/61/918661/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/4"}}},"commit":{"parents":[{"commit":"78088c57170a452f885133b454cf16797a8ab44e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78088c57170a452f885133b454cf16797a8ab44e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-09 06:57:46.000000000","tz":0},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f35b016e7130db8c6510b7e21207278e263ebe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f35b016e7130db8c6510b7e21207278e263ebe6"}]},"branch":"refs/heads/master"},"4f1f5f678c7c15a8687e164bd48a699c3f058d14":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-05-09 07:10:35.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/61/918661/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/5"}}},"commit":{"parents":[{"commit":"78088c57170a452f885133b454cf16797a8ab44e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78088c57170a452f885133b454cf16797a8ab44e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-09 07:05:27.000000000","tz":0},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0. Users can use sqlalchemy-migrate if they want by\nremoving alembic configuration file.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4f1f5f678c7c15a8687e164bd48a699c3f058d14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4f1f5f678c7c15a8687e164bd48a699c3f058d14"}]},"branch":"refs/heads/master"},"38dff130ff28973eed68ae6e791b26fa918072cb":{"kind":"REWORK","_number":6,"created":"2024-05-13 16:41:00.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/61/918661/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/6"}}},"commit":{"parents":[{"commit":"78088c57170a452f885133b454cf16797a8ab44e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78088c57170a452f885133b454cf16797a8ab44e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-13 16:40:38.000000000","tz":0},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0. Users can use sqlalchemy-migrate if they want by\nremoving alembic configuration file.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/38dff130ff28973eed68ae6e791b26fa918072cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/38dff130ff28973eed68ae6e791b26fa918072cb"}]},"branch":"refs/heads/master"},"dc3cb68e02c780b998ad2d0f3dee5b8e74cce41a":{"kind":"REWORK","_number":7,"created":"2024-05-13 22:59:59.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/61/918661/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/7"}}},"commit":{"parents":[{"commit":"78088c57170a452f885133b454cf16797a8ab44e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78088c57170a452f885133b454cf16797a8ab44e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-13 22:59:34.000000000","tz":0},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0. Users can use sqlalchemy-migrate if they want by\nremoving alembic configuration file.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dc3cb68e02c780b998ad2d0f3dee5b8e74cce41a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dc3cb68e02c780b998ad2d0f3dee5b8e74cce41a"}]},"branch":"refs/heads/master"},"a9b67bde9af269b36a5c23dd64436f651b1e19c5":{"kind":"REWORK","_number":8,"created":"2024-05-19 15:16:26.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/61/918661/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/8"}}},"commit":{"parents":[{"commit":"78088c57170a452f885133b454cf16797a8ab44e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78088c57170a452f885133b454cf16797a8ab44e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-19 15:12:25.000000000","tz":540},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0. Users can use sqlalchemy-migrate if they want by\nremoving alembic configuration file.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a9b67bde9af269b36a5c23dd64436f651b1e19c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a9b67bde9af269b36a5c23dd64436f651b1e19c5"}]},"branch":"refs/heads/master"},"f81a87a2de282fd6f1976d6a173f05d894de5e7d":{"kind":"REWORK","_number":9,"created":"2024-05-19 22:28:56.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/61/918661/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/9"}}},"commit":{"parents":[{"commit":"78088c57170a452f885133b454cf16797a8ab44e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78088c57170a452f885133b454cf16797a8ab44e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-19 22:28:21.000000000","tz":540},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0. Users can use sqlalchemy-migrate if they want by\nremoving alembic configuration file.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f81a87a2de282fd6f1976d6a173f05d894de5e7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f81a87a2de282fd6f1976d6a173f05d894de5e7d"}]},"branch":"refs/heads/master"},"8c348c2ef08b1eeb81d6d44e8986c56dabe8997b":{"kind":"REWORK","_number":10,"created":"2024-05-21 21:44:51.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/61/918661/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/10"}}},"commit":{"parents":[{"commit":"78088c57170a452f885133b454cf16797a8ab44e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78088c57170a452f885133b454cf16797a8ab44e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-21 21:41:14.000000000","tz":540},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c348c2ef08b1eeb81d6d44e8986c56dabe8997b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c348c2ef08b1eeb81d6d44e8986c56dabe8997b"}]},"branch":"refs/heads/master"},"a18c5a0455fe2681a22b4bf59e8c575e87787a0a":{"kind":"REWORK","_number":11,"created":"2024-05-21 23:32:23.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/61/918661/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/11"}}},"commit":{"parents":[{"commit":"78088c57170a452f885133b454cf16797a8ab44e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78088c57170a452f885133b454cf16797a8ab44e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-21 23:31:58.000000000","tz":540},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a18c5a0455fe2681a22b4bf59e8c575e87787a0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a18c5a0455fe2681a22b4bf59e8c575e87787a0a"}]},"branch":"refs/heads/master"},"6aec38d63d87fa2e2696003300ee87f5c7add834":{"kind":"REWORK","_number":12,"created":"2024-05-31 15:56:45.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/61/918661/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/12"}}},"commit":{"parents":[{"commit":"78088c57170a452f885133b454cf16797a8ab44e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78088c57170a452f885133b454cf16797a8ab44e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-31 15:55:16.000000000","tz":540},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6aec38d63d87fa2e2696003300ee87f5c7add834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6aec38d63d87fa2e2696003300ee87f5c7add834"}]},"branch":"refs/heads/master"},"81c7b2c61eb2773aa3bd4d5a4f5f71fab70b73c7":{"kind":"REWORK","_number":13,"created":"2024-06-05 03:21:07.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/61/918661/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/13"}}},"commit":{"parents":[{"commit":"78088c57170a452f885133b454cf16797a8ab44e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78088c57170a452f885133b454cf16797a8ab44e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-06-05 03:20:44.000000000","tz":540},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81c7b2c61eb2773aa3bd4d5a4f5f71fab70b73c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81c7b2c61eb2773aa3bd4d5a4f5f71fab70b73c7"}]},"branch":"refs/heads/master"},"54eeb33e2d5b8ce2cd3d6831439298c9eb35b2ba":{"kind":"REWORK","_number":14,"created":"2024-06-05 05:47:00.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/61/918661/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/14"}}},"commit":{"parents":[{"commit":"78088c57170a452f885133b454cf16797a8ab44e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78088c57170a452f885133b454cf16797a8ab44e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-06-05 05:46:10.000000000","tz":540},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/54eeb33e2d5b8ce2cd3d6831439298c9eb35b2ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/54eeb33e2d5b8ce2cd3d6831439298c9eb35b2ba"}]},"branch":"refs/heads/master"},"60069ef4003aaff363c25672c005effd8f20721c":{"kind":"REWORK","_number":15,"created":"2024-06-05 23:18:19.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/61/918661/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/15"}}},"commit":{"parents":[{"commit":"78088c57170a452f885133b454cf16797a8ab44e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78088c57170a452f885133b454cf16797a8ab44e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-06-05 23:17:50.000000000","tz":540},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/60069ef4003aaff363c25672c005effd8f20721c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/60069ef4003aaff363c25672c005effd8f20721c"}]},"branch":"refs/heads/master"},"e594651b6322d6bb73c4df016235064adaa0432d":{"kind":"REWORK","_number":16,"created":"2024-06-06 01:58:55.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/61/918661/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/16"}}},"commit":{"parents":[{"commit":"7501804a9f321525e7f5e68f4c51ff5cbd45be70","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7501804a9f321525e7f5e68f4c51ff5cbd45be70"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-06-06 01:58:01.000000000","tz":480},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e594651b6322d6bb73c4df016235064adaa0432d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e594651b6322d6bb73c4df016235064adaa0432d"}]},"branch":"refs/heads/master"},"0c85e57a198b1c4f49ea3685506b53c34c80c430":{"kind":"REWORK","_number":17,"created":"2024-06-08 06:18:44.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/61/918661/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/17"}}},"commit":{"parents":[{"commit":"7501804a9f321525e7f5e68f4c51ff5cbd45be70","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7501804a9f321525e7f5e68f4c51ff5cbd45be70"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-06-08 06:16:33.000000000","tz":480},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/921589\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0c85e57a198b1c4f49ea3685506b53c34c80c430"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0c85e57a198b1c4f49ea3685506b53c34c80c430"}]},"branch":"refs/heads/master"},"7a83ad010a2f591c1ed646a9e0ec6367f7c8ec3b":{"kind":"REWORK","_number":18,"created":"2024-06-08 14:42:30.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/61/918661/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/18"}}},"commit":{"parents":[{"commit":"7501804a9f321525e7f5e68f4c51ff5cbd45be70","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7501804a9f321525e7f5e68f4c51ff5cbd45be70"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-06-08 14:40:34.000000000","tz":480},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/921589\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a83ad010a2f591c1ed646a9e0ec6367f7c8ec3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a83ad010a2f591c1ed646a9e0ec6367f7c8ec3b"}]},"branch":"refs/heads/master"},"45623c27bb063106dff0de680c5c53267ae47107":{"kind":"REWORK","_number":19,"created":"2024-06-09 15:07:00.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/61/918661/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/19"}}},"commit":{"parents":[{"commit":"7501804a9f321525e7f5e68f4c51ff5cbd45be70","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7501804a9f321525e7f5e68f4c51ff5cbd45be70"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-06-09 02:25:31.000000000","tz":480},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/921589\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/45623c27bb063106dff0de680c5c53267ae47107"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/45623c27bb063106dff0de680c5c53267ae47107"}]},"branch":"refs/heads/master"},"999872554e3cf4c9fa8e42fa67e452f047a9a8d6":{"kind":"REWORK","_number":20,"created":"2024-06-12 08:13:22.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/61/918661/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/20"}}},"commit":{"parents":[{"commit":"7501804a9f321525e7f5e68f4c51ff5cbd45be70","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7501804a9f321525e7f5e68f4c51ff5cbd45be70"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-06-12 08:13:14.000000000","tz":480},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/921589\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/999872554e3cf4c9fa8e42fa67e452f047a9a8d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/999872554e3cf4c9fa8e42fa67e452f047a9a8d6"}]},"branch":"refs/heads/master"},"114601edba9fc736af43a3f2705552d8c746115c":{"kind":"REWORK","_number":21,"created":"2024-06-13 08:21:04.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/61/918661/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/61/918661/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/61/918661/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/61/918661/21"}}},"commit":{"parents":[{"commit":"7501804a9f321525e7f5e68f4c51ff5cbd45be70","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7501804a9f321525e7f5e68f4c51ff5cbd45be70"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-05-08 12:25:36.000000000","tz":0},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-06-13 08:20:52.000000000","tz":480},"subject":"Introduces Alembic database migration tool","message":"Introduces Alembic database migration tool\n\nThis PR introduces Alembic database migration tool. Trove currently\nuses sqlalchemy-migrate with which SQLAlchemy-2.0 does not work. By\nintroducing the new migration tool, we can prepare for migrating\nto SQLAlchemy-2.0.\n\nThis change will not affect the default trove-manage\u0027s usage while\nthe database schema versioning will be changed because alembic uses\nnon-linear, dependency-graph versioning while sqlalchemy-migrate\nuses integer based versioning.\n\nDepends-On: https://review.opendev.org/c/openstack/trove-tempest-plugin/+/921589\n\nCo-Authored-By: wu.chunyang \u003cwchy1001@gmail.com\u003e\n\nStory: 2010922\nTask: 48782\nChange-Id: Idd63f470a2b941720314b6356fe28cd8e394427e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/114601edba9fc736af43a3f2705552d8c746115c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/114601edba9fc736af43a3f2705552d8c746115c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
