)]}'
{"id":"openstack%2Ftrove~923787","triplet_id":"openstack%2Ftrove~stable%2F2023.2~I232f53bdbba0d0a17dfda395b89300419879575e","project":"openstack/trove","branch":"stable/2023.2","topic":"2023.2","attention_set":{},"removed_from_attention_set":{"12932":{"account":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"last_update":"2024-07-16 21:28:47.000000000","reason":"\u003cGERRIT_ACCOUNT_12932\u003e replied on the change","reason_account":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-07-16 21:30:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I232f53bdbba0d0a17dfda395b89300419879575e","subject":"guestagent: implement \u0027rebuild\u0027 for postgres manager","status":"ABANDONED","created":"2024-07-09 19:36:00.000000000","updated":"2024-07-16 21:30:46.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9473fb21a32dd8ceab87aa84a6d5c9f89c77e383","_number":923787,"virtual_id_number":923787,"owner":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-09 23:23:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-09 19:57:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-09 23:23:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-11 02:20:59.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"}],"messages":[{"id":"0f3d979010f591493de48fa71c6c25e7b1c1b94c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-09 19:36:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7712660752275712bf9a30064553edc03fc7a2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-09 19:37:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"349b446c4f07909cf1843aac9ae5eaf8deb4ece3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 19:57:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7267b4eff66a4a8a9c81fdd1d96dac57\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fce0f0c71d943e6850816e99d84b8cb : SUCCESS in 15m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f03efba2c1984fce97043107f3d85b48 : FAILURE in 4m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/85eb3fc8b0694ccb9a99d554e6cb6f2f : SUCCESS in 17m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c258ed077f164c1793c6f5d5bbc14457 : RETRY_LIMIT in 1m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8aac9990b156c12994117c850156fe17a005e4bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-09 20:27:07.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e5b8bb16ac57b6c50c867efcf4d2683167248b3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 20:38:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cddeb791e9d4bed81521bcfdc796558\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec66b85315574aa69632a52f936981b1 : FAILURE in 4m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd060934cd8d4f278aa8bd3f7fd280b8 : FAILURE in 3m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0f4e2c73d17e4a8c9312a4563c012ac0 : FAILURE in 5m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0d9a78d685994253812134e613a31daa : RETRY_LIMIT in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f16049e970dd98e490a667a9a340216917d470c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 23:23:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93a9d99083c744dc9e219d70483fb9b3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e772ee1dfd146e1b0311d9c34b694ff : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d80671bf5f24387b787a655fcb4449f : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66f4e88abe05417aa3f115f4d3230835 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3153358a623c480181b2207a3f92c3d3 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5975e7ec121f4472bffdd4f78422ee00 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ca4a460013348cba6d0fe5287290b4f : RETRY_LIMIT in 1m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4138e5778e234b52a4f73cab7e250f71 : SUCCESS in 4m 48s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eb0d0c14799d4b338b6d7033dc44ca25 : SUCCESS in 4m 29s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/76bf1353c0eb41f0af87f3b54352d558 : FAILURE in 2h 54m 34s (non-voting)\n- trove-tempest-mysql-network-isolation https://zuul.opendev.org/t/openstack/build/4e4a0bc21feb49b3a9cfb244ed7fa0f3 : SUCCESS in 1h 56m 31s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/c9764bcd83364ee3935b5a1d29eae5d9 : FAILURE in 2h 52m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f3d69078827d87f6d9f6a1a3804111dea31b8952","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-07-11 02:20:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dec3e6270e4f5b743ff70875eb278c4ddeef22b0","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-16 21:28:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9473fb21a32dd8ceab87aa84a6d5c9f89c77e383","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-16 21:30:46.000000000","message":"Abandoned\n\ndropping in favor of https://review.opendev.org/c/openstack/trove/+/924285","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"65547c9747af89de4740e595ea3e69aa9ff1910b","revisions":{"4c06e686829142883012185cb2c339d3cc98cc67":{"kind":"REWORK","_number":1,"created":"2024-07-09 19:36:00.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/87/923787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/87/923787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/87/923787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/87/923787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/87/923787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/87/923787/1"}}},"commit":{"parents":[{"commit":"885df8df9a44b30726d13c30e9422879bc5e3a2f","subject":"Merge \"Enable network_isolation by default\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/885df8df9a44b30726d13c30e9422879bc5e3a2f"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-09 19:34:20.000000000","tz":-300},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-09 19:35:04.000000000","tz":-300},"subject":"guestagent: implement \u0027rebuild\u0027 for postgres manager","message":"guestagent: implement \u0027rebuild\u0027 for postgres manager\n\nThis is almost but not quite a copy of the same function\nfrom  mysql_common.\n\nChange-Id: I232f53bdbba0d0a17dfda395b89300419879575e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4c06e686829142883012185cb2c339d3cc98cc67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4c06e686829142883012185cb2c339d3cc98cc67"}]},"branch":"refs/heads/stable/2023.2"},"16e876188b3ad9d139b808553f4ae1c0c53ecc95":{"kind":"REWORK","_number":2,"created":"2024-07-09 19:37:50.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/87/923787/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/87/923787/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/87/923787/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/87/923787/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/87/923787/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/87/923787/2"}}},"commit":{"parents":[{"commit":"885df8df9a44b30726d13c30e9422879bc5e3a2f","subject":"Merge \"Enable network_isolation by default\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/885df8df9a44b30726d13c30e9422879bc5e3a2f"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-09 19:34:20.000000000","tz":-300},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-09 19:37:38.000000000","tz":-300},"subject":"guestagent: implement \u0027rebuild\u0027 for postgres manager","message":"guestagent: implement \u0027rebuild\u0027 for postgres manager\n\nThis is almost but not quite a copy of the same function\nfrom  mysql_common.\n\nChange-Id: I232f53bdbba0d0a17dfda395b89300419879575e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/16e876188b3ad9d139b808553f4ae1c0c53ecc95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/16e876188b3ad9d139b808553f4ae1c0c53ecc95"}]},"branch":"refs/heads/stable/2023.2"},"65547c9747af89de4740e595ea3e69aa9ff1910b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-07-09 20:27:07.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/87/923787/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/87/923787/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/87/923787/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/87/923787/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/87/923787/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/87/923787/3"}}},"commit":{"parents":[{"commit":"885df8df9a44b30726d13c30e9422879bc5e3a2f","subject":"Merge \"Enable network_isolation by default\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/885df8df9a44b30726d13c30e9422879bc5e3a2f"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-09 19:34:20.000000000","tz":-300},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-09 20:26:54.000000000","tz":-300},"subject":"guestagent: implement \u0027rebuild\u0027 for postgres manager","message":"guestagent: implement \u0027rebuild\u0027 for postgres manager\n\nThis is almost but not quite a copy of the same function\nfrom  mysql_common.\n\nStory: 2011172\nChange-Id: I232f53bdbba0d0a17dfda395b89300419879575e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/65547c9747af89de4740e595ea3e69aa9ff1910b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/65547c9747af89de4740e595ea3e69aa9ff1910b"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
