)]}'
{"id":"openstack%2Ftrove~924285","triplet_id":"openstack%2Ftrove~master~I232f53bdbba0d0a17dfda395b89300419879575e","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2024-08-21 04:47:45.000000000","reason":"\u003cGERRIT_ACCOUNT_28691\u003e replied on the change","reason_account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}},"12932":{"account":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"last_update":"2024-08-21 07:31:10.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-08-21 06:46:49.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I232f53bdbba0d0a17dfda395b89300419879575e","subject":"guestagent: implement \u0027rebuild\u0027 for postgres manager","status":"MERGED","created":"2024-07-16 21:29:47.000000000","updated":"2024-08-29 09:29:17.000000000","submitted":"2024-08-21 07:31:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":34,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"924285","meta_rev_id":"658d39b7e4d3a36d51270eb3b2cc74825c215bee","_number":924285,"virtual_id_number":924285,"owner":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-21 07:31:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"all":[{"value":2,"date":"2024-08-21 04:47:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-21 06:46:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-21 06:46:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-16 21:37:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-16 22:09:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-23 07:24:53.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2024-08-09 05:49:36.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2024-08-12 03:16:01.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"CC"},{"updated":"2024-08-12 07:02:23.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"}],"messages":[{"id":"8fcfb82e1b442c5ecef2032f08f10f847e5ecac9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-16 21:29:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e5fb4a6bb6b42c2883194dad66a204b7efdbeed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 21:37:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4d2eb73c68341fd8904ba65b6fc8fdf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3debc2516dfd40b79050e0e6c31c4489 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7c0e9bf9a8ee4df2877e3e0346e41c11 : SUCCESS in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b4ad23ef47ca90e792e8482518a2ad85d0141cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 22:09:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2990e152c5914c9499f6d8c7fc246196\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/412776a5efc34ec4843b7a497edec8c5 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae41ef6b3ccc4dfd839f95e513dd1888 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f979e6f457a14a879f8a33647494aa1f : SUCCESS in 6m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/24998156499f4b088f3b167b7655d2c5 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d89e2593b9b648e8bd27b8db499aca22 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34b1c35f5d404d909175dba031314eee : SUCCESS in 4m 50s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/3c14cea836824f988ba5cf62a4a171ef : FAILURE in 26m 29s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/96a5d58b12154fd1a00e7c57a2b32e43 : FAILURE in 38m 35s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/4ad4097d29ae4e95b60fde6ebdd88300 : FAILURE in 38m 54s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/894d3f3f8a2c4f6791668387b3e415c2 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"dfc1ada0b6aab6de08611bf057665554dd5799b0","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-17 13:04:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a0d6fde1221d456eef297bfbfa250ed5ad4b124","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 13:10:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9a11f29f7c64f2abbde9b069f07fa78\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf63f6c63cc842ac9632412aa7c4979e : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/28df2dfad7da4faaa2a09003068eb5e9 : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f3e5d0381b2cd3c431220ba8ee8925016566e64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 13:44:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e42f5c61c83a4c47b83e6c1d67f9470b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/738295e726bc4e9cb8f663c19088f4a7 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6048cd963d64261b643affda153f254 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4c09841404a4e8b96bfe696e195e076 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/222efb0e58a24567b78498bf02f6fbbc : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb8eb172f6294525b2554f39cbd1370c : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e73e1927b72b4b34bdc21903ed02d7af : SUCCESS in 5m 51s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/40abdb0bc7b044499a727910c6e10dc3 : FAILURE in 29m 33s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/d60e9038d4e34647abff35e1567efef7 : FAILURE in 38m 55s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/ba0485ac496f4a48990cabfc047ac3ed : FAILURE in 34m 13s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bdd7e49349eb4cfdbbf85cbcd16abb2b : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"717c8f9b60974b3779b9972bf6e6c643bf488f0d","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-18 19:45:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5155a103d1c65abd4f51cd321221c16d23f2ec3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 19:51:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/765842010917471eb2695d2abfabd251\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67dee4ee581e486facb792c7feef862a : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2795b9613627420686620da7240610c6 : SUCCESS in 5m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a8cf263a8d8cd03cb5221796374cc3bc863ad5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 20:36:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b99dc38202ed452292a0c94c38f32ffb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8ae1871ae4541698f7876aad77e9cd6 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3428a6542e2343d7ab49bd9eae625916 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f42e690a8c044a9a435072eeb4f22de : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b1c0ccacd57425795d081271dfff486 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16946d014a914f728df35b03435eee25 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e0baca51c44df78de44ac55ff2c7cb : SUCCESS in 4m 48s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/8758f5d0ba024656b4309f5a50e884cf : FAILURE in 23m 37s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/2af52873a19a4d98a4966b1e3a7684e1 : FAILURE in 48m 18s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/517e315bfce04e2aae5e22057873cac4 : FAILURE in 48m 17s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bf96e508a0214b959e58031de09e037d : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9cfa250ff52bff5f587cc193eaa856f2c812dd3d","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-19 00:21:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60a5d49e8835b0d46f2ed82f124428f834323f78","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-22 13:40:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"edf185eee50ad9b52e3660cc20f7551d42b47b16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 13:50:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de8a8bbce5c848c6bf0501fece4ea169\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/942d2fc2c5854d5fbe4225e0ec3c8068 : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a63dd22951d54f10be8a6e983844698c : SUCCESS in 7m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ee956bf797381a3d519ea39cf6ddab72b235a67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 14:28:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/435b69aa3dc146ab8242b1d44ffc0653\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71cb96fc5598415d86f65830f6378d78 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f8206477c104493b2a4bbcf3ec29043 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84d1b55a629343eaa53243c9561eb2ef : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/00533e1397d5429eb5c407cc99e42b3a : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b35156a20fa94fc496f9dead07e25df7 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b280f9fd8a634a408fcbfb9de7983cea : SUCCESS in 6m 06s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/34f23d0a15ab405f8a616e6feab7aa88 : FAILURE in 28m 53s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/181bd615cc86420ab148eda13f7d3bc0 : FAILURE in 46m 23s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/5ba667578a794ba6a525261134c08f86 : FAILURE in 39m 12s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/47604c46b6b24ece8af35483121556f3 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f0b8b6b4756b61e756beb4b01eacd61d3e9991bf","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-07-23 07:24:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9da94ae85ac352fbb650f270d234882fb023abfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-07-26 05:35:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_12932\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"}],"_revision_number":2},{"id":"a9d8a6e38b557a70bbac0e215450515083f539c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 05:42:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84a450a2310e447fb576830c5a5bd466\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04f5e026fbb642519113ea81520a1816 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9e06fb0de13240eebc70cc344feefe99 : SUCCESS in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"592d7a751fcbabe4bf8c1690f7b0eb9da01edaa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 07:00:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea1f6bb9073f42e4a43f9a0ebb9b34ff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2364e95e564142fb8033a4d52ab5601b : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9487e617c1804c3298bb0656221b679c : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67d0b06c3aae43d0a3b2db3bbad467e8 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af95c53eff454457a783be4018a5b40e : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ebc671b346f54e2a8bc72f9a84c670f0 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6af59ace9d34530adcf46777622956d : SUCCESS in 4m 05s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/e03d4c84021344b78d63b741af55c129 : SUCCESS in 41m 01s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/00fd428f5c904999aab732822c6df6f4 : SUCCESS in 1h 23m 31s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/31a2cf26685c4efa97ad6736f26d5fae : SUCCESS in 1h 12m 18s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/509c3bdf08714301905de9c6fa9e98fa : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"551529cf712199feec0ffdc6164db235c60c958e","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-29 14:08:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e4c8662689c9bcfc13be0af9f2c7b74a22a7c8a2","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-07-30 15:11:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"35bfe631d043c101da7d9dc80537eb4ba55de922","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-09 05:49:36.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84d23866e2ee85044660b8d2eef8f19b05425481","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-08-10 01:48:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"244b9d60828e84f3985595d5108d6d5ab12756e6","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-08-10 01:49:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a2fc7b9d7efb73faf5b46790ac0555eb2ce848f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-10 02:03:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/150410ec38aa485fad483d6880c61fc3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b0a2b6579bd4169aeef6920b7a88448 : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a0b6382d54f34476b8d50b7a5baf3819 : SUCCESS in 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a6503eccecd8b2e34477b7c6da2a1c00c61bf13c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-10 02:16:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ffeb4a9607b40f5b91646acba5c2ad1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e34a051921cc4a918b13596e0af481a3 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db5fbbe53bd54f6890a1e01b9f83288d : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0344545e9874f3b9bbae49338a67bdd : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7000b936be534c67a2e373684ec25388 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5daee884f094177b5ce4843ad0c865e : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5348e493402f4751b2494a1a021960e8 : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1145834d7ca04fe18e7680c67c063b5d : SUCCESS in 5m 39s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5e9d63f97d054b8eb818aa1ebfdd51de : FAILURE in 27m 06s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/982f1194521a42c986d59ad2ecaf325f : FAILURE in 27m 26s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/4e3e88e3b6d64b5cac26755d3a7ff3d9 : FAILURE in 27m 15s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d946c709339e4723bd3d86038597afea : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":3},{"id":"c3db56fe42c5fd72204b769ac555043e07a987d4","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-12 03:16:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"558375df61c3d2ca2597580c30133713b52b812a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-12 06:28:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f0c2d15c62322a2b88c933a1a8a041ff7bb403a8","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-12 06:35:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"33e73635ab41ad143f770cf8cfdc3ca603d7c42a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-12 06:39:38.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ae86c609898925d56ed86569ce31e18a7e4d6bcd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 06:54:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27a69b55f81641808b72e8a7b777c8c2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60c43bd27afa4663b67cf4b940900a6b : SUCCESS in 13m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/07e6db696291428e8acdcd3d6c298a7b : SUCCESS in 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7f8be6670a7fe1bb0f167558b430c708bbb33c8c","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-12 07:02:23.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"acdc4382837b763fc20e5aea2fbd1ca633762e85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 08:10:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74d31a0b051a4c96b950ac60ca2b21a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebaaa01b789444f686fc2c3d8f398adc : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f32428275a9d4d89bcc8a3a4f674264c : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e20e1713fa4f491182ad5197b79578e4 : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c7dbb5dfcff4533987c995c89fb5474 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7682eb0cbcea45c4bb35408d939dee51 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a61a42eca7d485ca974935cbd2d1efd : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/333e137e9fcb4620887004c9c806b132 : SUCCESS in 6m 39s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/3ba616b6501f439f858a0d5da5894556 : SUCCESS in 43m 16s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/2bb7238c97d8436f98635a8a7fd86580 : SUCCESS in 1h 29m 08s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b824d1f666744bba8b63c8965c4a20ed : SUCCESS in 1h 10m 26s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c454258feb51417196c893d005ad98f2 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":4},{"id":"0a1b9b4d9897b9f629207b904739727eee06bbab","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-12 13:03:00.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1d91d83b6707675669afd0379c801cf9a3b9303d","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-13 01:33:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7926a6f857e7c6b3453eef312fb5c888d9628484","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-13 01:46:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e36b42ae1fa76b16ee95dfa4e7ebb021acb97437","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-13 01:52:46.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"93e28b3c266a268fe7e5cc1df904909a14cff0dc","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-08-13 14:11:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4c2a92f5bb8a67b9d4336a8f4fdc0cd99c255018","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-08-13 14:11:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9629274e9b4de20096f3afe0a15e864b888f6129","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-08-13 14:20:45.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"494506f9d7836f6e1c578c9f4404df41cc0e3ed8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 14:27:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1e6f88ccd394330a1b656a6d5e1dbbd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/86fb1d7a548f457dad707c3471e05ac2 : SUCCESS in 13m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/26c5a248bb614338b4714c55876931e7 : SUCCESS in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ee40c0201d15a76be5f00ecfc4021922ef14b035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 15:52:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e18c947bd484fc1936369e9432fde8c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4fd027dbfd640ad83ebd1059d7c7d8a : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c4d4f2986794e63ac833ec41055d89e : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59c6ec8b3475419e9885ab972538ff18 : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12a49a59fdfa4c06bb71ba921b7aa7d6 : SUCCESS in 6m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d7ea2ecf1e3d43a289049f187d6469a4 : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c09acecca204ac1b785403cba7bc3a3 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9f3f4e1bd1f4da794299725356b91f4 : SUCCESS in 5m 53s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/3f56c89c885d4b6b91d1f8dabb3daf1e : SUCCESS in 1h 01m 52s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a790baa195f143f29e079e3189597076 : FAILURE in 1h 30m 29s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b409d166a5bc46a9a5194fae2c84aa23 : SUCCESS in 1h 31m 42s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/83f1a75c042942588f4a97ada9d97ed3 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":4},{"id":"bd32a96c39c984120beeaf892a9e653a56ebc965","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-14 06:14:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"547e5f50aa342d4e80b59ceb24ae51c99d592e60","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-17 06:09:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d5ae042836838f8fcade53ace62e59e5aad803f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-17 07:45:55.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"cd1e0fc7d61c235a0894e8a0e5f34e99a27fbc45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-17 07:46:47.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"e4773876a0eb91542ec72ae6242bccc5e3064f7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-17 08:02:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/871802332f31456ab2f4a3e3163d1bb0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b471f7652db2465f87891042b49b7e7a : SUCCESS in 13m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/99bdd4042ed7462bbf441c0f9b8a2936 : SUCCESS in 12m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"093195328fbe98c003d1bbd79848c0deda0db189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-17 09:19:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d30b466275ec4acc8dde1eee2d08374a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bb3406a0d7441e0b4f13a3cf784704a : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09d234f7a5d14ae6bb83d91faabb6e13 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79bfb28e287642cd91882e25a1e57b80 : SUCCESS in 9m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/beb1f4429f5a410eaf7570406fcc9c90 : SUCCESS in 6m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73ce651913bb42b997e93c7d23e62445 : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/275c22fd1be741ba96c57d0ac7fefd63 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69be86720ba34ecd93f47ebd92b5c547 : SUCCESS in 7m 25s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5bc68bafbf82466a95702ccbe5d94026 : SUCCESS in 46m 56s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/46dd68e10c394f75802f92b338510185 : SUCCESS in 1h 29m 48s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/81cf1aa6323745b8ae41c274fa607ac3 : SUCCESS in 1h 16m 38s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/55f1230eea77428e997fbaa2ad758181 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":6},{"id":"adba1d314396a328e8f9f6344cc15ca121a69b41","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-19 03:44:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6fd5442ea3b730552f2773e158b0572e8bbc6af9","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-08-20 03:03:03.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ae2d74dbe5d22a1d49a39252d8d9078afaae84b5","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-20 03:10:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6d5f15d7ced25f3974b8b4bfc4628f40b144c6c9","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-20 03:17:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ea8e8e36655a2cc83075536bccdebbd3d6b49c6f","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-08-20 14:14:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e04086a0369ee0f5ef033a8cd7926249cc943608","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-21 02:20:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"31b40a96f44d4523e068440818b2299134d9eda4","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-21 04:47:45.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0a956c6245bc429e4068247ae5c651da9ffdc7eb","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-21 06:46:49.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c57bbdea7f4588b41162d2256ed89d5792261e11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 06:47:18.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"408cf4367a511dc71378f71243009dbe07d06106","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 07:31:09.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b9dcafa5e2e444cbf3a0c8bfefe8597\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad0c3877b584f9b8673f3dde7ca0635 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/833b15cb12f444798fc72401f34c5882 : SUCCESS in 6m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/147d3abacb5d45238f72397f69e9d591 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74fccc6da6334fd9a6215c78aef0066e : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/891413665db6445194d304dd2944f503 : SUCCESS in 6m 50s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/14b10dccb60f4e2bab47150a39644730 : SUCCESS in 41m 58s","accounts_in_message":[],"_revision_number":6},{"id":"d029ab4f04f4f35a9ecf62f1aec98f4c81f2b23c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 07:31:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"a07e10348c54c4ba6e198c5624178292c85923a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 07:32:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70ec51b2baff4fc3b1f8b02b9c328b17\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a37aa1f0c88045f5b2a95a66426c049a : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f64a46de8724da5844c0ef75631e69d : SUCCESS in 44s","accounts_in_message":[],"_revision_number":6},{"id":"6ce71958d44201432a95b26820e412c6048f2bd3","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-08-21 18:55:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"658d39b7e4d3a36d51270eb3b2cc74825c215bee","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-29 09:29:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b1c74bf6bff6be1402bf42648113e62f09ec2475","revisions":{"3b32e0611eccc3f6cfe7459486c7ba9f89c6433b":{"kind":"REWORK","_number":1,"created":"2024-07-16 21:29:47.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/85/924285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/85/924285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/85/924285/1"}}},"commit":{"parents":[{"commit":"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155","subject":"Use policy to determine admin context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-09 19:34:20.000000000","tz":-300},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-16 21:29:34.000000000","tz":-300},"subject":"guestagent: implement \u0027rebuild\u0027 for postgres manager","message":"guestagent: implement \u0027rebuild\u0027 for postgres manager\n\nThis is almost but not quite a copy of the same function\nfrom  mysql_common.\n\nStory: 2011172\nChange-Id: I232f53bdbba0d0a17dfda395b89300419879575e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b32e0611eccc3f6cfe7459486c7ba9f89c6433b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b32e0611eccc3f6cfe7459486c7ba9f89c6433b"}]},"branch":"refs/heads/master"},"efc4b7e58231f0e0b2eaa69aa79bdf9a4fbf1613":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-26 05:35:04.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/85/924285/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/85/924285/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/85/924285/2"}}},"commit":{"parents":[{"commit":"3f521cb7f577635ee74b2a48c70f874dab2f9e30","subject":"[CI] fix trove gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f521cb7f577635ee74b2a48c70f874dab2f9e30"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-09 19:34:20.000000000","tz":-300},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-26 05:35:04.000000000","tz":0},"subject":"guestagent: implement \u0027rebuild\u0027 for postgres manager","message":"guestagent: implement \u0027rebuild\u0027 for postgres manager\n\nThis is almost but not quite a copy of the same function\nfrom  mysql_common.\n\nStory: 2011172\nChange-Id: I232f53bdbba0d0a17dfda395b89300419879575e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/efc4b7e58231f0e0b2eaa69aa79bdf9a4fbf1613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/efc4b7e58231f0e0b2eaa69aa79bdf9a4fbf1613"}]},"branch":"refs/heads/master","description":"Rebase"},"ca39d9540f2a7a82291fc0088132e8765321c49f":{"kind":"REWORK","_number":3,"created":"2024-08-10 01:48:19.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/85/924285/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/85/924285/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/85/924285/3"}}},"commit":{"parents":[{"commit":"3f521cb7f577635ee74b2a48c70f874dab2f9e30","subject":"[CI] fix trove gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f521cb7f577635ee74b2a48c70f874dab2f9e30"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-09 19:34:20.000000000","tz":-300},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-08-10 01:48:10.000000000","tz":-300},"subject":"guestagent: implement \u0027rebuild\u0027 for postgres manager","message":"guestagent: implement \u0027rebuild\u0027 for postgres manager\n\nThis is almost but not quite a copy of the same function\nfrom  mysql_common.\n\nStory: 2011172\nChange-Id: I232f53bdbba0d0a17dfda395b89300419879575e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ca39d9540f2a7a82291fc0088132e8765321c49f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ca39d9540f2a7a82291fc0088132e8765321c49f"}]},"branch":"refs/heads/master"},"a97d3b46b9432d862c9bb325561603742addf505":{"kind":"REWORK","_number":4,"created":"2024-08-12 06:39:38.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/85/924285/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/85/924285/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/85/924285/4"}}},"commit":{"parents":[{"commit":"3f521cb7f577635ee74b2a48c70f874dab2f9e30","subject":"[CI] fix trove gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f521cb7f577635ee74b2a48c70f874dab2f9e30"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-09 19:34:20.000000000","tz":-300},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-08-12 06:39:38.000000000","tz":0},"subject":"guestagent: implement \u0027rebuild\u0027 for postgres manager","message":"guestagent: implement \u0027rebuild\u0027 for postgres manager\n\nThis is almost but not quite a copy of the same function\nfrom  mysql_common.\n\nStory: 2011172\nChange-Id: I232f53bdbba0d0a17dfda395b89300419879575e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a97d3b46b9432d862c9bb325561603742addf505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a97d3b46b9432d862c9bb325561603742addf505"}]},"branch":"refs/heads/master"},"96f363b508863af2b33ac2fd8d79949cb8df811e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-08-17 07:45:55.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/85/924285/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/85/924285/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/85/924285/5"}}},"commit":{"parents":[{"commit":"7358330c14aee1ba3a449cb17607338e3b61e802","subject":"Merge \"Add UID/GID configurations for each datastore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7358330c14aee1ba3a449cb17607338e3b61e802"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-09 19:34:20.000000000","tz":-300},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-08-17 07:45:55.000000000","tz":0},"subject":"guestagent: implement \u0027rebuild\u0027 for postgres manager","message":"guestagent: implement \u0027rebuild\u0027 for postgres manager\n\nThis is almost but not quite a copy of the same function\nfrom  mysql_common.\n\nStory: 2011172\nChange-Id: I232f53bdbba0d0a17dfda395b89300419879575e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/96f363b508863af2b33ac2fd8d79949cb8df811e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/96f363b508863af2b33ac2fd8d79949cb8df811e"}]},"branch":"refs/heads/master","description":"Rebase"},"b1c74bf6bff6be1402bf42648113e62f09ec2475":{"kind":"REWORK","_number":6,"created":"2024-08-17 07:46:47.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/85/924285/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/85/924285/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/85/924285/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/85/924285/6"}}},"commit":{"parents":[{"commit":"7358330c14aee1ba3a449cb17607338e3b61e802","subject":"Merge \"Add UID/GID configurations for each datastore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7358330c14aee1ba3a449cb17607338e3b61e802"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-09 19:34:20.000000000","tz":-300},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2024-08-17 07:46:46.000000000","tz":0},"subject":"guestagent: implement \u0027rebuild\u0027 for postgres manager","message":"guestagent: implement \u0027rebuild\u0027 for postgres manager\n\nThis is almost but not quite a copy of the same function\nfrom  mysql_common.\n\nStory: 2011172\nChange-Id: I232f53bdbba0d0a17dfda395b89300419879575e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b1c74bf6bff6be1402bf42648113e62f09ec2475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b1c74bf6bff6be1402bf42648113e62f09ec2475"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
