)]}'
{"id":"openstack%2Ftrove~924686","triplet_id":"openstack%2Ftrove~master~I5d20ad8e8f4e262b9db34b1572eca71e9bb93357","project":"openstack/trove","branch":"master","topic":"noop","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2024-07-23 12:58:09.000000000","reason":"Change was abandoned"},"12932":{"account":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"last_update":"2024-07-23 12:55:34.000000000","reason":"\u003cGERRIT_ACCOUNT_12932\u003e replied on the change","reason_account":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-07-23 12:58:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5d20ad8e8f4e262b9db34b1572eca71e9bb93357","subject":"postgresql guest agent: no-op change to validate tempest","status":"ABANDONED","created":"2024-07-22 23:44:26.000000000","updated":"2024-07-23 12:58:09.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f00be9107708a37dc03a66be0b9deab844ab140b","_number":924686,"virtual_id_number":924686,"owner":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-07-23 03:11:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"all":[{"value":-1,"date":"2024-07-23 02:59:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-22 23:49:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-23 00:22:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-23 02:31:44.000000000","updated_by":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2024-07-23 02:59:38.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"}],"messages":[{"id":"bde78f827950e214b4021d25335a092ab540a01e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-22 23:44:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0da8e2588db4d3304f5225747873aca189f4b89f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 23:49:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c37d558904874876ab11a2e259408ddb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5687b97b8fe34ebd9e52b404f1181d8d : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/079847a53ce84919a79f4df4582f4c70 : SUCCESS in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dba144809e7db566c213622830a82101cc7bf4f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 00:22:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f83b3548f3e14f7aa9d3a3d24b944484\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfdbc5b9bc16494fb147b0e0564a1735 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6e373d0f52a4f60b5f6fcd2ac843521 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/886ec3cfcb294717875219e07607784d : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e2553d0aa054f3190000bb9638fdeee : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/728518a89a36420c83d25478ba8067f1 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ac02a5d9ffe40e6a24b9ec2dfb65b72 : SUCCESS in 4m 10s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/44737bc440334c46930a03f0fac5855e : FAILURE in 25m 45s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/ee21e498307642c39b4177b377b38a11 : FAILURE in 37m 19s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/0dd79eb775154a69a8cb99bfa143e7dd : FAILURE in 37m 30s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bbb1d08a4c2b4bc1ae4e0fdea053773c : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0f2960eb27ada8aa8206adafb5ad894d385db02d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-23 02:30:16.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"49038f75696422ca02a29e0592d19ed940d80a35","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-23 02:31:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4faf71dd5699e5da2d24af25771902e06cf5a0e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 02:37:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daee8c828cf1404aa160976f53908389\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/540896a968ff45dfa7e3587d2a6388ea : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9e00602c5a30474ca8cb86d154b55235 : SUCCESS in 5m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd4ed1c01235422843661967921e3cccbd46c887","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-07-23 02:59:38.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"582287a4d9e4b8b765896a2a2abd3a6adf614c4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 03:11:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ce37312b3cb4865a817be5434fd5f77\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/196b02c44edc44bbb6b1e95e56e4935e : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/898a703ac82f4355a06188948e5c2505 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b8209be877d44e6a85e8cb534899c4d : SUCCESS in 8m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ebbbafbe821d44838f25e0fa0567faea : SUCCESS in 6m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8d38130fb82451aa16ddab9a44cc3a1 : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36ffc91d5a3649129ee065a827a480c3 : SUCCESS in 5m 59s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/b19bf84ed92349f0a4b16834cf1a90be : FAILURE in 25m 06s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/6a3905472d7a48569d7b9f24ec4d79c8 : FAILURE in 39m 55s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/a8bf84d8896a49329e9b811f15d12565 : FAILURE in 39m 54s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/628a961ea7ff4c1b8d77953b1ffa53ac : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"bc076e058b2312c2f7d633baf8144737391fadbc","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-23 12:55:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f00be9107708a37dc03a66be0b9deab844ab140b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2024-07-23 12:58:09.000000000","message":"Abandoned\n\nThis has served its purpose :)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2d2c7b12dfcb1adf170c42f106e9ee3caa928f08","revisions":{"d8c19dae33a518a470c00038a3d328a7e3d242a8":{"kind":"REWORK","_number":1,"created":"2024-07-22 23:44:26.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/86/924686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/86/924686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/924686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/924686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/924686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/86/924686/1"}}},"commit":{"parents":[{"commit":"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155","subject":"Use policy to determine admin context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-22 23:44:09.000000000","tz":-300},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-22 23:44:09.000000000","tz":-300},"subject":"postgresql guest agent: no-op change to validate tempest","message":"postgresql guest agent: no-op change to validate tempest\n\nChange-Id: I5d20ad8e8f4e262b9db34b1572eca71e9bb93357\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d8c19dae33a518a470c00038a3d328a7e3d242a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d8c19dae33a518a470c00038a3d328a7e3d242a8"}]},"branch":"refs/heads/master"},"2d2c7b12dfcb1adf170c42f106e9ee3caa928f08":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-07-23 02:30:16.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/86/924686/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/86/924686/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/924686/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/924686/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/924686/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/86/924686/2"}}},"commit":{"parents":[{"commit":"be9bd3f3a034d9e31b32fa194db8e42ed0ee4155","subject":"Use policy to determine admin context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9bd3f3a034d9e31b32fa194db8e42ed0ee4155"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-22 23:44:09.000000000","tz":-300},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2024-07-23 02:30:16.000000000","tz":0},"subject":"postgresql guest agent: no-op change to validate tempest","message":"postgresql guest agent: no-op change to validate tempest\n\nStory: 2011172\nChange-Id: I5d20ad8e8f4e262b9db34b1572eca71e9bb93357\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2d2c7b12dfcb1adf170c42f106e9ee3caa928f08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2d2c7b12dfcb1adf170c42f106e9ee3caa928f08"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
