)]}'
{"id":"openstack%2Ftrove~926466","triplet_id":"openstack%2Ftrove~master~Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2024-09-26 08:14:43.000000000","reason":"\u003cGERRIT_ACCOUNT_31737\u003e replied on the change","reason_account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2026-08-20 21:34:13.000000000","reason":"Change was abandoned"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-12-25 08:26:47.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e","subject":"Supports kernel options when starting containers","status":"ABANDONED","created":"2024-08-16 12:03:25.000000000","updated":"2026-08-20 21:34:13.000000000","total_comment_count":26,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"df20165a8c2e804ac37d86b90a343b3fb8b7aeea","_number":926466,"virtual_id_number":926466,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-06 04:51:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-12-25 08:26:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-16 12:20:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-16 14:03:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-17 01:20:57.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"CC"},{"updated":"2024-08-28 07:52:21.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2024-09-20 03:14:16.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"4321052072de8d30ee590de086656161f85ea552","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-08-16 12:03:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa14b3c15dd4528079c14a92b32d625f699132a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 12:20:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39f97942a6d540eaa34ca96a060fadd4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3566e08d70643e690ad2b47e42a9fab : SUCCESS in 15m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c957750d9a2b4b8db8ed54ff9e27da44 : SUCCESS in 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4dadfdd80bc0b3740ad541d44421c086ee951316","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-08-16 12:28:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7dc39c5874634f1e07eb8ae63b77287b6ccb149a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-08-16 12:31:59.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"923cb94c1b796fc67e9a943f54e8b39d658f474f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 12:56:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cf28bf055064255af19916da68cfc82\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/acaf8dfdcc0a49c888d53673c8729a9a : SUCCESS in 22m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/26d23a6bf4924e1abab6ecc2bf25c80e : SUCCESS in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c4ff13189072282efef5b88954697fbd9aa4c8bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-08-16 13:03:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d1eb29638c04a4975ecbef7976a3a9b0bbf5ff2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 13:28:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1660c77a9e14119849d2e646218616e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41c2c9b9be5e416a8cd1c794e364a977 : SUCCESS in 23m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b096921da2a14d81bb40d5f96269c02a : SUCCESS in 11m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"049493e3084c76f8fb8087aa6fcabdcd344913c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 14:03:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd1e97bc9a8b47c8be4bf4eced2714d1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/999a6f28034b4f6eba78b33e2b24cfaf : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4c082cfca8145fcb0fc7b6b3f2ab1e4 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41a2bbe0f05c4b908110bda33dcf4d99 : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f9924a0a2c64f3cb6896963e2278e6a : SUCCESS in 7m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f0af21df8df84f4cbea060276386420e : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa51d2a8f390474d9085a04cb2a256ae : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3cbdd654c394b8688540113e8bbd4f5 : SUCCESS in 6m 08s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/6a6bbf1a49fc4087aa4f0f8b88e235b6 : FAILURE in 36m 51s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/f79ae8f2dc6b402a84cfd63df0201c0d : FAILURE in 48m 27s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/48243b8ab98e44f89aa426455d16a827 : FAILURE in 55m 45s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e67cb2a0105242a6a72083003ff95aca : FAILURE in 41m 31s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0b2afd0c34af481ab85b05da21869d6d : FAILURE in 53m 25s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1a576828e1b64eebb92928a184c506ff : FAILURE in 54m 32s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/76717ec253a54b52b58cc352e3289d27 : FAILURE in 46m 12s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/544ce3ae806d4d54aef6dc28aead155b : FAILURE in 55m 45s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/9ebcd34763294e71a1f602751c1f5a9e : FAILURE in 58m 21s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8161d238cc1e4a13981ddef1a345c8c7 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":4},{"id":"87d39d4de6404838f451dfe9955dff8174305c68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-08-16 15:49:33.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"dff220aaf72e791cdf011cbf4cd0eb801c5facb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 16:16:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41b0920fb9d9473fae77b7473fa1e7f3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/83b7da854b104cbd923e05ad4e968b77 : SUCCESS in 22m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3e3f891883d5445bb526e520527d328c : SUCCESS in 13m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"64b431d0427efc82df1be510a8675c6752584400","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 17:10:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a77fee5387f447a988ebc56d753c899\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db6828803bfd443aac2581c10b20315d : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6746736b240a4feeb25c8ce6da6fc9f7 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f71ffc996d9e4980b8c5624937b357a3 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c030c0744fcf4ee59e410fba584b2e0c : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8da5b24374d4d45bb5cc4cfcffc7748 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24e9a58937034cc2a23387e3e11cd65b : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cad88ccf47c84b8289f186736a8a11df : SUCCESS in 5m 52s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/95770e1f0d5d44fbb4f92dcd760bb8d2 : FAILURE in 39m 05s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/f9ba36a4d17b413f8efcac1c942a545a : FAILURE in 1h 00m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/0c8762e2fc8f445c9d1ae830aafb0efe : FAILURE in 49m 34s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0c4fcc3440d54178ad1a22097ffb5d8d : FAILURE in 37m 33s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8461a876f83f49f0adea6afa19debed1 : FAILURE in 48m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5709ee0ad1d844f19bafbfa06792dbac : FAILURE in 48m 57s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/347448291996400587895bf6131a44c8 : FAILURE in 50m 34s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/6d36e3aeef28478085942b862ebbdea4 : FAILURE in 51m 15s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/dbadd1da2c5d4a0ba8cdfc6a1f2b8b3f : FAILURE in 57m 56s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ec3951d333c24603954a9fb7bdfbb099 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":5},{"id":"053c6b9e36cbd8dfcb4654731b26f8dd64c164ef","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-08-16 22:37:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"29628efede714fe63f53940bc46004b849f7fb8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 23:13:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04fb1801fcb54dfea57882d7b575b5a9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0f2f56e99b8439180b8424e0e0feb30 : SUCCESS in 32m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e2526b9510b74479811bd1ba2c596bfd : SUCCESS in 15m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"35918ae6297c06db4754579794f6cf68c6698659","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 23:36:49.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b8fb16b85d04b34a5ec528c358b912e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9015287694d54736a9e7b05d315e5dc4 : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02963590f3274ec2888429910c204ec6 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b2932f1971143c9b57207df61dfb8bb : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1df8524a73874a17847b713f1ee4a98f : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8689c8fdffb47dfb3a8c0a94750cf3c : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dceb2ac08cb4df0a7b0fc3f3863890e : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/997628f15a9f46e9bd95408d471c5c5a : SUCCESS in 5m 12s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/f7709fb06780447080becbb79722d0ad : FAILURE in 40m 24s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/309a24de50644728805343a9851c5af2 : FAILURE in 48m 30s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/f4ba0df3d5b84d3399a2628f925cc79b : FAILURE in 57m 05s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3838779ef8ec4ff0be23908564b60634 : FAILURE in 43m 09s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e5a4b6f1da0f481c9c76fb62d3a9dd9e : FAILURE in 44m 00s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7f726a1bff824e589aa88d58521a6eac : FAILURE in 47m 41s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/a309bf32c90f495ba3cb6eb7589ec924 : FAILURE in 39m 52s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/58901b6cb06142deb6fe8dea9dfada24 : FAILURE in 46m 59s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/729ae697d2694da5ad88e0cd2af6a8d7 : FAILURE in 45m 04s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4163e35c3a774ff7a401825e550041d1 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":5},{"id":"e1b8dd33530ff939a0ef7de19b7d633cda521a52","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-17 01:20:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b9cd5f4a469bf87b5e6973c5def0091579763ead","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-17 09:06:14.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_31737\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"_revision_number":6},{"id":"3796ca330a150348704bbb8467b9c00ea4f395fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-17 09:17:50.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac624b945c0e430898187c3e57dc2d4a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41a173e8e4904cfab52da2fb27062361 : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/922b88ef36a0478fa9596e92ee583dbc : SUCCESS in 9m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1d3ea05e7a85dd715aeaffbf9e8ca26b2d08424b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-17 10:02:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99a0c42ee5af4b2dbb2bcd4cd94241af\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/240efef5e6aa40c9bf7f48eefe080e83 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce3a5de8cbf64210b20f01e7da021b49 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13765858c20c46e38eff890fa8596181 : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c266f626426463790ba81e3f8c0fd9b : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95efa46e1e2e495085ce2d66c8d99f88 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86e6a1911e0043ee988d5bb7a0dac6d9 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/478a407e150f4d82b241f22f654be533 : SUCCESS in 4m 53s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/1ff88d8351ff41a59a71a5638e86b013 : FAILURE in 38m 15s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/4d3da58a474b4a9a8d93cd47ad944715 : FAILURE in 47m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/b81ee038ec934d6f850ff11046869748 : FAILURE in 49m 03s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2b21533ef8ff4bfc94bf19176627e056 : FAILURE in 35m 50s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2b43c432269b46a69a88ad3504176390 : FAILURE in 51m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c5bb6aef56b3427498f0acb195eea7b7 : FAILURE in 42m 47s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/c0132d5561174ffba8d214d050c0a980 : FAILURE in 43m 09s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/bcf5b39f69214c4098a9b11a6b61b905 : FAILURE in 51m 34s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/1764d79602e44d46a813c1e2991db900 : FAILURE in 54m 47s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e25415f29e48448b91c952c89248f3a3 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":6},{"id":"5677de34527ee9cc0f200979526930d4cf01a789","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-08-18 13:15:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"423c01f98f3392307d363f549128f268d9f7998d","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-08-18 13:17:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4ad7e7cd6e7abecfd64d77535404f843d4d8191d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-18 13:32:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e4eb5e8948f44da84701ed842c7dc8b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1207f6202ee849cb841aea4586e6e985 : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/21b57b6be2644707b2769f0d8ebff401 : SUCCESS in 10m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"89f997cbe3dd2a2eb73ef45cbb53ea61e25fea15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-18 14:25:16.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/191443808b984a72ac587d0395a701a4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2e5f3d5bac34b15895c590aadb46a10 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fc3b87867894f67b937e2bbe4b46b49 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8628fbd5caa466ca9eeb31b94eaef29 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d678f740e6e4b8797f2f7305e0c3a76 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d86bcb169d94068a23710737067153f : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e1959728c904882a8df79cb1f6d018d : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe95b95cd44543bcb14e7fd29e34f91f : SUCCESS in 5m 48s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/455026e2222640fd9be45f153f1a760c : FAILURE in 38m 24s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/cd50cd0d1f5c419ea18cc460f3ed3560 : FAILURE in 1h 06m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/4079c2152e7f4a2fb456797210a6610c : FAILURE in 1h 06m 15s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7f1dbbc4f31d4297af3e89996438d069 : FAILURE in 52m 05s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8d4a78e0232444e9bb6f3337aec5c3c6 : FAILURE in 1h 05m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/88dddc1013db4203801e894aff0ba3d2 : FAILURE in 49m 37s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/2a95a35b3bbf406c9cd3d8b19b18daff : FAILURE in 51m 11s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/e02c6b7618be4323b6bc64b1664fd84d : FAILURE in 1h 05m 23s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/751c41cf615743018c23d55674100c5b : FAILURE in 49m 42s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e8195cd11f98435bac1c8772f6afbddb : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":6},{"id":"d876cbd34d975f10ad10a5b4c8e722d3831a6089","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-19 01:24:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"051be095ea14c272b94f024e4a9c81af53f18594","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-08-19 12:32:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6a1eb6c88cd83f57e4d12f3a2bbc348be48d955f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-08-22 04:38:08.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"f8fe5c88e93b84d3d7391602e241f255c20d8cc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 05:16:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd2c8cd6f7534489b9b4d129603d0171\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8effbda9899f423bbeab0b8ca6c91093 : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/61376e84a4bb4fb8af1171c762e66664 : SUCCESS in 11m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"09f1bb9849c206f83879c3c150441eae1fa5c4af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 05:33:59.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4128eb48082341e5b5a1b565564b19ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf6e87d8846441488f3915081ca55a1c : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3d16215b326412eb70dfa8b71e0fd26 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73bfec92b9664168b754b5f1e4083b0f : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16f657cf5eb640c282a19d317701524f : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e60219392d0435588d049e2d0f61410 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0c845d7be3498d9e5793f927f175ee : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a38113a4ea724126848d730bb28b9149 : RETRY_LIMIT in 3m 22s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/af04a014e1b64285bcb573e95f90234c : FAILURE in 45m 10s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/7122a9dbba114357a2b1be2b58e68d75 : FAILURE in 54m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/4064789bda404aa2bc2b8a6e950a7fd1 : FAILURE in 51m 13s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2a0d2b23a3104a659ff2dcba905f1b3f : FAILURE in 36m 09s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/75b3c94f9bd64c07b78f12f3a3b2dbbc : FAILURE in 47m 41s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7a6b2cbd8690422ab8f76ecedba1322d : FAILURE in 45m 06s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6a026919180144e6bca3069d8ed60e3a : FAILURE in 38m 32s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/f22a6f4d81f149218eaf52dc5966dd55 : FAILURE in 54m 00s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/ba4c0bd6922e4163b4ae229a285590aa : FAILURE in 52m 15s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/35655779dea14f01bcb7e67de9945bd9 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":7},{"id":"b79d7738ab0b41661c0a89e36774399f4977f5c7","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-08-22 06:42:52.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"47ced428fe708279d956b0ac0ffe0ecfc880108d","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-08-22 07:35:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dc89b409d3f17bcb10db873af1d228ceba815686","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-08-23 02:54:06.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"fbbd0704507a8d67789f220ca7aa878250aa9872","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-28 07:52:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5df05d99b021d2edf99bd353be4150103e20af24","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-09-02 23:27:04.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"64379ee9dd620d41992d2e093ad8167bf4af8ede","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-09-19 15:30:11.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"368c78eaeda76db7694d5f7cd92c72337291328c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 16:03:40.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23bdfb18c9b84c4db74c50cc911b2d4f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb6b7fd3756b48379ad3b5d26d77a93f : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/749ac9a1c2fa4e81b450c40ba5cb52d7 : SUCCESS in 13m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2f3857cc4ce5d2d39c18fdfcb9c63a01d70308ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 16:40:27.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/397f5b7b217f4c4f95d62a8132a55c01\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bf4626d077f4a73af145aef3647a38f : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3ec8d4d9dba40b1b14ba6165e13e551 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49f01f306a0c4b16b527b2ca7403716c : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a76279e8a03645b2b26bfe9c1d483cf3 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48d9f68ca4594021b661612a6d33856a : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d11eecc91f6465087e72085b4d4612a : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4170c22e14740f18732004f4eb125fc : SUCCESS in 4m 47s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/8e3d749fd9264f80acfb779bda7e64c5 : FAILURE in 38m 15s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/64f9918ba80a4a82a3d77b5b0b40a059 : FAILURE in 1h 07m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/3c639461b2954f8f9866e7f4f74387cb : FAILURE in 51m 16s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fcedf3af3b58463db923816455c3a84f : FAILURE in 37m 07s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b567bbda774341d281b524a27c317162 : FAILURE in 47m 27s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/39ff0288f4894e7b9a0ef357887b6d73 : FAILURE in 50m 08s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/ddde26de98a4478f861484d558863ed4 : FAILURE in 38m 26s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/1c88e8fcdea74548b1d5059f9d00c21b : FAILURE in 1h 00m 51s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/700e9ede3a61410fb7c9286096a0d0a5 : FAILURE in 1h 03m 22s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e374c5184fac4d45a9bee54045415ef5 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":8},{"id":"d9f7a190c7e9403850ff354a66f01701b5e1236e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-09-20 01:03:22.000000000","message":"Patch Set 9: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"81fd94b75b96efa56d4edf6fdbb15bedc92bbfa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-09-20 01:04:19.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"59124fd77a210567a22f4c0218463635159391c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 01:23:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2be9f409d3549c2aad836d31b33e41c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a434d86466ed4bb7b953e69a0baaadf7 : SUCCESS in 16m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/db45be3a9a9648d9b25385b8aade196f : SUCCESS in 10m 09s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4fa148d33d3e5042daec2df8b7999b72646d619a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 02:08:31.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/687fb381954f40b3bc0b6e2916d05543\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5649df5a21f439cb9c320cbfb50c426 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2df44bbfefc0435ba90749bd6bccde23 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd8b143d642b4c958e8bee4314666f59 : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6eebd272474042ebac761691a623d800 : SUCCESS in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b26b6ba0148f4afba7b1f50100e0c4b3 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bfa81cebb294ca48cc0469ff04ca663 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05e73ac0d4da46d2a6eeaeb735f07308 : SUCCESS in 5m 26s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/f894852418e045ea9ac71d24547d5ed2 : FAILURE in 42m 53s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/2d6a1beb34244aa1b5e7469a35112fff : FAILURE in 50m 09s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/d4863e39e9304ae1b8157e261b236779 : FAILURE in 51m 55s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0c739e9a734d41b89524eb38a801a400 : FAILURE in 46m 56s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b9782d2443484f20872da32bc85f12fa : FAILURE in 44m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5a4792b437ee4f69a5769485127232f1 : FAILURE in 56m 35s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/c92b2b90ad4d4c30bdb65cbad773de6a : FAILURE in 31m 03s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/79c5cf1d4a284771b79cacb7dd01cc56 : FAILURE in 48m 07s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/53e1d92b03e4442e8a6518eb3cb57cab : FAILURE in 46m 30s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/76e605754f514526ad83c3e1387d49d5 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":10},{"id":"187cfad469659f21f4b7784418e75e0214aab822","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-09-20 03:14:16.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"91bb54e84dda78796d92231132fe0589200a051c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-09-20 03:22:14.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"84688ac3beb3169f8589abeaf29cc0725a2bd9c4","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-09-26 08:14:43.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b346b2c0b058ccb3b19a08d5f41b4dc9f73c2049","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-09-28 22:48:19.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"dadfc81c77297ee067c2820646534edaba2dce9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-28 23:02:16.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9e6f92915544c57bb9b9722d296227f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/713601b16b15466bbb8015f7fecc6163 : SUCCESS in 10m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/52bca50ea03343c1801715a3b6439004 : SUCCESS in 11m 55s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"26c493430a5948a75bf5a3b5526a1b86a1d2f5cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-28 23:42:45.000000000","message":"Patch Set 11: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44c5c70ecfab4d3c9a52d2c2a0718c2d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4010bee0da64448ad42c8b6274c0114 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6c6f6e637024c1b99ce01cca1b68a3a : FAILURE in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7571cf1329784ac4a8fd590feda176df : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6aa9f29b80564684b3c6c7acfe19a292 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b073702257f4f5f8bb3bcb03ee66986 : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33f4a91f1c5e448bad2b007735e2e615 : SUCCESS in 5m 46s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/5975fe8929394615bd8a15e5aa6f3b16 : FAILURE in 38m 43s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/cb3f6fe3c0d14bf8b7de26abd84abcb7 : FAILURE in 46m 01s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/0b9fc2aaf1e94b1ca987828e828ade9d : FAILURE in 53m 28s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f3ab5b28d9b2407b921f01bca9dc7943 : FAILURE in 34m 21s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bf24d6a9b3424292a3e7c88ae517dd35 : FAILURE in 46m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/23c43280af614ce3be8690e94370b115 : FAILURE in 47m 48s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5cf5df26f4d14e99a02b9b0b9d6b1e1c : FAILURE in 36m 21s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/7eb25d2d889647f8a45564b82a00c86d : FAILURE in 47m 49s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/83257be6094c42eb924fc023d3216022 : FAILURE in 51m 01s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8b47daf8884842b5a7637f9f9db95518 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":11},{"id":"c20c6f0e6dcfe3607b9a1d2cf8799e54d9e5de68","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-09-29 00:49:45.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e1c99ab09525b6160ffe80628b6f25b86c03cc65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-09-29 01:22:58.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"3c14a1b83408d4cd85bfb8cd2414d2bda050553d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-29 01:40:55.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/288007cfe9004c85b0be74fbaf9f182b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c67d1a091db342ef900cba85bd1db27f : SUCCESS in 15m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e27674fad8084b2e8031d13d73927c17 : SUCCESS in 15m 48s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8c1cbd8ca814af04bb2214f6fee1929c97cddd36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-29 02:10:25.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1315101d5a04ba49b8b2fd42ea417e3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5ab6eefccdd41b7b44e7c72aefda64e : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c56f9599d4e4fb583f2af959e7c2dad : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97ce0c10a32941d5ac075cef74b18c5a : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/83601b38152c433d8d0824bd8a5bb240 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48dbb61ac5e94045a844007977b64daf : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b43e4d7ac7c49c3927af2da1d575937 : SUCCESS in 5m 18s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/80bc69faeb2b492188dc4bacf7347d92 : FAILURE in 35m 27s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/96b4e0520ba043559cfe8a9c1700b620 : FAILURE in 45m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/445b39cce5bb4688a06f164626f79ca3 : FAILURE in 39m 47s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/103924fedacb404c996c203ed3fee6d5 : FAILURE in 31m 31s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/596723b712a244599a34056f76147e3d : FAILURE in 39m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/22ec900f046449eda4e4254ef49abde1 : FAILURE in 40m 02s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/67835648f3414adcb9cd6931cc2bd2a9 : FAILURE in 33m 00s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/d7fb821ce18140438f37bc9f83143ba8 : FAILURE in 40m 14s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/db2f3d801cd3498c96dd1bf5af8f0fda : FAILURE in 43m 39s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9873eb08ff094c19840e0e35ca0041fa : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":12},{"id":"1e693a6081def06bbd4e7b93ebebd9660ef1c463","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-09-29 06:01:26.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7a8f561a31114db3c15bbfb6dedc8d6bdf82d80c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-29 06:16:55.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3465e45833eb4bf79c8d7cde4973420d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/970f565237f34312af24568932aab9bb : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7c3350117c3f4e95b190a9b2bcc18b19 : SUCCESS in 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c087daac16e9bfdd3cc802532dd2531329a104fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-29 07:02:29.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b29f326c863a45c38fc930de681b5061\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89f919e1820848a98f636a20d50bed3b : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3fe4585064040d4abd68d1e5a22d847 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5c24bc40df54917a0a3500ed39c51d8 : SUCCESS in 6m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7161fc5440574d4688d3cc2636c5b3ad : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c071ce439a3248ff81b9fa2927a3f4b4 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1c6633893a245b087d0c83e31049e10 : SUCCESS in 5m 48s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/fa791d78fba649a4888dc704ec45a64b : FAILURE in 32m 08s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/ba84d3ae5d074d60b7d8828b3fca4c46 : FAILURE in 45m 31s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/085d86e97c584771ace14c91296a87bb : FAILURE in 40m 27s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6958d623d5504215b5a76205bf7fe8d7 : FAILURE in 37m 22s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/274da9c7a1064f72a324151eb4783ede : FAILURE in 44m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/366b865beca440c1bfb313713174c6a5 : FAILURE in 39m 21s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/4fba3e6761f340bc8d523191fc58ff32 : FAILURE in 39m 55s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/53cfec6f5c0947a39e4bc045914f4022 : FAILURE in 46m 05s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/02db6a0a40bf4d378577c00eddfa82db : FAILURE in 51m 10s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f088be5cc39b4ab4b700e2781e0b0990 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":12},{"id":"709e91be648eb7af1f76315774d78ae65aa4b977","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-09-30 01:08:39.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"0e5db2c81f0894b535acd14585372b0d531bc1b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 01:21:53.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3a0570ca2474377a3842c33f8812797\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c4b3e5001e64f729f204b1c9bcf5484 : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0eba4888d58445a1924f01aefe69b680 : SUCCESS in 11m 12s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f303e492389f73f6036b3d75caa8a36996790c0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 02:18:53.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/76ff901fe2184eff8601510bfe5dc191\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc6c178536dc443f9978112322281aed : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43a33be063524241a49cb82bf44c2a20 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b73ef68a972434f89b7b3ae7841bf12 : SUCCESS in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94f32358a1f1445aa455c2390b12252d : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c492e36070f540f9b66310ae2cb83148 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59681491331e4a23964fa368aaa2ca7b : SUCCESS in 5m 27s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/2e7715c19877411980ecc050d8be1e46 : FAILURE in 37m 49s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/71892e491cb24beb9793e1f41e7d9e87 : FAILURE in 1h 03m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/a615a4672885451fb8c287cb1e5a936d : FAILURE in 1h 02m 39s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9bf150145df84237ab4d6df2e2c02fdb : FAILURE in 46m 33s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5736fd89b8b1477fae717ef8c2829106 : FAILURE in 1h 00m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/806405016c9747f0b3fd9e5b9ec0f9c2 : FAILURE in 38m 59s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/f9074b18aa554127b50b7e263ff83724 : FAILURE in 47m 12s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a9c3f6d42e324012928093abfda9d6f8 : FAILURE in 40m 15s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/e79e7be9004d424ca9b8fd8998bef078 : FAILURE in 46m 04s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/99a1aa6f7dca43a8bd6ec6ff95dcda42 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":13},{"id":"88882fdf194096860112c3c95501289003f96eaf","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-10-15 03:44:12.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"a66b481d3bc0c01ec1f8e163681d009b9b4b1e79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 04:01:58.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5816028acf24a22baeb8e0c52a78c0c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d31fbd06b204f1284a568d46f0d7f0a : SUCCESS in 15m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f05eb242e89d4aa996afcb9e0abc66cb : SUCCESS in 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"632c980c43d53d2d4dbfaa942eacc57f5dbe1473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 04:49:14.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4011fc9c33e54bf2be5c231eb091d64a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2ac2897c8f345fdbb13af57ec05aee6 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06a3e5709ef7411e84d47329bff0a838 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c819a053ab384a4ebd0e886c4da4ec33 : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40bde8e867f046b09dbf9cf11c3f1167 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e504d4397c68425da032952ba4a387ab : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c18c4ffb78584fde9267f1f1fe2165d5 : SUCCESS in 5m 10s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/66414afb9d384f0a886b5f8ab6c7ca7e : FAILURE in 33m 39s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/affc8d873163470bad5216ce0f4a0738 : FAILURE in 54m 02s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/3052a2f4106b4e82a03c9d42f73d1def : FAILURE in 41m 48s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/737fef942e2b41298efb67c01a3116a3 : FAILURE in 50m 33s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1677f35abfdd438996f924cb4ccd0d66 : FAILURE in 45m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/295c36b472f244f5b6fc1fd0866abdd2 : FAILURE in 59m 10s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5a679d041d744573af91edcc74d50425 : FAILURE in 51m 27s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/1baa0bbe296843be9514939aa2c3b2dc : FAILURE in 44m 35s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/00feed46ec67492a874413f4ac617305 : FAILURE in 1h 03m 00s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d63d933fb590439ea7fb964ddeda55ba : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":13},{"id":"e266278fac95f30ba5339c0f050a0cd81c0dd467","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-10-17 06:37:38.000000000","message":"Patch Set 14: Patch Set 13 was rebased on behalf of \u003cGERRIT_ACCOUNT_31737\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"_revision_number":14},{"id":"d0e29b349af6ce977ad2e20c8170ee5bf83750f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 07:02:02.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe4a3bf7763c442d9ec4c03bdb0c7e3c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/958c5838c34c442693b3dce39849a112 : SUCCESS in 22m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b9de962521fd41db92d4078ad2d3d308 : SUCCESS in 21m 16s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"2f0e33537fa4e1cbcbb3fd6510584587eab97210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 09:41:45.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ba24aa60694420eb9fc299ef308cec3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b08bd82981249429a13a2f83e063157 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89ff6355892e4575a1ef07e0adae631d : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/566824addd3f40d3b929fb305e444ad8 : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6efc503eae734f9a80540fcd2a255121 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b11bac75c5534c06b14d6a4611d18cc7 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0801dd9c62094b06af520ccce3f7cb75 : SUCCESS in 4m 43s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/51caff75a17b421daa2cb344e2b618b0 : TIMED_OUT in 3h 02m 23s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/569b5742e3ff429184d48ad6ba41cbef : TIMED_OUT in 3h 02m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/d4a18be015404dc8abdee447c46ad0df : TIMED_OUT in 3h 03m 10s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/02c77e84994f48309ce2d5ec3b50525c : TIMED_OUT in 3h 02m 21s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/735b1f48e53d45519c3dc7691f202c98 : TIMED_OUT in 3h 02m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/732c30ba4d5b4714a263009ad75b98ac : FAILURE in 1h 12m 26s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/2a1cca881b3c4960afb9f638665a6382 : TIMED_OUT in 3h 02m 26s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/6d9b72e25a4b449bb1b2f94885886050 : TIMED_OUT in 3h 02m 31s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/5cb318f17d6b4ab6af95cb41b207f651 : FAILURE in 1h 09m 42s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ccbd8397f1a7458787bb62a6f927b7ee : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":14},{"id":"5fa8002336397fa62d74440fdc53cc3dd24c96e9","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-10-17 12:50:51.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"7623ba1a1562619380ba55098f0efb2cac00acc7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-10-17 12:57:33.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0e5cbd9074b7907dca99bc495b67cc1779fbf962","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-10-21 07:27:51.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"d2bb403518e8e7f4d3cd6c7a9debd7ce2f88f3c8","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-10-21 07:39:17.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3ce5ef66e97acd5641c77906d1c4ec75dd70554d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 07:42:54.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39cd6e5052724ac7ad75f6761dd0d245\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e915a5f84b0c42c3813f5f3f49abbd39 : SUCCESS in 12m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f03d559e04524315ace2b0f5da5f80d5 : SUCCESS in 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"1be129e9ae9226ab45574d8debee66bbdd324cde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 09:11:38.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a2e101abba84ff2805ff66daff88eb1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94bdd58db9e04bd5b141099641b1f2b4 : POST_FAILURE in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d59152accafc40db833ebf9644a78a9e : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16c97896a9a342ce90ab883d27dd05d2 : SUCCESS in 8m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9535ad7ea36448029214d46f00b5d894 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39928f0c463545f9878688789143c7dc : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bbe680c8fe34870bd510495bd946c5a : SUCCESS in 5m 45s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/e866165683524baa940808f3e7384b1e : SUCCESS in 49m 49s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/f2933397fd0849adbeddcec78901faca : SUCCESS in 1h 18m 05s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/820529bed7044af8815b15fba5fe72ec : SUCCESS in 1h 28m 56s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f7ffc3133d3d4b1ebd68721c33b47cdc : FAILURE in 32m 36s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/dffd3e81492d43ecb001114d7ffa9403 : SUCCESS in 1h 32m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/beec2a1c3ff447bcadb738c53dd083e2 : SUCCESS in 1h 27m 39s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/263c93f95b154d1bae1630ccc5414bef : SUCCESS in 36m 16s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/6a6fb3cad256414e81baf58f6b21a53e : SUCCESS in 1h 42m 11s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/37e58176ee054bc08aa03b43b9d872d0 : SUCCESS in 1h 16m 50s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ad46de408f2541d09d75be8e6734344d : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":15},{"id":"35027a0e5adf3371b9bc4614c369662cb2f2cf2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-11-06 02:26:47.000000000","message":"Patch Set 16: Patch Set 15 was rebased on behalf of \u003cGERRIT_ACCOUNT_31737\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"_revision_number":16},{"id":"166ef966394d15898e5fb88eceb1c9c9ab1638f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 03:03:22.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7877fbd8e5f84619afe4621e9ce591d3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/665491cb0e8c41f9b3359c4adff250f5 : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9c67c13c827041c7b0aad7ac26374d55 : SUCCESS in 17m 32s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"c113d6cf06d141cc1c712f4ff5e066aa43fdc77f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 04:51:25.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28004b0bf6f84f20805f8c2ac89a4080\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ecfc80f0f494c59bfada3f6c43b125a : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e16bb19a87374d40b2857bb210282869 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0557facf94141f59e9a5a879f4bc4f3 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e018d361183d4f429bd8e5a44273d61d : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f573c4c36be452cb18eb82e7fe1fd64 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc43fbc14e1c4ecf875082457f1ab404 : SUCCESS in 5m 31s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/677f91516af547338351793f65df18b6 : SUCCESS in 48m 24s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/8ae98b63cd5d4e6e9d70e142f10b6adb : SUCCESS in 1h 54m 01s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/5dbd5949b2bd4a12a251d67c87c923ef : SUCCESS in 1h 51m 08s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f259d0df092e44b69c3b28ab98e6875f : SUCCESS in 35m 45s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/70406079a2fe47a9a4fdedd90d5da5dd : SUCCESS in 1h 16m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f037bf26e5fc4b65b59cdc551f5e569c : SUCCESS in 1h 55m 08s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/b31a5ac6353b43aa9e8481bc9d5fbd90 : SUCCESS in 1h 02m 58s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/4026422162b64996ac74275ce2fe7147 : SUCCESS in 2h 15m 56s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/49d1d64f2b9742cf8665b6c285b9a4b4 : FAILURE in 1h 25m 23s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3f83b18e58d146d198e65c02a0393839 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":16},{"id":"d30642ec7a55d3e1817d3e68e67986e927add022","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-12-25 03:07:12.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"fa1789d2f5ed19fa92154effca3120e77b0c3de9","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-25 08:26:47.000000000","message":"Patch Set 16: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"df20165a8c2e804ac37d86b90a343b3fb8b7aeea","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-08-20 21:34:13.000000000","message":"Abandoned\n\nI will abandon this to restructure the patch.","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"04ffcc52647f09dc710bbed470798f1f89b5cd5c","revisions":{"91c12cd855c8f1fa2454638e7d1db55b9b5c2366":{"kind":"REWORK","_number":1,"created":"2024-08-16 12:03:25.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/926466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/1"}}},"commit":{"parents":[{"commit":"3f521cb7f577635ee74b2a48c70f874dab2f9e30","subject":"[CI] fix trove gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f521cb7f577635ee74b2a48c70f874dab2f9e30"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-16 12:03:13.000000000","tz":540},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1] when\nstarting database container. Setting kernel options is not required for\nall databases but, it is recommended to configure kernel parameters in\nsome database as PostgreSQL. Administrators can configure the option on\na per-database basis.\n\nThis patch maintains backward compatibility. Therefore, administrators\ndo not need to change any configuration files when applying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/91c12cd855c8f1fa2454638e7d1db55b9b5c2366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/91c12cd855c8f1fa2454638e7d1db55b9b5c2366"}]},"branch":"refs/heads/master"},"efaddcdaa25c3ec802271bf23171874f4852d1d4":{"kind":"REWORK","_number":2,"created":"2024-08-16 12:28:31.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/926466/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/2"}}},"commit":{"parents":[{"commit":"3f521cb7f577635ee74b2a48c70f874dab2f9e30","subject":"[CI] fix trove gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f521cb7f577635ee74b2a48c70f874dab2f9e30"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-16 12:28:12.000000000","tz":540},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1] when\nstarting database container. Setting kernel options is not required for\nall databases but, it is recommended to configure kernel parameters in\nsome database as PostgreSQL. Administrators can configure the option on\na per-database basis.\n\nThis patch maintains backward compatibility. Therefore, administrators\ndo not need to change any configuration files when applying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/efaddcdaa25c3ec802271bf23171874f4852d1d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/efaddcdaa25c3ec802271bf23171874f4852d1d4"}]},"branch":"refs/heads/master"},"9d5e9860a942e797443b30f0578fc583dce86ba3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-08-16 12:31:59.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/926466/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/3"}}},"commit":{"parents":[{"commit":"3f521cb7f577635ee74b2a48c70f874dab2f9e30","subject":"[CI] fix trove gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f521cb7f577635ee74b2a48c70f874dab2f9e30"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-16 12:30:49.000000000","tz":540},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1] when\nstarting database container. Setting kernel options is not required for\nall databases but, it is recommended to configure kernel parameters in\nsome database as PostgreSQL[2]. Administrators can configure the optio\nn on a per-database basis.\n\nThis patch maintains backward compatibility. Therefore, administrators\ndo not need to change any configuration files when applying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n[2]: https://www.postgresql.org/docs/current/kernel-resources.html\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9d5e9860a942e797443b30f0578fc583dce86ba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9d5e9860a942e797443b30f0578fc583dce86ba3"}]},"branch":"refs/heads/master"},"2bc3c98ee8a238a4caf661b3a78614ac924d0e02":{"kind":"REWORK","_number":4,"created":"2024-08-16 13:03:14.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/926466/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/4"}}},"commit":{"parents":[{"commit":"3f521cb7f577635ee74b2a48c70f874dab2f9e30","subject":"[CI] fix trove gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f521cb7f577635ee74b2a48c70f874dab2f9e30"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-16 13:02:52.000000000","tz":540},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1] when\nstarting database container. Setting kernel options is not required for\nall databases but, it is recommended to configure kernel parameters in\nsome database as PostgreSQL[2]. Administrators can configure the optio\nn on a per-database basis.\n\nThis patch maintains backward compatibility. Therefore, administrators\ndo not need to change any configuration files when applying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n[2]: https://www.postgresql.org/docs/current/kernel-resources.html\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2bc3c98ee8a238a4caf661b3a78614ac924d0e02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2bc3c98ee8a238a4caf661b3a78614ac924d0e02"}]},"branch":"refs/heads/master"},"b5bbdd1b559793e1f01cc59638268d803cc9690f":{"kind":"REWORK","_number":5,"created":"2024-08-16 15:49:33.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/926466/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/5"}}},"commit":{"parents":[{"commit":"3f521cb7f577635ee74b2a48c70f874dab2f9e30","subject":"[CI] fix trove gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f521cb7f577635ee74b2a48c70f874dab2f9e30"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-16 15:49:15.000000000","tz":540},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1] when\nstarting database container. Setting kernel options is not required for\nall databases but, it is recommended to configure kernel parameters in\nsome database as PostgreSQL[2]. Administrators can configure the optio\nn on a per-database basis.\n\nThis patch maintains backward compatibility. Therefore, administrators\ndo not need to change any configuration files when applying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n[2]: https://www.postgresql.org/docs/current/kernel-resources.html\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b5bbdd1b559793e1f01cc59638268d803cc9690f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b5bbdd1b559793e1f01cc59638268d803cc9690f"}]},"branch":"refs/heads/master"},"135cf12b58110fe79d6885c36938aaea23ad72d3":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-08-17 09:06:14.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/66/926466/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/6"}}},"commit":{"parents":[{"commit":"7358330c14aee1ba3a449cb17607338e3b61e802","subject":"Merge \"Add UID/GID configurations for each datastore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7358330c14aee1ba3a449cb17607338e3b61e802"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-17 09:06:14.000000000","tz":0},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1] when\nstarting database container. Setting kernel options is not required for\nall databases but, it is recommended to configure kernel parameters in\nsome database as PostgreSQL[2]. Administrators can configure the optio\nn on a per-database basis.\n\nThis patch maintains backward compatibility. Therefore, administrators\ndo not need to change any configuration files when applying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n[2]: https://www.postgresql.org/docs/current/kernel-resources.html\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/135cf12b58110fe79d6885c36938aaea23ad72d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/135cf12b58110fe79d6885c36938aaea23ad72d3"}]},"branch":"refs/heads/master","description":"Rebase"},"424246b472a2d40c77a05fde52ccade1c7fae3a9":{"kind":"REWORK","_number":7,"created":"2024-08-22 04:38:08.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/926466/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/7"}}},"commit":{"parents":[{"commit":"7358330c14aee1ba3a449cb17607338e3b61e802","subject":"Merge \"Add UID/GID configurations for each datastore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7358330c14aee1ba3a449cb17607338e3b61e802"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-22 04:36:10.000000000","tz":540},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1] when\nstarting database container. Setting kernel options is not required for\nall databases but, it is recommended to configure kernel parameters in\nsome database as PostgreSQL[2]. Administrators can configure the optio\nn on a per-database basis.\n\nThis patch maintains backward compatibility. Therefore, administrators\ndo not need to change any configuration files when applying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n[2]: https://www.postgresql.org/docs/current/kernel-resources.html\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/424246b472a2d40c77a05fde52ccade1c7fae3a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/424246b472a2d40c77a05fde52ccade1c7fae3a9"}]},"branch":"refs/heads/master"},"521136c8ce85ddd69c29180c59e56299f68e997e":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-09-19 15:30:11.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/926466/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/8"}}},"commit":{"parents":[{"commit":"af2bea2dcc03f91cfb6fc8e8a8217be6fad501da","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af2bea2dcc03f91cfb6fc8e8a8217be6fad501da"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-09-19 15:30:11.000000000","tz":0},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1] when\nstarting database container. Setting kernel options is not required for\nall databases but, it is recommended to configure kernel parameters in\nsome database as PostgreSQL[2]. Administrators can configure the optio\nn on a per-database basis.\n\nThis patch maintains backward compatibility. Therefore, administrators\ndo not need to change any configuration files when applying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n[2]: https://www.postgresql.org/docs/current/kernel-resources.html\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/521136c8ce85ddd69c29180c59e56299f68e997e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/521136c8ce85ddd69c29180c59e56299f68e997e"}]},"branch":"refs/heads/master","description":"Rebase"},"21d2341da162bdcb332e36d607109bf2bc9df03c":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2024-09-20 01:03:22.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/926466/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/9"}}},"commit":{"parents":[{"commit":"af2bea2dcc03f91cfb6fc8e8a8217be6fad501da","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af2bea2dcc03f91cfb6fc8e8a8217be6fad501da"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-09-20 01:03:22.000000000","tz":0},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1]\nwhen　starting database container. Setting kernel options is not\nrequired for all databases but, it is recommended to configure\nkernel parameters in some database as PostgreSQL[2].\nAdministrators can configure the option on a per-database basis.\n\nThis patch maintains backward compatibility. Therefore,\nadministrators do not need to change any configuration files when\napplying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n[2]: https://www.postgresql.org/docs/current/kernel-resources.html\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/21d2341da162bdcb332e36d607109bf2bc9df03c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/21d2341da162bdcb332e36d607109bf2bc9df03c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6fa87b53c28d9371ac91a36fd6c83dd383feca88":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2024-09-20 01:04:19.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/926466/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/10"}}},"commit":{"parents":[{"commit":"af2bea2dcc03f91cfb6fc8e8a8217be6fad501da","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af2bea2dcc03f91cfb6fc8e8a8217be6fad501da"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-09-20 01:04:19.000000000","tz":0},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1]\nwhen starting database container. Setting kernel options is not\nrequired for all databases, but it is recommended to configure\nkernel parameters in some database as PostgreSQL[2].\nAdministrators can configure the option on a per-database basis.\n\nThis patch maintains backward compatibility. Therefore,\nadministrators do not need to change any configuration files when\napplying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n[2]: https://www.postgresql.org/docs/current/kernel-resources.html\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6fa87b53c28d9371ac91a36fd6c83dd383feca88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6fa87b53c28d9371ac91a36fd6c83dd383feca88"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5357387117bec5b7a555a6bff56f75c1ac95d244":{"kind":"REWORK","_number":11,"created":"2024-09-28 22:48:19.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/926466/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/11"}}},"commit":{"parents":[{"commit":"7358330c14aee1ba3a449cb17607338e3b61e802","subject":"Merge \"Add UID/GID configurations for each datastore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7358330c14aee1ba3a449cb17607338e3b61e802"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-09-28 22:47:25.000000000","tz":540},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1] when\nstarting database container. Setting kernel options is not required for\nall databases but, it is recommended to configure kernel parameters in\nsome database as PostgreSQL[2]. Administrators can configure the optio\nn on a per-database basis.\n\nThis patch maintains backward compatibility. Therefore, administrators\ndo not need to change any configuration files when applying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n[2]: https://www.postgresql.org/docs/current/kernel-resources.html\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5357387117bec5b7a555a6bff56f75c1ac95d244"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5357387117bec5b7a555a6bff56f75c1ac95d244"}]},"branch":"refs/heads/master"},"3db9655306aea350f890e4ade1184b0619fdac58":{"kind":"REWORK","_number":12,"created":"2024-09-29 01:22:58.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/926466/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/12"}}},"commit":{"parents":[{"commit":"7358330c14aee1ba3a449cb17607338e3b61e802","subject":"Merge \"Add UID/GID configurations for each datastore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7358330c14aee1ba3a449cb17607338e3b61e802"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-09-29 01:22:38.000000000","tz":540},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1] when\nstarting database container. Setting kernel options is not required for\nall databases but, it is recommended to configure kernel parameters in\nsome database as PostgreSQL[2]. Administrators can configure the optio\nn on a per-database basis.\n\nThis patch maintains backward compatibility. Therefore, administrators\ndo not need to change any configuration files when applying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n[2]: https://www.postgresql.org/docs/current/kernel-resources.html\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3db9655306aea350f890e4ade1184b0619fdac58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3db9655306aea350f890e4ade1184b0619fdac58"}]},"branch":"refs/heads/master"},"694b55b5b0c148a6e1612d3722c52506373a6151":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-09-30 01:08:39.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/926466/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/13"}}},"commit":{"parents":[{"commit":"af2bea2dcc03f91cfb6fc8e8a8217be6fad501da","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af2bea2dcc03f91cfb6fc8e8a8217be6fad501da"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-09-30 01:08:39.000000000","tz":0},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1] when\nstarting database container. Setting kernel options is not required for\nall databases but, it is recommended to configure kernel parameters in\nsome database as PostgreSQL[2]. Administrators can configure the optio\nn on a per-database basis.\n\nThis patch maintains backward compatibility. Therefore, administrators\ndo not need to change any configuration files when applying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n[2]: https://www.postgresql.org/docs/current/kernel-resources.html\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/694b55b5b0c148a6e1612d3722c52506373a6151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/694b55b5b0c148a6e1612d3722c52506373a6151"}]},"branch":"refs/heads/master","description":"Rebase"},"6860b24bd21eb84e33f46f1bf98f29ff3beefd35":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-10-17 06:37:38.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/66/926466/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/14"}}},"commit":{"parents":[{"commit":"e28298000b400f6898eb581da53c239d67b04618","subject":"Merge \"Replace deprecated LegacyEngineFacade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e28298000b400f6898eb581da53c239d67b04618"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-10-17 06:37:38.000000000","tz":0},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1] when\nstarting database container. Setting kernel options is not required for\nall databases but, it is recommended to configure kernel parameters in\nsome database as PostgreSQL[2]. Administrators can configure the optio\nn on a per-database basis.\n\nThis patch maintains backward compatibility. Therefore, administrators\ndo not need to change any configuration files when applying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n[2]: https://www.postgresql.org/docs/current/kernel-resources.html\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6860b24bd21eb84e33f46f1bf98f29ff3beefd35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6860b24bd21eb84e33f46f1bf98f29ff3beefd35"}]},"branch":"refs/heads/master","description":"Rebase"},"f5943ed58c321ff87fa62282ce347a4b002efa18":{"kind":"REWORK","_number":15,"created":"2024-10-21 07:27:51.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/66/926466/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/15"}}},"commit":{"parents":[{"commit":"7358330c14aee1ba3a449cb17607338e3b61e802","subject":"Merge \"Add UID/GID configurations for each datastore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7358330c14aee1ba3a449cb17607338e3b61e802"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-10-21 07:27:21.000000000","tz":540},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1] when\nstarting database container. Setting kernel options is not required for\nall databases but, it is recommended to configure kernel parameters in\nsome database as PostgreSQL[2]. Administrators can configure the optio\nn on a per-database basis.\n\nThis patch maintains backward compatibility. Therefore, administrators\ndo not need to change any configuration files when applying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n[2]: https://www.postgresql.org/docs/current/kernel-resources.html\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f5943ed58c321ff87fa62282ce347a4b002efa18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f5943ed58c321ff87fa62282ce347a4b002efa18"}]},"branch":"refs/heads/master"},"04ffcc52647f09dc710bbed470798f1f89b5cd5c":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-11-06 02:26:47.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/66/926466/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/66/926466/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/66/926466/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/66/926466/16"}}},"commit":{"parents":[{"commit":"3ff639047eea34c416872014639258d9ecdf067b","subject":"Remove Python 3.8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ff639047eea34c416872014639258d9ecdf067b"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-08-14 13:43:28.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-11-06 02:26:47.000000000","tz":0},"subject":"Supports kernel options when starting containers","message":"Supports kernel options when starting containers\n\nThis PR adds an option to set kernel parameters using sysctl[1] when\nstarting database container. Setting kernel options is not required for\nall databases but, it is recommended to configure kernel parameters in\nsome database as PostgreSQL[2]. Administrators can configure the optio\nn on a per-database basis.\n\nThis patch maintains backward compatibility. Therefore, administrators\ndo not need to change any configuration files when applying it.\n\n[1]: https://docs.docker.com/engine/reference/commandline/run/#configure-namespaced-kernel-parameters-sysctls-at-runtime\n[2]: https://www.postgresql.org/docs/current/kernel-resources.html\n\nStory: 2010126\nTask: 45713\nChange-Id: Ib5ce722e0193c5d1ea4b9d00bd0a20f4fda9ab7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/04ffcc52647f09dc710bbed470798f1f89b5cd5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/04ffcc52647f09dc710bbed470798f1f89b5cd5c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
