)]}'
{"id":"openstack%2Ftrove~927033","triplet_id":"openstack%2Ftrove~master~I6365b37b7b31dbc7337aa4903e4ad4cf27df9faf","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I6365b37b7b31dbc7337aa4903e4ad4cf27df9faf","subject":"Remove outdated migration script","status":"MERGED","created":"2024-08-23 14:42:47.000000000","updated":"2024-08-28 06:54:34.000000000","submitted":"2024-08-28 06:53:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927033","meta_rev_id":"93f3995704be2eca86e0e1ba6891e9aac7d4b507","_number":927033,"virtual_id_number":927033,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-28 06:53:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-28 06:09:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-28 06:09:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-23 16:22:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91faed365aa381e43bfaf1028065c3885984e5e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-23 14:42:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cae6656fb78751a39a1de266ce761f538715538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 16:22:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/981ba11304814dc1ae90c2e7af22fc88\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6cb67f5eab7545dd94bfd6c306ffc6f4 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25d6b58f8e6d4ea8ace05f76cecc4eb1 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df1f95834a3a4fffa7afb714056eb66d : SUCCESS in 7m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a5731cb8bfa4d2aa477679c158492e1 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da44f54b1fd14aa48f952b9da5c86700 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f67d7e699e6548d3a7bbcedf5b74950c : SUCCESS in 5m 25s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/170a5ce4eca246ad96f8fd4d98d7828d : SUCCESS in 39m 49s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/817f15d29d2043c2a69d1afde58d59f5 : SUCCESS in 1h 38m 46s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/a9c02b6d4bbb41da98a2e6669cfb3d0d : SUCCESS in 1h 18m 00s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ecc79d4d773944a0b1ad6d020b0fee22 : SUCCESS in 40m 22s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b235ecf4a37f4a72a964b10ed7a80e7f : SUCCESS in 1h 28m 50s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e809fa56d69e45c4aff2c18f5187017f : SUCCESS in 1h 11m 13s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/7b76113f370e41d396f80fc1195cb1ee : SUCCESS in 40m 39s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/6781e8cf0a924b69b8844f5c369250ae : SUCCESS in 1h 26m 57s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/5d7323d1787243a3ae07fed2d8375fcc : SUCCESS in 1h 13m 15s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/19d3437d61874f2da2117c7085d194fc : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1da0e2bbc6c393938b2f5ede301e34534f28aab6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 22:37:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42a954175653446eacaf9e8bef548729\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d1555dfed0841cbb0144cf65c62559e : SUCCESS in 23m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ca6859d19bc14606a987e7f4a2ad261f : SUCCESS in 12m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32d9880578a02f98a0c10016d6190be20ad61aa0","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-08-28 06:09:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"032ca3e59c83a7f6a93f0d45da1b6312cc195a0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 06:10:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2c72b764f9b955a6831a8eaee8dac1418fefe1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 06:53:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8514a9e517624157ab990540b4143290\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39ba0996a10b4d11b7e1277d056a8c29 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e117a4964aa1453e9d86d182608e5dcf : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5cda74392d394e08b023930b26ec66fd : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0e2747160cc465182db42f7108405ed : SUCCESS in 5m 25s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c78facc39be8457c9f79fef17e5e9294 : SUCCESS in 39m 51s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b815328834a04fc48cd55fdb79bb73bc : SUCCESS in 37m 09s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/165db80ae9334d45a1663edc2725bbdd : SUCCESS in 42m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7a6c718383aa06fac46aa7b97f22efe6b13efd62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 06:53:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"93f3995704be2eca86e0e1ba6891e9aac7d4b507","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 06:54:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/163a4c311ea6447cb014ed5440c19c20\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe9e044be0042d3a526da1db504a9fb : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0759d68135eec4aba89f78c46c3df1928ea37111","revisions":{"0759d68135eec4aba89f78c46c3df1928ea37111":{"kind":"REWORK","_number":1,"created":"2024-08-23 14:42:47.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/33/927033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/927033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/927033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/927033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/927033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/927033/1"}}},"commit":{"parents":[{"commit":"b1c74bf6bff6be1402bf42648113e62f09ec2475","subject":"guestagent: implement \u0027rebuild\u0027 for postgres manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b1c74bf6bff6be1402bf42648113e62f09ec2475"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-08-23 14:42:06.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-08-23 14:42:06.000000000","tz":480},"subject":"Remove outdated migration script","message":"Remove outdated migration script\n\nChange-Id: I6365b37b7b31dbc7337aa4903e4ad4cf27df9faf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0759d68135eec4aba89f78c46c3df1928ea37111"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0759d68135eec4aba89f78c46c3df1928ea37111"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
