)]}'
{"id":"openstack%2Ftrove~933847","triplet_id":"openstack%2Ftrove~master~Ib615032747f89daf7f8500c7e2df4064fab05314","project":"openstack/trove","branch":"master","topic":"fix_error_add_datastore","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2024-11-14 10:39:35.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-11-14 09:55:01.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Ib615032747f89daf7f8500c7e2df4064fab05314","subject":"Follow up: Add Datastore Version Registry Extension","status":"MERGED","created":"2024-10-31 10:40:28.000000000","updated":"2024-11-14 10:41:27.000000000","submitted":"2024-11-14 10:39:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"933847-fix_error_add_datastore","meta_rev_id":"62a2bbb9664a0d97002640737882db7d7ee2e4bd","_number":933847,"virtual_id_number":933847,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-14 10:39:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-08 07:07:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-14 09:55:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-31 10:41:56.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2024-10-31 11:37:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-31 12:17:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2843de8fa9b3ad1388b3b7ecd19af7996bc5bc8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-10-31 10:40:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b7d00101fa5ee44dcb51d8559fcbe787775ea2f","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-10-31 10:41:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb4794a88e71c87d1d1a18d615d434206ca68eff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 11:37:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/129ede2f5083487fb03f82522a24e3b7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0d1c24e4fee34b7eb57c6f0106132194 : SUCCESS in 20m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3263241b3feb430486000f81b0e16648 : SUCCESS in 17m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b78734eb18d49894d8585085205b7647d84a1f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 12:17:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b44fc9c92fe948f58585d74818ef8afa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38b75e970d1a4cfabded7c494775499c : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dd105699f9b4f73804f5c24a7f252e0 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c4cd4f146f4463fa8788ca256400c94 : SUCCESS in 6m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f682b5fe0fb84a1798d7da79511911ea : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e04be357984547c388eb00e4350c0b78 : SUCCESS in 5m 00s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c27a5f6d888243bdb816d7e52ad92226 : SUCCESS in 40m 14s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/c9a378689d2d486d814796a8fb1d2064 : SUCCESS in 1h 34m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/673cdd90835e4fbaa37436bff73cf24f : SUCCESS in 1h 24m 35s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/51da7ce7638d46d3935a0a7b81a37a87 : SUCCESS in 44m 33s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/06f671d439e544f2a9eedd5d1ea2d4b8 : SUCCESS in 1h 34m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6c9c40e0bee441eaab2fdb7ba75fae61 : SUCCESS in 1h 19m 00s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6f558747eb774b01b4fa3d378de957df : SUCCESS in 47m 35s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a086212cd64a475787b645a4c2662dc9 : SUCCESS in 1h 30m 29s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/383e54005d7b47d78976eee24c7845b1 : SUCCESS in 1h 20m 04s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d727fb9b358149d4b01cd1dc530ef5b4 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6193cd47f69314650480686383429989297422ff","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-10-31 13:43:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"341dc995c2dde6c153cf658695696407bab61647","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-10-31 13:52:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c43cac19302537570f7b37588235d0b914039854","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-10-31 14:36:22.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"efcfd0df5107338c1c877744cff8cb0010f46077","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-10-31 15:24:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db5649ba2aef1211216e14045d9bdce27b857a20","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-11-01 08:28:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d456eb5f7dfac21f0a8c92d0b22d62a092c3821","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-11-05 02:54:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"946f6a992f1ce0cc0b6f807cb05ad9608e72d41c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-11-05 06:01:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a8c6e2491f98145ab9c046c82789aa0b94b3120","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-11-05 06:26:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c149f70ff4494ddeee789f1df3e8465be736e9fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-11-05 06:34:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d8e40b5e27748de183b07ddc95756b74703dcf89","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-11-05 06:36:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a612d08f150b95a1b5f99d08c0b8070554ad46b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 06:50:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfa4fcdfb00b450886379282f3153758\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96839765ba1544279653bdfd6749afb0 : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b8c558761a144f7e931e58a4e3f8eeb5 : SUCCESS in 14m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"454060771d3a9fa28220da8122463fb628338b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 08:11:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45414d71fc5e4fb295f195b09362a4d9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8f7ead3a20c24a69ad11714fd22300b6 : SUCCESS in 4m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a374cb2c8d64b878a25c183713ba9e8 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7c0f543e49d4768a9e37cf65e92bfac : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2d0e6fe60154eae99a36a6568f8de1a : SUCCESS in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb4faced73b54546afa83bcad518fa8c : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17d0205c8195450fa314635523c02a4e : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/204415672a094b3f8ea2aa6925097411 : SUCCESS in 4m 37s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/68a3778e9f8e47df84f08f2d514541b8 : SUCCESS in 40m 47s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/1f19401895d7457e905ec3733f3b51ff : SUCCESS in 1h 36m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/29402fae3b9c4918840b3de53d13d974 : SUCCESS in 1h 19m 43s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bd204cfbf85c4b8c910a31622bc9ab4c : SUCCESS in 41m 01s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/406cee5f32d047ac89c0124c2c23d327 : SUCCESS in 1h 15m 00s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/371d698c15fe40158713874055d0d157 : SUCCESS in 1h 34m 02s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/39f7c2d78c4f4239b8cc90ef2a00ceae : SUCCESS in 43m 32s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/c6c19fcba5334eeaa194bbe6c505d808 : SUCCESS in 1h 12m 36s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/82bcbb4440d74cef929d099f2fd45f3d : SUCCESS in 1h 31m 01s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4503521147de4f7cb8cac0b701c78238 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":3},{"id":"0c015af2520848c0aef897aa99e83d267d681509","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-11-06 16:35:03.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5f08960e6363a8a81be31c7f4ecd674ded2321ec","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-11-08 07:07:36.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8ec17d5a42bd8c5151176ddd2866b10ff798f946","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2024-11-14 08:27:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7d071ee2bd5a32fd100a5ac56a361161491dd96f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-11-14 09:55:01.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b73391cfeceb8b1aba6e420778bde568cd700b36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 09:55:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a5d9c163b146d0ac99a02085eae05b9de74445d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 10:39:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1508261ea402457b817f168fe278567c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2b3fc648bf7f44cbbf38b1fede87085a : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fec8268732364d21bd258d9f5c82ce66 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/920fc5f19fd4445295d7c8a28c5a0eaf : SUCCESS in 6m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d3123c7a4094fea88a3cf7a824aadae : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79d654f8ff4644178b841dfc261e6034 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9aa674cb52574a019502672d8b869c0a : SUCCESS in 5m 55s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/e7ec6db81d794a498679bfc8dac807fc : SUCCESS in 42m 37s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6fd27f4833a745b093629670f7e80625 : SUCCESS in 40m 01s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/eb1570a6bdb74ce1a8164bde85f16e31 : SUCCESS in 43m 07s","accounts_in_message":[],"_revision_number":3},{"id":"f3e88e76ddb0c1681137e59751d3288b0e3500fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 10:39:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"62a2bbb9664a0d97002640737882db7d7ee2e4bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 10:41:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52ab27e1483b4c6ebfc99b23b4711c1c\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/48c5ff621d08446e84b2f1425efdcb92 : SUCCESS in 46s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c6db10a1832466baef72dc0b2af010e : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0fc4e547e0b4fb5ba317965d092d526 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"596134e3ad3b96deca937b1dcb7b79dd80f3c32d","revisions":{"55e43402e18089e33678536cc384d14900c5b5b9":{"kind":"REWORK","_number":1,"created":"2024-10-31 10:40:28.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/47/933847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/933847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/933847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/933847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/933847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/933847/1"}}},"commit":{"parents":[{"commit":"3ff639047eea34c416872014639258d9ecdf067b","subject":"Remove Python 3.8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ff639047eea34c416872014639258d9ecdf067b"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-10-31 10:38:40.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-10-31 10:38:40.000000000","tz":420},"subject":"Fix error commit: Add Datastore Version Registry Extension","message":"Fix error commit: Add Datastore Version Registry Extension\n\nWhen we run upgrade version of trove, we can\u0027t migrate database\nof trove and we can\u0027t run cmd to list datastore.\n\nThis patch will help us fix above error\n\nChange-Id: Ib615032747f89daf7f8500c7e2df4064fab05314\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/55e43402e18089e33678536cc384d14900c5b5b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/55e43402e18089e33678536cc384d14900c5b5b9"}]},"branch":"refs/heads/master"},"21a68d507af5f5ca7bf5af051b8ad560ed21002a":{"kind":"REWORK","_number":2,"created":"2024-11-05 06:26:53.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/47/933847/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/933847/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/933847/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/933847/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/933847/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/933847/2"}}},"commit":{"parents":[{"commit":"3ff639047eea34c416872014639258d9ecdf067b","subject":"Remove Python 3.8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ff639047eea34c416872014639258d9ecdf067b"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-10-31 10:38:40.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-11-05 06:24:28.000000000","tz":480},"subject":"Follow up: Add Datastore Version Registry Extension","message":"Follow up: Add Datastore Version Registry Extension\n\nWhen upgrading Trove, the trove-manage script fails\nto upgrade the database when the DBDatastoreVersion\ntable is not empty.\n\nThis patch will help us fix above error\n\nChange-Id: Ib615032747f89daf7f8500c7e2df4064fab05314\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/21a68d507af5f5ca7bf5af051b8ad560ed21002a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/21a68d507af5f5ca7bf5af051b8ad560ed21002a"}]},"branch":"refs/heads/master"},"596134e3ad3b96deca937b1dcb7b79dd80f3c32d":{"kind":"REWORK","_number":3,"created":"2024-11-05 06:34:14.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/47/933847/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/933847/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/933847/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/933847/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/933847/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/933847/3"}}},"commit":{"parents":[{"commit":"3ff639047eea34c416872014639258d9ecdf067b","subject":"Remove Python 3.8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ff639047eea34c416872014639258d9ecdf067b"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2024-10-31 10:38:40.000000000","tz":420},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-11-05 06:32:25.000000000","tz":480},"subject":"Follow up: Add Datastore Version Registry Extension","message":"Follow up: Add Datastore Version Registry Extension\n\nWhen upgrading Trove, the trove-manage script fails\nto upgrade the database when the DBDatastoreVersion\ntable is not empty.\n\nThis patch will help us fix above error\n\nChange-Id: Ib615032747f89daf7f8500c7e2df4064fab05314\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/596134e3ad3b96deca937b1dcb7b79dd80f3c32d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/596134e3ad3b96deca937b1dcb7b79dd80f3c32d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
