)]}'
{"id":"openstack%2Ftrove~935447","triplet_id":"openstack%2Ftrove~master~I414b24355a2b750f5ae0ff92613cfbfc1f39f3ff","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-12-04 12:54:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I414b24355a2b750f5ae0ff92613cfbfc1f39f3ff","subject":"Remove pylint target","status":"MERGED","created":"2024-11-16 16:04:04.000000000","updated":"2024-12-04 12:55:58.000000000","submitted":"2024-12-04 12:54:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935447","meta_rev_id":"7cf3b47338d41f201b081b90217c68bb541f23ea","_number":935447,"virtual_id_number":935447,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-04 12:54:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-04 11:17:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-04 11:17:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-16 16:15:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-16 17:49:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-04 11:17:21.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"34aafd689de2089ad7b13bb05ad157642c64a636","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-16 16:04:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d91b4f3e4abd68d5c37c24229d8bbce25a2a1f04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 16:15:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a290c5efcde4f018e6cdca4e5195897\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a936ab9f6ce643be9d3dda4a9f034bcf : SUCCESS in 10m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6df51a28383e48b18df5e49a2f9382a3 : SUCCESS in 8m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03cf2dc7a3a7e3a0e4d1090ab1449b34b0a9327b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 17:49:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dae82c50c2e4a1c9c43887da7f6a756\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f55d2388d9b04202b72ebfad0a5102c6 : SUCCESS in 3m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4778f2b859a24eaf9dd1f7480a8f959e : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c74633b89df467d9c0e17b6f3220ef7 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06ba06c7747a46b889bc8184a64eb216 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e904935de3b146c8b49c3d927c3f9bd8 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed135deddb1f4a0b95b636b3e24cc95b : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e689edf44e54fd2a74ce0838d861f53 : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f1f6235de7348658fe56cce273c3e39 : SUCCESS in 4m 34s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/3cb3bb57492f4833bab66e976eb3c5fd : SUCCESS in 49m 56s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/ff39063708db4ad5953a30d4084f94c4 : SUCCESS in 1h 44m 05s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/eec7db5745bb41b4b4df546a960fd524 : SUCCESS in 1h 28m 34s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/65cda7f1bb5a49378253a62c4e44ad4c : SUCCESS in 37m 50s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/757a075c586a4fcbb7457bab8f3b23c8 : SUCCESS in 1h 37m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7d04bc9797ae4d668537190b6334c15a : SUCCESS in 1h 23m 24s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6b7ae80842ef40d393e0f62758cbbf20 : SUCCESS in 49m 34s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/41ea3f42b1654c0193aa3b2d1cbccb23 : SUCCESS in 1h 35m 44s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/e0847d47f23e423d99ad0334781b2ffd : SUCCESS in 1h 19m 51s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6e8f0460251b4fbebe063ed4b97df5e0 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"3675773f220d72a1c97d3aeb4afbe160013b498f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-04 11:17:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"240fac3b1535cdb68879100b27505ece7d604a26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 11:17:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eacb88b4a76441954d31ba1557a0326b1cd0ded5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 12:54:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/140038773d0744b6918af4e180c35605\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5b30e82a309f420ca31b5dfc032f5e69 : SUCCESS in 3m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/102a3acf25eb4d7e9d1f39fdb0800962 : SUCCESS in 2m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2744500b525a42949c271c56ff5de80c : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/289c260a7ce343319ee6e2fdb1738ef5 : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/53c5e9c505674f89bbd28657f1463d2f : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbde8de4cba24fa98a8c3be46bcc3b05 : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00bb91eb8f38495cbac7273211cef746 : SUCCESS in 5m 54s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/6dd1b0467b17405084fe9f7b41758b77 : SUCCESS in 59m 35s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/a6ecfe23c1e24b07994f3c2750579f6b : SUCCESS in 31m 57s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e14954e7d3284de1bee0ff7e0a02b1d9 : SUCCESS in 1h 00m 19s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/65d0433a8c634017902bc02e505ed6e0 : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":1},{"id":"4acb8be44c72ed998096c1744242ba932a2d002e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 12:54:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7cf3b47338d41f201b081b90217c68bb541f23ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 12:55:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/467b94d78c694e68bae1b6da5cb6f7d4\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/08b1640023964455bb6c288f7bc0c7f0 : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d4fdef29c34b08bffd2576aaf7c633 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7341402055a34de09ab29cc0ca9a1e92 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0057801f87f5150563d78bce4b7936bf491d9841","revisions":{"0057801f87f5150563d78bce4b7936bf491d9841":{"kind":"REWORK","_number":1,"created":"2024-11-16 16:04:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/935447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/935447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/935447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/935447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/935447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/935447/1"}}},"commit":{"parents":[{"commit":"e28298000b400f6898eb581da53c239d67b04618","subject":"Merge \"Replace deprecated LegacyEngineFacade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e28298000b400f6898eb581da53c239d67b04618"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-16 16:02:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-16 16:03:51.000000000","tz":540},"subject":"Remove pylint target","message":"Remove pylint target\n\nThis target hasn\u0027t been validated in CI. Currently running the target\nconsistently fails likely because of the quite old version pinned to.\n\nBecause hacking/flake8/pep8 is already used, drop the unmaintained\ntarget.\n\nChange-Id: I414b24355a2b750f5ae0ff92613cfbfc1f39f3ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0057801f87f5150563d78bce4b7936bf491d9841"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0057801f87f5150563d78bce4b7936bf491d9841"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
