)]}'
{"id":"openstack%2Ftrove~935840","triplet_id":"openstack%2Ftrove~master~I2ad2e692e5993d7cf8029f57c10538c854b06089","project":"openstack/trove","branch":"master","topic":"requirements","hashtags":[],"change_id":"I2ad2e692e5993d7cf8029f57c10538c854b06089","subject":"Remove diskimage-builder requirement","status":"MERGED","created":"2024-11-21 02:43:38.000000000","updated":"2024-11-21 08:17:48.000000000","submitted":"2024-11-21 08:16:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935840-requirements","meta_rev_id":"049812f2691543903fe66ad7ca140ff798851a26","_number":935840,"virtual_id_number":935840,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-21 08:16:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-21 07:15:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-21 07:15:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-21 02:59:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-21 04:28:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"15e21f41e5cc612855dc1698d88e38a9a6eabecf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-11-21 02:43:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2012afc2f328ea75b99333ac88a54d9750448b3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 02:59:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65320b60490f4a69bcb93d949a593951\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e4c603c7b6e4a9cb90f8632e08a1683 : SUCCESS in 13m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6396e7d41c3c40abba8ad1b25b975660 : SUCCESS in 13m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be926dbca0ed7cf9fe8b95f6104342c21d00dcd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 04:28:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53091425c2a94e3a894b0a7b60e31e33\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6acf8ea26e44476583cb117b422d61b1 : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdecd866d456447797ff434ae1588b70 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ff488ae061c4a99a1bdc0ac9c9def74 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ed9e2f5e6064c0fb33ca2767eb6c36b : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ebf99c8460134ce28ba690dbd6e8eac2 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70e6869ce70142baaf6668af1b329be1 : SUCCESS in 3m 42s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/d3dd196a64064f2ba9254f12c47744b9 : SUCCESS in 58m 07s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/1b6870a6fe3d4047ac35f6ec0390c194 : FAILURE in 33m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/69a3ffd9226744af8032a9b22583241a : FAILURE in 34m 37s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b8e34332453145d48b54b0ad6a1d9692 : SUCCESS in 48m 37s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b3c4fee520ad412b85fc57165add1fd7 : SUCCESS in 1h 42m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4e75016999a442aa8fd060c5243f8c09 : SUCCESS in 1h 24m 55s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5b657944c13c4f27acdb07b260d9c851 : SUCCESS in 33m 21s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/3385dd52d94849fbb4067b70b6b99bc0 : FAILURE in 32m 08s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/efbeeef3a0d7445db8e2a36281329185 : SUCCESS in 1h 36m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3272be6bb86527a73cf5ba2c079aa0e3630c6cce","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-11-21 07:15:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d1de049b715d054bb7e24c97aac437d6017f378c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 07:15:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89f50286647222ac11e17310d909824c040e5fc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 08:16:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a38538681321434da3bca90dbab57458\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/897b55c1042d4409a1021761eca72821 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b659490242764fdebdf231ce4f830898 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d594ad9eb68469aa72c1cd62bc66e3b : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11ae7e822804473dbddd6118ae4063c9 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2442bdc33a764bbca957f13f15f54d05 : SUCCESS in 5m 35s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/9a7a119808ea42268937383c8351d9dd : SUCCESS in 59m 06s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/75db6cf07f7348228ed3b612ae048488 : SUCCESS in 53m 16s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/cc15a49d314341f790e8a37436ec7932 : SUCCESS in 54m 56s","accounts_in_message":[],"_revision_number":1},{"id":"83615a5796e582853f5ea88f1fb2f519b6a553ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 08:16:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"049812f2691543903fe66ad7ca140ff798851a26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 08:17:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40e6ccd1b0934cbca5495cd07a2a088b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eef18b17d0bb48f9931bea1fe7996dd7 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d8501db5651518bb4fcaa6813525632497c23db","revisions":{"3d8501db5651518bb4fcaa6813525632497c23db":{"kind":"REWORK","_number":1,"created":"2024-11-21 02:43:38.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/40/935840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/40/935840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/935840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/935840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/935840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/40/935840/1"}}},"commit":{"parents":[{"commit":"cf5a9b803e9cdf799688cc339c9ed027e92a7d76","subject":"Merge \"Remove default override for config options policy_file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cf5a9b803e9cdf799688cc339c9ed027e92a7d76"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-11-21 02:42:55.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-11-21 02:42:55.000000000","tz":480},"subject":"Remove diskimage-builder requirement","message":"Remove diskimage-builder requirement\n\nthis package was not imported in trove project\n\nChange-Id: I2ad2e692e5993d7cf8029f57c10538c854b06089\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3d8501db5651518bb4fcaa6813525632497c23db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3d8501db5651518bb4fcaa6813525632497c23db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
