)]}'
{"id":"openstack%2Ftrove~936106","triplet_id":"openstack%2Ftrove~master~I6fe614e19b56dd20d503910afb48f349b103f8f0","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2025-09-08 13:03:27.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-08 10:24:24.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-09-08 06:00:19.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-09-08 13:03:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6fe614e19b56dd20d503910afb48f349b103f8f0","subject":"Remove pylint","status":"MERGED","created":"2024-11-24 15:51:25.000000000","updated":"2025-09-08 13:05:10.000000000","submitted":"2025-09-08 13:03:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936106","meta_rev_id":"c264d19e537e339295fd45e101f4913127732fc7","_number":936106,"virtual_id_number":936106,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-08 13:03:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"recommended":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-05-01 16:23:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-08 06:00:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2025-04-26 19:22:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-08 06:00:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-24 16:07:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-24 17:34:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-26 19:22:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-04-26 19:22:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-05-01 16:23:27.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-07-06 22:35:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"}],"messages":[{"id":"b2b6d676a709bfb3347178fe232f1e1c5534eba6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-24 15:51:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"318220dc999c26fe34ccf4b8dddad963ca68e441","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 16:07:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a85cc5c07204ff9a2d58df30dd66b22\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90b42cf220694c25a30c902b226f0d23 : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8d7aae536330438fa42881030c6f8d10 : SUCCESS in 13m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e3bad8ca341a8d22773002394893b7bbf2862f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 17:34:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3093e9edce634f59a96ec570e2d4b505\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/11308155d6744822ac63b35fd24d7ab9 : SUCCESS in 3m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/39015773e9a0463aa47b1d27a2cea158 : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72f6ed6615874d3098356d8d723e250d : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d75f8f50f0db4b97bfaabe67e362259c : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f5aa3ef38064f5cab16b75bbedf2e17 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25b90660adaf47069a5f7371b3ca56cb : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d10657dbc33d43d790222bf3f52bfb5e : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c76aae392464c92bfce4462f81e8d38 : SUCCESS in 4m 34s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/6fabd42ec6cf4e09bf208277e802dc5c : SUCCESS in 40m 28s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/4daaa61db408433683f1a608ea274d0e : SUCCESS in 1h 42m 07s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/3dca7e3976844872b600fe030982861d : SUCCESS in 1h 18m 53s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ffa4d632376a44f5a5d3f127df3fa0e2 : SUCCESS in 38m 21s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/dbf3fbc9f294471a9801cc2c9977749c : SUCCESS in 1h 25m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/26d5427a5c2142dc805267ca27a5a9c7 : SUCCESS in 1h 11m 39s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/8ed38fe18933471aa085fb5e39fa621c : SUCCESS in 41m 16s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/ac85dee7b2a14b37877b46b0b685d23f : SUCCESS in 1h 23m 05s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/e78493e191b04e24a490a295ea898449 : SUCCESS in 1h 08m 41s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6c4ed98c522648c79f06ca3c91cdb458 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"2e6b49f70b2b0f307fb67d715ced33534a66efd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-26 17:13:38.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d9656bd4a0501ca7d22977972f7ea78e9f34fc1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-26 17:13:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4bfa966b654451277426b1e6987adcaa01045390","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-26 17:17:49.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"875fadb3f9b06b2b868ec46a671787f7667a3a29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 17:32:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2021453435d042489d618a5e9068abd5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5a879e28eb11423fbdef53d7f676967b : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3a5b98c428be4688b775934cb1fce4c6 : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/87459d6ef3054254aae0944e7df4f6c8 : SUCCESS in 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b2340d0a4607e842205f5f1536603413d5732d8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 19:12:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fd2499e1d2d40e294d7284bdc3e8173\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/551a62034cb4450cb5562dd228c9dd72 : SUCCESS in 4m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1fa7fbf2558743eab36bc80500e03f2b : SUCCESS in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/978826cee8334227aa79a81824faf351 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cb5143d4c7046e4924094c42d73005d : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bd47411f23145e3842ebbdf8779bdaf : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f3cca2de9934516ab75facb1bbb201a : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a8ea4e965e044699083654d440b7a68 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce2ca7addfe0431c9b07326745fa872f : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54fa1b641fda442aa4d8f04c34043be1 : SUCCESS in 4m 23s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c19e72d3424048d7acbdbb9e6a2d6264 : SUCCESS in 45m 24s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/778025bf20f14950b167c91e649e63d9 : SUCCESS in 40m 38s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/a8b9bf1f9c3d4ddba1f261d75ad59538 : FAILURE in 1h 40m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/6b0e7826badb41919572b32b733014a6 : SUCCESS in 48m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/93b07874ec4248109623eedbaacfe3e0 : SUCCESS in 1h 10m 07s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/83bbad515ca54479bccab4caa2a4e4fb : SUCCESS in 42m 02s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9cbd07864ac34c5284d64e3cbc51e624 : SUCCESS in 1h 52m 57s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/93cbc81dd6f241dca783e085227746df : SUCCESS in 47m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/17d8c93aa51348158f5b2de25b89804d : SUCCESS in 1h 03m 55s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/1b9d2344e24144cf9761e513cb31ca69 : SUCCESS in 48m 18s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/e826ed5a78ae468f87b5faf4a5058654 : SUCCESS in 1h 36m 34s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/01565a7b36ef42b3b520263cc6a8c915 : SUCCESS in 42m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/4aa3d24603ac449fb03bafae47c60011 : SUCCESS in 1h 50m 16s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/096904c66d054d41861294eb7825e0b2 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":4},{"id":"92993fe943be91379a28abe9da0ce2ee8ae72cf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-26 19:22:11.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"83e309d74f916e7a8bd55d7d92fd4f37ec9d7c10","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-26 19:22:23.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"ab210840f0b8dcc30d19a76d42a008a023220388","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 19:37:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5523818efef24f5d9d8969c47494eaf5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f8f26cac746e4e4c9125c23fb300e2bd : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8f9c298b41434b4a85237e2523b20de1 : SUCCESS in 11m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6a0410bc079440bbb4b6dc7e82943448 : SUCCESS in 13m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1f45f3f51a433aed71716e27fcac991817b34fa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 21:40:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6df8b84350aa4c2aaef7185f17a76085\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8ce79fd4225e4ea993223c2f06738150 : SUCCESS in 4m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/19bc4aab21d345b68fd63f9ed13acbf3 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61b657537b1845389a21da0098791764 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3e8e9f62c14d45b9de1bcbaa0a9109 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4dba9ab82f494014906fab19698840a0 : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c563ac563b14281a3b40d1e28ad07e1 : SUCCESS in 5m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1daf6270536b4e198d2149c2957c35af : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/953c28fdd8ff4a72b4bf19c7bde831f9 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3323d91c510743b398201f7c3810f36a : SUCCESS in 2m 50s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c8dac4006d6f4a5cb52a82033f59eb62 : FAILURE in 26m 06s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/1e202aa1339447b398d537c1212088f0 : SUCCESS in 37m 29s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/d1cc39fd3ef24a829e9af0cc415acaa0 : FAILURE in 1h 58m 25s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/3f7f7c993483421cadae7e806ea22c69 : SUCCESS in 1h 51m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/255f20a336334edda223d5553c0ffd72 : SUCCESS in 2h 10m 43s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/23e4beafd15a45039a26179109aef0ee : SUCCESS in 56m 51s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/70dae66c511f4986a664defe23b10ffe : SUCCESS in 2h 08m 27s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8040b1f9106e4101aaffe0d538f2b02a : SUCCESS in 1h 36m 24s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e967eb8afbcb42848c7b11ab4467b02f : SUCCESS in 1h 59m 21s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/fe5ed274edcb46f5828b1e9a0f3fd78d : SUCCESS in 53m 24s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/556e13c54e704a13bde9812e34ce97ea : FAILURE in 54m 42s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/fccaba4622c94e6a914dc11ed8995862 : SUCCESS in 1h 48m 22s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/37941c18129a43a59b251d90c69efa30 : SUCCESS in 1h 36m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9c913e9d278f477eab0ad60664055e50 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":5},{"id":"d184be40b0b161b404230f8efe2f2247494ec483","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-26 22:05:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"52cedd6942130f199edac372e92dae9fcaef2f1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 22:20:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/480b66f3d2534d49a11a1e8b7f6f3d2e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/db9abce5038f4ba6b92824a0c9d3cd7a : SUCCESS in 10m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c6564291673e4e44b2bf7c072ae62331 : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/094c981830e044eeae34048df9bb78dc : SUCCESS in 14m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4839d52061c2baa9420b97171d313c5759142618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 23:56:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7d90c9cb0cf4fb6a796d3caef7dfb5c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/734d1cbb79a944589d9c65424eea1224 : SUCCESS in 4m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f7380f9930ef470a8b500e1a55c00a42 : SUCCESS in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ea64d10ce994889b8d978516e2d515e : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be70a9d42aee4f679006382042a496b8 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14714269928e4644990fdfc43b0606ca : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8910e3d896014f8da77f9ebf08dbfdf9 : SUCCESS in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c15bd109f1a4adfa4cfda49e57a3fec : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86587f3463d0492f84330a0ac119547a : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73c2a55063da4a81b10b469b76c1ceb9 : SUCCESS in 4m 30s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/e42e75ce946344b78484b0bec2fad9a4 : SUCCESS in 41m 44s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/82b51dd9ffac4d04b40a45aa6bc5f87a : SUCCESS in 55m 51s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/d1ce36b2f5b44b2d827239c5dc480969 : SUCCESS in 1h 39m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/b6f449bad501416fafa890f4cc4bca90 : SUCCESS in 1h 22m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/e198833a53b246c4bc8633cfbd68d2a7 : SUCCESS in 1h 50m 49s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b9275c77c844418cbd8af67336ba975e : SUCCESS in 41m 25s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/76660bb3525a45839330c4f0e46d1d9c : SUCCESS in 56m 03s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b4d6466d46164aaaa30aa7f9c43a45a9 : SUCCESS in 48m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f78fe67db05140c4aa1b6fb48557fb1c : SUCCESS in 1h 32m 26s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/a0c637549ee041d3b7c559ee52a5475f : SUCCESS in 43m 56s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a5e3cbd08d3d41518fd6e37de42c3738 : SUCCESS in 55m 40s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/bf914bf51ed447998aaa7db3296457fa : SUCCESS in 1h 13m 26s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/7076ca80254544e48a8922118f94d430 : SUCCESS in 49m 09s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/77c84f668dfd49da884cac4a41a8283b : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":5},{"id":"29e96f67f10ba9e17a35997ad90529d1346a79a7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 16:23:27.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"268639d57e28e45b1e9fd6542bd9c5793ce8d0a1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 03:13:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"06df599096ce2ff60bd2ca84c4b830d9e65c7eaa","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-09-08 06:00:19.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fa092ce1c70f22a2be6c5954216aebe3278b8b6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 06:00:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6fcea0f818e4d91ae80df427bf8dcce06770ecb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 09:03:49.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5996bafc6f9247d4a41532f9716bb60c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/89365d3d355442a8969e06fe3f7195ce : SUCCESS in 4m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c912c684b87d4765b7027dbe3ea5093e : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6da3173a56a94a168f985d70e76dea91 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ba7150ab18041fcb3d92c387851eb37 : SUCCESS in 5m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad02c8526ef94fecb4cd45f959a23e32 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ea73c31bde44b31a733ad4c7a033113 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05e7100698c34e2b9fbe40ad01bdaed5 : SUCCESS in 4m 32s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/a5237030eee643b88c56d291009e138b : TIMED_OUT in 3h 02m 42s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/fac9ae3e82204d58beaaa1813ae55f29 : FAILURE in 2h 01m 50s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/66dcd7367e544389bac92ac0a37b1364 : SUCCESS in 2h 02m 06s","accounts_in_message":[],"_revision_number":5},{"id":"c8bf2c46e2c4792c1d3e8cf35997674afbad902e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-08 10:24:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bed414bc2ee05e29eae369efc28db857372c2aa8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 11:27:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e28b614de9d14e648c4b91562f116ccb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/118aded93ed2431ca1626e423174088c : SUCCESS in 11m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8212b20a8cb44375a2300e4f4b7569db : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e4a8bf2ea02b48f49d4c6e04e6bcd59b : SUCCESS in 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7afce4f04e35233fcd65a6a222a7161657b77f7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 12:11:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f3b8e4c65824a9cbff350042e33650c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/69aa804611794e07b25a2406654069e5 : SUCCESS in 4m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2901a0aa22d6494f9c5b91e45e2df1d2 : SUCCESS in 3m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12dedc3fa2484b4897e73be2576735a9 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45d3b09970dd4b92abd696fb38ca11b1 : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/718580927bfd425f83d45cf750906950 : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11b09e96334d4629b128d9e4b5697434 : SUCCESS in 5m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5740b2441d9847d7afe733819ad41330 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93a281977083464b86f655f43c46d492 : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f242385b97d4e788e9e3614b7a260e6 : SUCCESS in 5m 08s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/04f3c3b48995480c8488353c5a90cee8 : SUCCESS in 48m 57s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/730d8c0dae7f4c75ac2c3becca775805 : SUCCESS in 1h 36m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/508a4c4589b5430ab534a1281c5b202f : SUCCESS in 1h 14m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/3c62118cf8f94c38856c68c7aea484da : SUCCESS in 1h 45m 46s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/968aabd6fada4e479c355e9b34a999d2 : SUCCESS in 46m 06s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/0f56d6c18f8547af8d699b9db2f4a223 : SUCCESS in 1h 24m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/141b357a13384a5fa3ea04da427733b0 : SUCCESS in 1h 12m 05s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/553673ca541e48b0944dfccd8b78be38 : SUCCESS in 1h 26m 33s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/8c772117a2914d36956712bf5ab99796 : SUCCESS in 55m 16s (non-voting)\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/db48983f30f641f880af3454492efed3 : SUCCESS in 1h 26m 50s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/1221862c3c1348e983b067cbeb440c7b : SUCCESS in 1h 10m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/28d383f09ec24d71be6c4d8d09f1433f : SUCCESS in 1h 26m 50s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4db4d0f94a724a6195fdfae49a373e89 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":5},{"id":"c481904616679b67d40d9835bc970c6dfd7d2802","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 12:11:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7e5feb9d4b27845d9991c6d541c9a0234c03fc99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 13:03:27.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/350db1ad283b4e77a4feaf684a5d9bc1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/794cab07d09e46ceaa49c3e6aa72fd64 : SUCCESS in 3m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6dc941ef265e4b19b853ea2f9dc5d6f6 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d7ae164e3fa4d9aabd3e1478dcc171c : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d977566bf8aa41faa8ed3b1d42f9a91c : SUCCESS in 5m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/816eaeab161e4fd4ab791876407ecf60 : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c6f45dda39c420fbf8d20a7dead64cb : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28ae5d86fc4c43eda4eca89d4b98dce4 : SUCCESS in 6m 20s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/67461d15dbed40beac797a009ae7c7ab : SUCCESS in 51m 33s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/2e8188f9cf824a3a9345f8904c1794ef : SUCCESS in 44m 55s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/8580418eed034e7bb09c9de98228d877 : SUCCESS in 47m 47s","accounts_in_message":[],"_revision_number":5},{"id":"b9841db2bdd0db01dc560f54c3cb216757cf829a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 13:03:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"c264d19e537e339295fd45e101f4913127732fc7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 13:05:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffd2699e587d4a2bbe7cc2bd80ae30c4\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/ef8dcb1026b744b28977aa06b708efbe : SUCCESS in 1m 00s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddfa8fd5b10a416983d6e36c890b37bf : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/114c08b93c24406da06f03b229b7ca3f : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4beb14657745bac15c745e955936e6dbd8f42e63","revisions":{"5b483637acf02ae2bb4b84fbe57d6e305a825b63":{"kind":"REWORK","_number":1,"created":"2024-11-24 15:51:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/936106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/06/936106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/936106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/936106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/936106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/06/936106/1"}}},"commit":{"parents":[{"commit":"44c96d9acb4eb4e52716f0eebca44cae8208382a","subject":"Merge \"Use venv to install diskimage-builder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c96d9acb4eb4e52716f0eebca44cae8208382a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 15:50:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 15:50:49.000000000","tz":540},"subject":"Fix pylint execution","message":"Fix pylint execution\n\nBump the versions so that pylint can run. Note that there are some\nfailures detected now, but these are not fixed by this change.\n\nChange-Id: I6fe614e19b56dd20d503910afb48f349b103f8f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5b483637acf02ae2bb4b84fbe57d6e305a825b63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5b483637acf02ae2bb4b84fbe57d6e305a825b63"}]},"branch":"refs/heads/master"},"5304449292517efd52a58ab7becac856c8438e21":{"kind":"REWORK","_number":2,"created":"2025-04-26 17:13:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/06/936106/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/06/936106/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/936106/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/936106/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/936106/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/06/936106/2"}}},"commit":{"parents":[{"commit":"44c96d9acb4eb4e52716f0eebca44cae8208382a","subject":"Merge \"Use venv to install diskimage-builder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/44c96d9acb4eb4e52716f0eebca44cae8208382a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 15:50:49.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-04-26 17:13:34.000000000","tz":0},"subject":"Fix pylint execution","message":"Fix pylint execution\n\nBump the versions so that pylint can run. Note that there are some\nfailures detected now, but these are not fixed by this change.\n\nChange-Id: I6fe614e19b56dd20d503910afb48f349b103f8f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5304449292517efd52a58ab7becac856c8438e21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5304449292517efd52a58ab7becac856c8438e21"}]},"branch":"refs/heads/master"},"172c4a9261b2e7ee70c6f28e783006361780703e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-04-26 17:13:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/06/936106/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/06/936106/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/936106/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/936106/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/936106/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/06/936106/3"}}},"commit":{"parents":[{"commit":"7ddb01cc86e4daee623e81e05ec0034c51bb81f5","subject":"Merge \"Fix trove install failed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ddb01cc86e4daee623e81e05ec0034c51bb81f5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 15:50:49.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-04-26 17:13:41.000000000","tz":0},"subject":"Fix pylint execution","message":"Fix pylint execution\n\nBump the versions so that pylint can run. Note that there are some\nfailures detected now, but these are not fixed by this change.\n\nChange-Id: I6fe614e19b56dd20d503910afb48f349b103f8f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/172c4a9261b2e7ee70c6f28e783006361780703e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/172c4a9261b2e7ee70c6f28e783006361780703e"}]},"branch":"refs/heads/master","description":"Rebase"},"a40a1f0f5b94fbbcc35e0f7d7726ce4fb88a15c0":{"kind":"REWORK","_number":4,"created":"2025-04-26 17:17:49.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/06/936106/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/06/936106/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/936106/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/936106/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/936106/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/06/936106/4"}}},"commit":{"parents":[{"commit":"7ddb01cc86e4daee623e81e05ec0034c51bb81f5","subject":"Merge \"Fix trove install failed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ddb01cc86e4daee623e81e05ec0034c51bb81f5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 15:50:49.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-04-26 17:17:49.000000000","tz":0},"subject":"Fix pylint execution","message":"Fix pylint execution\n\nBump the versions so that pylint can run. Note that there are some\nfailures detected now, but these are not fixed by this change.\n\nChange-Id: I6fe614e19b56dd20d503910afb48f349b103f8f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a40a1f0f5b94fbbcc35e0f7d7726ce4fb88a15c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a40a1f0f5b94fbbcc35e0f7d7726ce4fb88a15c0"}]},"branch":"refs/heads/master"},"4beb14657745bac15c745e955936e6dbd8f42e63":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-04-26 19:22:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/06/936106/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/06/936106/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/936106/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/936106/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/936106/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/06/936106/5"}}},"commit":{"parents":[{"commit":"7ddb01cc86e4daee623e81e05ec0034c51bb81f5","subject":"Merge \"Fix trove install failed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ddb01cc86e4daee623e81e05ec0034c51bb81f5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 15:50:49.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-04-26 19:22:11.000000000","tz":0},"subject":"Remove pylint","message":"Remove pylint\n\npylint is no longer in use and can be removed.\n\nChange-Id: I6fe614e19b56dd20d503910afb48f349b103f8f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4beb14657745bac15c745e955936e6dbd8f42e63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4beb14657745bac15c745e955936e6dbd8f42e63"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
