)]}'
{"id":"openstack%2Ftrove~937230","triplet_id":"openstack%2Ftrove~master~I7baf7391d2bc7f0f17f525792eff5c0c82e41b42","project":"openstack/trove","branch":"master","topic":"migrate-to-quay","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-12-10 05:49:04.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I7baf7391d2bc7f0f17f525792eff5c0c82e41b42","subject":"CI: push images to quay.io","status":"MERGED","created":"2024-12-06 08:11:55.000000000","updated":"2024-12-10 06:40:04.000000000","submitted":"2024-12-10 06:37:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937230-migrate-to-quay","meta_rev_id":"7a235d5b7ab7a9af66ecab21ae9c6e7e9533197e","_number":937230,"virtual_id_number":937230,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 06:37:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-10 05:49:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-10 05:49:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-06 08:12:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12ab5392a743ef7cbae7516cc5467dc33cfd48ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-06 08:11:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2ed464d63e0cb306fb29122fb0003a3acf59d77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 08:12:26.000000000","message":"Patch Set 1: Verified-1\n\nUnable to freeze job graph: Pre-review pipeline check does not allow post-review job sync-and-buid-images-to-quay","accounts_in_message":[],"_revision_number":1},{"id":"e55865c02273a3809b7e43aa3b3e5c4c1d7198d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-06 08:19:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7db52b9f8d4a086194a5ace8aa3c33a36fccf7e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 08:20:11.000000000","message":"Patch Set 2: Verified-1\n\nUnable to freeze job graph: Pre-review pipeline check does not allow post-review job sync-and-buid-images-to-quay","accounts_in_message":[],"_revision_number":2},{"id":"6448ff65f4f2b58482d8a815571705dc230410fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-06 08:37:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2adceea87a8003021d361e3aa30f23d396e20952","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-07 01:53:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ca109443c89e9d4d15dc389e522395ba3bb12452","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-09 06:47:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a4e2f4095bf02568d7ffb5db0e63fc6d775e8c87","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-09 06:50:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d2c391f7658b098442f9e369bd79b427cc694d6d","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 06:51:20.000000000","message":"Patch Set 4:\n\nUnable to freeze job graph: Pre-review pipeline experimental does not allow post-review job public-trove-images-quay","accounts_in_message":[],"_revision_number":4},{"id":"fda8b10ed10d5010feca9d9ef4a712e9543336fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-09 07:38:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cd1180ece5112235a7cfe92aa709c9caad6f6411","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 07:38:22.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch master.  The\nproblem was:\n\n  expected a list for dictionary value @ data[\u0027experimental\u0027][\u0027jobs\u0027]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      queue: trove\n      # templates:\n      #   - check-requirements\n      #   - openstack-cover-jobs\n      #   - openstack-python3-jobs\n      #   - openstack-python3-jobs-arm64\n      #   - periodic-stable-jobs\n      #   - publish-openstack-docs-pti\n  ...\n\n  in \"openstack/trove/zuul.d/projects.yaml@master\", line 2","accounts_in_message":[],"_revision_number":5},{"id":"7f32c7d425501ebb684425a43a0326a45a26138a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 07:38:23.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch master.  The\nproblem was:\n\n  expected a list for dictionary value @ data[\u0027experimental\u0027][\u0027jobs\u0027]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      queue: trove\n      # templates:\n      #   - check-requirements\n      #   - openstack-cover-jobs\n      #   - openstack-python3-jobs\n      #   - openstack-python3-jobs-arm64\n      #   - periodic-stable-jobs\n      #   - publish-openstack-docs-pti\n  ...\n\n  in \"openstack/trove/zuul.d/projects.yaml@master\", line 2","accounts_in_message":[],"_revision_number":5},{"id":"7b178118d50c2f7a156872e6e624cb8103d8b4ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-09 07:48:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0ca0901354b20a33ac912c106a6e4f0f11a81f56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 07:55:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18697c26cfcf43ceb3c93f69009995d6\n\n- test-trove-images-quay https://zuul.opendev.org/t/openstack/build/5dd8045cd3b24408983a8d70ead35037 : FAILURE in 1m 36s","accounts_in_message":[],"_revision_number":6},{"id":"d6898f61c9cc132528e97e7f501c3b8d015dd05e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-09 08:18:39.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"7f86ace8d8056a4ff0e1b5287701bd2afc6b517a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 08:22:00.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e66952c26403490a91bc3fcffd4843a4\n\n- test-trove-images-quay https://zuul.opendev.org/t/openstack/build/61cb28e6f729435784ea768f298c1e67 : FAILURE in 1m 33s","accounts_in_message":[],"_revision_number":7},{"id":"88c4e0668a10f2c4fb2ef3d74167e88a7653f349","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-09 08:37:20.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"6b3866f30daf0a4fb5a23b5caa55be5aa489012d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 08:39:06.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c63ad7e21bdc4651b904a8daec332ae6\n\n- test-trove-images-quay https://zuul.opendev.org/t/openstack/build/520b985a1b1a4666b9eb4da546f249b4 : FAILURE in 1m 14s","accounts_in_message":[],"_revision_number":8},{"id":"3ae2c17e635dbb0fc7beb06bbc356451b03c09d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-09 08:49:55.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"b78dcec68934e458e3ca268cf8911be6d6d7285a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 08:53:13.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aac1b7f6b231447ca4adb4c012c136e0\n\n- test-trove-images-quay https://zuul.opendev.org/t/openstack/build/934322eb83ea465b993b13150ed15193 : FAILURE in 1m 35s","accounts_in_message":[],"_revision_number":9},{"id":"3b5afc18e866039ee594e8af9a5ad83916cf5ef7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-09 09:06:55.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"2c0185cf88dcf9f3b4463cf607cca9ea21c649ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 09:14:13.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/23e386d8585846dfab55f44280994725\n\n- test-trove-images-quay https://zuul.opendev.org/t/openstack/build/5573bcbe01e04c52b8ee98da4f841f56 : FAILURE in 1m 24s","accounts_in_message":[],"_revision_number":10},{"id":"4ce6b8209515eed2235f8eb5cce6814ba553b890","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-09 09:22:51.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"608c327b12b2a452b372c0d143b589d5d0cadf2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 09:29:50.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b486dbb49da445208cfeea14ba798d01\n\n- test-trove-images-quay https://zuul.opendev.org/t/openstack/build/acbfdf323f1f4dd481ab7aa4455230fc : FAILURE in 1m 39s","accounts_in_message":[],"_revision_number":11},{"id":"63fcf594f20f1811c7d88f3c4161a7e8650c1834","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-09 09:33:01.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"8c76601d920361f9aac4f15735485c24020d70c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 09:41:23.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b613e28c589149d29a2f2672032d3522\n\n- test-trove-images-quay https://zuul.opendev.org/t/openstack/build/6bacac2ccb5747808985b0c36fbd72f0 : FAILURE in 2m 09s","accounts_in_message":[],"_revision_number":12},{"id":"d9d0e6583b60f618386c24c0a6ff1f4f075610c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-10 01:40:42.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"8c7d1e53b96893a41cc2e0264e073bfb2370cd00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 02:14:49.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb9e21ab112d42b09d837c3ba154ad24\n\n- test-trove-images-quay https://zuul.opendev.org/t/openstack/build/fa5f4c47fafb449c9fbc876404815dab : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":13},{"id":"0577e71dc22dc8aa8ccd076cb0f95dcef2aa05fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-10 02:29:38.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"6196d9398de10eefa5771ec42f0f77c904233808","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 02:44:01.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6335c463fedf4f0ebfa826271829b9a9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78711ed1530742d3ad2440aa318700ae : SUCCESS in 12m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/afd820acebe2466b8bd3d7879ec46437 : SUCCESS in 9m 45s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"30ca6d2acba0c708fc4878201bfd65a4547ff655","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 04:25:38.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3312b9d485334ada99c6294f5aa00c01\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc52ba5033d84253b8cb2e14b3fc1dcd : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2229e9b6e36b462c9081262437de1626 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06c75bf858e54d8bb832a1dea05cfe68 : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1033ff4b354a4d1387ab8cf1af49aa54 : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/415eb865c51c4ccfabdfdd125a7afb46 : SUCCESS in 5m 32s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/ff4d7399aca447c482e177761a714c71 : SUCCESS in 47m 19s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/346e1aec57c743c4b7a3dcb4580d2276 : SUCCESS in 38m 47s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/9183fcdb2c374376abf56277dab53315 : SUCCESS in 1h 15m 43s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/2702176233014ceb931078dc93728f88 : SUCCESS in 1h 31m 26s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9acb6aaeff654a0ea654d9d3bb3f168a : SUCCESS in 41m 33s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3b3075f2dfea45c2995b695e9dc73d7a : SUCCESS in 1h 31m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fd3d6a4d4dc14b91b49b5d3d32a34370 : SUCCESS in 1h 15m 05s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5ff6013794e54e6e995a3ba91630b6a2 : SUCCESS in 55m 09s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a8f46ce147534638b00c2e038c940be0 : SUCCESS in 1h 33m 07s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/ace8fadca302434f91411ad6ab8edbe8 : FAILURE in 52m 47s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ffe64e7261764cb982312b0d34e23e61 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":14},{"id":"696aadc3dd9b2ad625edbd95a089505a6dfd543b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-10 05:49:04.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"d993795eb1befe346fbdba0ab0c5015a95457553","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 05:49:40.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"2d64618892d200a7f69a0fc21909647f3c44a8bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 06:37:48.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66f1ca50a6594ad0a93b96228289509d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9213a1f51296499eac1c5ae0f579419e : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d5a2be686a24d108ac1c201ec364b29 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2833c7824020485d97fcb45a11492739 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6a46ca0431464582f4c09573035adb : SUCCESS in 3m 59s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/346103a492d74687aa9b2a32d4405317 : SUCCESS in 45m 40s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/48c6f3fa5cd740dd9aa371b60cc4f393 : SUCCESS in 42m 59s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/550e3879efdf4f2f8548a2d1d0013283 : SUCCESS in 43m 27s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/08cdcec4767f46c1addda641b3e464d4 : SUCCESS in 47m 37s","accounts_in_message":[],"_revision_number":14},{"id":"4b9813013f6227eed3ccd7fc76b9ebe471fdd7be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 06:37:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"7a235d5b7ab7a9af66ecab21ae9c6e7e9533197e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 06:40:04.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0cb482dc1974d93996c8600ad41b458\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/533e3c5aee264fec84a7f5c65dc13449 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"a74ea9794a3587f63788e92282c72f9e56b22195","revisions":{"dbcc3210220071e0528af8a97ee0241c4e93cbce":{"kind":"REWORK","_number":1,"created":"2024-12-06 08:11:55.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/30/937230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/937230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/937230/1"}}},"commit":{"parents":[{"commit":"3468240fae3dc9678f36c9955eb274af01775b73","subject":"Merge \"Fix replication detach error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3468240fae3dc9678f36c9955eb274af01775b73"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:11:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:11:31.000000000","tz":480},"subject":"CI: push images to quay.io","message":"CI: push images to quay.io\n\nChange-Id: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dbcc3210220071e0528af8a97ee0241c4e93cbce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dbcc3210220071e0528af8a97ee0241c4e93cbce"}]},"branch":"refs/heads/master"},"3f464d9b5c046417e7d9c03e3fdc71a32a3b4425":{"kind":"REWORK","_number":2,"created":"2024-12-06 08:19:48.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/30/937230/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/937230/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/937230/2"}}},"commit":{"parents":[{"commit":"3468240fae3dc9678f36c9955eb274af01775b73","subject":"Merge \"Fix replication detach error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3468240fae3dc9678f36c9955eb274af01775b73"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:11:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:19:41.000000000","tz":480},"subject":"CI: push images to quay.io","message":"CI: push images to quay.io\n\nChange-Id: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f464d9b5c046417e7d9c03e3fdc71a32a3b4425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f464d9b5c046417e7d9c03e3fdc71a32a3b4425"}]},"branch":"refs/heads/master"},"cb879562fa6c2d50cb89820059bb55b7f2c172b9":{"kind":"REWORK","_number":3,"created":"2024-12-06 08:37:08.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/30/937230/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/937230/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/937230/3"}}},"commit":{"parents":[{"commit":"3468240fae3dc9678f36c9955eb274af01775b73","subject":"Merge \"Fix replication detach error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3468240fae3dc9678f36c9955eb274af01775b73"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:11:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:37:01.000000000","tz":480},"subject":"CI: push images to quay.io","message":"CI: push images to quay.io\n\nChange-Id: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cb879562fa6c2d50cb89820059bb55b7f2c172b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cb879562fa6c2d50cb89820059bb55b7f2c172b9"}]},"branch":"refs/heads/master"},"99f46d246056f5ee2ac9ee366df17a251296e3c0":{"kind":"REWORK","_number":4,"created":"2024-12-09 06:47:26.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/30/937230/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/937230/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/937230/4"}}},"commit":{"parents":[{"commit":"3468240fae3dc9678f36c9955eb274af01775b73","subject":"Merge \"Fix replication detach error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3468240fae3dc9678f36c9955eb274af01775b73"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:11:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-09 06:23:32.000000000","tz":480},"subject":"CI: push images to quay.io","message":"CI: push images to quay.io\n\nChange-Id: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/99f46d246056f5ee2ac9ee366df17a251296e3c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/99f46d246056f5ee2ac9ee366df17a251296e3c0"}]},"branch":"refs/heads/master"},"93052c6b8c3a9cb0b86c06eed96bbd39dc5a673b":{"kind":"REWORK","_number":5,"created":"2024-12-09 07:38:01.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/30/937230/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/937230/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/937230/5"}}},"commit":{"parents":[{"commit":"3468240fae3dc9678f36c9955eb274af01775b73","subject":"Merge \"Fix replication detach error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3468240fae3dc9678f36c9955eb274af01775b73"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:11:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-09 07:37:31.000000000","tz":480},"subject":"CI: push images to quay.io","message":"CI: push images to quay.io\n\nChange-Id: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/93052c6b8c3a9cb0b86c06eed96bbd39dc5a673b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/93052c6b8c3a9cb0b86c06eed96bbd39dc5a673b"}]},"branch":"refs/heads/master"},"aa973a4f303ad26d9553f812ceefb215b2446d8f":{"kind":"REWORK","_number":6,"created":"2024-12-09 07:48:20.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/30/937230/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/937230/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/937230/6"}}},"commit":{"parents":[{"commit":"3468240fae3dc9678f36c9955eb274af01775b73","subject":"Merge \"Fix replication detach error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3468240fae3dc9678f36c9955eb274af01775b73"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:11:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-09 07:41:03.000000000","tz":480},"subject":"CI: push images to quay.io","message":"CI: push images to quay.io\n\nChange-Id: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa973a4f303ad26d9553f812ceefb215b2446d8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa973a4f303ad26d9553f812ceefb215b2446d8f"}]},"branch":"refs/heads/master"},"f2fa7ec6e024678474c1ada322ab3bef38b14b78":{"kind":"REWORK","_number":7,"created":"2024-12-09 08:18:39.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/30/937230/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/937230/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/937230/7"}}},"commit":{"parents":[{"commit":"3468240fae3dc9678f36c9955eb274af01775b73","subject":"Merge \"Fix replication detach error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3468240fae3dc9678f36c9955eb274af01775b73"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:11:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-09 08:01:16.000000000","tz":480},"subject":"CI: push images to quay.io","message":"CI: push images to quay.io\n\nChange-Id: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f2fa7ec6e024678474c1ada322ab3bef38b14b78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f2fa7ec6e024678474c1ada322ab3bef38b14b78"}]},"branch":"refs/heads/master"},"4d2a054d54dd9594005f6aee7f1a523edbbb5115":{"kind":"REWORK","_number":8,"created":"2024-12-09 08:37:20.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/30/937230/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/937230/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/937230/8"}}},"commit":{"parents":[{"commit":"3468240fae3dc9678f36c9955eb274af01775b73","subject":"Merge \"Fix replication detach error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3468240fae3dc9678f36c9955eb274af01775b73"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:11:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-09 08:37:13.000000000","tz":480},"subject":"CI: push images to quay.io","message":"CI: push images to quay.io\n\nChange-Id: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4d2a054d54dd9594005f6aee7f1a523edbbb5115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4d2a054d54dd9594005f6aee7f1a523edbbb5115"}]},"branch":"refs/heads/master"},"cdd120d033f459882c9d51e19a3f58b5262df0d4":{"kind":"REWORK","_number":9,"created":"2024-12-09 08:49:55.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/30/937230/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/937230/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/937230/9"}}},"commit":{"parents":[{"commit":"3468240fae3dc9678f36c9955eb274af01775b73","subject":"Merge \"Fix replication detach error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3468240fae3dc9678f36c9955eb274af01775b73"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:11:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-09 08:49:48.000000000","tz":480},"subject":"CI: push images to quay.io","message":"CI: push images to quay.io\n\nChange-Id: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cdd120d033f459882c9d51e19a3f58b5262df0d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cdd120d033f459882c9d51e19a3f58b5262df0d4"}]},"branch":"refs/heads/master"},"2822ab4e91ac511ec48c3a06f3f2b920be972fff":{"kind":"REWORK","_number":10,"created":"2024-12-09 09:06:55.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/30/937230/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/937230/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/937230/10"}}},"commit":{"parents":[{"commit":"3468240fae3dc9678f36c9955eb274af01775b73","subject":"Merge \"Fix replication detach error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3468240fae3dc9678f36c9955eb274af01775b73"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:11:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-09 09:06:43.000000000","tz":480},"subject":"CI: push images to quay.io","message":"CI: push images to quay.io\n\nChange-Id: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2822ab4e91ac511ec48c3a06f3f2b920be972fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2822ab4e91ac511ec48c3a06f3f2b920be972fff"}]},"branch":"refs/heads/master"},"735212b58468d9df539a5df7ec9db313ee0cc767":{"kind":"REWORK","_number":11,"created":"2024-12-09 09:22:51.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/30/937230/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/937230/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/937230/11"}}},"commit":{"parents":[{"commit":"3468240fae3dc9678f36c9955eb274af01775b73","subject":"Merge \"Fix replication detach error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3468240fae3dc9678f36c9955eb274af01775b73"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:11:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-09 09:22:34.000000000","tz":480},"subject":"CI: push images to quay.io","message":"CI: push images to quay.io\n\nChange-Id: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/735212b58468d9df539a5df7ec9db313ee0cc767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/735212b58468d9df539a5df7ec9db313ee0cc767"}]},"branch":"refs/heads/master"},"540f84c85b5fd627e6bfe799833da57aaa338386":{"kind":"REWORK","_number":12,"created":"2024-12-09 09:33:01.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/30/937230/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/937230/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/937230/12"}}},"commit":{"parents":[{"commit":"3468240fae3dc9678f36c9955eb274af01775b73","subject":"Merge \"Fix replication detach error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3468240fae3dc9678f36c9955eb274af01775b73"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:11:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-09 09:32:53.000000000","tz":480},"subject":"CI: push images to quay.io","message":"CI: push images to quay.io\n\nChange-Id: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/540f84c85b5fd627e6bfe799833da57aaa338386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/540f84c85b5fd627e6bfe799833da57aaa338386"}]},"branch":"refs/heads/master"},"d02f2f9708319dc8247085b57b46fb27515e412f":{"kind":"REWORK","_number":13,"created":"2024-12-10 01:40:42.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/30/937230/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/937230/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/937230/13"}}},"commit":{"parents":[{"commit":"3468240fae3dc9678f36c9955eb274af01775b73","subject":"Merge \"Fix replication detach error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3468240fae3dc9678f36c9955eb274af01775b73"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:11:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-10 01:40:30.000000000","tz":480},"subject":"CI: push images to quay.io","message":"CI: push images to quay.io\n\nChange-Id: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d02f2f9708319dc8247085b57b46fb27515e412f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d02f2f9708319dc8247085b57b46fb27515e412f"}]},"branch":"refs/heads/master"},"a74ea9794a3587f63788e92282c72f9e56b22195":{"kind":"REWORK","_number":14,"created":"2024-12-10 02:29:38.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/30/937230/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/937230/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/937230/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/937230/14"}}},"commit":{"parents":[{"commit":"3468240fae3dc9678f36c9955eb274af01775b73","subject":"Merge \"Fix replication detach error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3468240fae3dc9678f36c9955eb274af01775b73"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-06 08:11:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-10 02:29:04.000000000","tz":480},"subject":"CI: push images to quay.io","message":"CI: push images to quay.io\n\nChange-Id: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a74ea9794a3587f63788e92282c72f9e56b22195"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a74ea9794a3587f63788e92282c72f9e56b22195"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
