)]}'
{"id":"openstack%2Ftrove~937327","triplet_id":"openstack%2Ftrove~stable%2F2024.2~I76166f901f961a18ad4297d04e1a75a59b059e23","project":"openstack/trove","branch":"stable/2024.2","hashtags":[],"change_id":"I76166f901f961a18ad4297d04e1a75a59b059e23","subject":"Fix replication detach error","status":"MERGED","created":"2024-12-09 07:53:13.000000000","updated":"2024-12-20 13:37:04.000000000","submitted":"2024-12-20 13:35:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937327","cherry_pick_of_change":936218,"cherry_pick_of_patch_set":1,"meta_rev_id":"113aba44773b793628265b1a38bd90341fe08367","_number":937327,"virtual_id_number":937327,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-20 13:35:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-20 12:51:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-20 12:51:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-09 08:25:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-09 09:31:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"047333c6154e5c1f285bb7fc00ce571952e09417","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-09 07:53:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"460655196a475a6e2d60c4a6cd5e23ff07bbe172","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 08:25:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd58e6e9f3214e7f8dac34a4298ce025\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52b5d74391ce4c6696b03869686b7ce2 : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6304352ff708471b9442751d2e277890 : SUCCESS in 11m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"de9f124f05e1d9980ca0afd670235a4b7db52a28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 09:31:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b3bf04b3dd14e00933390fd1e66b68b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/654d6e5749cf4878939ae8445636f307 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e3c4888b2a46b4b04afd32ab242516 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0edce660dc0c495388c9e48c9e9d6cab : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/273e45fc13214246bd30fd07d273f269 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1df192c057314e36b88c0c3a29a6d171 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/139c4a7f5f4e4b68926adf124c653d19 : SUCCESS in 3m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3f98355b81c4b999f263b103c0ea357 : SUCCESS in 4m 46s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/f7227ef4509d4cc1929a86e691cdf9c4 : SUCCESS in 42m 46s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/5c7cc95d78b8434c8628444be8b2de40 : SUCCESS in 1h 35m 29s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/7adf9825c59f45b29c3ae373ec202aaa : SUCCESS in 1h 20m 01s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fd83cb5a47bd4a1caf7fc988efcaa786 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6250d18eb6605f4e5fb96a6039a68f6a8a462655","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-20 12:51:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e1d7763d97408f99feaf0cffd34b825fc6f9c9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 12:51:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1567eb1e5627c2b4dd8aca4076c5f2a7b3371546","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 13:35:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6afd633168bd4d38a9bfe633d48e3ea6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39cfbdb4e7c448a68593eb49da9b07e8 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cc00bff06b54c59a4ed6f25148d6c3f : SUCCESS in 6m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5a619b5e1aa4bd3a523ba32923d0f08 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b506249ec041d1b309b83b30c9c308 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5573c04042bf4f0fa984f4516f8f2bb0 : SUCCESS in 6m 37s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/f0961697bd444186b44fa1b604a7b28a : SUCCESS in 43m 33s","accounts_in_message":[],"_revision_number":1},{"id":"952c9eba41e87f03ad45334c4af3aac467ade332","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 13:35:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"113aba44773b793628265b1a38bd90341fe08367","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 13:37:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da1c85d7094a470c9dd4e96c0779ef62\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe70e3f5b3064bbbb1d79d6e9e83d6f2 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47786d1eabb54bde958cd62002e4a93a : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95e879d42544cec28322a21bd10fa0537f88e5de","revisions":{"95e879d42544cec28322a21bd10fa0537f88e5de":{"kind":"REWORK","_number":1,"created":"2024-12-09 07:53:13.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/27/937327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/937327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/937327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/937327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/937327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/937327/1"}}},"commit":{"parents":[{"commit":"4c191c30322a6fac2acfe6743daf151630bf8480","subject":"Fix MariaDB target dir for backup syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4c191c30322a6fac2acfe6743daf151630bf8480"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-11-26 06:42:09.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-09 07:53:13.000000000","tz":0},"subject":"Fix replication detach error","message":"Fix replication detach error\n\nTypeError: tuple indices must be integers or slices, not str\n\nthis is caused by the new version of the sqlachemy library.\n\nChange-Id: I76166f901f961a18ad4297d04e1a75a59b059e23\n(cherry picked from commit f0f186ef9a278c919cfba331a05f2f1b3fa6a572)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/95e879d42544cec28322a21bd10fa0537f88e5de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/95e879d42544cec28322a21bd10fa0537f88e5de"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
