)]}'
{"id":"openstack%2Ftrove~937841","triplet_id":"openstack%2Ftrove~master~I4fb7afd9f6a5d84578bb6226f8b2ff4b9908a526","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-12-18 06:18:43.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I4fb7afd9f6a5d84578bb6226f8b2ff4b9908a526","subject":"Using quay.io registry in project","status":"MERGED","created":"2024-12-17 06:38:59.000000000","updated":"2024-12-18 07:46:07.000000000","submitted":"2024-12-18 07:44:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937841","meta_rev_id":"a7d098999de38542e56aa40df36947f72068983b","_number":937841,"virtual_id_number":937841,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-18 07:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-18 06:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-18 06:18:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-17 08:00:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"448ed7a4d3b263644fd4f513fa2397e13bca5f28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-17 06:38:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fe34ce181fc5e8a37cd0f4ea609e7e61671d99a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 08:00:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f79690c6b2c4f5ea0c3616a5e10078d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7699ed0be3d142d2bdea986f398b91b5 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab6c4bbfd9084e429cd6e77eb83f2760 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2197cfed197641f08ad4205d6b583cf6 : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/18ffa74251164578bac762672be153cc : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c33a19a8d133480d9cc22a6f5332c8a5 : SUCCESS in 4m 43s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c2938f65c7d1429aa37df0060853c4b8 : SUCCESS in 54m 33s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/be929082294f47bda120f68acd6fa5da : SUCCESS in 40m 52s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/6e6a9b84f36749308a09ed8e4417e38d : FAILURE in 54m 54s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/6eb90146e88b4242a5de0ac0a78366d2 : FAILURE in 1h 03m 08s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b6afebfea51d47c08220dba4ce671507 : SUCCESS in 31m 52s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/88976ecc1ae5459483b0b85b7a54e5dc : FAILURE in 1h 03m 54s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c634c03507b84d828aafd118742beace : FAILURE in 1h 11m 43s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/3a1bfcb645d447b4ae185fc21b38bac9 : FAILURE in 32m 19s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/0de47677d60548f88beb987976d6225b : FAILURE in 1h 11m 51s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/d22f38e1ab874f9d9056fa9be2bbe68d : FAILURE in 37m 30s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c8f0cb32a5544c4fb7a8f6b932115703 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"dde5a631194a3ab3d57658b3618544cd96cf1aa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-17 08:26:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8a17af233e08a3f02e9dd5313357f957d553753d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 08:59:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16a79f38b360418f88fa11972414e241\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74aa0a56c6664770bc2fa6e51b1d3814 : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/86925873105e420d89e6681530ef6a63 : SUCCESS in 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"efa3e620eac691202a22b44bf7f594caea5bebce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 10:49:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/838f2dcac20944c8a759ed5a91c4cdea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b7bd06da673468bb9dce906aa30d99d : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0c7b4fc34d74cdfaec460894cc94d16 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/123d7cf6a8e4464d89943bb5146d8896 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8ba4955f8294d7a8d095e0b9deb013b : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25715ef927e842388d7a14c64ec6df0e : SUCCESS in 5m 35s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/d622af700ed24a4d8f63b051b441f924 : SUCCESS in 32m 15s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/c6ca8945d7dd440e9989706767f6720a : SUCCESS in 48m 42s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/c310a2a8ae234a86a62c61468abd7187 : SUCCESS in 1h 37m 53s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/76673374302842228ffaede71c5f1ceb : SUCCESS in 2h 01m 30s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/386963c0afa645e689118e85bb74c38a : SUCCESS in 42m 09s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4f241e81a8974c70bb85e87c0dcd5182 : SUCCESS in 2h 05m 55s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/154f10dfb79c4da690535f8e0442d33c : SUCCESS in 1h 46m 06s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/911731bc0a5f40b18f936c75d80a9c7b : SUCCESS in 55m 34s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/ff319df0e4a346478476d824ed6d7793 : SUCCESS in 2h 00m 55s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/f6661f6973ec45a0a9e1593d21c8f1a4 : SUCCESS in 1h 49m 40s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6c9e8b842c824694b4ee695f16db023a : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":2},{"id":"6f9db8982e7909eb5c9229b0274ac9a0b19da071","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-18 06:18:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"35d7d676b9495ddc32f5eee53336fe5d734a3023","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 06:19:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"12d82d6a3d5ddba87ac964ca475072b96ca88ada","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 07:44:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b66be5b7aa75400f8e07f954758e6e85\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e8f8971890c4724a2f6d6d84d1453e9 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad0196430d054e69915ddec9ee022bea : SUCCESS in 7m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd005126fc80470d9d69477a705e5ef7 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1547586e67d94d5982ad0a29c8ea7ed0 : SUCCESS in 5m 17s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/fbd86391316e4536a1fa2b611b857dec : SUCCESS in 1h 04m 59s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/aaef10221140465497c47c798a8dd154 : SUCCESS in 45m 02s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9c3eebb84d134c85a55b710ad52a2fd8 : SUCCESS in 33m 02s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/d380800128cc4bf2aac71beacf0917f9 : SUCCESS in 53m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b1530826a169aadb5762394b52774d492eb6c4bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 07:44:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a7d098999de38542e56aa40df36947f72068983b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 07:46:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed8f5c2379664351a0794235d410862c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bf0773011c34885a13c2733b26b17ae : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0138ce30e256a9f3a8fc3405a16051aa96294d57","revisions":{"83521a257f0d4ab163ac36e1553e66bdb59ae41f":{"kind":"REWORK","_number":1,"created":"2024-12-17 06:38:59.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/41/937841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/41/937841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/937841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/937841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/937841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/41/937841/1"}}},"commit":{"parents":[{"commit":"ff502ae696579ebe14a223fb309a5acdc963ebd7","subject":"[CI] Follow up: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ff502ae696579ebe14a223fb309a5acdc963ebd7"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-17 06:35:09.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-17 06:35:09.000000000","tz":480},"subject":"Using quay.io registry in project","message":"Using quay.io registry in project\n\nIn order to avoid docker pull rate limit, this commit\nreplaces images in the dockerhub to quay.io/openstack.trove\n\nChange-Id: I4fb7afd9f6a5d84578bb6226f8b2ff4b9908a526\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83521a257f0d4ab163ac36e1553e66bdb59ae41f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83521a257f0d4ab163ac36e1553e66bdb59ae41f"}]},"branch":"refs/heads/master"},"0138ce30e256a9f3a8fc3405a16051aa96294d57":{"kind":"REWORK","_number":2,"created":"2024-12-17 08:26:11.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/41/937841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/41/937841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/937841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/937841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/41/937841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/41/937841/2"}}},"commit":{"parents":[{"commit":"ff502ae696579ebe14a223fb309a5acdc963ebd7","subject":"[CI] Follow up: I7baf7391d2bc7f0f17f525792eff5c0c82e41b42","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ff502ae696579ebe14a223fb309a5acdc963ebd7"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-17 06:35:09.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-17 08:25:29.000000000","tz":480},"subject":"Using quay.io registry in project","message":"Using quay.io registry in project\n\nIn order to avoid docker pull rate limit, this commit\nreplaces images in the dockerhub to quay.io/openstack.trove\n\nChange-Id: I4fb7afd9f6a5d84578bb6226f8b2ff4b9908a526\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0138ce30e256a9f3a8fc3405a16051aa96294d57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0138ce30e256a9f3a8fc3405a16051aa96294d57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
