)]}'
{"id":"openstack%2Ftrove~938130","triplet_id":"openstack%2Ftrove~master~I74501c5d488108d9db44620025da5138be3da989","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I74501c5d488108d9db44620025da5138be3da989","subject":"Fix: trovestack failed to build images","status":"MERGED","created":"2024-12-20 12:46:41.000000000","updated":"2024-12-21 03:55:16.000000000","submitted":"2024-12-21 03:54:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938130","meta_rev_id":"7456d1e7a8f780f6b7d304986f35d52280ab4fac","_number":938130,"virtual_id_number":938130,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-21 03:54:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-21 03:00:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-21 03:00:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-20 13:03:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-20 14:39:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c2d9ad6ad10a006299d587a9668479f5d403ef0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-20 12:46:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad00522f35baddb66164724ef61e9bc1c82de6d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 13:03:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bb5673595d4420db471731985b18453\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ccaea45af3b9477bb7aa1b51d746fe76 : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/29b3a41d11c04e1f85d64f768d5a8486 : SUCCESS in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f1a0fdfbee7b8cbaf0ec3c2ee6ba24ca022bddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 14:39:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8652db2386c24557bd4307a7298be96a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad62472435414d3e9dcca44e2841472b : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01ce54431f7a4d819e0a11c6e84c78ea : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/498eae8d18d44196968da77bcc0c23d9 : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b53c5e89aa05430fb0f7d32f3b76c255 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6468416f87c74b3ba89523df994d9fde : SUCCESS in 3m 07s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/f348dbf39f4d48cca599d4fd9dfd9dcf : SUCCESS in 46m 44s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/d704dba7aace4f0f8bd6113068d2a463 : SUCCESS in 44m 30s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/4afa5fd3e13b47609c89d4fad7cb8b27 : SUCCESS in 1h 43m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/edec7193d32840db8267971ffd1b6253 : SUCCESS in 1h 19m 14s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/cf425011047c4121b7762eb3267eb452 : SUCCESS in 58m 59s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a0f22c36286340c6bffc687b7ba21c55 : SUCCESS in 1h 44m 09s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0f41a72977e94442b7e75014d6070dac : SUCCESS in 1h 20m 47s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/2523cfd51fb043eb9d1ab0bc42d46a6d : SUCCESS in 42m 12s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/5f75412273f74b518afbd08530664d60 : SUCCESS in 1h 04m 55s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/a332256054e64e679fdea0772840b7fa : SUCCESS in 1h 50m 44s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2d2d4307bad437b8bf8782292c88ecb : SUCCESS in 5m 00s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/107f3c44ec8a413c860551db89f3c575 : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9f61f833ae1d7acc81e67645c77a96e71f481976","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-21 03:00:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2cc0f693f1ff7b56acfe89e22c5ba2aff586ea35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-21 03:01:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"281cf558f65acf7e4760b04311277b9d7d6d3c6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-21 03:54:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/313d488ef84d4f4c9a46151d3d24c99d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd301856ba5490486c201156df3fb0a : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d9ecf39311f4e28bf29810808340391 : SUCCESS in 6m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb5027573b2a423fb78faa69b2ef6fc8 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeb27e7c23094eb688495c8cd9e751e6 : SUCCESS in 7m 45s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/6db2bb2b44204ce1aebc63211bb2bcff : SUCCESS in 51m 16s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/7f443501965b40c5a6206fec2f8b7588 : SUCCESS in 47m 49s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/79a5d1b3d09c46959085d9252c78bb0d : SUCCESS in 46m 23s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/870f2809dbd546a7af53098c889d4620 : SUCCESS in 48m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b3a45801849cd10b7f5af0a5e1b7c57e8b7b8325","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-21 03:54:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7456d1e7a8f780f6b7d304986f35d52280ab4fac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-21 03:55:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4c01e7598094b988331fe3ae30a0d1f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5672768f9954ab3938fb728c1242b2c : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1526b20826bb7bfcc3c62ec3ce7c44a90b78f46f","revisions":{"1526b20826bb7bfcc3c62ec3ce7c44a90b78f46f":{"kind":"REWORK","_number":1,"created":"2024-12-20 12:46:41.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/30/938130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/938130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/938130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/938130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/938130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/938130/1"}}},"commit":{"parents":[{"commit":"0138ce30e256a9f3a8fc3405a16051aa96294d57","subject":"Using quay.io registry in project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0138ce30e256a9f3a8fc3405a16051aa96294d57"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-20 12:45:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-20 12:45:57.000000000","tz":480},"subject":"Fix: trovestack failed to build images","message":"Fix: trovestack failed to build images\n\nfix a bug where trovestack failed to build images when $HTTP_PROXY\nis set\n\nChange-Id: I74501c5d488108d9db44620025da5138be3da989\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1526b20826bb7bfcc3c62ec3ce7c44a90b78f46f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1526b20826bb7bfcc3c62ec3ce7c44a90b78f46f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
