)]}'
{"id":"openstack%2Ftrove~938236","triplet_id":"openstack%2Ftrove~master~Ice103b430444eed797b6f193df88481b6839d452","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2024-12-26 02:00:57.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2024-12-25 06:15:49.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Ice103b430444eed797b6f193df88481b6839d452","subject":"Remove python3-venv if is_fedora is true","status":"MERGED","created":"2024-12-25 01:51:55.000000000","updated":"2024-12-26 02:05:55.000000000","submitted":"2024-12-26 02:00:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938236","meta_rev_id":"6623107638ac22a7e89014205ea939411e821719","_number":938236,"virtual_id_number":938236,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-26 02:00:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-26 01:14:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-26 01:14:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-25 02:03:06.000000000","updated_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2024-12-25 02:05:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-25 04:32:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54d7e6213b53a68952e8f81be3cd389318f193dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-12-25 01:51:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6de34f375008e3880bc713abf85ad73149021df7","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-12-25 01:58:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e253f6c182df116e49d193dc418172d4f40689de","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-12-25 02:01:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90c9db0552f4a0f86e4e6c5373cc6c42e5780746","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2024-12-25 02:03:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ea9e43dabcd37da9dfd5e082325d666737c6e16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 02:05:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f6d1bf458504535ad5df75a971d4093\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf9e12a4292e4e87959b4c484bb3c93b : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3a040eafd46d48b2bb89f28453d63a64 : SUCCESS in 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a361561d1b95712428fb48b066170ee11eac283a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 04:32:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d648385bb2454365b415b631250b9de4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d7740b983df4126b4913934fd6bac62 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b916d3954c354fe396ad9bb8edcba1c3 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7016c5358e334ccc84bc4257b51d8c11 : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/afe48cf0e9b04a299d633031d8d75f23 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf4c08ea6c894a2da164fbb30f23d555 : SUCCESS in 4m 54s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/1790f0e960d84c729c865bebc0bfb4b6 : SUCCESS in 57m 48s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/d1c6a91e35784d049c1c1ca61e081aaa : FAILURE in 53m 00s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/b57bd78da9b142c48c8f73779387cb0c : SUCCESS in 2h 16m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/df81e0a56aca4feabb559d6e2baece3e : SUCCESS in 1h 45m 46s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0e079084a9e3448b933fc2f12ee9866f : SUCCESS in 55m 52s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9de0d12bc3d242f3b59e9e069ac6a5a6 : SUCCESS in 2h 01m 09s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0e338b27f44c49beaa6e04a3a6ac937b : SUCCESS in 1h 44m 31s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/b4b6dde82e734dedb5e28918539f2a84 : FAILURE in 1h 05m 18s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a122a28956c448abbba956e650288997 : SUCCESS in 1h 32m 21s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/73233ec12428422a97f4c371844ecd9b : SUCCESS in 1h 43m 47s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/362cfd8782144c14a17af9e8b57862cc : SUCCESS in 5m 11s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/50af4815c3fd4634a1929a8fd2c0257b : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":1},{"id":"0e2843ccdb982a768154d0f5fd101da603bd1303","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-25 06:15:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"86f5ba34b26018cbd41402497a8526416c4b7c6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-25 14:08:39.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_31737\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"_revision_number":2},{"id":"70bc876578bfb4f4b0b02493e8a0c32f7fb483d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 14:19:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1de721331af043b7b86da679f6b0e38c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d00a1526060146aaa6dbbd4e861a2e68 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/86fefccafdca4812898305e074fbfef2 : SUCCESS in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad6f48ec880fa8f1541e9f9ce11dc3691f803777","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 16:25:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be147ed6057241899a94c90a7b00761e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60c48c065fc44be9b2847ca2b0882ba8 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff59a94ed4994c2193012c995af86583 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86b0f5d45a144c2ba92dc9f6f696e21f : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30953f9c545144c58955d1e1d62d07d3 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f1d78f743d4ca48279c6fff448974d : SUCCESS in 4m 12s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/358b7c88e378454da6c9acc03d9d1efc : SUCCESS in 41m 55s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/18c381e85349479ebd214a5295bc3953 : SUCCESS in 40m 56s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/3dd3d2795a4e4ea0992918d00c42e19f : SUCCESS in 1h 09m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/02622ac2c538441fb4add64b221b423c : SUCCESS in 1h 22m 46s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d8905953ccca4b5e8ba1eb21a6959719 : SUCCESS in 40m 34s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e9ab3ea4146f4512bc4c9a8a06a52eec : SUCCESS in 1h 06m 43s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4636faa3573045e794fd153e7bc68581 : SUCCESS in 1h 19m 27s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/af60e5ea753f48e5b4c68adf82e22368 : SUCCESS in 1h 01m 28s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/5d906cfde5604de0816a5c243194f186 : SUCCESS in 1h 57m 10s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/11a357e7fd1940f5a85f87b63a1da81e : SUCCESS in 1h 41m 04s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1876558ac54140bd8066d041e189f9e0 : SUCCESS in 4m 44s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/423897cbf20242a3b08f30f394117360 : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":2},{"id":"56b4fc10157eebe20509eed3d54482ecdbfa05eb","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-12-26 01:14:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c822cd4b2cede84f1e9f7fa40e6e3fbffe873b96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-26 01:15:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"809cf4b842a20aadfbc2facf690d508fd67a7956","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-26 02:00:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30e7e23ad8524b89b3fefcd024cf2c4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac1a60f5af3e4e40ab6cc44e7952f97d : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da127668fd46430d83d601b29c93cf08 : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/262bf99583db447fa32b39cb92c14a33 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80995c599bfc45dab857597d95229c10 : SUCCESS in 4m 18s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/ad32bcd81f324a7aa6bae26d806a594a : SUCCESS in 42m 21s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/63eb5fda7c7c4be5bfea5617639144c9 : SUCCESS in 38m 32s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3dccb9bda5b947ee905993f69575977c : SUCCESS in 40m 59s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/16dbadc1c6de490ba8403ebb41f82b84 : SUCCESS in 45m 14s","accounts_in_message":[],"_revision_number":2},{"id":"52077698dfcdb5584cff313aee09d69d5f591497","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-26 02:00:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6623107638ac22a7e89014205ea939411e821719","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-26 02:05:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ac08aa0b7724bf3b04142c87ae70b57\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7824badbf8c747988432ddfedd76d468 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9c92df8152d84485226364cc6eb4ee1078bbb238","revisions":{"57a0305c3e20a4b224fccc3d63af55ff77ae0b8f":{"kind":"REWORK","_number":1,"created":"2024-12-25 01:51:55.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/36/938236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/36/938236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/36/938236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/36/938236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/36/938236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/36/938236/1"}}},"commit":{"parents":[{"commit":"1526b20826bb7bfcc3c62ec3ce7c44a90b78f46f","subject":"Fix: trovestack failed to build images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1526b20826bb7bfcc3c62ec3ce7c44a90b78f46f"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-12-25 01:50:26.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-12-25 01:50:26.000000000","tz":540},"subject":"Remove python3-venv if is_fedora is true","message":"Remove python3-venv if is_fedora is true\n\npython3-venv does not exist in Fedora family\u0027s OS.\n\nChange-Id: Ice103b430444eed797b6f193df88481b6839d452\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/57a0305c3e20a4b224fccc3d63af55ff77ae0b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/57a0305c3e20a4b224fccc3d63af55ff77ae0b8f"}]},"branch":"refs/heads/master"},"9c92df8152d84485226364cc6eb4ee1078bbb238":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-25 14:08:39.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/36/938236/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/36/938236/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/36/938236/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/36/938236/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/36/938236/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/36/938236/2"}}},"commit":{"parents":[{"commit":"d09de01fdf95fd60f8b38c39899275177f4b3dbf","subject":"Merge \"Working on missing postgresql features\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d09de01fdf95fd60f8b38c39899275177f4b3dbf"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-12-25 01:50:26.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2024-12-25 14:08:39.000000000","tz":0},"subject":"Remove python3-venv if is_fedora is true","message":"Remove python3-venv if is_fedora is true\n\npython3-venv does not exist in Fedora family\u0027s OS.\n\nChange-Id: Ice103b430444eed797b6f193df88481b6839d452\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c92df8152d84485226364cc6eb4ee1078bbb238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c92df8152d84485226364cc6eb4ee1078bbb238"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
