)]}'
{"id":"openstack%2Ftrove~939082","triplet_id":"openstack%2Ftrove~stable%2F2023.2~If7420fe75b9016c7b13ed97f953ca3d9c34f58c2","project":"openstack/trove","branch":"stable/2023.2","hashtags":[],"change_id":"If7420fe75b9016c7b13ed97f953ca3d9c34f58c2","subject":"[CI]: run perodical jobs on master branch only","status":"MERGED","created":"2025-01-13 12:58:28.000000000","updated":"2025-01-14 02:11:22.000000000","submitted":"2025-01-14 02:07:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939082","cherry_pick_of_change":938244,"cherry_pick_of_patch_set":1,"meta_rev_id":"e66026292a0a4f802649d166aaf6d3d8ed888ded","_number":939082,"virtual_id_number":939082,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-14 02:07:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-14 01:21:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-14 01:21:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-13 12:59:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-13 12:59:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f8d510ec22e95d98e8db4a70e2d4413a1a39cd3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-01-13 12:58:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* zuul.d/projects.yaml","accounts_in_message":[],"_revision_number":1},{"id":"393a5c43c193b8b727dcc099c733223adb60fa2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 12:59:03.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch stable/2023.2.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/trove/zuul.d/projects.yaml@stable/2023.2\", line 44, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/trove/zuul.d/projects.yaml@stable/2023.2\", line 45, column 22","accounts_in_message":[],"_revision_number":1},{"id":"e7f5d4b1c7d2afe2d79dbba2edd8c373fe2d5ee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 12:59:12.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch stable/2023.2.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/trove/zuul.d/projects.yaml@stable/2023.2\", line 44, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/trove/zuul.d/projects.yaml@stable/2023.2\", line 45, column 22","accounts_in_message":[],"_revision_number":1},{"id":"f2ae1a1f47bbbfc5a171b1bd0f53cb8fbee9b152","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-01-13 13:03:42.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"727e6ece2e45ed250d71de796b6bfa6bbc8625f1","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-01-13 13:03:48.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"51cab764cf3f22e5972a884d22c12f847b9f2a04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 13:19:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e79fa3f6ff3142df8ce50d1e731f0782\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d91c063721164e3c8b9cde845e5af143 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e10b5a7cc9f4d3a9154ee213fc02d84 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ca2fac7133e4274bebfee6863b57867 : SUCCESS in 8m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3d9749b759c4d67a5b08846ab2a724a : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45c25814afd44982a4df94d8dd5a38d1 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fefd70ef99a841d38a6c4cee84fac080 : RETRY_LIMIT in 1m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8c4865939734bddbd8dd45efc5aa320 : SUCCESS in 4m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a5a594f767cd43b18cd7155533483dc6 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":2},{"id":"3b62389aebca5315cb5cfcb915213cc1541e65b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 13:27:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7884ddb538a843058990f5078fa2f374\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e85c69756234f6fa8c4f694cbcb162c : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0731f422d414faba72d7a84c61f22c3 : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/54876ccda2c34ab483653c39b83d38d8 : SUCCESS in 12m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/46a50b774292456c89ed4bf981318034 : RETRY_LIMIT in 1m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ecb0ff6280ec9d81f1afb3ae290a426fe03948da","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-01-14 01:21:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"82fd89f6c298bf25facb112861900a6fe59be92e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 01:21:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"35af43ce2580f2e9f079b3053076e51bdde764aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 02:07:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d163d5911e249e8906c95c0f4a55930\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b8e42f43c20479c9e49a8e26a75d3b2 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6afc28da637e4b1c9576fd75605b176b : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d26f83ed026049ee909f215d79ce51ad : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df309e6de2944505b3f469889abf2965 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fdbe17b477a40c8ac07206753b94ec9 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"f595ca17c17ed61f7b8661c5dde701f3ad66d3b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 02:07:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e66026292a0a4f802649d166aaf6d3d8ed888ded","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 02:11:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3703191430e747c780451cb17d628411\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f794ef74e89046e4b78515d59fd61539 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a391f6c4c30003cf8a53acc3275ada97620a1c78","revisions":{"366d1d7ef324395a880ef5ce69eecb54dcae36d1":{"kind":"REWORK","_number":1,"created":"2025-01-13 12:58:28.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/82/939082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/939082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/939082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/939082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/939082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/939082/1"}}},"commit":{"parents":[{"commit":"885df8df9a44b30726d13c30e9422879bc5e3a2f","subject":"Merge \"Enable network_isolation by default\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/885df8df9a44b30726d13c30e9422879bc5e3a2f"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-25 08:47:41.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2025-01-13 12:58:28.000000000","tz":0},"subject":"[CI]: run perodical jobs on master branch only","message":"[CI]: run perodical jobs on master branch only\n\nChange-Id: If7420fe75b9016c7b13ed97f953ca3d9c34f58c2\n(cherry picked from commit 0f1fc5dbcc109af52dae60260213934a367253f7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/366d1d7ef324395a880ef5ce69eecb54dcae36d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/366d1d7ef324395a880ef5ce69eecb54dcae36d1"}]},"branch":"refs/heads/stable/2023.2"},"a391f6c4c30003cf8a53acc3275ada97620a1c78":{"kind":"REWORK","_number":2,"created":"2025-01-13 13:03:42.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/82/939082/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/939082/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/939082/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/939082/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/939082/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/939082/2"}}},"commit":{"parents":[{"commit":"885df8df9a44b30726d13c30e9422879bc5e3a2f","subject":"Merge \"Enable network_isolation by default\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/885df8df9a44b30726d13c30e9422879bc5e3a2f"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-25 08:47:41.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2025-01-13 13:03:42.000000000","tz":0},"subject":"[CI]: run perodical jobs on master branch only","message":"[CI]: run perodical jobs on master branch only\n\nChange-Id: If7420fe75b9016c7b13ed97f953ca3d9c34f58c2\n(cherry picked from commit 0f1fc5dbcc109af52dae60260213934a367253f7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a391f6c4c30003cf8a53acc3275ada97620a1c78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a391f6c4c30003cf8a53acc3275ada97620a1c78"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
