)]}'
{"id":"openstack%2Ftrove~939084","triplet_id":"openstack%2Ftrove~stable%2F2024.2~If7420fe75b9016c7b13ed97f953ca3d9c34f58c2","project":"openstack/trove","branch":"stable/2024.2","hashtags":[],"change_id":"If7420fe75b9016c7b13ed97f953ca3d9c34f58c2","subject":"[CI]: run perodical jobs on master branch only","status":"MERGED","created":"2025-01-13 12:59:01.000000000","updated":"2025-01-14 02:11:15.000000000","submitted":"2025-01-14 02:07:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939084","cherry_pick_of_change":938244,"cherry_pick_of_patch_set":1,"meta_rev_id":"3eb5b91a0ab2b786ca46c3e1631462eb4003977c","_number":939084,"virtual_id_number":939084,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-14 02:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-14 01:21:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-14 01:21:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-13 12:59:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-13 12:59:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b9486b31de0db1da4fa1a5262b6c5db0a5a00fb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-01-13 12:59:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* zuul.d/projects.yaml","accounts_in_message":[],"_revision_number":1},{"id":"dd72e5abbe21fa82f583476619c520c71f54d568","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 12:59:36.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch stable/2024.2.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/trove/zuul.d/projects.yaml@stable/2024.2\", line 47, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/trove/zuul.d/projects.yaml@stable/2024.2\", line 48, column 22","accounts_in_message":[],"_revision_number":1},{"id":"c879c3f61e9b3e1e0ccbdf5dcfc7dc8b30bfd616","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 12:59:43.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/trove on branch stable/2024.2.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/trove/zuul.d/projects.yaml@stable/2024.2\", line 47, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/trove/zuul.d/projects.yaml@stable/2024.2\", line 48, column 22","accounts_in_message":[],"_revision_number":1},{"id":"cd7a97de43b1856bda17eee648c95159b5823fa7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-01-13 13:02:12.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"afdfd8fe7e8b066f513add1af90a190327dde8f8","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-01-13 13:02:25.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"abb4e0d123c5899de9c26cca285375393f07248f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 13:14:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/139a1a673cab4c28bd70963dfbcdc865\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a404cdb702f5450ba3e0cb6217d558fb : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d66a625789104993bd7173badcf935e1 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3dda2c35399242abaa1c0c2557ab512d : SUCCESS in 6m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63c40c231960418d914ddd0d87e407f8 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46d17c9ab15743c29f0c6be8d1412f10 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/994468a2ed5345e0a28176265182531f : SUCCESS in 5m 23s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d40aeed407fd48f8a8db37c18cdb22ed : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":2},{"id":"b54611788d27f57d465343217050859cf2c9025b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 13:16:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7098c0c9e454f3b9551d4e1bb1abbcd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9cad7edd4c84449b8e9752526e3dba1 : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5d58388ec6ad4882890918d42b526224 : SUCCESS in 11m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"334c6c47fc56574173aac4db2d056663af1aebf2","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-01-14 01:21:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"11ab5aa07486433a3fcffbe46077726063cba20b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 01:22:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5cc25f06480c30220e0706f721777d82635a8baa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 02:07:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02a3b3c0e2074da1b5df50e7bf6c6497\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4c8e4688a2e4a8f8743dfa00e270b23 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f84fec9526f84091b08b21598e934ac8 : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/96f364f55bab4858b01d5bc9494f6703 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e45ee97b29434a00aede2a632721724b : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":2},{"id":"42161a57887ce07b3860811a6b0c190d1d1f56cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 02:07:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3eb5b91a0ab2b786ca46c3e1631462eb4003977c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 02:11:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1919a443ea044e36969017be9deca394\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5fdea8f27ed4d038829385002ea0377 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"93efb5f57f7bc6f683c5a5829ba9d7c955b30219","revisions":{"394d8e949043fae313b9237a5f3ac06791a688c4":{"kind":"REWORK","_number":1,"created":"2025-01-13 12:59:01.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/84/939084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/939084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/939084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/939084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/939084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/939084/1"}}},"commit":{"parents":[{"commit":"95e879d42544cec28322a21bd10fa0537f88e5de","subject":"Fix replication detach error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/95e879d42544cec28322a21bd10fa0537f88e5de"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-25 08:47:41.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2025-01-13 12:59:01.000000000","tz":0},"subject":"[CI]: run perodical jobs on master branch only","message":"[CI]: run perodical jobs on master branch only\n\nChange-Id: If7420fe75b9016c7b13ed97f953ca3d9c34f58c2\n(cherry picked from commit 0f1fc5dbcc109af52dae60260213934a367253f7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/394d8e949043fae313b9237a5f3ac06791a688c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/394d8e949043fae313b9237a5f3ac06791a688c4"}]},"branch":"refs/heads/stable/2024.2"},"93efb5f57f7bc6f683c5a5829ba9d7c955b30219":{"kind":"REWORK","_number":2,"created":"2025-01-13 13:02:12.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/84/939084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/939084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/939084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/939084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/939084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/939084/2"}}},"commit":{"parents":[{"commit":"95e879d42544cec28322a21bd10fa0537f88e5de","subject":"Fix replication detach error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/95e879d42544cec28322a21bd10fa0537f88e5de"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2024-12-25 08:47:41.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2025-01-13 13:02:12.000000000","tz":0},"subject":"[CI]: run perodical jobs on master branch only","message":"[CI]: run perodical jobs on master branch only\n\nChange-Id: If7420fe75b9016c7b13ed97f953ca3d9c34f58c2\n(cherry picked from commit 0f1fc5dbcc109af52dae60260213934a367253f7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/93efb5f57f7bc6f683c5a5829ba9d7c955b30219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/93efb5f57f7bc6f683c5a5829ba9d7c955b30219"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
