)]}'
{"id":"openstack%2Ftrove~939319","triplet_id":"openstack%2Ftrove~master~Ic796eea59647533a58c6af77abd8e67fe3852751","project":"openstack/trove","branch":"master","topic":"support_snapshot","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-02-10 01:12:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic796eea59647533a58c6af77abd8e67fe3852751","subject":"[DNM] test swift snapshot driver","status":"ABANDONED","created":"2025-01-15 07:29:12.000000000","updated":"2025-02-10 01:12:14.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2917425e26bc6c80255c924289d9f92209a11464","_number":939319,"virtual_id_number":939319,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-06 06:08:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-15 07:44:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-15 09:28:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f21ed58209749cd96ef696680ce36daa70376c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-01-15 07:29:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e568c563e2feb75fa3ee62e9ee993b443ab136b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 07:44:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a06dc0e4b98747d994c0b6f10aa4915e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cccf4465524d4fd68294221c3f70a5d6 : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/edce121079164338a8008f14e07e7bad : SUCCESS in 11m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf7757d8ef1e17369fbe8788e78e04f55b0e0595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 09:28:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6a8e8755293474aa64d2e75fbedb6fb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b4395b9e899411faa6d22eade3757dd : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c03faa787c548c588b95f4124456863 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/408a83151fea45eaabce6a0d737b5343 : SUCCESS in 5m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/482551b093044ddaa69a55f357bf0f66 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a20cebe7bef4acc89436332ad9e3c8d : SUCCESS in 4m 39s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/b08a3d213a9f4feb8bfe976e6d595ddf : SUCCESS in 1h 07m 18s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/090bf5332c1947b5912d2b71e3be9e0a : SUCCESS in 48m 57s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/e34c3e12482d4b5c9eb7c1c3d3f0a621 : SUCCESS in 1h 50m 43s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/4090f92071cf48c49f8801c4dce2cf2e : SUCCESS in 1h 39m 55s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0eb72dbcf659436bbaefbb85c8a9f798 : SUCCESS in 42m 23s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/cfe054853349454db6b5e58deeecd97c : SUCCESS in 1h 04m 04s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/172d6181110d4817b3a5a239bd98291c : SUCCESS in 1h 57m 40s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/bd254286517945f3bf4420a11536d6f7 : SUCCESS in 45m 48s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/b7a070d49bde48dd96af53544b1348fc : SUCCESS in 59m 35s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/d3ef50a40efb4128bdb6cb8bf88c76f4 : SUCCESS in 1h 58m 15s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ebe2ca69751b4a1d96f06166ce6f16ef : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7ec414920a4eb3e0f4ae5291a6b49b2e986dec28","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-02-06 04:23:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14d33035857334c677cc4f4fba1ebf97c4d1aa1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 04:37:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d0bd003b71c4b24b58236de93a51842\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12ec1e631a5f4abf8805bdf5261bf7ec : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/88ebb17afef341ebb44cf50a90d93268 : SUCCESS in 11m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d728d96b9197c2090d9ccc17bfadf2c82721ffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 06:08:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cd343f6ae2442909efc294ca56737f3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb8170e18303473790d5df2b4c39aade : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f7d10341b2b4ba2a3804c3b26e527d3 : FAILURE in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b91ceff1e90747579484404613d5047d : SUCCESS in 6m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/10918c9c17c54f50b0bc34cfa2674f21 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ccab02b5fed4b2daf8b75dccb731f18 : SUCCESS in 4m 47s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/3e1298c782044cd0aaa770553a8c2bbe : SUCCESS in 49m 02s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/b1c27e535e1e48168029b69a481aef55 : SUCCESS in 44m 00s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/ae2f0ddb54af40ddb474c34b65a4e41a : SUCCESS in 1h 41m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/e53d82be050a4609ba14c1acff6301ba : SUCCESS in 1h 33m 52s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/362ab9b4f0ff4d2494ca7b2ec96d77f1 : SUCCESS in 40m 58s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f5b80f5dc9474d7f97767d17a3250911 : SUCCESS in 1h 18m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f6e924842efb44f9af0f666e4034369f : SUCCESS in 1h 16m 43s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/749350e49282496ab343d6c0aadcd573 : SUCCESS in 46m 23s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/8b6ff7b562c34682872fd0a8c747a3f1 : SUCCESS in 1h 15m 46s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/499aaa1a4cea41eb95d5d2763e73d936 : SUCCESS in 1h 13m 12s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff3e91c1534942599b992a93fc39a824 : SUCCESS in 4m 54s\n\nWarning:\n  Comments left for invalid file trove/guestagent/api.py\n  Comments left for invalid file trove/taskmanager/models.py","accounts_in_message":[],"_revision_number":1},{"id":"2917425e26bc6c80255c924289d9f92209a11464","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-02-10 01:12:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"429f65012f7c6a4db3992b73c4ea7d7fa1c11991","revisions":{"429f65012f7c6a4db3992b73c4ea7d7fa1c11991":{"kind":"REWORK","_number":1,"created":"2025-01-15 07:29:12.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/19/939319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/19/939319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/939319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/939319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/939319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/19/939319/1"}}},"commit":{"parents":[{"commit":"e9fa651ac0c06d16240ac597830020e75026b0e9","subject":"[CI]: fix unittest failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9fa651ac0c06d16240ac597830020e75026b0e9"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2025-01-15 07:27:31.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2025-01-15 07:27:31.000000000","tz":480},"subject":"[DNM] test swift snapshot driver","message":"[DNM] test swift snapshot driver\n\nthis tests the default stroage_strategy is cinder while the\nreplicas_snapshot_driver is swift\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/903440\nChange-Id: Ic796eea59647533a58c6af77abd8e67fe3852751\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/429f65012f7c6a4db3992b73c4ea7d7fa1c11991"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/429f65012f7c6a4db3992b73c4ea7d7fa1c11991"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
