)]}'
{"id":"openstack%2Ftrove~939320","triplet_id":"openstack%2Ftrove~master~I529eb76076021d02e8f15546484674c7b3c4d4d3","project":"openstack/trove","branch":"master","topic":"support_snapshot","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-02-10 01:12:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I529eb76076021d02e8f15546484674c7b3c4d4d3","subject":"[DNM] test cinder snapshot driver","status":"ABANDONED","created":"2025-01-15 07:31:56.000000000","updated":"2025-02-10 01:12:15.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b870881d595023f25b310de4757d93820ac84286","_number":939320,"virtual_id_number":939320,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-06 06:17:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-15 07:47:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-15 09:36:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c097b1799550158247391ffb55b135ff93c4ff9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-01-15 07:31:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90c4d9dae20061c71babe17cfcb416bef94ddbf7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 07:47:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb30cb1031424882869f6f801d1f28fb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1cfc9027eacf466bb2f169bc61ef9224 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9f61e8725a3244479833b2f97006b684 : SUCCESS in 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e71bbb80ca2e76b64b63fe691d36d23cb414815c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 09:36:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77a2923960b84007842e98c812e2850f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e20a03fb65f84318ba8072a3e8502f76 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b001b4bd55f4ef584cd4cd55b90e6d7 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dc6064c1d2a43e2bbc09839a45e85a0 : SUCCESS in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5678c2dbdb0e4ac3b83f24400705866a : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c75c92725e4dc992e4ce2d9b71d5f4 : SUCCESS in 5m 19s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/0f08d2d3b0d54a29afb4fd0e57492530 : FAILURE in 36m 47s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/98e0483429514e6a8ab71dcade0e791e : SUCCESS in 59m 52s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/a40a4d12f1c94704a3061f498ec08a9b : FAILURE in 2h 01m 27s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/f261f12b042b49fda2c7568a9bf3ae7e : SUCCESS in 59m 57s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/34c58f1266c34c6a9d3fd71b2ccc31a0 : SUCCESS in 1h 03m 49s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ad20605ad3d24ad0a4784ff92f0604ee : SUCCESS in 1h 13m 31s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ed955850d7b348549ec5f236a72f7694 : SUCCESS in 1h 36m 01s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/3d299c1442014fc28d6191a58a696c0f : SUCCESS in 1h 06m 45s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/0b992e9522db4bc8abbffbde39a3add5 : SUCCESS in 1h 10m 53s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b4f5a642c5c54a7ea3bf297dc44b3641 : SUCCESS in 1h 38m 39s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ccd2afcf8e154b43b1c755cc6d2b13dc : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"317422a580450359d8a0415ff2be032d71ef475c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-01-15 14:36:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"feda8ec65f1086e23fffa6de1ac334cabf0c597b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 14:55:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea50fa6da5f3414ebbef746dca050bcc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db9af37f70c846228eaf0ffe84fea7f2 : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1e2d732c7193422a878c355bf025324c : SUCCESS in 14m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a62fa7e159dc0d39aa1302ff60ed4f60777722dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 16:54:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc74b75ffc234e50854b4d7519889eb5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d24403e8c7c412fb7fe5493ba148793 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35811af1c7254756b2f9f10d180212c5 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aacd99549ba04c45a5507c05c574388c : SUCCESS in 7m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bfb2902f5d742e09d9a9c1bf335b29d : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c44382e9160648e5be54de8dc11aa361 : SUCCESS in 5m 22s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/8d4ff17eeb174558b77c3c66f222cc63 : SUCCESS in 1h 10m 56s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/37b24b3cd7fc4ecd94332540fd61929e : SUCCESS in 57m 23s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/b022a2f0286b4fbeb800aaf50a538658 : SUCCESS in 1h 46m 49s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/1d18dafed5e045e19871dc093e7b2c94 : SUCCESS in 1h 22m 38s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a1c4583db44e4735b6e7af4ab2a968f3 : SUCCESS in 59m 08s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/30fdb2c480964bbfbd12d8759d9594be : SUCCESS in 2h 17m 04s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/cd9fafe055e14328be5c61d2a5458d99 : SUCCESS in 1h 34m 08s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5029058528044366badf030173392f8b : SUCCESS in 36m 16s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/06d5d2fbd60b4aa59558848126b173d0 : SUCCESS in 2h 11m 31s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/c039ca5879af4cdd90eb736bccd2d156 : SUCCESS in 1h 26m 02s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0060129b7c2c4213820805b2a029654d : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"523a544b8acd1d6f0d9dcf2dc968a3fe095fbf3e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-02-06 04:24:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0335dfcaac836b84bd208150c3e7f7e77becacbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 04:37:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d11946002325464b94b5389426419edb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71c3dedfdac7410b9c2331f646f0cc71 : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5fde1a66997c40a889674a15dc2d4a64 : SUCCESS in 10m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f498632296ff6f7a616a073cc1e973181fa6d258","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 06:17:34.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/903440 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"b870881d595023f25b310de4757d93820ac84286","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-02-10 01:12:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59230a85e364b36345e58972fe65e3cba97a45be","revisions":{"59230a85e364b36345e58972fe65e3cba97a45be":{"kind":"REWORK","_number":1,"created":"2025-01-15 07:31:56.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/20/939320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/20/939320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/20/939320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/20/939320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/20/939320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/20/939320/1"}}},"commit":{"parents":[{"commit":"e9fa651ac0c06d16240ac597830020e75026b0e9","subject":"[CI]: fix unittest failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9fa651ac0c06d16240ac597830020e75026b0e9"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2025-01-15 07:30:58.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2025-01-15 07:30:58.000000000","tz":480},"subject":"[DNM] test cinder snapshot driver","message":"[DNM] test cinder snapshot driver\n\nthis tests the default stroage_strategy is swift while the\nreplicas_snapshot_driver is cinder\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/903440\nChange-Id: I529eb76076021d02e8f15546484674c7b3c4d4d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/59230a85e364b36345e58972fe65e3cba97a45be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/59230a85e364b36345e58972fe65e3cba97a45be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
