)]}'
{"id":"openstack%2Ftrove~946331","triplet_id":"openstack%2Ftrove~master~I9292132aeffe3d36eb4a0d993f9feeebe81c1a9a","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-04-24 09:00:36.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-05-27 11:28:30.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-04-24 08:39:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9292132aeffe3d36eb4a0d993f9feeebe81c1a9a","subject":"Increase max_file_size for container","status":"MERGED","created":"2025-04-04 09:19:24.000000000","updated":"2025-05-27 11:28:30.000000000","submitted":"2025-04-24 08:39:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":45,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946331","meta_rev_id":"e4d55619ec7fbe04dd4dfea72552ad2db9d2dc4c","_number":946331,"virtual_id_number":946331,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-24 08:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-24 08:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2025-04-24 07:54:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-24 08:35:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-04 09:22:33.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-04-04 09:26:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-04-04 09:36:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4ec988a5934ab42f774885cf9218a57f0acfee0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-04 09:19:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46f6ce4319795956f45fecd104b0989fd56b2e75","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-04 09:22:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2049932aedb961e2b938f2294b754fd2fb291e82","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-04 09:22:24.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"3f4685d7602f2e6dc6b267327c7c76d26120eefc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-04 09:26:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2297f3bd0ff75a374f7c365f56e2e466d170b800","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-04 09:29:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"243961b115903a0331e9996f491f53f1d94bdad7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-04 09:29:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"058978e32065487497203560db56bf4312bc9092","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-04 09:35:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ffeb97f749a52dbddb27cc5946962c42a8b8e3db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 09:36:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41cb0757aa874847b79dcc626d8e7699\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53e642f9125743ea9ed964587e81b3b7 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bab13fbf3bb4cbcaf596c26e12c1920 : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":2},{"id":"c8b966451fa9832590cec2ca9250ab206f8ee665","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-04 10:11:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4aa7370b66ec8d0b68208c9981f25903714863c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-04 10:12:28.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"aa90f4e80cb8c328ee13a771c20d18429c4aa5bb","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-04 10:15:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9a784e2891f3f1fe3283c1d212a3e8e1f9345331","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-04 10:17:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a6f22c7953b9fc64c5838c5a7871ee73c3febb96","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-04 10:19:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b3fe0539f07a744916a4dc207bbae05f4bcd4db4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-04 10:24:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8e1e099bc6f224bdd08091a60254365a14fd62a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 10:25:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75b3e7fbae9f41b288de78c1900b1130\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b150b266c9c44ea800607c77841441c : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/367ac0c2211e4161b400d70c9114583f : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":3},{"id":"001dad6a0c602876d2302f267a361d1138c81c5d","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-04 10:39:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6d28d44717c393494e3e9ac383141f2e3e03432c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-04 12:26:28.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"63a2309414c7a4fc6533d2b56c8e604e86da1a26","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-04 12:27:28.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"34b1810294fb1c4960cb43061352529ebb5a7403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 12:36:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37ba307fdcfa48caab386bbf7e406b07\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b69ff82b1c394b829270fea35fbeff58 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a0cf286d0ae477dbf48a482948bcbb6 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":4},{"id":"71e5a86f397f7179c5d74164af491b7f410fd5a9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-04 14:22:00.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"403330b0b36e3145ba8b880728543a4d04f3a2b4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-04-08 14:10:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0ca421f50b78dee3981b524e728042c8695da2d0","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-08 22:41:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d7ac8a196f212176c39ccd11b64be0a3b2a28d47","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-04-14 12:08:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f69d76633a0f0d2763150f284101f01c259ed2d5","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-14 12:28:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"46040745eadab906693a5266ca823483524f3ad3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-04-17 07:37:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7b540767a157fe7237ab4af2030a6306bd727f5c","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-17 14:49:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"00a4ca471f6e4de4e6eeb4020b21928c8c0885e0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-17 14:51:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"700ea33de031ac3f483b520e4e124af9a442b3cb","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-04-22 14:07:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b5547bf26b32be94dc648c072cc4295cd2b2af81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-23 07:17:43.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8797ad585fb4884176460e423ec6ef1ee8b4c986","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-23 07:18:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c89553300c63f0258f6ccaf5c7e059a5511f6127","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 07:27:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e87b726523f4c8b8060db4cea89c3a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a443fc888b114804997b9906f271037c : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d4327ea634643b2bbcd2e4618713cb4 : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":5},{"id":"fb49618e051267ab656c9a2825260601ba1d9a96","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-23 07:37:25.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"82d04d1c95f66428bf3b19efabb4b14b85469ee6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-23 22:37:16.000000000","message":"Patch Set 5:\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"180ee704d4ee997c5d382e0fa2d5b3f8cf689978","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-24 07:28:40.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"c13487ae96e3736cee6c3c126a7e5eb75d0124a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 07:28:45.000000000","message":"Patch Set 6: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 946331,6","accounts_in_message":[],"_revision_number":6},{"id":"6ebdf64870c58e938a0e6af82a73081c27c87dbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 07:28:45.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/trove for 946331,6","accounts_in_message":[],"_revision_number":6},{"id":"29a3c9894b6a5760e208a2d623abf2dbdeaf813d","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-24 07:30:55.000000000","message":"Patch Set 6:\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e2b218c654fb6aaad5a825e3dae6664a9834cece","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-24 07:31:21.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nThe following files contain Git conflicts:\n* doc/source/install/install-devstack.rst\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"7e60f845eb7efbb2bba204cad4b72e22e2be0030","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-24 07:31:25.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"d317d2fd43f6fb34a38f374d7b0101590d4035fe","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-24 07:32:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"daec6a1adfddf272be1078f34fdac2e847d0fd72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 07:34:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/409ea4eab76c47ae8986d02207755ec4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49f45107baca4b5ab83e3c437173f378 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d50e4a6145774e7ab3f114f6c03e8bf2 : FAILURE in 2m 46s","accounts_in_message":[],"_revision_number":7},{"id":"59562cb887e03337159264667e87ed9df8754f24","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-24 07:39:57.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"003e08a619f4c477d28ddccea44623f9795be811","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-24 07:40:05.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"80e57da9eae60be3479685ce9fb40e0b6abea022","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 07:49:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8952a82b7b954311adb8bc08294a9ee0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8edef98a81e54fe78ad284ded841efa4 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bece4212b97745e99fccb32597cfa1b8 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":8},{"id":"75a709ba4c08a60c25b1599eaa33169ea8cd56b1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-24 07:54:22.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0c248e80c2471439fc21350410cbf62c19d8b436","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-24 07:54:47.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"750362b7a343e5445914dd56591e5383212501c7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-04-24 08:35:24.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"76663045964d8766beba81b842e103622b4c5234","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 08:35:34.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"70122338613912593818a5602ea1a3cd6860f956","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 08:39:00.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d03bed60dea47dc8ffd973071521399\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4b03025965f4e44bb8c0e5757aa38f0 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beac1a7f71af44c797506ad5f06c76e2 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":8},{"id":"4e413d03bd56d11b84642a010cd07696c28de7a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 08:39:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"6bf62399b04b5f1d6502ba50a2f53661d538bda3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 08:39:51.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c725adb9fc14992b81a18e0843651a5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/884ff32bf22d4276bfee586d909fde0f : SUCCESS in 36s","accounts_in_message":[],"_revision_number":8},{"id":"b558906cd7b6ecd4898eb528c3819a062b91dabe","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-04-24 08:44:14.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"faf6b5c41e6a5b9fef28d5dd786a5061f9dc212f","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-24 09:00:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"84539ad9d26834824555d965a821bf1ab62038a7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-24 17:48:21.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"a341465ae9bb644f9ae5f19768db9f2107cb3161","revisions":{"cf9cf04fef0956ca485867605c3e29b39fa57a61":{"kind":"REWORK","_number":1,"created":"2025-04-04 09:19:24.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/31/946331/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/946331/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/946331/1"}}},"commit":{"parents":[{"commit":"54b241a3f854c0b65f45f7e4f4aadac6f1340729","subject":"Merge \"Do not install ceilometerclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/54b241a3f854c0b65f45f7e4f4aadac6f1340729"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-04 09:19:24.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-04 09:19:24.000000000","tz":0},"subject":"Remove swift from devstack","message":"Remove swift from devstack\n\nChange-Id: I9292132aeffe3d36eb4a0d993f9feeebe81c1a9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cf9cf04fef0956ca485867605c3e29b39fa57a61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cf9cf04fef0956ca485867605c3e29b39fa57a61"}]},"branch":"refs/heads/master"},"afd0a8d2a305061ccd58e776e5bc8e782706ad2f":{"kind":"REWORK","_number":2,"created":"2025-04-04 09:22:18.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/31/946331/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/946331/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/946331/2"}}},"commit":{"parents":[{"commit":"54b241a3f854c0b65f45f7e4f4aadac6f1340729","subject":"Merge \"Do not install ceilometerclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/54b241a3f854c0b65f45f7e4f4aadac6f1340729"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-04 09:19:24.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-04 09:22:01.000000000","tz":0},"subject":"Remove swift from devstack","message":"Remove swift from devstack\n\nSwift is no longer required for DevStack when using Trove.\nIn fact, enabling Swift can cause Trove to crash, so \nit\u0027s recommended to disable it during setup.\n\nChange-Id: I9292132aeffe3d36eb4a0d993f9feeebe81c1a9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/afd0a8d2a305061ccd58e776e5bc8e782706ad2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/afd0a8d2a305061ccd58e776e5bc8e782706ad2f"}]},"branch":"refs/heads/master"},"854381a218a60209912219691a8533d509b21de2":{"kind":"REWORK","_number":3,"created":"2025-04-04 10:12:28.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/31/946331/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/946331/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/946331/3"}}},"commit":{"parents":[{"commit":"54b241a3f854c0b65f45f7e4f4aadac6f1340729","subject":"Merge \"Do not install ceilometerclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/54b241a3f854c0b65f45f7e4f4aadac6f1340729"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-04 09:19:24.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-04 10:08:53.000000000","tz":0},"subject":"Remove swift from devstack","message":"Remove swift from devstack\n\nSwift is no longer required for DevStack when using Trove.\nIn fact, enabling Swift can cause Trove to crash, so \nit\u0027s recommended to disable it during setup.\n\nChange-Id: I9292132aeffe3d36eb4a0d993f9feeebe81c1a9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/854381a218a60209912219691a8533d509b21de2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/854381a218a60209912219691a8533d509b21de2"}]},"branch":"refs/heads/master"},"45a99a7434adbf78696f848b133a04aaf6637231":{"kind":"REWORK","_number":4,"created":"2025-04-04 12:26:28.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/31/946331/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/946331/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/946331/4"}}},"commit":{"parents":[{"commit":"54b241a3f854c0b65f45f7e4f4aadac6f1340729","subject":"Merge \"Do not install ceilometerclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/54b241a3f854c0b65f45f7e4f4aadac6f1340729"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-04 09:19:24.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-04 12:24:01.000000000","tz":0},"subject":"Remove swift from devstack","message":"Remove swift from devstack\n\nSwift is no longer required for DevStack when using Trove.\nIn fact, enabling Swift can cause Trove to crash, so \nit\u0027s recommended to disable it during setup.\n\nChange-Id: I9292132aeffe3d36eb4a0d993f9feeebe81c1a9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/45a99a7434adbf78696f848b133a04aaf6637231"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/45a99a7434adbf78696f848b133a04aaf6637231"}]},"branch":"refs/heads/master"},"b771bbd02bbacdcdd875cf80ffa58584f4c37abe":{"kind":"REWORK","_number":5,"created":"2025-04-23 07:17:43.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/31/946331/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/946331/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/946331/5"}}},"commit":{"parents":[{"commit":"54b241a3f854c0b65f45f7e4f4aadac6f1340729","subject":"Merge \"Do not install ceilometerclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/54b241a3f854c0b65f45f7e4f4aadac6f1340729"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-04 09:19:24.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-23 07:17:16.000000000","tz":0},"subject":"Increase max_file_size for container","message":"Increase max_file_size for container\n\nIncreasing the default MAX_FILE_SIZE in Swift to \nsupport larger image uploads, specifically \nenabling the successful upload of the Ubuntu Trove image.\n\nChange-Id: I9292132aeffe3d36eb4a0d993f9feeebe81c1a9a\nCloses-Bug:#2106545\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b771bbd02bbacdcdd875cf80ffa58584f4c37abe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b771bbd02bbacdcdd875cf80ffa58584f4c37abe"}]},"branch":"refs/heads/master"},"6bdfcebde3cec210d84e958930d096aadd926b4c":{"kind":"REWORK","_number":6,"created":"2025-04-24 07:28:40.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/31/946331/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/946331/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/946331/6"}}},"commit":{"parents":[{"commit":"54b241a3f854c0b65f45f7e4f4aadac6f1340729","subject":"Merge \"Do not install ceilometerclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/54b241a3f854c0b65f45f7e4f4aadac6f1340729"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-04 09:19:24.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-24 07:26:43.000000000","tz":0},"subject":"Increase max_file_size for container","message":"Increase max_file_size for container\n\nIncreasing the default MAX_FILE_SIZE in Swift to \nsupport larger image uploads, specifically \nenabling the successful upload of the Ubuntu Trove image.\n\nChange-Id: I9292132aeffe3d36eb4a0d993f9feeebe81c1a9a\nCloses-Bug:#2106545\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6bdfcebde3cec210d84e958930d096aadd926b4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6bdfcebde3cec210d84e958930d096aadd926b4c"}]},"branch":"refs/heads/master"},"81b69210808c6810926a83e4a212a2440f72b86f":{"kind":"REWORK","_number":7,"created":"2025-04-24 07:31:21.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/31/946331/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/946331/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/946331/7"}}},"commit":{"parents":[{"commit":"136fadddbf06b404a4a4c15d0443dfcee965bd33","subject":"Merge \"Remove unnecessary +x mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/136fadddbf06b404a4a4c15d0443dfcee965bd33"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-04 09:19:24.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-24 07:31:21.000000000","tz":0},"subject":"Increase max_file_size for container","message":"Increase max_file_size for container\n\nIncreasing the default MAX_FILE_SIZE in Swift to \nsupport larger image uploads, specifically \nenabling the successful upload of the Ubuntu Trove image.\n\nChange-Id: I9292132aeffe3d36eb4a0d993f9feeebe81c1a9a\nCloses-Bug:#2106545\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81b69210808c6810926a83e4a212a2440f72b86f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/81b69210808c6810926a83e4a212a2440f72b86f"}]},"branch":"refs/heads/master","description":"Rebase"},"a341465ae9bb644f9ae5f19768db9f2107cb3161":{"kind":"REWORK","_number":8,"created":"2025-04-24 07:40:05.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/31/946331/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/946331/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/946331/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/946331/8"}}},"commit":{"parents":[{"commit":"136fadddbf06b404a4a4c15d0443dfcee965bd33","subject":"Merge \"Remove unnecessary +x mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/136fadddbf06b404a4a4c15d0443dfcee965bd33"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-04 09:19:24.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-24 07:38:59.000000000","tz":0},"subject":"Increase max_file_size for container","message":"Increase max_file_size for container\n\nIncreasing the default MAX_FILE_SIZE in Swift to \nsupport larger image uploads, specifically \nenabling the successful upload of the Ubuntu Trove image.\n\nChange-Id: I9292132aeffe3d36eb4a0d993f9feeebe81c1a9a\nCloses-Bug:#2106545\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a341465ae9bb644f9ae5f19768db9f2107cb3161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a341465ae9bb644f9ae5f19768db9f2107cb3161"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
