)]}'
{"id":"openstack%2Ftrove~948048","triplet_id":"openstack%2Ftrove~master~Ib65e5d4a120b6db2818afc97d4f2e742d262a35b","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-04-24 12:00:33.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-04-24 11:55:41.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Ib65e5d4a120b6db2818afc97d4f2e742d262a35b","subject":"Fix typo in link for documentation ","status":"MERGED","created":"2025-04-24 08:57:34.000000000","updated":"2025-04-24 17:46:32.000000000","submitted":"2025-04-24 12:00:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948048","meta_rev_id":"07afa2768c6e559dc1031bf6b2d18f3850c9bb8a","_number":948048,"virtual_id_number":948048,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-24 12:00:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-24 11:55:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2025-04-24 17:46:32.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-24 11:55:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-24 09:00:03.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-04-24 09:09:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-24 17:46:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"aaf54c569471d29582b366e8ec95e6481ddd6ea3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-24 08:57:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42ab7e33dc4f5589ad7edf51731737a197e7b7c6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-24 08:59:40.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6256b9ea03d6fc93a45c243d8f4b3d955e2d4208","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-04-24 08:59:43.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"45db40ea8a8936575b9274cc1aaa6b85ceb30f90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 09:09:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b5780acc15c4a3192a81303e20501bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/509a05a080af48ba85ca2b954f744bcb : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/329b4623d0eb4389aef799f4d834b5ae : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":2},{"id":"912042ad00766fcc8925f839af503b395702625d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-04-24 11:55:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ab29e8af3f195f6c408169eb30ef66c704038027","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 11:55:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"292890b9ac208b34422eb922b385e2d585faca43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 12:00:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b95922abd828404b9f9a0aefdec569aa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/150851299a484c19a66df6cca92a5d9e : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c081772e4634995958f19ec85997820 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"ff585d5fa32739881441d756b735b56ce9a7c07f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 12:00:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ef08d60e3118b4b8b0437c792da941aa3ef63122","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 12:01:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/331f020e1e8d489da49d8aeeeaaba180\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5272d8e6585a482481e8e91cdb24bb0e : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2},{"id":"07afa2768c6e559dc1031bf6b2d18f3850c9bb8a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-24 17:46:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f9b73bac306c9b0f65a46b2d6854628fe49317ba","revisions":{"dacbf473ac45862ba4c9b205d3c79008ebe67e33":{"kind":"REWORK","_number":1,"created":"2025-04-24 08:57:34.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/48/948048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/948048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/948048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/948048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/948048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/948048/1"}}},"commit":{"parents":[{"commit":"a341465ae9bb644f9ae5f19768db9f2107cb3161","subject":"Increase max_file_size for container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a341465ae9bb644f9ae5f19768db9f2107cb3161"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-24 08:57:34.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-24 08:57:34.000000000","tz":0},"subject":"Fix typo in link for documentation ","message":"Fix typo in link for documentation \n\nFix link syntax for documentation in README\n\nChange-Id: Ib65e5d4a120b6db2818afc97d4f2e742d262a35b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dacbf473ac45862ba4c9b205d3c79008ebe67e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dacbf473ac45862ba4c9b205d3c79008ebe67e33"}]},"branch":"refs/heads/master"},"f9b73bac306c9b0f65a46b2d6854628fe49317ba":{"kind":"REWORK","_number":2,"created":"2025-04-24 08:59:40.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/48/948048/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/948048/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/948048/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/948048/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/948048/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/948048/2"}}},"commit":{"parents":[{"commit":"a341465ae9bb644f9ae5f19768db9f2107cb3161","subject":"Increase max_file_size for container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a341465ae9bb644f9ae5f19768db9f2107cb3161"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-24 08:57:34.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-04-24 08:59:12.000000000","tz":0},"subject":"Fix typo in link for documentation ","message":"Fix typo in link for documentation \n\nFix link syntax for documentation in README\n\nChange-Id: Ib65e5d4a120b6db2818afc97d4f2e742d262a35b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f9b73bac306c9b0f65a46b2d6854628fe49317ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f9b73bac306c9b0f65a46b2d6854628fe49317ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
