)]}'
{"id":"openstack%2Ftrove~948282","triplet_id":"openstack%2Ftrove~master~I7b4fc6ed8771f119eaac19d8643e78d2a36bb02f","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2025-05-15 13:54:14.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-05-15 13:54:14.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-05-15 12:42:47.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-05-01 16:39:18.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I7b4fc6ed8771f119eaac19d8643e78d2a36bb02f","subject":"Add build of modern Ubuntu distributions","status":"MERGED","created":"2025-04-26 10:10:35.000000000","updated":"2025-05-15 13:57:12.000000000","submitted":"2025-05-15 13:54:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948282","meta_rev_id":"9edb5056fb2159314997153b79c841e6a962af2d","_number":948282,"virtual_id_number":948282,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-15 13:54:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-15 12:42:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2025-05-01 16:39:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-15 12:42:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-26 10:26:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-26 12:15:27.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-04-26 12:15:27.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-04-26 12:43:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-01 16:39:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-05-15 07:58:40.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"}],"messages":[{"id":"1431c14ce933f8bd79753a2f6fea6fe66a2a9f9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-26 10:10:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"301fd730c10634d33c77615784b30c53905c84bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 10:26:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/436ac9097b3f4c2e9cefb9b56acbddf5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca1678c71ae54c97b231c7f99956a6cf : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4cc5da2228a14e3591a56a8801ed4d49 : SUCCESS in 11m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4d989cb249134d62ac6fe870d29e78f0 : SUCCESS in 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60bbee1d44d69a5678692ce38f362b0db5f6b514","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-26 10:34:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6dda9ae91bdc91a260f3bfd6185214a4820530f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 10:49:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/604b73ac53e34acdad3149009aa9d83b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc84f754666d4e1e961a50788867728c : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f5397f3617a8454c88d506e0fb5d6590 : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/83d7dd75fb844d02bffd6daabfb3c72c : SUCCESS in 13m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5e53fdcd44ec39a09b130386d9a250104e3c51d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 12:43:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2c8fbcaff5040dc85c8f9773abfc5b5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/105b23443d2840f0850d47f88cba4a03 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b98b637b0c4f418393a2403d7de67658 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a486a122de94ad59518a798d98ffdcc : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b96e34ff2020412f8526b0d9c48e8b01 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/edf9bf6b0efd40bab91bdd4a68979839 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08e554f168074d039a2ee0d2bda12ba1 : SUCCESS in 3m 15s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/54b6250fd5ae4127a10b9ae8791ef031 : FAILURE in 26m 22s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/b60d2cd73d0549ef8d1fe382d8f2a9d9 : FAILURE in 29m 02s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/b0dda1e7d62447b790bb2d336cd0d715 : SUCCESS in 1h 00m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/cc878991fbde4004a73c02b2c4512e64 : SUCCESS in 1h 30m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/c308e35742a74590a2113e506a859e8f : SUCCESS in 2h 08m 44s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/956e8401e7d2462c91a6ffb69cb82acd : SUCCESS in 41m 21s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d4fe3601e84c41b3a5c1f6834f380b20 : SUCCESS in 1h 41m 28s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bf121818c8e44ca68552b7368b960a59 : SUCCESS in 1h 46m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/22765ed0be334ac2b73a5785596e5626 : SUCCESS in 1h 37m 17s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/1e878d152c2142808b43717866c0cc84 : SUCCESS in 45m 04s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/79ca2c6d14704d89b483fe936547980e : SUCCESS in 1h 34m 25s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b0ad2ab143ea484bb0d7105ed981cd7b : SUCCESS in 1h 17m 47s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/6972c04e34b84d1ea706e1b311cebfe2 : SUCCESS in 1h 39m 31s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0f174555a7974ab8b536403e40fd00a7 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":2},{"id":"f94aacdd36485ccf67e72fc502fa31e85a4c6a03","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-26 17:04:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"441ccf449a635ed5782e6e03cbd46fc63c31cce1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 17:20:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be16c1ceedb847fcb8048a3389950c72\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/554a91b6910e40eebfd7ed26e2278252 : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4cabab82618643519537f34525d841e2 : SUCCESS in 10m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/594a3b4705fd427bbe268b4ba864aa26 : SUCCESS in 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f34f5aba1cad92e3105a9398fd2128fdc3093a23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 19:09:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1872e6fe423b46efa48e3323b3f9a894\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f09be48af3a447b8a1a649f6771eb8aa : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f306487be9d2431083095a2c753a876c : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68d718dd7573439392c312e5e05bd9b3 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e77d89589c814cd396777b6202678e50 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf65b08ba94048c8a462f474f2c701ac : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2592d282c48b4a1fa634b8864440ed19 : SUCCESS in 4m 13s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/09da83b36a0a4261b0aec944e0bdf134 : SUCCESS in 25m 56s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/366f8a9b14c9419288bb93ff8dfbdd72 : FAILURE in 25m 28s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/cc4c10447e1c46e184a0f91fcaf85cd7 : SUCCESS in 1h 50m 49s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/1bd3bc2cab3d417787a22f93946abf22 : SUCCESS in 1h 30m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/43d56abdf4a7412daa39a4fe661ccb20 : FAILURE in 45m 03s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e8bd1575abe84b8788a36b2de9705b07 : SUCCESS in 30m 08s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f7beb9b451c14e1ebb64c0881f0258c2 : SUCCESS in 54m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f0e2dc30891c42e4a46bda9e26999160 : SUCCESS in 1h 38m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7fe61f2ee97444638671fbede5f5d49b : SUCCESS in 1h 55m 29s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/b2355c41b85a419296827827b72b2fe5 : SUCCESS in 47m 36s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/f1f9232474ac4dbc99accc2cdf50daac : SUCCESS in 1h 37m 58s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/7baa745326954a4686fd9863d2d68676 : SUCCESS in 1h 20m 48s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/c6f56f6a205e4b42af9feeb0510365bf : SUCCESS in 1h 30m 39s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f240d1d1db0d4b4f8840511cafd04739 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"632bdd58cb93e8407ad5ca730ee2a080fd7ac92f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-26 19:20:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2de9fc81db9de9dab8a3b58b4472598cc487688f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 19:38:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3424726d935141c78fa46eb15ccdeed3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3fa4abd95aac4be0bf16bbb976c2a3b2 : SUCCESS in 10m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/46409241008a4380b167f72350facd46 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a5f69c5da9834bdbab84c8b432477bef : SUCCESS in 15m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"33684b4a89150ff3865b4f12b65e85df0af49b4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 21:17:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f56726538f241278140f6680e58967a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4241e6f62143441b9f3ca4ca1fa146b1 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9995f5636b544a99ce475ff66325a0a : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a9ec370a8144ce09dfa21ee90839b6d : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/82b74703905341a1902a300015df1fe6 : SUCCESS in 5m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f81f96995904fa2941985c29de94a76 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00c7f515522d41839cc34b478b246ec6 : SUCCESS in 5m 04s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/9b3b061a936b446e8b7aaa88e12a5740 : SUCCESS in 50m 09s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/fd1908d1fedd484583d202715bcecd61 : SUCCESS in 45m 48s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/2530b5ea44824163a391fa96c3a94868 : SUCCESS in 1h 46m 19s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/51d58ab0b61c43e39c27bc71d578ff2f : SUCCESS in 1h 37m 39s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/0bbea4238c274c6683a6084cd043d10c : SUCCESS in 1h 55m 52s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6d0d2330ad3a46cda98f0e7d5030c879 : SUCCESS in 42m 22s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/29dfcbd0869f44e1b84f01a97308ddc4 : SUCCESS in 1h 45m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1ea7f258bb53497489efab63e554e149 : SUCCESS in 1h 29m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6d5c72d9c8ba4ac892fb8737e649568e : SUCCESS in 1h 41m 12s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/e510f8da8c71426d96e052ca81070eff : SUCCESS in 48m 09s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/6a31bb9962064c76945aa4014883f7e6 : SUCCESS in 1h 42m 26s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/10196d515dc24759b03bf67cfd6455e8 : SUCCESS in 1h 26m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/c915a87cc3ce42e5b7cea7208752d41f : SUCCESS in 1h 36m 23s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b80ce4f22d064322b24c79ac1c265044 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":2},{"id":"ec4b808a3d14f25da224b46e5aa583f670b3f231","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-04-27 02:21:09.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"eff5c733058aef356372b190befb93d80ec13af8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-27 13:07:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b82b3efa5450b9243684f8ea33b5471b0aad664e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-04-28 02:06:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c3908aeea8e8feaf126dd03c1d82524bac22923e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-28 21:18:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"70fb1e93c1374e15ab65ddac249cc906510a2cb0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 09:38:49.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"584a2ad6f847be3e5dc65c74ff49f8cb61fded75","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 09:40:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"61eb5a32ab6fd9adc2cecdb3dc46d62bc79bbce3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 09:57:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c7a56bd6ea2d3bf1d0b0ad00c421def86b2211ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 10:01:24.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d78ac58c9c28b522193cff3cc3515d9fd6fc31b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 10:03:50.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"575483d0e9acc2f9c89cf541f9de5a449cba50c5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 10:13:18.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"83bfb3081f365878d4da04a5384f4df8befe9b57","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 10:15:14.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9ec21dcba9ecbe96550704ec24bc185a63aad9fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 10:17:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7af75feec00f40a1831673ea9ea3d131\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3095e6cb3ee94e73a21b8de77b85a9a2 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4c6475be6a1646bb87cfbb23ff6afb15 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e3dd449ca23247859b9cd7ed63dfd705 : SUCCESS in 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f271af77e13f621d0e7a2ae5c6c2e53e5f3b228f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 10:25:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c2d6cbea83728d2ee0186826dfb10dc4b480b776","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 12:22:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8bb58950151d90c94bd3cff4965b72b80b5aae61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 12:23:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26724614322747958a11bbb79a95664b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9120ecc0b6d64a06959f3ca13fb2e6f1 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb1804f6d7ce45bb829089c46be16b22 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a13cd41d18c34068b410b4e577466f0b : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8a8d753726e4a4cb7c8c2eafe043bdb : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bdabd1a444334cdc8ccdd81c3b7fde77 : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/007eb5e461e844d6a11132a79fbae95e : SUCCESS in 5m 19s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/7115a88aa7f94de2bcbeea3e7c355f45 : SUCCESS in 25m 03s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/a5e2359b44df41bd80c5d59e7113a072 : SUCCESS in 43m 54s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/b155cb09fbbf445fb96650f50564fed8 : SUCCESS in 1h 45m 53s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/141cb45cb53f440e91274c62578285f2 : SUCCESS in 1h 30m 38s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/7c7d6cbe20da4b53b3cab452ed39218c : SUCCESS in 2h 18m 29s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f641d9e3e3ed429ba70262c19a5ab395 : SUCCESS in 45m 18s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fd4016f7298441ffbc58eee9943dd3cf : SUCCESS in 2h 08m 08s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4a874d60a30544b4bb21924952760283 : SUCCESS in 1h 20m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/441c66e8f8b64bd5bbb9b940cd925486 : SUCCESS in 1h 52m 09s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/554a5b73470645d08150705a4babce2b : SUCCESS in 48m 43s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/cbadf923f6c64bb29bf3af3cc0eb822f : SUCCESS in 1h 34m 39s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/e0b79a34849c4d10b2ba36ea31f4db8c : SUCCESS in 1h 26m 53s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/23389de7cb9548d1bb558bb44d2db943 : SUCCESS in 1h 36m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aadeca93a8854080988ed50eca0199e4 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":4},{"id":"dd6dae9e51b28c8dff9b4ea83700bf454d2f1ea1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-01 16:39:18.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"bad36d33c4b756d49d120e763ee878842262f461","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-09 07:59:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"279eb9b97c64115f035f1de7eba2ee297c2ef47d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-05-09 09:03:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6960d9287df1646153361559b6b60c640cc0caa","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-05-15 12:42:47.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1485abb623210b8b17c2828c153686d1b20d4177","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 12:43:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0c80f9d868db6e44a5032ef0f6c395a802a89b4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 13:54:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/686c51ce4dea4673b0e7c3c63ff0151e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/661e52e8aef94d87b3dd5ab600f1b43b : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e93515671c34f86b147552ff9b63dfe : SUCCESS in 5m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a2ffcaa289864d26920683dd906d9c9a : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b12a1f143a434b80b680fa3727df8b57 : SUCCESS in 7m 16s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/b782c165f9194a1a9ac15481435fde5e : SUCCESS in 1h 02m 35s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/44ab58d59a7c403c9441223a184806ea : SUCCESS in 24m 56s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/82a9741b97c04b8bb38c858d1c0993fd : SUCCESS in 1h 10m 27s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/72b11fb0a8d24cc18d80bac6320a7c4f : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":4},{"id":"5cadd9dc91a4e20c55c10983a258c491861fb7c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 13:54:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"9edb5056fb2159314997153b79c841e6a962af2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 13:57:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/499626ecb9034c228c4aff05e3ece65f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c012cf79ff1b460aaa94bed9518d1eaa : SUCCESS in 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"76cd5a7bf941f22da8395b1cbc9e3ac7c0607758","revisions":{"1ec89c281976dcbe94b3789ceb63a426e7029e4e":{"kind":"REWORK","_number":1,"created":"2025-04-26 10:10:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/82/948282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/948282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/948282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/948282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/948282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/948282/1"}}},"commit":{"parents":[{"commit":"7ddb01cc86e4daee623e81e05ec0034c51bb81f5","subject":"Merge \"Fix trove install failed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ddb01cc86e4daee623e81e05ec0034c51bb81f5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-04-26 10:10:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-04-26 10:10:27.000000000","tz":120},"subject":"Add build of modern Ubuntu distributions","message":"Add build of modern Ubuntu distributions\n\nWith world moving forward, we need to assure that more modern\ntools and images are going to be built for DB Backups.\n\nFor that we are starting with allowing to provide a different OS\nfor the Dockerfile as well as add OS version to the promote\njobs.\n\nThis should be the first step before moving forward with adding more\njobs/images  on modern distributions.\n\nChange-Id: I7b4fc6ed8771f119eaac19d8643e78d2a36bb02f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1ec89c281976dcbe94b3789ceb63a426e7029e4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1ec89c281976dcbe94b3789ceb63a426e7029e4e"}]},"branch":"refs/heads/master"},"fdf3b9dc319de1e5b00a323b42e8a5901ef3b6ab":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-04-26 10:34:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/82/948282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/948282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/948282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/948282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/948282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/948282/2"}}},"commit":{"parents":[{"commit":"7ddb01cc86e4daee623e81e05ec0034c51bb81f5","subject":"Merge \"Fix trove install failed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ddb01cc86e4daee623e81e05ec0034c51bb81f5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-04-26 10:10:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-04-26 10:34:09.000000000","tz":0},"subject":"Add build of modern Ubuntu distributions","message":"Add build of modern Ubuntu distributions\n\nWith world moving forward, we need to assure that more modern\ntools and images are going to be built for DB Backups.\n\nFor that we are starting with allowing to provide a different OS\nfor the Dockerfile as well as add OS version to the promote\njobs.\n\nThis should be the first step before moving forward with adding more\njobs/images  on modern distributions.\nChange preserves current behaviour, as new ubuntu images need to be\npublished before proceeding further.\n\nChange-Id: I7b4fc6ed8771f119eaac19d8643e78d2a36bb02f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fdf3b9dc319de1e5b00a323b42e8a5901ef3b6ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fdf3b9dc319de1e5b00a323b42e8a5901ef3b6ab"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f19c95859cc3df333d6a28270fc7ad028831b0b5":{"kind":"REWORK","_number":3,"created":"2025-05-01 10:01:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/82/948282/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/948282/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/948282/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/948282/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/948282/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/948282/3"}}},"commit":{"parents":[{"commit":"3ba7bb3efe132612e534c5aaa105a7f4bdf35ad7","subject":"Update backup image maintainer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ba7bb3efe132612e534c5aaa105a7f4bdf35ad7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-04-26 10:10:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 10:01:03.000000000","tz":120},"subject":"Add build of modern Ubuntu distributions","message":"Add build of modern Ubuntu distributions\n\nWith world moving forward, we need to assure that more modern\ntools and images are going to be built for DB Backups.\n\nFor that we are starting with allowing to provide a different OS\nfor the Dockerfile as well as add OS version to the promote\njobs.\n\nThis should be the first step before moving forward with adding more\njobs/images  on modern distributions.\nChange preserves current behaviour, as new ubuntu images need to be\npublished before proceeding further.\n\nChange-Id: I7b4fc6ed8771f119eaac19d8643e78d2a36bb02f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f19c95859cc3df333d6a28270fc7ad028831b0b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f19c95859cc3df333d6a28270fc7ad028831b0b5"}]},"branch":"refs/heads/master"},"76cd5a7bf941f22da8395b1cbc9e3ac7c0607758":{"kind":"REWORK","_number":4,"created":"2025-05-01 10:03:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/82/948282/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/948282/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/948282/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/948282/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/948282/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/948282/4"}}},"commit":{"parents":[{"commit":"3ba7bb3efe132612e534c5aaa105a7f4bdf35ad7","subject":"Update backup image maintainer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ba7bb3efe132612e534c5aaa105a7f4bdf35ad7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-04-26 10:10:27.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 10:03:50.000000000","tz":0},"subject":"Add build of modern Ubuntu distributions","message":"Add build of modern Ubuntu distributions\n\nWith world moving forward, we need to assure that more modern\ntools and images are going to be built for DB Backups.\n\nFor that we are starting with allowing to provide a different OS\nfor the Dockerfile as well as add OS version to the promote\njobs.\n\nThis should be the first step before moving forward with adding more\njobs/images  on modern distributions.\nChange preserves current behaviour, as new ubuntu images need to be\npublished before proceeding further.\n\nChange-Id: I7b4fc6ed8771f119eaac19d8643e78d2a36bb02f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76cd5a7bf941f22da8395b1cbc9e3ac7c0607758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76cd5a7bf941f22da8395b1cbc9e3ac7c0607758"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
