)]}'
{"id":"openstack%2Ftrove~948284","triplet_id":"openstack%2Ftrove~master~Ia1167890dc3890b908125ee300068e1e1397475a","project":"openstack/trove","branch":"master","attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-05-23 09:25:24.000000000","reason":"A robot voted negatively on a label"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-04-09 22:05:49.000000000","reason":"Reviewer was added"}},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-06-23 17:24:11.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}},"hashtags":[],"change_id":"Ia1167890dc3890b908125ee300068e1e1397475a","subject":"Remove deprecated get_backup_image overide for MySQL","status":"NEW","created":"2025-04-26 12:20:34.000000000","updated":"2026-06-23 17:24:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"42f522dbdcbe67541262d72a2b60f61c6c0e66c1","_number":948284,"virtual_id_number":948284,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-08 12:23:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-26 12:39:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-26 14:16:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-28 02:22:24.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2026-04-07 20:34:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-04-09 22:05:49.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2026-04-09 22:06:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"5aecf64e229454c44f947b8c3c1a379c74991a09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-26 12:20:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"545512a7c3b71d81a1a55a67ac5c2c0b81fbe2b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-26 12:20:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"91a5a8e79a4dad1d8dbac8124a31d5ac5e998afd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 12:39:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07f8531ce2e8476c8e5b3da2bb6cf6ab\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/67b26847c9314f1ca7f8bcb6ef480f80 : FAILURE in 11m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/11c0eaeae61a48d4b658b2ed986b1618 : FAILURE in 9m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e0c2cf9c8452403199f4ee119e207cee : FAILURE in 16m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa6eac84ed1267180e5a725800203fb4cb8db5bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 14:16:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1d4ca844ce244a8b488ad886dfed1b7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9d9fe40f71c4a2dbd082afdb1654f40 : FAILURE in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e78b562b36b544db9fd5c0ddfd9aa327 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23583c38829249bbbf91f9a9735e9bb3 : FAILURE in 6m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88f967a082a6415b88d4338cf95afa89 : FAILURE in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd27c836ce334f1aafe7ba81927079aa : FAILURE in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b8c6674fbf40988978ce825feb94ae : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ec73f939f55454399a33b0b63724d4c : SUCCESS in 6m 26s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/981444a1468e44cbac45b417c68ed1b0 : SUCCESS in 45m 21s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/383b0b5265904ca7b0d5c0457b8914ef : SUCCESS in 45m 55s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/655ca226e2334daf8b13fe32306d1b09 : SUCCESS in 1h 50m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/c585674752934114b6acf8b142e3d271 : SUCCESS in 47m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/dadf5f0ec4eb4c10a81c319e39a3dabd : SUCCESS in 1h 45m 31s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f7b706e807b744c0aa78a8e464f39282 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"e580d3004a8826901d73e9c2a8d4952c17826ee6","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-04-28 02:22:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"194510c19ac06d8e45b1d32258300425d03e7fb4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 09:46:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a2840a9fca33c3653bb935e75a74b7e036940d76","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 09:52:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"007794ef5080fb467871f33d19ee3cfb6c2a0def","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 09:54:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"adccf01e65545f4f3922fc6b8aebd4bc7a441cfc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 10:10:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2643b4c948754e4484c0c5c22826deba\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/95782465c7704845a6fbd2903f3e8cf9 : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4144c47a77ef4d7a95cdf353c2255d66 : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9072492075224403986905b7de939243 : SUCCESS in 13m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0e681a95dd663edb1d385025a1b0928f7c54ce2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 11:53:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb6460edad3946828aeb294a3ddd5658\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d8cf7da24ee4bbb9386401583223d3c : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/720c1021d3434f6c91a676f1fa4bbdb1 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92cdc09049824a98bfb56e05cfa7e4e1 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/83b2184ef3374578b029d8932c9fccaf : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e300ca5445bc4c4d9dc9a4c351260981 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/879b347879804893bcad543ff1cb95f6 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1eb39df2f8b5461eb869063d41b82a79 : SUCCESS in 4m 23s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/d1c81050bdea4750a09524abc462296e : SUCCESS in 51m 05s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/658d50f0d91a41e9b5864ec801eb0e06 : SUCCESS in 48m 54s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/0f9ea77b61f042d090a3d1c9b680e3e1 : FAILURE in 1h 32m 28s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/53dbe8ad4a5c44d8aa04317436403215 : FAILURE in 1h 10m 09s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/ca29ccf8ef004336bd50805c1159e49a : SUCCESS in 1h 57m 25s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f4f48e954e054512af8b271d59d496c6 : FAILURE in 24m 11s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/06d13fa08be544e29e92758579256956 : SUCCESS in 1h 51m 27s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7e798770ae3e400ab091973f12b3e580 : FAILURE in 15m 39s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8ba69c716e54456f92389675425bff41 : SUCCESS in 1h 44m 57s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/89064371a4e641dab51e77dfc563a8c0 : SUCCESS in 26m 32s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/750140ecb0ae4cc4ae3e2044f9a331bc : SUCCESS in 1h 46m 02s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/3601342e624942b3af1ebee56d33780b : FAILURE in 32m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/db28b461773b4c9a890e359d8f4aeb31 : FAILURE in 47m 05s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/951718fae3134713a0feedb0e926e111 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":3},{"id":"0c64ca78c684307a9d7e0b8165b51bcda2a71df2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 12:17:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3d6b401357f163bc68721d2bbeaadfbebffb71e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 12:33:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f51dff060a746ab9976ecae44c35a02\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5518d684f8d94fefb8efd826da40a7f2 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5b6939d40ded4912874875fa8977f756 : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/372d6a1167e34198a9dce95699e61791 : SUCCESS in 14m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"382dd36ae378c2a4fad8d8b1cf9c7f0f814c2dd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 14:12:51.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5147469e24f444cbc4a6531ed2cf9b5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbc76cc34e8d4ff0b903cfb7f9745913 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37a0eb648cb94da6a01f4b6513328025 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e3ee920657b43cfa5902207d7ed2996 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ca517a3d8154d4f9ce77f3f1c4280aa : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ecc6b06981b488d84506cea8ae7d5a2 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfd20afbb7fe4741b52e3de77695c18f : SUCCESS in 2m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/779fae9b171d4d25ad7c97eb3efbc792 : SUCCESS in 2m 45s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/46f18a6922db4361979fa90ee4fe349c : SUCCESS in 25m 56s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/5a8a45b6ee614644bf2e6bdf697c1420 : FAILURE in 23m 48s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/1a554b4255b646ff8025d4737b34baf5 : FAILURE in 31m 16s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/b31be0de8e274a329c90ed3b32c97f01 : SUCCESS in 48m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/bf51d16b18974bc7aefd29c452bf5134 : SUCCESS in 1h 54m 40s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/64ef42fe9e54431b9d06264ed7bcad05 : SUCCESS in 47m 09s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ef7c9a4c0b844ce2bff2c3cfd9eb5e46 : SUCCESS in 1h 47m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/22d86cc665c746d496f9a015f520cf05 : SUCCESS in 1h 25m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8ea5129b89a44af590122d3bed5d1534 : SUCCESS in 1h 39m 11s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/367648ac74f8494588fc2a9df12cdf23 : SUCCESS in 51m 55s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/eb350e2c08d54c1a9bba8507d3ca8965 : SUCCESS in 1h 31m 14s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/4a5ab5c93d4f458aa8aa5cca97989e1d : SUCCESS in 1h 26m 52s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/0217698303f74ae88e542a175ad79b68 : SUCCESS in 1h 38m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e021e12b7efc428395dea5f7c7ee3dd1 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":3},{"id":"9b3344be95bfe229379a11305d3cebeb8e928883","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 14:43:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"83ae331ea98b5edf94332bd115b52d0cd61f0a91","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 14:45:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fd35cf5436b94ecb69c3dec9994fcb7d976cec91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 15:00:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d09edc76fa2d403a9497acb0a82f6e03\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2931205a6e5e446d8f9f1826b4ce6cb5 : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c0bdc59a5c7d43b6b8f52201204f52f8 : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e05b4ebc260c4170a6301ab047409c78 : SUCCESS in 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4bffc4ff183b4503c643084ad8721d459531294b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 16:25:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"821de471b8a15f5ac124cf8d1b350d013fa6711c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 16:37:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8d03899efb04199aee2336b4277121d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/37c34645ce4947c68de25aa7a0a327a6 : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/53258a158c7449f6b151faf52e0c66db : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c9fbf72beb1f48018f40a8967adb363f : SUCCESS in 9m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8c61dca2e314a01df5f70f70e4b0f654f0fc17a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 16:52:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9b3bb4091bf420fb8addb033948c158\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/813a291f34cd41b1b1754eb8f094057b : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1d3e8360703492aa710a854e8053f18 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a928e712412342daa8f347cc7a3a56ad : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ae15aaf453e4979b16151a5f3924bed : SUCCESS in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b15ab9f9b1ca488a8765bf5f10bac476 : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04fe9cb270954ce4aa55cadd025930c1 : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b221cf6ab5b438cbe71753e194954cf : SUCCESS in 6m 19s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/4e6cb5226a484bc0b895b4d6b981edfe : SUCCESS in 25m 06s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/d177d33ae21047c19acbf3f795582bdf : SUCCESS in 56m 57s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/ef0289403deb4363bcbcb0817d33da79 : SUCCESS in 1h 55m 48s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/682c965595994c648ee4343a0d3f210c : SUCCESS in 47m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/05e4aaffa42b49adbb9b7013d86f8f35 : SUCCESS in 2h 08m 30s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7719aa658cfd4cc4983fc9a5cc83df61 : SUCCESS in 23m 52s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/437b05b78d0745c798df0085387b646e : SUCCESS in 1h 41m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fc5b160d91074b5db74456a244e2b102 : SUCCESS in 1h 25m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/68c539482db940d39e1d9dcd8df27be8 : FAILURE in 40m 15s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/04da3cde9f2c486994b5686e2f7b709a : SUCCESS in 1h 01m 54s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/3cc441a499ee4dfbacc3cba9b192bc4c : SUCCESS in 1h 37m 29s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/feea4cc3506547338de5aa33a5f2b6bd : SUCCESS in 1h 14m 00s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/191aeccd42d74f64b53683359bd86092 : SUCCESS in 1h 35m 37s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cdb085b9c9114c64b69cb3a97a71c7e4 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":3},{"id":"62a315f573ddee1930cbb7446001988ddaf4c839","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 17:39:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e7d4b8c65a955ec778c7e4fe4ebbb37e8d884d4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 17:53:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91473faa9d8540fbb105ad1831ff83da\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7e89c520234144968027af18ed9a677e : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/56ccd85500d44d5b939f91cd30111122 : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8dafa9c26a964550a2930ecccb822571 : SUCCESS in 12m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bf69d8d77f4f304b1c35d87d4c6b91d37ba9086e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 19:38:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f4df36977f444eaad42c06318dd7d1d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c3b61e2ddfe4f578b11a84dd4bb193d : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82f7f30aba2e466c84911db184017938 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54d2c0687b1748d1a734518cf3f276eb : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e6f7e34e4fc247758ed51fcde8799ce1 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a186ccaf8634d2da4d23d0e2b5fc542 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f7aff7225794803b23b24f40073bc73 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51a04281066c419eb949addb861e7407 : SUCCESS in 2m 47s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/07b4859ba01045e588883d4571ed362a : SUCCESS in 24m 40s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/5e0d2db6c2b7497fb500524582cc03c0 : FAILURE in 22m 58s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/6d2955520353482cbaf7577d64522a55 : SUCCESS in 56m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/8d9d3450d1164d2a9398589646564967 : SUCCESS in 1h 37m 55s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/0114fa55e1da41e9a056c1506f4563c5 : SUCCESS in 1h 58m 16s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/649ae088a7514d6293c2f7efca7debff : SUCCESS in 44m 10s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b1bbdeae474149279fbe6b39546d556d : SUCCESS in 1h 54m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/760d8d3ce6ee4a33a4513d503bb63a5b : SUCCESS in 1h 30m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/29440000df66433a8f8367c0fc539a3a : SUCCESS in 1h 22m 28s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/3580df9c926a4f999e3291a640a9d5a5 : SUCCESS in 43m 38s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/dcbfae3b763144bd98b7b5d7a34a17e9 : SUCCESS in 1h 45m 24s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/2fd86a13a4b1415993bdfb699e4727fa : FAILURE in 1h 08m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/cfd545fe5487428885bd0a810bcbbf7c : SUCCESS in 1h 37m 46s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4a3774ba39ff40789799630530bd1384 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":3},{"id":"f052aba71db496aa005ce9d345ff4db5bb7ebe8b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-04 06:16:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"446921827dc14f410787ea36ef6eec0251c035f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 06:32:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6ef1d0c1cd04be39b21cf6e27c3c735\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/01ac001d68c141eaa6e2fe1635834aef : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/93b07786ca464041bba2e73df56c3aa0 : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/263feb5748ca4eaf9bd756d5581a2a7b : SUCCESS in 13m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b72ea6f42b04f37d5800e495c12e5a7f73b9dddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 06:52:29.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/264a0f8025cd4ea4af70927ad1ce53f3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/908072c53e2345eab17d2dc7ff8e0b8e : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/037650d1bc4b4a3395b1fd53e44399a8 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/105d627e7f834a718b423fbd5e2f4b0f : SUCCESS in 5m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3654b10eb29948d7915583510021e12f : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80d054ef68cf493c8c6a4ee8a2e6e13d : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840d89953a984567a0b5619a5b48801a : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ca46ac0403e455f9669717185b91a16 : SUCCESS in 5m 38s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/8a6208f66da442029290b6dc4ca468a5 : FAILURE in 6m 31s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/513b81b7125c4255bb7ed414308c0996 : FAILURE in 6m 37s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/2c328ecaaf514c4cad42c2400a315307 : FAILURE in 7m 25s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/abd264f240b64145969ec7c4eab6bd74 : FAILURE in 6m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/fa2475f4b6654f7586a327d4d3af63be : FAILURE in 7m 43s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/368f9d08055744d5a3bf8277cd8febe4 : FAILURE in 6m 27s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7ce9321cc8aa40bb868fdb82b4f3ae2d : FAILURE in 6m 24s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fd699b151ed449dd8653781dcc8455ce : FAILURE in 6m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/981cc03bf4174cfe92160207d950be8f : FAILURE in 6m 30s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/cc94900d99854e84a027ae43975a568d : FAILURE in 8m 29s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/09ff3cd8b7234f14a948df5cb4912dfa : FAILURE in 6m 53s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/a4511251d88c410d8a52babeb2836327 : FAILURE in 8m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/a5619c4361994b99b0acae596c01e8bd : FAILURE in 6m 30s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0b7d379e9b004016ac04f5d0be8eebbb : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"b1c1941105176ccf4edec8cd86bead0b5ad25353","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-04 10:33:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"92a0d77f05828c89f3773ca2e89f6c3fbd195558","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-23 07:30:08.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ae7377a6b3ea7658857a966284c877b161aa6a31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 07:48:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0279d9e08a0948feb7575b6c5c517b1c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9fc25dc38be946ba940712f23d0f2c60 : SUCCESS in 13m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e0975461e87443b7ba9ff1593817fdb1 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dda359460fd240779cffd3da75fd852a : SUCCESS in 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d9bba916155bdf35b49f2b05e20f12754fcbe4a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 09:25:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c071b759449414fa9a95bf85416b11f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85a6fa6a02524be2a626ee6bcfaba1a2 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b17af075250448cd96b7251c7b2a6d28 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ec1d2531bc64f43af32cf4d7391b31e : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5420d098cc3c4c02a295824a1b10ac36 : SUCCESS in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5808a0ed3a348c5b71ddd05759de8de : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e15b04fe31334c76bb6ea16aceab2c8d : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7e2de26f749468bad81d1f9501f10a1 : SUCCESS in 5m 23s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/8ece45baaef4433584b80e1f5c6d3e3d : SUCCESS in 58m 03s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/3087d2de8b7246bfa7ed5e9180814606 : SUCCESS in 1h 07m 59s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/6082fe262d924532b7e67750c57b3701 : FAILURE in 30m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/6077ab3459ca4a7aa0482f77ebf13804 : FAILURE in 37m 55s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/45da54eb632c48b1868fc8e1724cf85b : SUCCESS in 1h 36m 00s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b8bf5ad5d9164338accc598b885142a0 : FAILURE in 47m 27s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/59831c0f9c7f42ea8486995eef385585 : FAILURE in 42m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/68c2bfc566e840fcbd352e5de767e56b : FAILURE in 29m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6c2a086c3b1e479fb312599cb5f2f641 : SUCCESS in 1h 25m 13s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/9cdf93e353824e47b5d29e809cc51e24 : SUCCESS in 1h 05m 30s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/4d60a6f4e7ea457f8881bd8429e0964b : FAILURE in 41m 16s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/edefadd17b0e40da8b8bc7be3e7ac7f1 : FAILURE in 30m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/1aa4def25a724149ab036352c0adf10c : SUCCESS in 1h 52m 22s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/68c347bd1f4e43dfb7720e4e6fe424b6 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":4},{"id":"692d76690687dad74d9b14b7bc26b046a42207e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-07 08:54:17.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"b9ec0021bb12e0070c41178785bc6a32547108cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-07 08:54:20.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"2ae0c4c6716d08d5438ffdb9daa9e4de1d3be7ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 09:05:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f4102603cf0417fb5eb8d8f390d887f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f8b7cfb1c3bd4511931c3a23f5a94b46 : FAILURE in 6m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/829f689eeaf247948415427af94f0bc4 : FAILURE in 5m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4ceee6497e494b8d911c6ff20266ce3c : FAILURE in 9m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5a8097100fde34426a29f9a23cdd52331b6a965a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 10:47:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e659e94a6dda43438f910b9ab86bb930\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0ae3c5644134a8588b7dadef23417c6 : FAILURE in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b10fb490524a8dadf8f9d3b21ee369 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/477e0e39dad74bb2aec7e34e5b5e680f : FAILURE in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/977ce68df96842e49b91529123a62676 : FAILURE in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7f4977c7565e433da98851484e927d3b : FAILURE in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9280c66849844b6a68043d841ba02f0 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/926f38be83c248679e806fe494cd5f26 : SUCCESS in 5m 12s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/2bd7bbb98b2a4408b9667178d9bf84f1 : SUCCESS in 47m 25s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/71fc5fd1a2064eb781ef09933725675b : SUCCESS in 1h 50m 13s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/fbecf7e0625f47e0a972cd8e8a1c939c : SUCCESS in 1h 26m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/dd52fae392b84844bc3cc77204ba132f : SUCCESS in 1h 45m 27s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/70cbd21cf46f42208f7b2e955bd0cc35 : SUCCESS in 42m 20s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/83e09f8bf0244683807ccff9beb39888 : SUCCESS in 1h 51m 24s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/80271f09553e47379320b6094371819e : SUCCESS in 1h 21m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c528c20ffed445f6805d7c479047044b : SUCCESS in 1h 29m 36s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/36265c57e9f747448e3afc71b3040a81 : SUCCESS in 45m 43s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/4baa7b2e6ebf458ab9e2d46ccf25aed2 : SUCCESS in 1h 29m 09s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/46af99b7b3244ccd90ed3b64830e153c : SUCCESS in 1h 13m 40s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/28ee6074184044a988f3d2e4fc2e312e : SUCCESS in 1h 18m 35s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b837ac8040c1403ea9947145d898f64f : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":6},{"id":"5104d308c502c8fb6076652abe78cfbca058e914","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-07 20:34:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7fe257cf19617141f67f578c3328b28a27898a1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 21:16:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13e94b5e989c458f8bfe0268e7fc2223\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f5d11cd3d5f04b8898516333fc4b7f79 : FAILURE in 7m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bbd0b9529c24479ba6963ced7996d476 : FAILURE in 7m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4034c7c33c5a414ead9fe6c54e281896 : FAILURE in 10m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7a9f0c7b5f641f076da276b9e5a8083a9a8efc79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 22:21:01.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/886b3911a3254d3bbcd80910f42f7733\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/263947d3f4d54baf989355a332d5dcd1 : FAILURE in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b52a186f4dd5413a9ec731b424cc89c7 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a7d38f4919d455fa2e214d198554a9a : FAILURE in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ccc5741d92424e77bebe68c4191159dd : FAILURE in 4m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/33bf95721c2e417a8ebbdb8f3146dd91 : FAILURE in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4181b66249e431c893e9d654f6f80dd : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07005ffbe6754024b7ef8707407153fc : SUCCESS in 5m 20s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/eb3c29e1514c469aa846b615971bddb2 : SUCCESS in 48m 02s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/13a5e6832a384a32bbd0b29c02778d84 : SUCCESS in 1h 44m 38s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/40f310fda6ea44d8a6ceaacdeedd6108 : SUCCESS in 1h 29m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/41e0da5db0f94868868bb572c96bbbc1 : SUCCESS in 1h 41m 54s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/338171a03c8b48098d9a93861ee59f95 : SUCCESS in 43m 30s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/6d8c7c2a1b8b4f9a831b98c22bf7e1c3 : SUCCESS in 1h 35m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/2a0723cf34f34c49b89d06b3ec94a03d : SUCCESS in 1h 19m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/9600e2860c0843a085d01e3bd6566b88 : SUCCESS in 1h 31m 31s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/93468a14fc6a48c6b56f9193c4c6de78 : SUCCESS in 47m 19s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/b2fecbaadf894529b9fee56817c86ba6 : SUCCESS in 1h 28m 56s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/4aeb5b22f7734ac0a8e1c4da90c80728 : SUCCESS in 1h 13m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/6933e0d9ddbf45a8a47e9d4b26dc83a5 : SUCCESS in 1h 18m 57s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ad521b1e20ea4d90a3d6ea421b290aae : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":6},{"id":"e725df4509425e38f68fb2b253fb3c229993cc43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-08 10:11:14.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"ff52c45bcd6fba02e357b067391987b2200fe229","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 10:38:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8c2c4d3675646cb9922e8f563d43f1f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/017e97d8182346dd9f9fff95cc74690b : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3f54c8240d314a3fb298de3f2f8be338 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0406345aef7a4b8b8107b9824de626b5 : SUCCESS in 9m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c58db41628d5e7cfa1ed358ea6993a281e6d64c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 12:23:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dad25b2e7c3d45bdb0ba6d5091c8ac39\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9750e3940e1b4837886c2887293b1bfe : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7603d4d76ce41898707763227dc5eb2 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a45b6212dab746979594905abe4ec37f : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f432f01aad7144ac9ce21ee13d474011 : SUCCESS in 2m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0e3eccf6290f42ff8d3e9c72638e1a86 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20d4adcb202145febcf9a420ecf20ad5 : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09ce7779d79843cba17360c921cd68f4 : SUCCESS in 4m 06s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/b5eb66cf68984f84b74497283fcb47c0 : SUCCESS in 51m 26s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/2056bd4f46c1478ab534560adb473a73 : SUCCESS in 1h 51m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/a684b667998d4384a45ee87b0816021d : SUCCESS in 1h 53m 22s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/d57611fc979f4c088823ecd3c826bdf5 : SUCCESS in 2h 03m 51s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/3b07db8fe1414a7eaf9b77f1acf58227 : SUCCESS in 46m 20s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/ba06cba80a2f4b1a8304a2af02217cf6 : FAILURE in 42m 03s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/9db1f9a615424ead98de92abcf7ac410 : SUCCESS in 1h 38m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/eb8d526306124d39aeff36aa7885a4b8 : SUCCESS in 1h 30m 31s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/3e269b4e755041a5b405ad7c6457a567 : SUCCESS in 47m 59s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/62647522057a48018dbf16e863974b78 : SUCCESS in 1h 36m 46s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/2312d0e187894c1a8e19c35b5d3f1b24 : SUCCESS in 1h 22m 36s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/c2f8bbf67a7c483e884f28bcc1b78b5c : SUCCESS in 1h 22m 52s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b5715e19e18845ddb5005bd31773852f : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":7},{"id":"42f522dbdcbe67541262d72a2b60f61c6c0e66c1","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-06-23 17:24:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6f940bf0e2b738cd2600f0a5c73c4606b25ad56e","revisions":{"24bcdc455d27004aee6ab37dcfedd0b9112d6314":{"kind":"REWORK","_number":1,"created":"2025-04-26 12:20:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/84/948284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/948284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/948284/1"}}},"commit":{"parents":[{"commit":"1ec89c281976dcbe94b3789ceb63a426e7029e4e","subject":"Add build of modern Ubuntu distributions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1ec89c281976dcbe94b3789ceb63a426e7029e4e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-04-26 12:20:18.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-04-26 12:20:18.000000000","tz":120},"subject":"Remove deprecated get_backup_image overide for MySQL","message":"Remove deprecated get_backup_image overide for MySQL\n\nAt the moment MySQL backup image is trying to append a version\nto the image name itself, which has been deprecated by [1] in\n2024.1 cycle.\n\nAs we do not publish a MySQL images with such naming convention\nand MySQL is the only outstanding store with this, we remove\npreviously deprecated method.\n\n[1] https://review.opendev.org/c/openstack/trove/+/884165\n\nChange-Id: Ia1167890dc3890b908125ee300068e1e1397475a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/24bcdc455d27004aee6ab37dcfedd0b9112d6314"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/24bcdc455d27004aee6ab37dcfedd0b9112d6314"}]},"branch":"refs/heads/master"},"7581b12135579b255af40aa029e410783b5b3d7c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-26 12:20:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/84/948284/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/948284/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/948284/2"}}},"commit":{"parents":[{"commit":"7ddb01cc86e4daee623e81e05ec0034c51bb81f5","subject":"Merge \"Fix trove install failed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ddb01cc86e4daee623e81e05ec0034c51bb81f5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-04-26 12:20:18.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-04-26 12:20:50.000000000","tz":0},"subject":"Remove deprecated get_backup_image overide for MySQL","message":"Remove deprecated get_backup_image overide for MySQL\n\nAt the moment MySQL backup image is trying to append a version\nto the image name itself, which has been deprecated by [1] in\n2024.1 cycle.\n\nAs we do not publish a MySQL images with such naming convention\nand MySQL is the only outstanding store with this, we remove\npreviously deprecated method.\n\n[1] https://review.opendev.org/c/openstack/trove/+/884165\n\nChange-Id: Ia1167890dc3890b908125ee300068e1e1397475a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7581b12135579b255af40aa029e410783b5b3d7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7581b12135579b255af40aa029e410783b5b3d7c"}]},"branch":"refs/heads/master","description":"Rebase"},"214dd12a79bdf5aa283564d03e9b4b4f79739522":{"kind":"REWORK","_number":3,"created":"2025-05-01 09:54:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/84/948284/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/948284/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/948284/3"}}},"commit":{"parents":[{"commit":"7ddb01cc86e4daee623e81e05ec0034c51bb81f5","subject":"Merge \"Fix trove install failed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ddb01cc86e4daee623e81e05ec0034c51bb81f5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-04-26 12:20:18.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 09:54:47.000000000","tz":120},"subject":"Remove deprecated get_backup_image overide for MySQL","message":"Remove deprecated get_backup_image overide for MySQL\n\nAt the moment MySQL backup image is trying to append a version\nto the image name itself, which has been deprecated by [1] in\n2024.1 cycle.\n\nAs we do not publish a MySQL images with such naming convention\nand MySQL is the only outstanding store with this, we remove\npreviously deprecated method.\n\n[1] https://review.opendev.org/c/openstack/trove/+/884165\n\nChange-Id: Ia1167890dc3890b908125ee300068e1e1397475a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/214dd12a79bdf5aa283564d03e9b4b4f79739522"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/214dd12a79bdf5aa283564d03e9b4b4f79739522"}]},"branch":"refs/heads/master"},"fc2729b9e41d8b80a26275e5beb25e4cac55197c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-05-23 07:30:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/84/948284/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/948284/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/948284/4"}}},"commit":{"parents":[{"commit":"d5bccd7bb66adcc5f23ec10731ebb10a47cdbede","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d5bccd7bb66adcc5f23ec10731ebb10a47cdbede"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-04-26 12:20:18.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-23 07:30:08.000000000","tz":0},"subject":"Remove deprecated get_backup_image overide for MySQL","message":"Remove deprecated get_backup_image overide for MySQL\n\nAt the moment MySQL backup image is trying to append a version\nto the image name itself, which has been deprecated by [1] in\n2024.1 cycle.\n\nAs we do not publish a MySQL images with such naming convention\nand MySQL is the only outstanding store with this, we remove\npreviously deprecated method.\n\n[1] https://review.opendev.org/c/openstack/trove/+/884165\n\nChange-Id: Ia1167890dc3890b908125ee300068e1e1397475a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fc2729b9e41d8b80a26275e5beb25e4cac55197c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fc2729b9e41d8b80a26275e5beb25e4cac55197c"}]},"branch":"refs/heads/master","description":"Rebase"},"b32d2ef3dd00a05ff7cfdabd0b1ec77881f39d73":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-04-07 08:54:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/84/948284/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/948284/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/948284/5"}}},"commit":{"parents":[{"commit":"d5bccd7bb66adcc5f23ec10731ebb10a47cdbede","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d5bccd7bb66adcc5f23ec10731ebb10a47cdbede"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-04-26 12:20:18.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-04-07 08:54:17.000000000","tz":0},"subject":"Remove deprecated get_backup_image overide for MySQL","message":"Remove deprecated get_backup_image overide for MySQL\n\nAt the moment MySQL backup image is trying to append a version\nto the image name itself, which has been deprecated by [1] in\n2024.1 cycle.\n\nAs we do not publish a MySQL images with such naming convention\nand MySQL is the only outstanding store with this, we remove\npreviously deprecated method.\n\n[1] https://review.opendev.org/c/openstack/trove/+/884165\n\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\nChange-Id: Ia1167890dc3890b908125ee300068e1e1397475a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b32d2ef3dd00a05ff7cfdabd0b1ec77881f39d73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b32d2ef3dd00a05ff7cfdabd0b1ec77881f39d73"}]},"branch":"refs/heads/master","description":"Edit commit message"},"95b69d1cb71f07ae35b1682506b3a374bc7ea68f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-04-07 08:54:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/84/948284/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/948284/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/948284/6"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-04-26 12:20:18.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2026-04-07 08:54:20.000000000","tz":0},"subject":"Remove deprecated get_backup_image overide for MySQL","message":"Remove deprecated get_backup_image overide for MySQL\n\nAt the moment MySQL backup image is trying to append a version\nto the image name itself, which has been deprecated by [1] in\n2024.1 cycle.\n\nAs we do not publish a MySQL images with such naming convention\nand MySQL is the only outstanding store with this, we remove\npreviously deprecated method.\n\n[1] https://review.opendev.org/c/openstack/trove/+/884165\n\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\nChange-Id: Ia1167890dc3890b908125ee300068e1e1397475a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/95b69d1cb71f07ae35b1682506b3a374bc7ea68f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/95b69d1cb71f07ae35b1682506b3a374bc7ea68f"}]},"branch":"refs/heads/master","description":"Rebase"},"6f940bf0e2b738cd2600f0a5c73c4606b25ad56e":{"kind":"REWORK","_number":7,"created":"2026-04-08 10:11:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/84/948284/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/948284/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/948284/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/948284/7"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-04-26 12:20:18.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy@adria-cloud.com","date":"2026-04-08 10:11:08.000000000","tz":120},"subject":"Remove deprecated get_backup_image overide for MySQL","message":"Remove deprecated get_backup_image overide for MySQL\n\nAt the moment MySQL backup image is trying to append a version\nto the image name itself, which has been deprecated by [1] in\n2024.1 cycle.\n\nAs we do not publish a MySQL images with such naming convention\nand MySQL is the only outstanding store with this, we remove\npreviously deprecated method.\n\n[1] https://review.opendev.org/c/openstack/trove/+/884165\n\nSigned-off-by: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\nChange-Id: Ia1167890dc3890b908125ee300068e1e1397475a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6f940bf0e2b738cd2600f0a5c73c4606b25ad56e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6f940bf0e2b738cd2600f0a5c73c4606b25ad56e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
