)]}'
{"id":"openstack%2Ftrove~948626","triplet_id":"openstack%2Ftrove~master~Ie7c49c75b7546787be588cc1ab7f61fd36d52559","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-05-22 09:34:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie7c49c75b7546787be588cc1ab7f61fd36d52559","subject":"Switch backup build to Ubuntu 24.04","status":"MERGED","created":"2025-05-01 10:11:46.000000000","updated":"2025-05-22 09:36:20.000000000","submitted":"2025-05-22 09:34:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948626","meta_rev_id":"6bc31f42f62b00249e350c0f0b312c25fade76e1","_number":948626,"virtual_id_number":948626,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-22 09:34:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-22 08:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-22 08:32:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-01 10:27:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-01 12:30:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-21 13:04:19.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2025-05-22 08:32:14.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"a611ea6ae840139f9d46a981befbc68fdd807b42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 10:11:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"522e1c07dbcf5cc3e36e31fa73beaaf66827f664","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 10:12:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"47c0b889aee0a9b64ea1fbbb315428353087a20c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 10:12:23.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf0f1c35c174b2062b17725797d8248583a9bef6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 10:27:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba913ffedc734e47b0b073f321d9caee\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8dfcbd85e942404dbc9e2b6b1acb21ee : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b5cb725bec2b494a984b98e330f2c3e8 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cc1e98d695e54a53b063a6dd01dc3976 : SUCCESS in 13m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"41b657d9489c8d6c3c6a65b62eb59fa7d5343747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 12:30:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a15a23422ae34ce1a1b02a482fea44af\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b59a498fc9ff43c6a0f95c0f1ad3c6b4 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09fdef55ce8b48ebbf78b81bba4da0f0 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9b9bbc192f4401a8c614dbba9907487 : SUCCESS in 5m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e47b83aaaff7436691a217311dc11dec : SUCCESS in 5m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a53dbe38e9242c5a52b2160293f17e0 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f042fa17c2d34a5a85279bdb7729a937 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ca82d2750104c27b10452c1d7c837cd : SUCCESS in 4m 28s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/22069c820ea54a5faf3b29f7ad845ea4 : SUCCESS in 1h 01m 31s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/e2c1f25c319a42879ab1b2a7784a5b03 : SUCCESS in 41m 20s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/8f0cfbf874ad40608a5bd9646a44fc37 : FAILURE in 42m 03s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/3bbad6780cb24bbe8b410dedefab3e56 : FAILURE in 30m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/3b0d0e566f0b47a2898ceb76e7e59b5d : SUCCESS in 2h 11m 40s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c267e37656fc4532aa34a2be36445fac : SUCCESS in 46m 04s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1a592776a5864ab093d70f9dbcd77212 : FAILURE in 31m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/72e7326644bc4832a3bcbead4c977631 : FAILURE in 41m 17s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/da686a083dd14b8bb7c7d1153d7c80f0 : SUCCESS in 1h 42m 30s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/1f709e417dd84374b092ac90c16dc481 : SUCCESS in 53m 49s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/6efca4ef215f442992675abb5be54a82 : FAILURE in 16m 59s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/bf8c78689dfa4b56ab7737ea07839feb : FAILURE in 29m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/9166963e1bae4b479e6dc9c56c36db35 : FAILURE in 1h 33m 27s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/111bbcfa9ac342e8a67fe1a9d7364284 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"7462e2c72fd46ecdedac9751479b6126f821fff8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-15 15:06:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf59a673092a3bbf3f2b3427e6d1860dfad609ef","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-05-21 13:04:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3662d4ca098c230884cc99ff174c542bcd5a1ad0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 13:30:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0959457b0a224e95afbc721c23b4c192\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/37edd691890143eeab7685932ec35ff8 : FAILURE in 12m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eae8e3f68a994d23888135bb0a7fa734 : FAILURE in 12m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aa4b2facdf89406b87498df333d1bd2d : FAILURE in 18m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a7ab117d8a2549a3fe1133f12b15b1d77a2539f1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-21 14:17:00.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"771f7e5204c1c6edf60cabb27c053e02bc033920","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-21 14:18:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5248c9d36f8513725382c51a6855e1523fb0f83","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-21 14:19:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b7ec1d65a0db71ad3b7cf9b593e7ded24664a20d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 15:03:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9adfdc0acd24484b6a5ba3983b75f5d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b23c52494a074403a8209be4144d5265 : FAILURE in 8m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b9f547e850aa47b293646a4fa7095299 : FAILURE in 12m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fd41216b4042463797543837021a40ab : FAILURE in 20m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"10a3fd10baff00efe3acc227caff40b9c75cdd80","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-05-21 15:10:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a419f186d453cf3dfc33aa6fed07a0fc66e6a02e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 15:10:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8063fdf4cef94aa1b0dc598144d2c57e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/436f0d192b6e4791b82809fc27bf966b : FAILURE in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6b136613f384966a10e745f5306c068 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4aa7836d9e3740c494c5f5bd4737dbfa : FAILURE in 5m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2129a6d0d2ea4c2e8bd48489eb1b8bd8 : FAILURE in 6m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92051189569c407094af152adfbdb818 : FAILURE in 9m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/585f56002fb0454f833af48071f54c4e : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b989f0998ce42e494f27f8c0c68c7b3 : SUCCESS in 5m 40s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/2c63c34cdfa748e1a4b95570aaffda4b : SUCCESS in 41m 40s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/6615948e104648cabc47c11c4eeb7d18 : SUCCESS in 37m 10s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/ad222cb2706649d8bd3641f2323e0985 : FAILURE in 37m 09s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/d543d41d88c149f1a55f00d54bc2a186 : FAILURE in 36m 52s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/67c0c62ffd5a4326a81231798075dd04 : SUCCESS in 1h 35m 01s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0f76b4bb0001492ba5e889ee9d8ea9e1 : SUCCESS in 47m 36s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/383f36ac7a0e478b897f4244220cd179 : FAILURE in 30m 25s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/048dfaf1ba894ce28a4067838ac8fc5a : FAILURE in 29m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8ec6ffe108c7470aae9a58440bcfd5f3 : SUCCESS in 2h 03m 32s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/77d863824d824d569fd99b607d3d5c5e : SUCCESS in 43m 42s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/eb907a27eb2647ee89b28b89330c27b5 : FAILURE in 31m 12s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/70502570e9f44e22a036c0a09530acf3 : FAILURE in 37m 33s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/719e0613115b4c32b92e527b2e7629cd : SUCCESS in 1h 14m 58s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/555207d06b624d74b1ba8a7771644c49 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"6a0250bd6a8d42e1d7c77895b10ba012b50fc3ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"real_author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-05-22 06:04:14.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_28619\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"_revision_number":3},{"id":"71068544d37f019a31f18eb2c2209bed9bf0918b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 06:27:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ccf154c452241ac87ceebffc273bcdf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ac6e5bcc7cc4eb387b416aa356bd1d1 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bafca2050d954addb47910313535eff9 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a47873cf470c4aaeaa0ab7ac894668a6 : SUCCESS in 9m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"154035186433cf97c7637f0700363536e315efe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 08:30:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49a6be6d64c147539cb076c8bb172db5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08b800ce4e7c497f96bc617bdc7d0634 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b60c279446944dda3aed73f4721c19e : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f65c08404a84bc6a49c28c7ead0a3d3 : SUCCESS in 5m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fcbb237cf95f491dbebb9a6eeb8f630c : SUCCESS in 6m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d73853ecb8024a9f88093aab55c7d5c6 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5072e50e68cf4a8bbd97e143414f66ad : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d3e81267c24444ebc09dd5fa4c2a84f : SUCCESS in 5m 19s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/2619e3e3832e4e01b95b90f26cb307a0 : SUCCESS in 57m 25s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/030d7458e56a4e9693026a25767cca73 : SUCCESS in 43m 18s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/3115dd4a46444ede917644ab82079ec1 : FAILURE in 32m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/cfb2dc1ccd5041348eb945d2964403b9 : FAILURE in 31m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/7ae68b845a964d78af6e8ebf4bad526a : SUCCESS in 2h 21m 43s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a18475174fe04c6da103c0bc6602fa59 : SUCCESS in 55m 31s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e3c5444e13084cbe9ebe70582e8b01f7 : FAILURE in 17m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/457f97f78cae4ab5ba7d9347a850cf55 : FAILURE in 41m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bb4da863422f4e3a994341a9673234cf : SUCCESS in 1h 23m 38s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5e7b47e001d94e7894512588fd2cfdb4 : SUCCESS in 44m 31s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/8bd101a425d44b228ab8355f59d4631a : FAILURE in 40m 53s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/bf8fd345a1694c8ab4648864ea5cb458 : FAILURE in 39m 37s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/c982540816e44ae8bac4a600bded2130 : SUCCESS in 1h 54m 08s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/404759c054c34d90835365f6b77c7d8d : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":3},{"id":"8a68c66489ef2b7cd12ad90d9c3b0bd2287eaa62","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-05-22 08:32:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7163f5ba75b236f1f3705869ef0f546949ef9b60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 08:32:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"30cd909483cf60c7ec157a7a7ccad7fb1aaafa61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 09:34:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b8a7f750a294e66b70b173741ec34ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81edd7a96a044586970192a6ae2cac14 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b7062f138b24bb1b3118b5299ab833c : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ac134a084ac452da39e123dc724bfb8 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d67e689bf857403a83621351f1e44518 : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3336c2e17bb24bb8a1a4048f4e5a10ba : SUCCESS in 5m 40s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/eb692e3b9c284da89be828d589529adf : SUCCESS in 45m 01s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/4838842206cb4038901678cba271e098 : SUCCESS in 41m 31s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/922cb48f63634821b981344bdc157821 : SUCCESS in 44m 43s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/49577747d425489b918c096b35cbc00d : SUCCESS in 48m 12s","accounts_in_message":[],"_revision_number":3},{"id":"5c4a13f778ae704a48296bc036f86c3e4182a9c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 09:34:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6bc31f42f62b00249e350c0f0b312c25fade76e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 09:36:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/141249d2384d4a1297ce051792f8ca66\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4250181f9cc4df4b6b3228649d6e21a : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a548ed32c3a498b83d10ebd3eedf631 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5939477fa36dcb86cc96b8e7a82539ebbbb91551","revisions":{"b7e1e3403820f782a9af8d244abb3f0d54bbd4fb":{"kind":"REWORK","_number":1,"created":"2025-05-01 10:11:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/26/948626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/948626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/948626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/948626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/948626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/948626/1"}}},"commit":{"parents":[{"commit":"f19c95859cc3df333d6a28270fc7ad028831b0b5","subject":"Add build of modern Ubuntu distributions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f19c95859cc3df333d6a28270fc7ad028831b0b5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 10:11:37.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 10:11:37.000000000","tz":120},"subject":"Switch backup build to Ubuntu 24.04","message":"Switch backup build to Ubuntu 24.04\n\nThis effectively switching container build and promotion for backups\nfrom Ubuntu 20.04 to 24.04, which is prescribed under current PTI.\n\nChange-Id: Ie7c49c75b7546787be588cc1ab7f61fd36d52559\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b7e1e3403820f782a9af8d244abb3f0d54bbd4fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b7e1e3403820f782a9af8d244abb3f0d54bbd4fb"}]},"branch":"refs/heads/master"},"3fe13d05167e1a5a593375d64e86a76e595261f4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-01 10:12:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/26/948626/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/948626/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/948626/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/948626/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/948626/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/948626/2"}}},"commit":{"parents":[{"commit":"76cd5a7bf941f22da8395b1cbc9e3ac7c0607758","subject":"Add build of modern Ubuntu distributions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76cd5a7bf941f22da8395b1cbc9e3ac7c0607758"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 10:11:37.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 10:12:02.000000000","tz":0},"subject":"Switch backup build to Ubuntu 24.04","message":"Switch backup build to Ubuntu 24.04\n\nThis effectively switching container build and promotion for backups\nfrom Ubuntu 20.04 to 24.04, which is prescribed under current PTI.\n\nChange-Id: Ie7c49c75b7546787be588cc1ab7f61fd36d52559\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3fe13d05167e1a5a593375d64e86a76e595261f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3fe13d05167e1a5a593375d64e86a76e595261f4"}]},"branch":"refs/heads/master","description":"Rebase"},"5939477fa36dcb86cc96b8e7a82539ebbbb91551":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-05-22 06:04:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"real_uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/26/948626/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/948626/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/948626/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/948626/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/948626/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/948626/3"}}},"commit":{"parents":[{"commit":"d23ffece32fc9b029ab107266fbebad28023f3d4","subject":"Merge \"[CI]: Fix the unittest failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d23ffece32fc9b029ab107266fbebad28023f3d4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 10:11:37.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-22 06:04:14.000000000","tz":0},"subject":"Switch backup build to Ubuntu 24.04","message":"Switch backup build to Ubuntu 24.04\n\nThis effectively switching container build and promotion for backups\nfrom Ubuntu 20.04 to 24.04, which is prescribed under current PTI.\n\nChange-Id: Ie7c49c75b7546787be588cc1ab7f61fd36d52559\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5939477fa36dcb86cc96b8e7a82539ebbbb91551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5939477fa36dcb86cc96b8e7a82539ebbbb91551"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
