)]}'
{"id":"openstack%2Ftrove~948627","triplet_id":"openstack%2Ftrove~master~If622bb88d97a1d5850d6068668ea9a1d09c4cd69","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-06-30 05:56:54.000000000","reason":"Change was abandoned"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-06-30 05:56:13.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"If622bb88d97a1d5850d6068668ea9a1d09c4cd69","subject":"Stop promoting 20.04 image","status":"ABANDONED","created":"2025-05-01 10:14:42.000000000","updated":"2025-06-30 05:56:54.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"68d214a5411ef8892cce1cd670a9c0afa12eecaf","_number":948627,"virtual_id_number":948627,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-06-27 09:08:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2025-06-26 09:02:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-01 10:28:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-01 11:54:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-01 12:08:49.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-05-01 12:08:49.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"b9568bdc86222e1a53b6358fa3c060eaa9b22fdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 10:14:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65f1cc1ca5153781e1e23719255ee46e97509a24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 10:28:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7ecc1dd468b443d8a4be3dd44f7e628\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d0f75f9f4f6940b4a2166e23b49a7f81 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/10e07ea8fc5e4424b8c38a4262073e5e : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9acf80fafde6403ea208c4e5f6251c10 : SUCCESS in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a1f3306c5783ded0c7a372e2a7b9c53217e94812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 11:54:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff7afb7cbb6648f38ce86d84df81ebe1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85d751a204ca4438b8f7b3a262868662 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8daf5a547ff4ec2b9ec2eb6c5903032 : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cb91c8a92e94110a0d6c8061b178623 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e1e060615b494c41819979301ef9e048 : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/769edd60683a418ca0a42fd9bbb261b3 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ef96fece1844b0888ee93bc43ceb6dc : SUCCESS in 4m 59s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/0d954b9c0f43416c86bfa345fb6e89f7 : SUCCESS in 43m 08s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/303f6e3aca224b5697ab8504c9ea895a : SUCCESS in 40m 54s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/a955a019da854f9c8f888ef4ceb8a859 : FAILURE in 34m 46s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/e0bd074efaf94552a0ce1b6b964db354 : FAILURE in 30m 36s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/926c610b74b84eab800b03395ef31686 : SUCCESS in 1h 38m 17s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/aee7265acab94b91abd894026db92913 : SUCCESS in 42m 37s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bf332395b90b4788890768674e859c75 : FAILURE in 28m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6baeb236ca97448cb0f738beab7db66c : FAILURE in 30m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c608f8016383455fa4b0af616af7e53a : SUCCESS in 1h 23m 31s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/34da0107c0b74709a4c51574ee7ad321 : SUCCESS in 42m 56s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/2b3325c74d234293b4380fb71a2f970a : FAILURE in 30m 22s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/5acaa20b0ad843f0b9334000eae60d83 : FAILURE in 28m 44s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/8b8d8dca2c3f482585b1a8a0d22f2eef : SUCCESS in 1h 15m 39s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/841d71f08e7d4a3696e57f67a9cfc516 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0e38571060d12308fe4f9c96798e55a376a6a7c2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-01 12:08:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3d2bcfb0014406fa228cfb5d2368753dee86cdb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-22 11:28:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"20bf33615a0901925cf9fba0dc66e23fc2e7a2dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 11:53:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62d1a9ccbd664ab2a238e74718e11d74\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f49890149a4b4ec8839ca72fa85aba54 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7c4e42b7d7d244c5b8db06169d7f46b5 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e26ffded3b6e462ea6d386120e31cee3 : SUCCESS in 8m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d19a8c94ef0c781a7406d281ee67e744f629c2bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 13:35:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a783c2184494282ae632083fd97563c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6eac6077f48450484fbdc1f2baf13d9 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43ea09e809cb428ca3333b49fb3e9fb5 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/521fc8b51b434217b1a1441990e79ed2 : SUCCESS in 5m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0efdee0b020144a3b49b3614e4f1f875 : SUCCESS in 6m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d1c1b01b6af4a3982a70606a7eda970 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2193bcb214024e41965ad9f7cd24a439 : SUCCESS in 5m 00s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/cee7c3342f5948afb423c0367c0aa7e6 : FAILURE in 1h 03m 27s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/d19b0e5c00a44892afa3880f3e874e94 : SUCCESS in 1h 00m 24s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/d0af9e70822d4ca2ba40cf5eeffa5be1 : FAILURE in 18m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/e78b1528f1b04de8a0af2bb36737a725 : FAILURE in 35m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/94a78d65a0eb419e8e36a972d735a4fb : SUCCESS in 1h 31m 26s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0fb963767c9947e3abc792644045efbf : SUCCESS in 59m 34s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fe02d2c550d8453291612fff4f7ba510 : FAILURE in 34m 48s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b5f058129dfc4321aa246bbbb938a75f : FAILURE in 35m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ad8aaffda51845fa913c1e8dffa5a5be : SUCCESS in 2h 02m 58s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6817f66385a2425fa80e5ad3f48637b6 : SUCCESS in 57m 11s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/2b36b14a63a44caf9f20a8325edc7766 : FAILURE in 34m 25s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/312ad9e31d0944029d167b2dc5fbcc16 : FAILURE in 43m 36s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/937c1f27fe064a688e0e92ff3f75db4a : FAILURE in 1h 51m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a6d37357d6d4a39be8fca339db1a8d6 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":2},{"id":"dc4f7663033f103338a62f934ea243ae6fc03ba6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-28 08:34:45.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"28acc8eb490a4c41f0019013ef62ce43c5d14977","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 08:54:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e94697b89a3a4804952440ba6274a333\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49a98047815245d0832e61bb4b5dd2e6 : SUCCESS in 14m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bb022e3076d64b3299eee278b2e71e96 : SUCCESS in 13m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fc98508661fd4c0ba5308a17af13b3d0 : SUCCESS in 17m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"562d4607b5d6ce617286af9e8f408a4555f2da98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 11:08:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8f17bbf69144f4498c41467bef2558c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1495d81a7fc4c279313940a5ddc0379 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d13cb6522be0429da00fdedfb1a05b9a : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a816b545652e4593abcf3943963ca775 : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8aae23bd8f394f56b93145617691068a : SUCCESS in 7m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be7493d6b0f04e118fed6d953a1f9068 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b02ef975ccd64223a0a6fac340b10548 : SUCCESS in 6m 49s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/a5a5088f863243f9bcccd361fd54cfbb : SUCCESS in 57m 33s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/47f97ff9d7bd486da8d708075b5c167f : SUCCESS in 1h 02m 23s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/a589b9183af248f486f710365c634b89 : FAILURE in 33m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/89b3390098bd46e392c1d234188ce7a9 : FAILURE in 34m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/b98374adab6a406d8c43c78fd35955f6 : FAILURE in 2h 07m 53s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bc532854dff2411e8427e2589f236648 : SUCCESS in 48m 03s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2bced8b8481141bda1f7e6806a3d46e2 : FAILURE in 33m 01s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4744895685494c24a6ace21f2d131ea3 : FAILURE in 34m 22s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c64748aceae64c7e83d1894bba727750 : SUCCESS in 2h 30m 41s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/9ec2c7f9dd0c4d4cad53afe8d7c2ca97 : FAILURE in 51m 11s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/ee936fa36e424d56b7600b9809dd97f5 : FAILURE in 37m 53s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/e1427181c6eb4f8e91e409b452e38eab : FAILURE in 40m 50s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/d1703501229040d491a61f8e52bd0785 : SUCCESS in 2h 12m 29s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/51c21a44eeb74ea790c04992c5034577 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":3},{"id":"4a1ba151d1c0f7f84695fc978b2e0f81a18e6679","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-29 13:03:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aa4a9149a1e5dbaadba12512774b9fba2fc70df9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 13:18:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abb6dfec154d4115b42481c195c222fa\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/58fee4e37b7a489c98a5e59c73c31277 : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/79a0d62d8a224f8d8c62801c93300cfe : SUCCESS in 12m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/93a6a731ff1642ffa6d72c7635bf1e06 : SUCCESS in 12m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"62a0b9a7fb417aea5d69310b105b604b138a2c1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 14:40:42.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7573bf58492b4f1f9ba9a11e49124ec8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/197756052bcf421da025aa772eff80e0 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2c0d20ea3d94ab4bb3a942f6e70155f : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7d2c0df32a84a4fb19534fa14f2153d : SUCCESS in 8m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/82ce0f2968ea4d99b032c3f40c823657 : SUCCESS in 5m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42e18d20831f4e60a1e30d509d641cf6 : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1aec519df374873b0fea97b05d821bf : SUCCESS in 5m 25s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/9d1b772565f34e8ab54e6a078e3e9582 : FAILURE in 27m 36s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/36f7e668be264fe0ae3081f4e9904184 : FAILURE in 27m 33s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/05d631e8379d4715a5030cf2c61ede93 : FAILURE in 38m 58s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/d6a9d192507a43f3819d558e032bf8c4 : FAILURE in 22m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/46a0b40d116e471abc496c7f4419ad3c : FAILURE in 33m 38s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/58cc1d75ebe0441ca3dc9349de0a2dae : FAILURE in 22m 34s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/073c5a591b264275b35e453faae05c5a : FAILURE in 24m 12s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c5ef2275089c42f78329a0a68242d85f : FAILURE in 22m 26s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/96d02dca991c46bdaf692b26d8935669 : SUCCESS in 1h 35m 03s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/e86598bf283a46fba591c99dc4d6a63c : FAILURE in 27m 40s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/ccc01c4c4e604c7c9ee8edb017bf2d28 : FAILURE in 26m 19s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/4d618a8c923b4ebda2df08220b5861f8 : FAILURE in 28m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/46929d9937f140869871766a6a59a9ca : FAILURE in 22m 53s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4e46cf6bbfa34ad6a991371a87b28719 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":3},{"id":"74aac2e2eca4aaaf2083124ebcc7058e0a818885","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-26 09:02:32.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"336fb6891c8a9eb667bfada923a50d56c5f21868","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 10:18:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d280de46f1c745e1bd1bff630ff65320\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dcb5bb8141cd4cf491fe5ac9a730fb6a : SUCCESS in 13m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ffe194cffa724d7794ebbbe5e8834c98 : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd2b1746a9ae484381057c23a390a16b : SUCCESS in 17m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a756fc7689b68707c1255c0a30f5d5550aef9143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 11:59:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d961bbc6832f49e5a78493f1ac7ab2e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/817b7935b8d241be9347d8e4148417a3 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afcbdd6a5e7147b79284d2cf7265e146 : SUCCESS in 6m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/347f24926c084b5c9d34676b68a37c24 : SUCCESS in 5m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e82abff48e4847e7bcf2f3bc3155e815 : SUCCESS in 6m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d710d5eeb01045f4afd9ab816381e839 : SUCCESS in 10m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac7da3a0831e4df2bdfea4aeaa60521c : SUCCESS in 5m 47s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/1692d266ac5b49d0b89bcc8cfc7ff609 : SUCCESS in 1h 11m 13s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/587d407a514f416694e47e15065d9dd4 : FAILURE in 1h 01m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/413eff31b3f04970a011a2174abc2754 : SUCCESS in 49m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/3db57056b555443aa61735b17eba3aed : SUCCESS in 2h 47m 18s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/515d6e3344a44525818d8fc713aba004 : FAILURE in 27m 35s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/883a9565a446475393f434f7630397d7 : SUCCESS in 1h 40m 55s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/de38faa885d34ed395d4ed2959a980b8 : SUCCESS in 1h 40m 00s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c2d56bb2c6434e5099207afd4955ad81 : SUCCESS in 1h 35m 38s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/2063bee6451e4283a0f0d477083021f2 : SUCCESS in 1h 03m 26s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/d3c2ed17c3e640499ef09a8c99f70702 : SUCCESS in 1h 59m 59s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/99a0d3f24c134670892aecbfd09c220f : SUCCESS in 1h 17m 52s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/99484553447d456088061141545ba8cc : SUCCESS in 1h 58m 05s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3d565f6032e047c9822c7df24693e52f : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":4},{"id":"4acc8905385ea57b67f277f597794ad158746c71","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-27 07:03:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"57989caea846db2b609cc9af2d5d5f7d069352e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 07:43:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bc777c575314f9d856d025a7aef739b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/017031715e47422193ab4eb23025096d : SUCCESS in 13m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8e10025d5d084a86ac2483aa802e8f73 : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7e06cd4463624ce29a01f648b815f53d : SUCCESS in 21m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5dcff7b6bb6f22511c0e2b63802d6050661eac64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 09:08:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae818be4c15245808b7b1a5fb94eb599\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe77ed88565042eda27a190e4e860265 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/017e8c0317dc48a5a887f0092962203f : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/995f4012869b42af8f1d4b3e63a60c1d : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/364432aadf0b4f6dbebad303d5749b91 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5852e8936e54271b0ab21abd93a8a75 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e44ad4aa090a44469621cd9d65e63cde : SUCCESS in 5m 20s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/ba90e95a99b9404e93390926d267223e : SUCCESS in 33m 12s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/e8ddf433332e45af8832dd058cf871a3 : FAILURE in 1h 01m 05s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/1eb9817710df42559188e11333175104 : SUCCESS in 1h 04m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/60c312a2a0774b2c85dffc6ff4fb8b1f : SUCCESS in 1h 19m 39s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0e402e55b96c4b5ea7299f0e84bd0ce2 : SUCCESS in 38m 34s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f2a3ed82ce794cfa8a4869017f2f40f1 : FAILURE in 1h 48m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7ef146eba1a2484dafe94e51acfd9535 : SUCCESS in 1h 38m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3e75e653ba764e178dc7809252173d6b : SUCCESS in 2h 02m 22s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/4859d534cc7b4fce908e3492af43f975 : SUCCESS in 1h 05m 08s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/9a6fb447cd1e4452868f4ca6b3e32072 : SUCCESS in 1h 55m 13s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/3ef6a7bd471041799cf45a2ce18ad951 : SUCCESS in 1h 53m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/2ea358e10d17417a8f9e44cac13e1c0b : SUCCESS in 1h 59m 30s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8bacff97610e4a99aa1a5becc69c2ecd : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":4},{"id":"218bf5a8695ef28073f244c2a2bd45617491493d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-30 05:56:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"68d214a5411ef8892cce1cd670a9c0afa12eecaf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-30 05:56:54.000000000","message":"Abandoned\n\nremoved in other commit","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"282956c32ea0a6e9dcdc12310640b6016aeb3d5d","revisions":{"ba575aadde5853f90d24188e23e2a6f9a8e9657f":{"kind":"REWORK","_number":1,"created":"2025-05-01 10:14:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/27/948627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/948627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/948627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/948627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/948627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/948627/1"}}},"commit":{"parents":[{"commit":"3fe13d05167e1a5a593375d64e86a76e595261f4","subject":"Switch backup build to Ubuntu 24.04","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3fe13d05167e1a5a593375d64e86a76e595261f4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 10:14:33.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 10:14:33.000000000","tz":120},"subject":"Stop promoting 20.04 image","message":"Stop promoting 20.04 image\n\nWith backups being switched to Ubuntu 24.04 we no longer need\nto promote Ubuntu 20.04, so it can be safely removed from the list.\n\nChange-Id: If622bb88d97a1d5850d6068668ea9a1d09c4cd69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ba575aadde5853f90d24188e23e2a6f9a8e9657f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ba575aadde5853f90d24188e23e2a6f9a8e9657f"}]},"branch":"refs/heads/master"},"d1973d204755f3964836ae45a3a96ab52fd7398a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-22 11:28:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/27/948627/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/948627/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/948627/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/948627/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/948627/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/948627/2"}}},"commit":{"parents":[{"commit":"39bcc306850b5bd8a2a3eebe03503ee29a1ca546","subject":"Merge \"Switch backup build to Ubuntu 24.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/39bcc306850b5bd8a2a3eebe03503ee29a1ca546"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 10:14:33.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-22 11:28:26.000000000","tz":0},"subject":"Stop promoting 20.04 image","message":"Stop promoting 20.04 image\n\nWith backups being switched to Ubuntu 24.04 we no longer need\nto promote Ubuntu 20.04, so it can be safely removed from the list.\n\nChange-Id: If622bb88d97a1d5850d6068668ea9a1d09c4cd69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d1973d204755f3964836ae45a3a96ab52fd7398a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d1973d204755f3964836ae45a3a96ab52fd7398a"}]},"branch":"refs/heads/master","description":"Rebase"},"a14a4b6c253e95fa02cb143f6297ee93dcab6456":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-05-28 08:34:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/27/948627/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/948627/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/948627/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/948627/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/948627/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/948627/3"}}},"commit":{"parents":[{"commit":"00aa57453a5f0631d3ce052ef0f9206b5b73c2ae","subject":"Merge \"Introduce mariadb_backup backup_strategy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/00aa57453a5f0631d3ce052ef0f9206b5b73c2ae"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 10:14:33.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-28 08:34:45.000000000","tz":0},"subject":"Stop promoting 20.04 image","message":"Stop promoting 20.04 image\n\nWith backups being switched to Ubuntu 24.04 we no longer need\nto promote Ubuntu 20.04, so it can be safely removed from the list.\n\nChange-Id: If622bb88d97a1d5850d6068668ea9a1d09c4cd69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a14a4b6c253e95fa02cb143f6297ee93dcab6456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a14a4b6c253e95fa02cb143f6297ee93dcab6456"}]},"branch":"refs/heads/master","description":"Rebase"},"282956c32ea0a6e9dcdc12310640b6016aeb3d5d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-06-26 09:02:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/27/948627/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/948627/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/948627/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/948627/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/948627/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/948627/4"}}},"commit":{"parents":[{"commit":"70001bc70bdc44a7043b8f8b8e3fa0ac913e623f","subject":"Merge \"Fix PostgreSQL compatibility issues for versions greater than 15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/70001bc70bdc44a7043b8f8b8e3fa0ac913e623f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 10:14:33.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-06-26 09:02:32.000000000","tz":0},"subject":"Stop promoting 20.04 image","message":"Stop promoting 20.04 image\n\nWith backups being switched to Ubuntu 24.04 we no longer need\nto promote Ubuntu 20.04, so it can be safely removed from the list.\n\nChange-Id: If622bb88d97a1d5850d6068668ea9a1d09c4cd69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/282956c32ea0a6e9dcdc12310640b6016aeb3d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/282956c32ea0a6e9dcdc12310640b6016aeb3d5d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
