)]}'
{"id":"openstack%2Ftrove~948657","triplet_id":"openstack%2Ftrove~master~Ia9a0d032033a6f99c2faf42a7f8c79214b247904","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2025-06-27 07:19:59.000000000","reason":"removed on reply"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-06-27 08:32:38.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-06-27 08:32:38.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-05-02 08:52:45.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ia9a0d032033a6f99c2faf42a7f8c79214b247904","subject":"Fail the backup properly if no backup image is supplied","status":"MERGED","created":"2025-05-01 16:10:01.000000000","updated":"2025-06-27 08:33:39.000000000","submitted":"2025-06-27 08:32:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948657","meta_rev_id":"86beaf69cad439fd772f5e1da4d663f99c7b17f7","_number":948657,"virtual_id_number":948657,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-27 08:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"all":[{"value":2,"date":"2025-06-27 07:19:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-26 08:59:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"all":[{"value":1,"date":"2025-06-27 07:28:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-01 16:21:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-01 17:38:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-05-01 17:38:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-05-01 17:45:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-28 08:35:30.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"}],"messages":[{"id":"bff910ddf0b02c1e0061ffdd8c0a72bdeacaefa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 16:10:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b1375ae439dd7e06f3efa0fa5e9eab0ec0b51e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 16:11:21.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"7261b5e776959f4b3f3d7ead792b25de9fa1d133","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 16:21:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c1d39cda1cb43fdb185362789929271\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/acad27191ee948318f575cd2a4c7b939 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5a7411f010ec46e79e92df99dd8a2ca4 : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bee6132dae024be8b3e53c0602ef7c3d : SUCCESS in 8m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1739e5ddaa236ac99d66323ff66beabd0322f89","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-01 17:38:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"be7731e65f068b8efc5caef5af7c43f186a46729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 17:45:26.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7429c801676545e9ac21241b32d51d40\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8912599d96da4c8e9dc57cea8df29dd5 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8780a27cdb7d4c2ebd993991f533d2f5 : FAILURE in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65909d3277c74b65ab771a1491b2f49d : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd57722101d0473298d2dfaab15ec9c6 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a1b33a0f1544bb4bbe5f87282b3461b : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80760ebb501c4fb8b321698e480e155d : SUCCESS in 6m 33s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c86f07ff06c843b680a2061391fbe04c : SUCCESS in 45m 30s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/069f47f0c7b9484c82176c45eef4336a : SUCCESS in 46m 09s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/9353ef70f9ca4f7dad49818305abe0ce : FAILURE in 28m 05s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/fe12dcf5daa94ea698c725aff0035eb2 : FAILURE in 26m 36s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/e707efec1ec348c5acb9df8d7c44058b : SUCCESS in 1h 33m 11s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c00f22d886a145e59f312743012a8be9 : SUCCESS in 41m 20s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/23cef5e7188c40e8abf1d9496010aa04 : FAILURE in 32m 37s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fe7c48b99e2f41ddbae17c667ebaea9a : FAILURE in 28m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7a93f524ed324687a5ff67cac531ad75 : SUCCESS in 51m 23s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/59b9d8a89ae1491b8016e6d596ff7399 : SUCCESS in 45m 48s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/1f60d84f0e8a4fa5a5d67da93a888780 : FAILURE in 16m 59s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/409780bd5f0344c99bfd7ec8b8e3f8d0 : FAILURE in 27m 19s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/6d94083cee7c46d687aadebb3fe0b158 : SUCCESS in 1h 16m 33s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/69cad68845cd4c7d8af862bf44e0d775 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"50ba68425760468c446b862c1267678175e3a264","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-02 06:54:41.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c979594706b8aa81f02760a4a69d4e4752f3702e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 07:06:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a79c90b6302047488b2439a2e0b4084f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f5275f3623c4bff8dd3c07261ada1ec : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a66f31be03ab4ed681f749fd957c86ea : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d66a55fef4c347f88bc5172320d43804 : SUCCESS in 9m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7dece5a31b2632d98561e10cdca51e457df64a1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 08:28:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/53028bc1f20c43ce82dcab838bbea27e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d5f7075308447b59fab199d8d009616 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c3f397108b4b81ac05bfa70514de49 : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de921a3afadb4c79805e701ac6742719 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31e4f1475d854ec3b062003ccd8c5df4 : SUCCESS in 5m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f82e172a2d34ae7aca47fbbddb4ef42 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/876a2351800e43019cb22f56cb5141e3 : SUCCESS in 2m 42s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/ad1c1638e6ff4d3ca02ceedfb92d5f1b : FAILURE in 25m 00s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/923485c6b6874677b660ec735ea8656f : SUCCESS in 41m 59s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/da7588268ca243d3b390bdf7bf5c9596 : FAILURE in 33m 39s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/72aa3cf6d03a4f99aa79ceb436d0a1f8 : FAILURE in 34m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/175bc862c6e44ace9ba32c15bca9d409 : SUCCESS in 1h 30m 50s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bc7f60ca4ec242acb7b6e50bb671445f : SUCCESS in 44m 12s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/98bc19efe76d4f24887f5df500f09955 : FAILURE in 26m 15s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6b160ed8adf54e8ea50a978ec70474fb : FAILURE in 26m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/63a0239b5695408cac766420e3dc823a : SUCCESS in 1h 33m 19s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/c3f2d4aee4694e03aa3ca9f3ea070f2e : SUCCESS in 47m 45s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/c0d83b3c0ef348b5b5aec94154103ffc : FAILURE in 27m 52s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/7d8cce39fa6e4ea588a6ae4705aabc1d : FAILURE in 28m 44s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/97391c6a3d584b2081cbd551337b763a : SUCCESS in 1h 24m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d49dd77a6808413bbe820fb6ee0582cd : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":3},{"id":"85a08340bff8f0c11b9e11149f193a935a02e290","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-04 06:17:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c83e07a03a9719feaab0655484833bc1a1202525","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 06:32:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa4583839a2347ad92e2d4b42ffc2094\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/930a4161e66947dc98a14319016cb93f : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/97cf31e0a4c6445cbe5fa3c238d6ce80 : SUCCESS in 10m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/08d7f428dc4240efb9396152c9bd0ad5 : SUCCESS in 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b2682d568c56037a32c1e968767116eca94df929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 06:45:29.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de7259976c364168999cfb7ad5972d53\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/456dd9de5cee42e5b9ccdd1a763894d1 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b680759975ac45e7875fc7315005ae2f : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/536227523a784431a364304f3dc7fcf1 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69212da15ebd47d88279d02b88aea9df : SUCCESS in 5m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2769985154b545c6b752031fd507eac6 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9be2d091ed4044659f9d5e91ee3a2aee : SUCCESS in 5m 16s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/81deab036f474f45b17c271c34745d13 : FAILURE in 6m 18s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/b1b1fe9933e54cabbe9b04f94dc8b665 : FAILURE in 8m 59s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/e20794229cb6425da4481d8cb65c3be3 : FAILURE in 8m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/b657caae831d478b9e1e69eaae078b64 : FAILURE in 8m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/4f9c4a57f4d045e3bb9a590337824fbd : FAILURE in 6m 39s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a5b7bdd8a5684159a38ba666bf66fc32 : FAILURE in 6m 00s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/727de2002e264869877ef4c118b4a433 : FAILURE in 6m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1c15e362878c4e4081d34bb0520d2c7c : FAILURE in 6m 33s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/31626c76d0d74551a1fa31e72314ac7f : FAILURE in 6m 23s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/a90a622d13704fc8b7f42ef133a3fe4b : FAILURE in 6m 32s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/4adc83a31c024391ba353893761ae1cc : FAILURE in 6m 28s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/ae1580cfc1e0471390c05ee35cc83c2a : FAILURE in 6m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/9d8aa6c2aaf4431f812a5c30b661c64d : FAILURE in 6m 13s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/63e4641f073746c0bbfda5e572df3775 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":3},{"id":"5e8f97180152aeb9835b7814182e23689fc0dbf7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-04 07:56:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bfdbd0774e79af4425c0ca8d1a3a0e95ee1e43d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-04 18:06:40.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_28619\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"_revision_number":4},{"id":"a079b94168bc6a0787b3d22c9bba6bec15805e8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 18:21:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b10b9b0fa08949aab080dbc349a0dbb1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89a7d65c6dcd425a9c8c1a1f6e00c69b : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f3eaed63041448b7b707dc6441caf01a : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c0dab38e641d4574866560eabbefa241 : SUCCESS in 13m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e0f27eca8880cfb37b63192ad27420098146a6af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 19:55:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/429c42672b414a1bab75590b7faa0cff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bb48ef59b534a61b72f509cc2322257 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa7c256c888c4db4abf36ebfa7fb8ff1 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb8a31d4d1c948aca2130e150f3b9e27 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a93aa92ebd94421bea5ac4feeb1ec76 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc30a341ed8949089ae27cdc921a9dee : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0e946bceeed4f648a25db57123d68ec : SUCCESS in 4m 10s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/781fdd37f56e41a8870db51be20161c1 : SUCCESS in 50m 51s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/324ab23dda9c47eeb1b894c4c9a43439 : SUCCESS in 40m 43s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/314adae5d0384a418e1d5c7286c22e05 : SUCCESS in 1h 43m 02s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/9b7620ff2f224dc2b5fc7f3df5b2943b : SUCCESS in 46m 40s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/d3a15b69712d4c59b9def2d7afb47dd8 : SUCCESS in 1h 39m 58s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b6aeb2092f8a497683c3aea348409ea9 : SUCCESS in 48m 55s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1f4b581807a045bbad07d588ec23c3e6 : FAILURE in 30m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2e3d0e12306e416ca6baee1ba1596d1b : SUCCESS in 1h 16m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/688f7b0a99ae4caa9e32ce60054a8e08 : SUCCESS in 1h 29m 17s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/c3683875535047ebbe852ed278faa5ad : SUCCESS in 44m 06s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/0339e7d24a6942c78729cc5a7717fe9f : SUCCESS in 1h 47m 35s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/3202bfcdc18149ae91e0e4ab06f8b732 : SUCCESS in 1h 21m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/eca32e42f3804b3ca94ee94293899122 : FAILURE in 46m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a06bb329d454dd781e80250122986d6 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":4},{"id":"bd72c439dea0da1d3f4d88be540d775ab20aa4e9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-04 20:01:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7c81fa6a9f108a9d14bbf1f5fb18fd6a01610919","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 20:16:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42c11c31510e4f93b1b78f9fa0e208f2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f73d2d02b19f4a04be8160c3523c11e0 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4daa7435a3d145799421650bc500d49e : SUCCESS in 10m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/af790e4863684643bbd401e1274aafe7 : SUCCESS in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"254d4c939a9d290ec546fc15462822029ac05503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 21:48:40.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93f556432ff3492988041d79742ef186\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6099ad35724b43848e9708b629d9629f : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cf1d1817d7040e598d17ea06b6ebf49 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bcdf6efaffd41d6975fb8911de7244b : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/594cb34f168945ea8507235f4d877e26 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d22b91742af4a20ba7f5614ddb2ff62 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7badc4c4b0ef4796b800c85c859a4dcc : SUCCESS in 4m 39s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/be51580523754faba4e2d692f41e12fc : SUCCESS in 41m 06s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/2f4cd8804d984044ac4ce9547f7fb4a3 : SUCCESS in 47m 09s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/91025f998d01457eaa08cd74292ebb5b : SUCCESS in 1h 46m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/0e2b9d77780b42588da4f56de1672bad : SUCCESS in 1h 18m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/9f4284c20c154f069cf007068ef8bb2c : SUCCESS in 1h 44m 16s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a7174cb6c867421c8158233bd83aef7f : POST_FAILURE in 50m 59s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2ef2ee8be14540eb84b7ed71bd1d1ef8 : FAILURE in 52m 28s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c2350474e4e84edc965bc00e9116a437 : SUCCESS in 1h 26m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a77dfea7825d48a7bf42c14e4c906d1d : SUCCESS in 1h 32m 40s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/ecf37ac0598c4fdba733a1c3d0172b14 : SUCCESS in 55m 19s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/58efa352d75a48cc81ae0384d5142fb6 : SUCCESS in 1h 31m 41s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/b090c29eb98b4fc4a1724657b60fcef0 : SUCCESS in 1h 24m 17s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/9058d7c1c51046ce992be19ee90f18d9 : FAILURE in 38m 24s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e5008234ba9f4badbbc8b32a01661dd5 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":4},{"id":"857b356b1a8038af38633cade27860446c794264","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-05 07:59:57.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"833718881b0613e2a54fbe5cf16cf6f3469aaedc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 08:25:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14dc9263f76e427f884ae0df60f8b994\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7b3c538a8ba44cdea7543be6b16bbe33 : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3171ffcece054a3f829da387db0d66c3 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f2de38e64eb7447fa3367a9ed900330c : SUCCESS in 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b372586bd51214a88476bf3551a8372d5c2500e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 10:20:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a992aef153c342fb92eee6079f779625\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5af9a28dc4a94947ada18d2e9958662c : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d286f4da5ad4feeb552958c4ca9bc00 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18fabe6788014e76979a997774647601 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/10d9e85e8e904a3b9100097b477c63e5 : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac366981471148dd9555ced1b634adda : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087107ea822e48d7819024a41f80c18f : SUCCESS in 2m 39s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/63601a910de84ad7a5a6daeb8f50ea62 : SUCCESS in 24m 58s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/e3f209d7e8244317a2c48aa4af9dd0e2 : SUCCESS in 49m 09s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/e62d519126614efd9fbb1704d44e22bd : FAILURE in 1h 08m 19s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/5c3cf0469f3046d4a2565d58363bf42d : SUCCESS in 1h 40m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/dacc485025e84d858a18c4a8c48f10bf : SUCCESS in 2h 20m 02s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3f0a78e780564ff68f6ad2252b1d30f5 : SUCCESS in 47m 31s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fdd08507d0c1491a8ce0c43cf869f52c : SUCCESS in 1h 48m 50s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a8beab16a54847128df27814ff8da5c3 : SUCCESS in 1h 21m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7cf0a7210fbf4551b6b8066dcc450e81 : SUCCESS in 1h 25m 49s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/b465c9879f644c38b5f93c4a17ec82e5 : SUCCESS in 48m 24s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/2d94950accc3435fb0e2b278f75be9c6 : SUCCESS in 1h 28m 18s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/2ee7856a47064b958bf9d87c2212b4d0 : SUCCESS in 1h 16m 47s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/6c7fce7a5665408db23b7ab55dfd7497 : SUCCESS in 1h 19m 12s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d830cda796e741fea8b59874c287a2c8 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":5},{"id":"cc052e6561aecc31b8371df36d6d541e12cbf2a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-22 11:29:57.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"df8e0e698bab22b4d7f671f4e0ed382824b2848b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 12:09:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee1a002c4d5340318edb2fe7d0a87070\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2fccd014d3b145e7902d707713d992e0 : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/995d98d1b32644fd978e4aae7f29b2bb : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/256f90424fe3408ab2cfa2df547f60f3 : SUCCESS in 14m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2b1a3b0c76dd82e602e1177a2b0d17a0f346f190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 13:44:45.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef4d602aa2a340598dd96051e5063405\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f76a7ccae7ee4393a7a5cb1886292165 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79bc241612174b21ade14e58cbb54b1b : SUCCESS in 7m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a9ee6717c974cb98f1e9b909f6bd4cb : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/250aaead55fa47fe91daba1533c5cd0f : SUCCESS in 6m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f20b3ec933254a7c9f47356761d6270b : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bc9ff9305b14d3d8c5b8f9ea125da43 : SUCCESS in 5m 03s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/82079c4340b84bbdaa02884f7f035d13 : SUCCESS in 41m 14s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/e2e6395d1143470f86838defd254dd82 : SUCCESS in 39m 11s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/d209a48f03a5428b9d68d828926752a2 : FAILURE in 35m 44s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/1cfc6b7d04c645c3b63c1135eab92eed : FAILURE in 42m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/a38140b612d042e5b134ca2c679a3468 : FAILURE in 17m 33s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/317ccfd27cfc420f9a84908730df3da0 : FAILURE in 16m 51s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/424bb15df907484684368a4b9a3ec678 : FAILURE in 40m 27s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3274425e37414f8b86135b3fe3344345 : FAILURE in 38m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/19cbfd6df86948c5b3ba2c92fb1fee01 : FAILURE in 1h 20m 26s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/911cf23f3636462e96fcd780d2f2383d : SUCCESS in 59m 12s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/458d21de7aa541a9ab2e979315371e56 : FAILURE in 41m 01s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/3de167c921464d008b0dcafd49e6e8bb : FAILURE in 37m 17s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/a7994531dc7041d280c2062896845908 : SUCCESS in 2h 10m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/12cdadae7bd1465195b7d74f0388babc : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":6},{"id":"b1809dc1dd57b8bdd06e00564c06102553a04653","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-23 11:15:57.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"c1f511b37530850c961b114a721f9aa67cd3500c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 11:33:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91fc8bea25b94c53864412bf79ab1d5c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/20341bb41a1243ed9ff8f888ce2edc6b : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/61b4af0b6b344ae3aad20040de0f35c5 : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/adfd32b6b3204f9f8036b436cf649931 : SUCCESS in 14m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"58095e95dd17211dbe03b96d55cf85d9daeda94a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 13:42:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d2859906304492686ed1b56f3105747\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c9d902e017a416a93d4211ec48e6ad1 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aee7745b50f41c7a4c45b427a00eab7 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ca1377c6cd1410db3d01c143a3257f5 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d7057a32d6f4fdead4f57c96d1b4c03 : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ce808b12a3a4d6bbd69d75652a9483e : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29f77c6632174d6399a79529f6c746d9 : SUCCESS in 2m 52s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c87bd17977a04182bb5342152f0d7508 : SUCCESS in 42m 47s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/2df2fb341b424fdbadecd7066c2dbaa4 : SUCCESS in 46m 45s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/fb2803f717014247b084a1cd13e3f041 : FAILURE in 30m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/7bc5d8f39eb6471088d2f7d9e44ba2cc : FAILURE in 32m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/8649509c974c44db9b3f517b5158ddc7 : SUCCESS in 2h 23m 22s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3a4ef24b0ee445c189403339cdcd405c : FAILURE in 59m 56s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1a2fdb6753f841c0a6b779adb39d5a88 : FAILURE in 32m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/50bba4cfad8445ed9eef872d29e861f2 : FAILURE in 32m 10s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/916dcc96f8f347dbb17e150ccb8e7735 : SUCCESS in 2h 10m 13s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/8e11d942d7f84f75b2a6e8c1cec67132 : SUCCESS in 47m 26s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/8c3b4b5fe2b74ae2a715fe2087c81998 : FAILURE in 44m 10s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/6678e9a57e6149e18f001b2f6c5efd13 : FAILURE in 27m 50s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/58380637b12842caadeb87fd50cd86d3 : SUCCESS in 1h 59m 46s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bc3f719bbbea4470aef1f273fac1bccc : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":7},{"id":"065ec2148a7eb814b194058f9f05b4b3dfed5ffd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-23 17:26:40.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e07a772e51c75e3efd8cd4f6efc7e4a3ba062140","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 17:44:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afeddf1acd3a46a594a8909d7d1bebbe\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f249e4d309164b31bae751696770ab16 : SUCCESS in 11m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/525370f4980c4832b2968dc6fbb03986 : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/42ad612ee3df4417bdab6547d09ac047 : SUCCESS in 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f4381c30487fb38f66ecc71c95dffb1314c1ff47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 19:52:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95a82724020f415ca04c493511dfcf8b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c75d69620a2f4049adf69d45248a1ba2 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1991ea03dac74ba49692073c7cbf1df1 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f65f733351d245a2985bc1cf22e4e6a4 : SUCCESS in 5m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ce068cfea6f474c99c2b22fab645259 : SUCCESS in 6m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14218abc48c345d6a0f3083f5b3f39ed : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13cb3b4f4c944e879dd0200e3540d9c9 : SUCCESS in 5m 19s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/4ce1a972ac2b4974ac5cf5af5020054c : SUCCESS in 43m 17s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/042b429c048844f7af410c53ec6290e1 : SUCCESS in 39m 42s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/cc8ff07643614197bba7b0c721cc0483 : FAILURE in 29m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/8c6b8d89584244c9861e6186d9fcc000 : FAILURE in 39m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/0bb5a6140eb543a4aab9719986f23da1 : SUCCESS in 2h 22m 35s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ce77dfc1f09c4942b68410423949b1fd : SUCCESS in 1h 01m 09s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/62ecc730666344708399802b703e4d28 : FAILURE in 33m 57s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ef2d912b8ed744fc8464fc3eddc7b5c9 : FAILURE in 18m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/442a8cf4bf3e47edb059b8d5029ab88f : SUCCESS in 2h 14m 56s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/e56fc3f6e8524871aad1001b6217cf78 : SUCCESS in 1h 03m 26s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/b5b719fbe0fa4da98da928dd2d80aa06 : FAILURE in 34m 44s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/8310f2c8609a46958a033b2d2f8c0517 : FAILURE in 34m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/2aa3cb60f0264a7bbd68493bea9b02f2 : SUCCESS in 1h 50m 43s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d5991a3afd4644a0b47971cdeace00f2 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":7},{"id":"22162baa22bcc8d116f866d7f7ec8f59dff05d48","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-23 19:56:17.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"edfbfdf9f03982892a0f2fb67a31afc695ebb699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-26 08:59:42.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"d8933dbd7fb7e6aa6135f2162ee79197cf0f62db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 09:39:42.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90f7f39def5e4709b8f786de14d1b5d7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c39a0f842804092aa1a201ce7f7e896 : SUCCESS in 11m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6749019cbbd2487db1dd30bd9077547a : SUCCESS in 13m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/55e7096d42ac4f9a9aa029d32f2d519e : SUCCESS in 18m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"aef16061a559a9ab45ccc068667ee5909471fb3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 11:18:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b65eb0ba7e984924aaffc238acb196d9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff70f929a5964ae999134a7e582b369b : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a9992a1644542549a587aa6b0b30d18 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d015c14cb85469796a94b760c5c88d5 : SUCCESS in 6m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88900c1224cf4ac48b21e1e98fb02733 : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2287310bd28d4d19b9ca2324f806d2e6 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af19250924d341a4aa4eda10066bc36a : SUCCESS in 4m 57s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/28b153df8e814a5ab0dbf1ce77777dbe : SUCCESS in 28m 26s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/951568758ecc4033b201ec0b203ea3aa : FAILURE in 1h 54m 04s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/6c2ad6815b034a65bdb3509024082f80 : SUCCESS in 2h 00m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/4ba149a2b0d249c8a3134f6d4139e7af : FAILURE in 2h 14m 08s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b502274e6d4d49e9ade02405e6ee731d : SUCCESS in 50m 58s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e75b3ef08d0048899e33c9028a9bbd2e : SUCCESS in 1h 53m 55s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f00db81d3b0a48e99a63596844fd29de : SUCCESS in 1h 29m 44s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7e2e67a909844484b07a68dba3f8816d : SUCCESS in 2h 03m 04s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/65f641ebaa484baf835943896ace1bbc : SUCCESS in 56m 39s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/f482a56ea4444f988faf8c321d8b55cd : SUCCESS in 1h 40m 21s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/46849793435d4c13ab57efbcb1cb9f7c : SUCCESS in 1h 24m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/e5323bf700a146f6a27b16cc1a43a1d4 : SUCCESS in 1h 57m 27s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3c1157b16ed94ca3aeafeeff241c738e : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":8},{"id":"a48a46765654b570b1132b0d02b7461425d24f84","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2025-06-27 07:19:59.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"a16a2f782ae5a1453e6ab7fa68227ee2d0944b2f","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2025-06-27 07:28:56.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"0499f4e496ec6da3e3651cf152c214b07a85cca8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 07:29:13.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"d006591570ace4df6ecfe53576fb4ca8918b5d34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 08:32:38.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd6aa139198641b4b71f29662d73590b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9583fbed60764be3a7d461fecf271a94 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8ccb4ff3c2742bfa0abca1acf284cf4 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85fa2145a8754379bada968973de5d3f : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b55c005bdd44230b2689f790155b746 : SUCCESS in 3m 53s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/589405237c114adca1ba6a11856eca8c : SUCCESS in 43m 07s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/744324f0084442d1bcac90065910008f : SUCCESS in 1h 00m 45s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/23820d020f9446fd90a443b52877934a : SUCCESS in 40m 56s","accounts_in_message":[],"_revision_number":8},{"id":"df42fecba543681efbf3d2ac23cc45531d75a81b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 08:32:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"86beaf69cad439fd772f5e1da4d663f99c7b17f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 08:33:39.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/486405fd9c4249ffb27d14eeea638533\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e90f82a142145319d6535433cf1f06c : SUCCESS in 51s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"fd434c2538d9cf82335e902450f0dcf696d6dd44","revisions":{"f43278540cb08d685c8d41934e8d3e3eaaf50f76":{"kind":"REWORK","_number":1,"created":"2025-05-01 16:10:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/57/948657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/948657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/948657/1"}}},"commit":{"parents":[{"commit":"7ddb01cc86e4daee623e81e05ec0034c51bb81f5","subject":"Merge \"Fix trove install failed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ddb01cc86e4daee623e81e05ec0034c51bb81f5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 16:09:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 16:09:53.000000000","tz":120},"subject":"Fail the backup properly if no backup image is supplied","message":"Fail the backup properly if no backup image is supplied\n\nAt the moment if no backup image is supplied in configuration,\nthe creation will fail with AttributeError on _image_has_tag\nverification as it will attempt to split NoneType.\n\nWith that, such backup will stuck it the New state, as\nget_backup_image is called from outside of try/except. This also\nblocks any further attempts to create backup as well as does not\nprovide any result to the user, and requires a manual intervention\ninto database to reset the backup state to Error.\n\nThis patch fix error handling as well as add a human-readable error\nwhen configuration does not contain backup images.\n\nChange-Id: Ia9a0d032033a6f99c2faf42a7f8c79214b247904\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f43278540cb08d685c8d41934e8d3e3eaaf50f76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f43278540cb08d685c8d41934e8d3e3eaaf50f76"}]},"branch":"refs/heads/master"},"e47fbe70b1e9136222d5fa3143cc2bca158d7a7e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-01 16:11:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/57/948657/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/948657/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/948657/2"}}},"commit":{"parents":[{"commit":"58887118a54c0b9a7cd82e57bef2cdb2f4a45b0a","subject":"Install backup inside of a virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/58887118a54c0b9a7cd82e57bef2cdb2f4a45b0a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 16:09:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 16:10:46.000000000","tz":120},"subject":"Fail the backup properly if no backup image is supplied","message":"Fail the backup properly if no backup image is supplied\n\nAt the moment if no backup image is supplied in configuration,\nthe creation will fail with AttributeError on _image_has_tag\nverification as it will attempt to split NoneType.\n\nWith that, such backup will stuck it the New state, as\nget_backup_image is called from outside of try/except. This also\nblocks any further attempts to create backup as well as does not\nprovide any result to the user, and requires a manual intervention\ninto database to reset the backup state to Error.\n\nThis patch fix error handling as well as add a human-readable error\nwhen configuration does not contain backup images.\n\nChange-Id: Ia9a0d032033a6f99c2faf42a7f8c79214b247904\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e47fbe70b1e9136222d5fa3143cc2bca158d7a7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e47fbe70b1e9136222d5fa3143cc2bca158d7a7e"}]},"branch":"refs/heads/master"},"359958d11d2995ae8466656aa24e372df0ebd25a":{"kind":"REWORK","_number":3,"created":"2025-05-02 06:54:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/57/948657/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/948657/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/948657/3"}}},"commit":{"parents":[{"commit":"58887118a54c0b9a7cd82e57bef2cdb2f4a45b0a","subject":"Install backup inside of a virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/58887118a54c0b9a7cd82e57bef2cdb2f4a45b0a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 16:09:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-02 06:54:41.000000000","tz":0},"subject":"Fail the backup properly if no backup image is supplied","message":"Fail the backup properly if no backup image is supplied\n\nAt the moment if no backup image is supplied in configuration,\nthe creation will fail with AttributeError on _image_has_tag\nverification as it will attempt to split NoneType.\n\nWith that, such backup will stuck it the New state, as\nget_backup_image is called from outside of try/except. This also\nblocks any further attempts to create backup as well as does not\nprovide any result to the user, and requires a manual intervention\ninto database to reset the backup state to Error.\n\nThis patch fix error handling as well as add a human-readable error\nwhen configuration does not contain backup images.\n\nChange-Id: Ia9a0d032033a6f99c2faf42a7f8c79214b247904\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/359958d11d2995ae8466656aa24e372df0ebd25a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/359958d11d2995ae8466656aa24e372df0ebd25a"}]},"branch":"refs/heads/master"},"e59d9d5880e5898a709db3df605dfd9e72a57e83":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-05-04 18:06:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/57/948657/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/948657/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/948657/4"}}},"commit":{"parents":[{"commit":"f1e3f514e2b0204db87803d1c2a60e54daa73cce","subject":"wip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f1e3f514e2b0204db87803d1c2a60e54daa73cce"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 16:09:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-04 18:06:40.000000000","tz":0},"subject":"Fail the backup properly if no backup image is supplied","message":"Fail the backup properly if no backup image is supplied\n\nAt the moment if no backup image is supplied in configuration,\nthe creation will fail with AttributeError on _image_has_tag\nverification as it will attempt to split NoneType.\n\nWith that, such backup will stuck it the New state, as\nget_backup_image is called from outside of try/except. This also\nblocks any further attempts to create backup as well as does not\nprovide any result to the user, and requires a manual intervention\ninto database to reset the backup state to Error.\n\nThis patch fix error handling as well as add a human-readable error\nwhen configuration does not contain backup images.\n\nChange-Id: Ia9a0d032033a6f99c2faf42a7f8c79214b247904\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e59d9d5880e5898a709db3df605dfd9e72a57e83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e59d9d5880e5898a709db3df605dfd9e72a57e83"}]},"branch":"refs/heads/master","description":"Rebase"},"90a5d999937a0667cae0bfd146c248cd832bb8a0":{"kind":"NO_CHANGE","_number":5,"created":"2025-05-05 07:59:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/57/948657/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/948657/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/948657/5"}}},"commit":{"parents":[{"commit":"e629220ab44a13fadc90907cf8500e592a0ca921","subject":"Add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e629220ab44a13fadc90907cf8500e592a0ca921"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 16:09:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-05 07:59:57.000000000","tz":0},"subject":"Fail the backup properly if no backup image is supplied","message":"Fail the backup properly if no backup image is supplied\n\nAt the moment if no backup image is supplied in configuration,\nthe creation will fail with AttributeError on _image_has_tag\nverification as it will attempt to split NoneType.\n\nWith that, such backup will stuck it the New state, as\nget_backup_image is called from outside of try/except. This also\nblocks any further attempts to create backup as well as does not\nprovide any result to the user, and requires a manual intervention\ninto database to reset the backup state to Error.\n\nThis patch fix error handling as well as add a human-readable error\nwhen configuration does not contain backup images.\n\nChange-Id: Ia9a0d032033a6f99c2faf42a7f8c79214b247904\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/90a5d999937a0667cae0bfd146c248cd832bb8a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/90a5d999937a0667cae0bfd146c248cd832bb8a0"}]},"branch":"refs/heads/master","description":"Rebase"},"058efc04dc377a9227a5715a7541046524e73a57":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-05-22 11:29:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/57/948657/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/948657/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/948657/6"}}},"commit":{"parents":[{"commit":"39bcc306850b5bd8a2a3eebe03503ee29a1ca546","subject":"Merge \"Switch backup build to Ubuntu 24.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/39bcc306850b5bd8a2a3eebe03503ee29a1ca546"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 16:09:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-22 11:29:57.000000000","tz":0},"subject":"Fail the backup properly if no backup image is supplied","message":"Fail the backup properly if no backup image is supplied\n\nAt the moment if no backup image is supplied in configuration,\nthe creation will fail with AttributeError on _image_has_tag\nverification as it will attempt to split NoneType.\n\nWith that, such backup will stuck it the New state, as\nget_backup_image is called from outside of try/except. This also\nblocks any further attempts to create backup as well as does not\nprovide any result to the user, and requires a manual intervention\ninto database to reset the backup state to Error.\n\nThis patch fix error handling as well as add a human-readable error\nwhen configuration does not contain backup images.\n\nChange-Id: Ia9a0d032033a6f99c2faf42a7f8c79214b247904\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/058efc04dc377a9227a5715a7541046524e73a57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/058efc04dc377a9227a5715a7541046524e73a57"}]},"branch":"refs/heads/master","description":"Rebase"},"9ab2eaf09c48d15f82f946955e71cd1ab53e6c5e":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-05-23 11:15:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/57/948657/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/948657/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/948657/7"}}},"commit":{"parents":[{"commit":"d5bccd7bb66adcc5f23ec10731ebb10a47cdbede","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d5bccd7bb66adcc5f23ec10731ebb10a47cdbede"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 16:09:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-23 11:15:57.000000000","tz":0},"subject":"Fail the backup properly if no backup image is supplied","message":"Fail the backup properly if no backup image is supplied\n\nAt the moment if no backup image is supplied in configuration,\nthe creation will fail with AttributeError on _image_has_tag\nverification as it will attempt to split NoneType.\n\nWith that, such backup will stuck it the New state, as\nget_backup_image is called from outside of try/except. This also\nblocks any further attempts to create backup as well as does not\nprovide any result to the user, and requires a manual intervention\ninto database to reset the backup state to Error.\n\nThis patch fix error handling as well as add a human-readable error\nwhen configuration does not contain backup images.\n\nChange-Id: Ia9a0d032033a6f99c2faf42a7f8c79214b247904\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9ab2eaf09c48d15f82f946955e71cd1ab53e6c5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9ab2eaf09c48d15f82f946955e71cd1ab53e6c5e"}]},"branch":"refs/heads/master","description":"Rebase"},"fd434c2538d9cf82335e902450f0dcf696d6dd44":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-06-26 08:59:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/57/948657/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/948657/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/948657/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/948657/8"}}},"commit":{"parents":[{"commit":"70001bc70bdc44a7043b8f8b8e3fa0ac913e623f","subject":"Merge \"Fix PostgreSQL compatibility issues for versions greater than 15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/70001bc70bdc44a7043b8f8b8e3fa0ac913e623f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-01 16:09:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-06-26 08:59:42.000000000","tz":0},"subject":"Fail the backup properly if no backup image is supplied","message":"Fail the backup properly if no backup image is supplied\n\nAt the moment if no backup image is supplied in configuration,\nthe creation will fail with AttributeError on _image_has_tag\nverification as it will attempt to split NoneType.\n\nWith that, such backup will stuck it the New state, as\nget_backup_image is called from outside of try/except. This also\nblocks any further attempts to create backup as well as does not\nprovide any result to the user, and requires a manual intervention\ninto database to reset the backup state to Error.\n\nThis patch fix error handling as well as add a human-readable error\nwhen configuration does not contain backup images.\n\nChange-Id: Ia9a0d032033a6f99c2faf42a7f8c79214b247904\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd434c2538d9cf82335e902450f0dcf696d6dd44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd434c2538d9cf82335e902450f0dcf696d6dd44"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
