)]}'
{"id":"openstack%2Ftrove~948765","triplet_id":"openstack%2Ftrove~master~I66b075b15a83e8d30803e92c36683aff569f3d7b","project":"openstack/trove","branch":"master","topic":"mariadb-backup","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-05-23 03:56:27.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-05-23 02:36:36.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-05-23 03:56:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66b075b15a83e8d30803e92c36683aff569f3d7b","subject":"Add pyproject.toml to support pip 23.1","status":"MERGED","created":"2025-05-04 15:12:22.000000000","updated":"2025-05-23 03:57:18.000000000","submitted":"2025-05-23 03:56:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948765-mariadb-backup","meta_rev_id":"558318c09e5bd52e4d9311aaf78f275d141e1db3","_number":948765,"virtual_id_number":948765,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-23 03:56:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"recommended":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-05-22 13:28:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-23 02:36:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2025-05-22 15:52:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-23 02:36:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-04 15:26:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-04 21:53:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-05-05 07:57:26.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2025-05-05 10:00:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"5648fa2e314d81aeddc499f5c42e2c85a429f599","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-04 15:12:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22f7e39ca0c6eb65376f66c63c77c28535d769d6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-04 15:13:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d23155a674565cd9532863e9d278e807eaccc11c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-04 15:14:43.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ce1140f710fab83f2e81168c32f28554c5b4626c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 15:26:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4b7074aca6f4d13916ae3da0f748447\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9ae29ff27e9c4b4cb4b3bdf2a32204f5 : SUCCESS in 1m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8d7e56dda5a4916b9bf23617f633698 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95734114f77f4a75b980d1a68a7a38c9 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bdc7c6509fe487d8c5e2c3dffb47c6f : SUCCESS in 5m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0402f801444e41a6b061cfd064f99df7 : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e77c214451b6455888417541375ecdea : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e26f51ab69ed45a6aff754ba636e0d45 : SUCCESS in 4m 30s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/5dd3fe5da987494fa4ee111608dc8067 : FAILURE in 6m 13s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/ae1c955077f4416eba9aac54ba0e8150 : FAILURE in 7m 35s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/dceb628628d248e8849d6d5ceab51c3f : FAILURE in 6m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/94cf6bac4aa94222ae3cc5dff0aa6f78 : FAILURE in 8m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/1aff990b7e614e3eabeab998e77f9968 : FAILURE in 8m 01s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/85663795b5c940089935afd77f3bdc91 : FAILURE in 4m 41s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c6a1842ff0db4db8a073e59d670cd87e : FAILURE in 8m 19s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/75382b1a603449738eaf5af3c8089c2f : FAILURE in 4m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1e8cd8f7866642c4b58a385f735ac86c : FAILURE in 5m 23s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/d35c3d8d02534242a95b63c5efccfa83 : FAILURE in 8m 02s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/586532bfdf854f6bbbd31f5cd9276efc : FAILURE in 4m 33s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/240e97c84c554468bd91a912dc678a1e : FAILURE in 6m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/f3199c16e0d84cd3ba17aa703a657dfb : FAILURE in 5m 26s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/42ea0b0ec03e44c88ef8339458b84a38 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":3},{"id":"c4e90087a0aced22ae00460517631f57243c1291","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 15:35:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8436db870f74bd1a8098f74e48c5d70\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0979686224da41b5b36c6ba4faa07ccd : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/74981887a07d4c7cab2f6ce72ac987c4 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/65ea9b2da38c405db14201ef6f771437 : SUCCESS in 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a51ae0d557d49203479f772b1cf4756f25f82060","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-04 16:09:45.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"8e89723d8a5ac552d94eb2002dfd9f264480b973","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 16:23:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fab73fc5c512467d94dcd7d5bcba175c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/34908b560486408f928a6f29531ff8ed : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e81a6a481eef4cba91d9ca4b8343c1fa : SUCCESS in 10m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4dd4e2c008e3434e982dc73a43d060a9 : SUCCESS in 13m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9ad051bb48d6a690cb7bf946a5cb2f9a91ba734a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 18:05:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b45e1244bff042cba4ba382a94943d44\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d7fb0ded6816488e9d91c2ccc8a7c7b8 : SUCCESS in 2m 34s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5fff5d0ae8d34da0b980e7dc6daddd59 : SUCCESS in 2m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aea04c6909b648d091cf867a3c7b9e0a : SUCCESS in 1m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3b922b796ff4d539e00000270680240 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c4a4d5284c84339a039c8712b8693d0 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9104aebdcf34aa7a5267fb835609c67 : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb4111e99ccd4e99b639bfa4947864bc : SUCCESS in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51a7246e0e2e45e99f6b191c8d3405b7 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d6c7355ca474a8f99d5d9f2583885c8 : SUCCESS in 2m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c3655af6b834de9b13d606f0d4d4676 : SUCCESS in 4m 18s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/623f0ab04fc047e6a4acafc52b20110e : SUCCESS in 46m 17s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/0150bd765bed421496937b79b461f590 : SUCCESS in 24m 23s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/a021858b9a1741ed912c63b296a32e61 : SUCCESS in 1h 47m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/9fe899dbf8bb47f6b0be2bab3e16e6f7 : SUCCESS in 1h 28m 28s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/811b806ab8e942918e5500e1efbedf9f : SUCCESS in 1h 46m 24s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/00e6029ec0994d628c8b2c0b1e498c39 : SUCCESS in 43m 02s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6a36400251ff49b4b75fe5edc7b562ae : SUCCESS in 1h 38m 55s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/131b60ef880d4d67aa9572399c562038 : SUCCESS in 1h 22m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c2b02eaefc0e4a44bb4e0cf4870c48c3 : SUCCESS in 1h 34m 53s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/51702488bec74443895709cb261c6972 : SUCCESS in 54m 12s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/49d1dc57f8984ba3862ca37e66f7dd13 : SUCCESS in 1h 48m 54s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/fef595d094bf426c9c686625cf254422 : SUCCESS in 1h 20m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/3e764ccf7885485689fc03ed7099115a : SUCCESS in 1h 30m 18s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5912cd5c8fe848b7ac24f01ab60b1d70 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":4},{"id":"836ba8c96cdcb98d1a729b8e4393590a0ebf6f4c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-04 21:53:21.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"b69aedc573f27993db42aa2d9f3155b99ac31160","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-04 21:53:28.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"37032fe2830d640d951bc064e1b6a52168edcd26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 22:08:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6beca841116c49a18c364783b2c3893c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f4e95e08e3f4009a969d282b3cbcf7d : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/376dbb455484475ba978a79e6f4e9cfd : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1284223909454da0b2537a795c2c2270 : SUCCESS in 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4b1c03dc66e12468c4991d50293e496bba5093e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 23:45:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9b04816f40d44bcb12c589bc73d3de1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b1fa3266d26843dab8dd7d71a8af6bdd : SUCCESS in 3m 56s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e56ae77c4371443686f4696698eb0ac7 : SUCCESS in 1m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d2a9fc7f23864e1c89d83688836a7f52 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b12c7d051224fa5839201e6337bae51 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f15253e733243c2bd87c6f9a0d112a6 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be45243bedd7423a84318ecfe223b9e6 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1c2b092fc87f41688260280c7637e2c6 : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad53346577fe45259ead6e8a70fe8e9d : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27b342e598514ba88e5398f27a09f2c1 : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abb49f344b334e0bb443c3e584d6817f : SUCCESS in 5m 54s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/2862cde75dca46298d7c3b219912029c : SUCCESS in 42m 10s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/2fa1eaaade634d5486b1e22733c8d703 : SUCCESS in 24m 25s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/13739be3812d4be987de2e02417190e4 : SUCCESS in 56m 44s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/fcc3fd9608624a78979b545e72e3aefe : FAILURE in 30m 53s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/3711650c56ac4b7a92a86f8272d1b256 : SUCCESS in 1h 50m 56s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/50dd45d846f345cab5fbe28a7522e1e6 : SUCCESS in 42m 19s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4e7625b306844cc89f7d9d0ed16331e3 : SUCCESS in 1h 39m 09s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/32cb0f829e894f45bd0f73fff5adb0e7 : SUCCESS in 1h 19m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/66cb84375ef04f30be38155bc10003f8 : SUCCESS in 1h 30m 04s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/e755a390691b450baf8d93b716d2d453 : SUCCESS in 45m 37s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/ea878373cece4387b659f2c91f7d64e5 : SUCCESS in 1h 34m 21s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/ace67621b93b487fbd6b5cf53aef94cf : SUCCESS in 1h 10m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/15b43d08db57433e8a633a21df54d824 : SUCCESS in 1h 27m 43s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/84e9f061868d4cadb83ca126a1f34546 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":5},{"id":"d3bdae8c4c8053577b0825dbc3ad3a0bc5fea380","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-05 07:57:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0c431a9e1639060c255679c9a328159a449a3663","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-05 08:23:27.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"129dcb5715aa86973fc38ca55e9db0cf214f9728","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-05 08:23:37.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":6},{"id":"12af447b0d7a4bf2356272a58f50b32a4d90256d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 08:51:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/074eb9b994904abbbea3bf7c6766953d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a783abd90647460da75a7331de257308 : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/856a627dc5f645b594b7161466c1633e : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/df7d62c91d9e45ba9f6241b48019c6c8 : SUCCESS in 19m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5c0d3435f52f8bd1b7249fa06bd57c8581e97642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 09:58:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f51ec4515f84b12a65045835563ee58\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0bb108c36a0c42d2926c71f7f37c0f8f : SUCCESS in 4m 19s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c8108ec9f417430ca3a0113e5f9ee01d : SUCCESS in 1m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca1d0c0cdb7047ebb26fd271ac737a06 : SUCCESS in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/186663efc1d94f15a875e6271978e8ab : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e75779c55ddb4a469f49886d51bbb99f : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d63ace982a74fd2a3e63e7c85d6a1ab : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/225b039ac7bf4980bd3a8f86783d6459 : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18de7d55458f4996a992ad9440aed201 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/904c074ae801430dab5480d47789392d : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00e9f50b93ec4e99b8a59a52695d3e95 : SUCCESS in 5m 23s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c36c1dff5f824870a7dc69ca36138545 : SUCCESS in 41m 54s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/9ce12d87055e461199a92a1a1218cdbc : SUCCESS in 37m 55s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/75aad04a02de426ea829cc560a66b58e : SUCCESS in 1h 34m 15s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/4a20ba7db0b24070a32f45163f7b9ce4 : SUCCESS in 1h 20m 48s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/f917ea95f9f24b1ba5d308e8d48cbba7 : SUCCESS in 1h 30m 53s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3ff74cdd023b4c9fa310a35aef262966 : SUCCESS in 42m 03s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d5d466ced65041d19a9727e682da59f0 : SUCCESS in 1h 31m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/39445b6fb9b846a7b4824fc0eecf551b : SUCCESS in 1h 28m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e7531428b56e40e89ec67c692e9aa3a2 : SUCCESS in 1h 23m 45s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/748857384c57496281d2adc53da4e618 : SUCCESS in 45m 09s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/5185ad4972e9492e823abc50185ab502 : SUCCESS in 1h 30m 09s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/73eb5d26181546af8194f0d66a728350 : SUCCESS in 1h 14m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/83b053c283144d32952a1d50e81e47ba : SUCCESS in 1h 15m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8f3d6d614e274049b663824cc606bb0b : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":6},{"id":"10f21bc15473bef63f5356d8ad9cddb2dcd19d14","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-05 09:59:58.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"55e3f9db8cc9a7ef4daa0691fb7557ad7c8dcfb5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-05 10:00:12.000000000","message":"Patch Set 7:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":7},{"id":"511c061354a2eac20cc607d74782d201e930aaa7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-05 10:01:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4a2edfa7fca258d3afa723883c6eb965656b110f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 10:16:21.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a845b2ce23846158f9f4ca5b2aeaf43\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/86bdca7786274bc4beb931e2035bbac2 : SUCCESS in 13m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6f67dffe3b8f48f59b8f4a9c08e982cc : SUCCESS in 13m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a5cf655559ad45ffba34746cf0fbb863 : SUCCESS in 10m 31s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"65a9f92b4982d11f2e4beaa8ae65a94b191dec08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 11:56:46.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/736216504ab448d0ad527d12e95511f5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8e5f28f314464d73ad3ca8cada2e4569 : SUCCESS in 2m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/95962cb2878a4ac8afac2a63be26f46f : SUCCESS in 3m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05cd6470a4db4d198d5e9e2a99708b0b : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72961008a79a4a73a5bb8701d55c08a9 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b48840a26854bd4a5a515f3d216898d : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9001eed83c32452697103bf76e340649 : SUCCESS in 5m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e210e7316a64fcd94b1d471616dbbdf : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b650aae19f2c4522a02d35e6e7c1870f : SUCCESS in 5m 09s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/6768e475d28c4d6d8c65672e0c05dea6 : SUCCESS in 51m 41s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/1f7ad72739ac477aa703cace85cf0248 : SUCCESS in 37m 48s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/f036b375ed2045338527fe0ef2c31518 : SUCCESS in 1h 55m 50s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/fc075f53623a4f4bbf7c99df1ee9c77b : SUCCESS in 1h 20m 55s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/84a5824332f5474199c7be51e848595d : SUCCESS in 1h 54m 39s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2a56c5576a014dcd916e909758edb3cd : SUCCESS in 47m 26s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d4effe2f39954f1ab545f75106c53176 : SUCCESS in 1h 48m 05s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/cc642e5a03aa49378ab9c67e14898cf9 : SUCCESS in 1h 14m 43s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b8713d9c3cf048b6a74b3f2bc4f1338f : SUCCESS in 1h 24m 16s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/f8e58fc9647e4708bdd77cf5f74df8e0 : SUCCESS in 44m 37s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/19d0dbcfa32742f4a29020a133878a03 : SUCCESS in 1h 27m 35s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/d534321f64a2448bb9b578734ef1cafd : SUCCESS in 1h 17m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/9d60778e02584aa5a02b6ceb6dc7bbde : SUCCESS in 1h 14m 45s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9a5ba14e83864c60a9fa3fb9cb2edd55 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":7},{"id":"6c9dea41572497b4ca002fd9171edc4162a88099","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-15 08:00:30.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"3e89c4475d4e942b4172615365d8439e4a8400df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-22 11:19:05.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"1796d724a3154916369ab6926b38f92115cafb5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 11:55:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d085b22626ca4deda9541245da57a49b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/98e3a4c3dc8848058da30a4c60093a8b : SUCCESS in 11m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ec97b10d2f3f40a6ad8634573142d51b : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e6dc6997cbcc4ab793a9aab79f6b4b2a : SUCCESS in 13m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"876744d28f7f9e507b01aedeb03d8e7248a78cf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-22 13:28:09.000000000","message":"Uploaded patch set 9: New patch set was added with same tree, parent tree, and commit message as Patch Set 8.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":9},{"id":"57849d3959d904dd9426e85c30eb08e9ff6ee982","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-22 13:28:50.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8994183ba2703e0eaa7b04aeb8019fb870b4108c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 13:57:25.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a934473468e847bead0da3f4742d75e6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/22072f6a10a0443a950cb867ca28d7cc : SUCCESS in 12m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/67dbb72dc08e43fe8965036b84f6e179 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d9831ca3e4214501a7cf5dcacbca70d0 : SUCCESS in 18m 01s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"de3e55d4c8a3eaf032d64e6f8f8a94bffb4aec66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 15:41:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b812c6aa41345a7a628b7714b083ef5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/21008a7381794fd98c11f9b5b021c04b : SUCCESS in 2m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9a7c238744a44c35bfe57927edb4ccad : SUCCESS in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38dd09a78aa04764acdfe7bdb95cfc7a : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88974324fff04cada41bece1075f0447 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62cdb36fafec46ef97ab80b84863bf0b : SUCCESS in 6m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d773e39e0bd9452ca864de2054f7d891 : SUCCESS in 7m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d205ca7d69ef4937b7ae68914636e318 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c558eb29e93481b9589c0757bb884dc : SUCCESS in 7m 19s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/49c2a0e590f34f4598262f1c9e0b1a66 : SUCCESS in 1h 13m 08s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/613e89b665c0495baa4ff49f3c760f4e : SUCCESS in 52m 39s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/e3de9fe4ee3245ab9dfc069f5e15dd22 : FAILURE in 40m 43s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/8f2774c97178488cbc0b4beae1dd330f : FAILURE in 46m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/ad7ffc55489e4d1a9425ec33a3a305c9 : SUCCESS in 1h 34m 16s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f85d8dfad0b94a93bd11c7353fda8869 : SUCCESS in 53m 57s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5a7a1bd711de42d2b3b41c55142b3ee5 : FAILURE in 32m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2acdac66ae3149aea9a680912c8175a8 : FAILURE in 35m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a527e3845a6c4e3f85f3be92e678e78b : SUCCESS in 2h 09m 46s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/845a30eee857443c96a9305e08078f3d : SUCCESS in 1h 03m 51s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/fac2867d2dc34db09369db7269340f41 : FAILURE in 34m 46s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/5d71510b62674f46a95c56d6256f34b2 : FAILURE in 29m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/49078c2464224823b25eab79f8265f5e : SUCCESS in 2h 01m 53s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9c8f3f9d39df41e68859a38a8f36abfe : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":9},{"id":"64f1f5a5cb2d395b3df712f62f4d890288faa5d6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-22 15:52:26.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"06afc28fe9655ef97bf507c1ad0f919b62c400a9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-22 15:52:41.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"c332fa61b7bad73369ebb63e9df67fd46fa1ed9e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-05-23 02:36:36.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"c334512e889cab81c69c5e9aad4ae04cbc1019cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 02:36:59.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"0a4c078915f6225278059093e2fe507fef2b2b9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 03:56:27.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/357f059a58f447a38c2c95740f319f61\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/47867dfd761045b6b6f1190c75b42126 : SUCCESS in 4m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/51ba1d90d5a144fe8df06c1cba3a20bc : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/764371c3afbc44a1942b4bd86f69a819 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bc47c5fa70c4c09a9c68040112d5f98 : SUCCESS in 7m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/983afcf154e345788674b43dd329c585 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce6b58112c4f42f7be5f9cfffe1b78e0 : SUCCESS in 4m 55s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/dd9d9d210ed040e596db6a325745821f : SUCCESS in 59m 44s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/07cd9a95d7ea4aa28606d9f0e2dd424e : SUCCESS in 1h 01m 02s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/87b1396da02d45e6bc9b16187350a56a : SUCCESS in 24m 07s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5c35b5e8200a466086a5f7c66ae6b9ea : SUCCESS in 1h 18m 14s","accounts_in_message":[],"_revision_number":9},{"id":"febb8e0b0d3ac636ad5a0e99c80dda39f1940a9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 03:56:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"558318c09e5bd52e4d9311aaf78f275d141e1db3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 03:57:18.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f930fa17f7284ae0a146846ee489577d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/becea1cf85d7466982ba19214c9f443c : SUCCESS in 40s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"3e0b163d800afcb4d020dbf0c5f68886fa69bc23","revisions":{"b6e2a0eca8dd46ba50aee5a380278b07f0f8b09d":{"kind":"REWORK","_number":1,"created":"2025-05-04 15:12:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/948765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/948765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/948765/1"}}},"commit":{"parents":[{"commit":"7ddb01cc86e4daee623e81e05ec0034c51bb81f5","subject":"Merge \"Fix trove install failed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ddb01cc86e4daee623e81e05ec0034c51bb81f5"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-04 15:12:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-04 15:12:22.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I66b075b15a83e8d30803e92c36683aff569f3d7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b6e2a0eca8dd46ba50aee5a380278b07f0f8b09d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b6e2a0eca8dd46ba50aee5a380278b07f0f8b09d"}]},"branch":"refs/heads/master"},"2c4e68ec1b57ea520e75652684d48b9bf002e108":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-04 15:13:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/948765/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/948765/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/948765/2"}}},"commit":{"parents":[{"commit":"ab9115a46c0fd6911588c97afe8f40ba767f949c","subject":"Cleanup setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ab9115a46c0fd6911588c97afe8f40ba767f949c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-04 15:12:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-04 15:13:56.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I66b075b15a83e8d30803e92c36683aff569f3d7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2c4e68ec1b57ea520e75652684d48b9bf002e108"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2c4e68ec1b57ea520e75652684d48b9bf002e108"}]},"branch":"refs/heads/master","description":"Rebase"},"c27bb7c5c05441404e2e384c923c9b9903cfaaa0":{"kind":"REWORK","_number":3,"created":"2025-05-04 15:14:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/948765/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/948765/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/948765/3"}}},"commit":{"parents":[{"commit":"ab9115a46c0fd6911588c97afe8f40ba767f949c","subject":"Cleanup setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ab9115a46c0fd6911588c97afe8f40ba767f949c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-04 15:12:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-04 15:14:42.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I66b075b15a83e8d30803e92c36683aff569f3d7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c27bb7c5c05441404e2e384c923c9b9903cfaaa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c27bb7c5c05441404e2e384c923c9b9903cfaaa0"}]},"branch":"refs/heads/master"},"f1e3f514e2b0204db87803d1c2a60e54daa73cce":{"kind":"REWORK","_number":4,"created":"2025-05-04 16:09:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/948765/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/948765/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/948765/4"}}},"commit":{"parents":[{"commit":"ab9115a46c0fd6911588c97afe8f40ba767f949c","subject":"Cleanup setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ab9115a46c0fd6911588c97afe8f40ba767f949c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-04 15:12:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-04 16:09:45.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I66b075b15a83e8d30803e92c36683aff569f3d7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f1e3f514e2b0204db87803d1c2a60e54daa73cce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f1e3f514e2b0204db87803d1c2a60e54daa73cce"}]},"branch":"refs/heads/master"},"e629220ab44a13fadc90907cf8500e592a0ca921":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-05-04 21:53:21.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/948765/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/948765/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/948765/5"}}},"commit":{"parents":[{"commit":"ab9115a46c0fd6911588c97afe8f40ba767f949c","subject":"Cleanup setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ab9115a46c0fd6911588c97afe8f40ba767f949c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-04 15:12:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-04 21:53:21.000000000","tz":0},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects that do\nnot have pyproject.toml and now uses a pyproject.toml which is vendored\nin pip [1][2]. pip 24.2 has now deprecated a similar fallback to\n\"setup.py develop\" and plans to fully remove this in pip 25.0 [3][4][5].\npbr supports editable installs since 6.0.0 so we can get ahead of this\nby adding our own minimal pyproject.toml to ensure we are using the\ncorrect build system.\n\n[1] https://pip.pypa.io/en/stable/news/#v23-1\n[2] https://github.com/pypa/pip/issues/8368\n[3] https://pip.pypa.io/en/stable/news/#v24-2\n[4] https://github.com/pypa/pip/issues/11457\n[5] https://ichard26.github.io/blog/2024/08/whats-new-in-pip-24.2/\n\nChange-Id: I66b075b15a83e8d30803e92c36683aff569f3d7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e629220ab44a13fadc90907cf8500e592a0ca921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e629220ab44a13fadc90907cf8500e592a0ca921"}]},"branch":"refs/heads/master","description":"Edit commit message"},"565d6eebb49804b4f66b31fcd27868f6149c3862":{"kind":"REWORK","_number":6,"created":"2025-05-05 08:23:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/948765/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/948765/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/948765/6"}}},"commit":{"parents":[{"commit":"ab9115a46c0fd6911588c97afe8f40ba767f949c","subject":"Cleanup setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ab9115a46c0fd6911588c97afe8f40ba767f949c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-04 15:12:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-05 08:23:27.000000000","tz":0},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects that do\nnot have pyproject.toml and now uses a pyproject.toml which is vendored\nin pip [1][2]. pip 24.2 has now deprecated a similar fallback to\n\"setup.py develop\" and plans to fully remove this in pip 25.0 [3][4][5].\npbr supports editable installs since 6.0.0 so we can get ahead of this\nby adding our own minimal pyproject.toml to ensure we are using the\ncorrect build system.\n\n[1] https://pip.pypa.io/en/stable/news/#v23-1\n[2] https://github.com/pypa/pip/issues/8368\n[3] https://pip.pypa.io/en/stable/news/#v24-2\n[4] https://github.com/pypa/pip/issues/11457\n[5] https://ichard26.github.io/blog/2024/08/whats-new-in-pip-24.2/\n\nChange-Id: I66b075b15a83e8d30803e92c36683aff569f3d7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/565d6eebb49804b4f66b31fcd27868f6149c3862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/565d6eebb49804b4f66b31fcd27868f6149c3862"}]},"branch":"refs/heads/master"},"02c4f82729b180aaef03fbc30fc497917047fea9":{"kind":"REWORK","_number":7,"created":"2025-05-05 09:59:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/948765/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/948765/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/948765/7"}}},"commit":{"parents":[{"commit":"ab9115a46c0fd6911588c97afe8f40ba767f949c","subject":"Cleanup setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ab9115a46c0fd6911588c97afe8f40ba767f949c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-04 15:12:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-05 09:59:58.000000000","tz":0},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects that do\nnot have pyproject.toml and now uses a pyproject.toml which is vendored\nin pip [1][2]. pip 24.2 has now deprecated a similar fallback to\n\"setup.py develop\" and plans to fully remove this in pip 25.0 [3][4][5].\npbr supports editable installs since 6.0.0 so we can get ahead of this\nby adding our own minimal pyproject.toml to ensure we are using the\ncorrect build system.\n\n[1] https://pip.pypa.io/en/stable/news/#v23-1\n[2] https://github.com/pypa/pip/issues/8368\n[3] https://pip.pypa.io/en/stable/news/#v24-2\n[4] https://github.com/pypa/pip/issues/11457\n[5] https://ichard26.github.io/blog/2024/08/whats-new-in-pip-24.2/\n\nChange-Id: I66b075b15a83e8d30803e92c36683aff569f3d7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/02c4f82729b180aaef03fbc30fc497917047fea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/02c4f82729b180aaef03fbc30fc497917047fea9"}]},"branch":"refs/heads/master"},"f99bc30e362d179b5680931aa3d2dba0a4d09c00":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-05-22 11:19:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/948765/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/948765/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/948765/8"}}},"commit":{"parents":[{"commit":"39bcc306850b5bd8a2a3eebe03503ee29a1ca546","subject":"Merge \"Switch backup build to Ubuntu 24.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/39bcc306850b5bd8a2a3eebe03503ee29a1ca546"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-04 15:12:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-22 11:19:05.000000000","tz":0},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects that do\nnot have pyproject.toml and now uses a pyproject.toml which is vendored\nin pip [1][2]. pip 24.2 has now deprecated a similar fallback to\n\"setup.py develop\" and plans to fully remove this in pip 25.0 [3][4][5].\npbr supports editable installs since 6.0.0 so we can get ahead of this\nby adding our own minimal pyproject.toml to ensure we are using the\ncorrect build system.\n\n[1] https://pip.pypa.io/en/stable/news/#v23-1\n[2] https://github.com/pypa/pip/issues/8368\n[3] https://pip.pypa.io/en/stable/news/#v24-2\n[4] https://github.com/pypa/pip/issues/11457\n[5] https://ichard26.github.io/blog/2024/08/whats-new-in-pip-24.2/\n\nChange-Id: I66b075b15a83e8d30803e92c36683aff569f3d7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f99bc30e362d179b5680931aa3d2dba0a4d09c00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f99bc30e362d179b5680931aa3d2dba0a4d09c00"}]},"branch":"refs/heads/master","description":"Rebase"},"3e0b163d800afcb4d020dbf0c5f68886fa69bc23":{"kind":"NO_CHANGE","_number":9,"created":"2025-05-22 13:28:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/65/948765/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/948765/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/948765/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/948765/9"}}},"commit":{"parents":[{"commit":"39bcc306850b5bd8a2a3eebe03503ee29a1ca546","subject":"Merge \"Switch backup build to Ubuntu 24.04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/39bcc306850b5bd8a2a3eebe03503ee29a1ca546"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-04 15:12:22.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-22 13:05:50.000000000","tz":120},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects that do\nnot have pyproject.toml and now uses a pyproject.toml which is vendored\nin pip [1][2]. pip 24.2 has now deprecated a similar fallback to\n\"setup.py develop\" and plans to fully remove this in pip 25.0 [3][4][5].\npbr supports editable installs since 6.0.0 so we can get ahead of this\nby adding our own minimal pyproject.toml to ensure we are using the\ncorrect build system.\n\n[1] https://pip.pypa.io/en/stable/news/#v23-1\n[2] https://github.com/pypa/pip/issues/8368\n[3] https://pip.pypa.io/en/stable/news/#v24-2\n[4] https://github.com/pypa/pip/issues/11457\n[5] https://ichard26.github.io/blog/2024/08/whats-new-in-pip-24.2/\n\nChange-Id: I66b075b15a83e8d30803e92c36683aff569f3d7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3e0b163d800afcb4d020dbf0c5f68886fa69bc23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3e0b163d800afcb4d020dbf0c5f68886fa69bc23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
