)]}'
{"id":"openstack%2Ftrove~949915","triplet_id":"openstack%2Ftrove~master~I5336c8f98d9f1f33810113b9d7a84f1bef3deb66","project":"openstack/trove","branch":"master","topic":"mariadb-backup","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2025-05-27 14:32:08.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-05-27 14:32:08.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-05-27 13:51:17.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I5336c8f98d9f1f33810113b9d7a84f1bef3deb66","subject":"Introduce mariadb_backup backup_strategy","status":"MERGED","created":"2025-05-15 16:28:03.000000000","updated":"2025-06-13 01:45:33.000000000","submitted":"2025-05-27 14:32:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949915-mariadb-backup","meta_rev_id":"bd7a13a5cff7cb2532cf22641eb8085947deea61","_number":949915,"virtual_id_number":949915,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-27 14:32:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-27 13:51:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2025-05-23 07:52:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-27 13:51:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-15 16:36:08.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2025-05-15 16:36:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-05-15 16:36:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-05-15 16:55:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-15 18:43:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-23 07:52:26.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"fb02653537499103b4d030c4310bd5a3822f2c91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-15 16:28:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b6cb190245b181efd66f52d38a1de5f325928fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-15 16:34:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"de05fae135a10cb6ae42b67ea0db579cc85b0f24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-15 16:38:13.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"8f74d4c0681a7ec7414caa43e5dcd09f08ffe9e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 16:55:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c7d14345bc544f7bd7341ee00a96ec6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/97a102712c3147f2adefc6a7c0d6ffde : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9c25f96369814d059cd282de559fa9af : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0d7eda3ab9e64c7ea9a05d321087d13c : SUCCESS in 15m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"02b4aca4639653a243dc3ba5889d3b5feff3ea18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 18:43:20.000000000","message":"Patch Set 3: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2238b892f84474b81f56a442ba2955a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ff875575421489287f69b8256737f1e : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/925a784c1c2c4108907cf239acbdf068 : FAILURE in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58fd6621dc03454eafdff7968f174215 : SUCCESS in 5m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d1aea61741b442b8e0a1c2f72ff3ce3 : SUCCESS in 5m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/019d9fc78144425ebe9106af824ecbae : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2704c727a0db4c15a13ded62e36b2a8b : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d03392eaef4c40ffa61fa720c6046583 : SUCCESS in 6m 49s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/4ad79859cb7f4834b2c7e552d268d168 : SUCCESS in 55m 37s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/a3b64875c0634de48bfaff8d13e9d2e0 : SUCCESS in 1h 02m 45s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/241b58af189549cfaf4a02788b83e39d : FAILURE in 1h 40m 44s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/cabd58aa0bed4d0889b7fb2bdc772271 : SUCCESS in 1h 58m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/3ad8bc34e0084b7bbe360e2bd40a4509 : FAILURE in 1h 59m 37s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/638bf37a0f384373b1b4c0586b0c24b8 : SUCCESS in 27m 38s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d3bfb607c00646d199212c13e841cefd : SUCCESS in 2h 04m 15s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/cd4190cf76e74004b71c47e83cb1c163 : SUCCESS in 1h 15m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f0c8f355ffb94013a5a7c2137b61617a : FAILURE in 17m 50s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/30c733f02d114f999faa6642723cd6da : SUCCESS in 1h 11m 57s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/3ea91e5f80964a83b69d604f8cb692b4 : FAILURE in 1h 25m 00s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/3063f59fc22745398270998d0e5e8500 : FAILURE in 44m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/8a99db619dee411abc1dd667687d3df6 : SUCCESS in 1h 16m 01s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7418694cdeea48629ad0a984c2de53f9 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"ded4d47bb3c3c1c776fe98fc3bfb46b195ee2bfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-15 19:17:55.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9a0e7884cb43e318b2205d8d13e2dd0c9b33e9fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-15 19:18:59.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"8af151108bbdb2044664f0bf91ce3a2c1ce5abb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 19:32:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c319b6bf37b4428b24012bee82f456f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1d569f1c47ca46b384f9ddf48681539e : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/474f434debd741b1a960b9a205d40362 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/76855ab07ed844f4bb06f157ad77354b : SUCCESS in 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"45f81917624bc0ba4eff002491f54157eda08afc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 21:40:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa612fa9a0364990a3efc71f953cb250\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a14860d341e418f9dcbb2453a2c8262 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/097f6c0a0beb4a0a9b51114699bac00b : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6be6b60dbb7e4f2084a0b9c1813481dc : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ded62b1f7ee4f378088ed828c84c986 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1965c3427d445d990f742cbe2cc9981 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93452f5d112542b0965b5b7ae075feb9 : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9c626dd110d42ba95d6e4a643dfb27e : SUCCESS in 3m 04s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/88c45f4282bb4723865accc3e83a1911 : SUCCESS in 41m 21s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/2d734b654cbf4bc7a282e7f7aef5bd45 : SUCCESS in 38m 21s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/09e35bc2974c475e8ddfca5ee1aec8c3 : FAILURE in 1h 50m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/8b7e0472e1c64906bbb1d5a29bada880 : SUCCESS in 1h 19m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/19056c888fbb4200a7ad1cb9e93f2366 : SUCCESS in 2h 19m 51s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/82500e1c7380439c80980dea1c045a0f : SUCCESS in 48m 25s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/18cb1ec42f6944c2ba24cf3ec6044d09 : SUCCESS in 1h 57m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a80ee80a39b54eefb9c49bc02d6c4cb4 : SUCCESS in 1h 43m 36s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/592de35af3d34eeb8d756e1e35c5f7ff : SUCCESS in 1h 55m 33s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/215872e6d0064d368cd9e194d20e5fdc : SUCCESS in 1h 03m 39s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/840214f834ee47019ae9d03be65c2c60 : FAILURE in 1h 27m 55s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/f6222c3af0604c479295258e2131402f : SUCCESS in 1h 54m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/e7a2542020814975bba16de9db065c5c : SUCCESS in 1h 45m 09s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/898bbca2b73f41dd850a263f0c00ad64 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":5},{"id":"1a131b68b7fa010886239e425fb28b6ec6714c01","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2025-05-16 03:15:24.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a91d5380fb0646971b76af00a3c104d0565b58e8","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2025-05-16 03:16:46.000000000","message":"Patch Set 5: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5b5f31806edfd9cb31326bd296f766dc19576ba5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 03:42:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/137ac09a779c42c6b1c3a6a3049265cf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7a5923b6c0b9462e81899e668061ee5a : SUCCESS in 12m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d0f4602b56da4e7482694a3615bee8f4 : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5a02745c2c5d4df3aa4203bae4a589b9 : SUCCESS in 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fb794131d43dc3a1e8a3607bb60d73b585624421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 05:32:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8cf7baf9b68d4e1aa8ce7a269d914917\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37dc8d8d1d4247c293f9c9e802ab5f0e : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1c9f7a72c6146feaa85f8eed65b79ad : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b38c08fe6024f89874d587c483f4b4a : SUCCESS in 7m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65fd0ddffa4449e3bae989bd580116cd : SUCCESS in 7m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/64392a9e87454b50bd7c1683410852b4 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a57498792cee4315aea257ec14ed2d53 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f1b95e659bc4a5f9b1fd47d58bfbc04 : SUCCESS in 5m 01s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/9a13fe08375e4492b3c172b16512819e : SUCCESS in 1h 03m 22s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/7b1cbf6cbacc4231a4fa2bf71227d90f : SUCCESS in 48m 27s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/e783a38e1d1342d381a649b84bec4e9f : SUCCESS in 2h 13m 35s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/e63e693f8b9a467f875886808c286737 : SUCCESS in 1h 31m 05s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/2f546856626247d990ebba65c355c57f : SUCCESS in 2h 00m 53s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a1f3befb2cda44f0a5448ea7e5302a01 : FAILURE in 57m 17s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/699faf5fddee4fb1a1683b24b0e93922 : SUCCESS in 2h 06m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/977085ebdc0c40e49fdfbac1fe39ef2f : SUCCESS in 1h 25m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4629bd90107148368b0ca655c85589b3 : SUCCESS in 1h 44m 07s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/4b6e028e868a42e78e8cfabbee504601 : SUCCESS in 57m 00s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/92f57e8159934b598b7528c9d34c4bf5 : SUCCESS in 1h 56m 33s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/e21632c0a1b54b448f303b2f8f030af1 : SUCCESS in 1h 31m 50s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/3e5826d414174086ba5facf12a427077 : SUCCESS in 46m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/98d713d2b85b4a0dbbb306251cd36ea7 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":5},{"id":"bb1002f8f1df0cc52e6156b0b19c918413d7c010","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-16 07:54:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"40e5e5dd3684facedbd88f4eb80582895c66dbee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 08:07:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9665896fe3f48d0b448f41c8cceb76b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/456fe7adbffa4310bf34df704cb43284 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9b743c1deacc4448bb8f17f8be75bffd : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9b33a716956c452bae4b49d4510d5de5 : SUCCESS in 9m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b0faef8202534a4177b3aec1725cd37fdd956deb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 10:24:48.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5a51e9fa56c4e3fb038b67d00bbd12c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1208501a425044b3becc5b373b491d5b : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b76ba01736cc41959bf754620db7d338 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2deea5bdfd3440b88e1b7fa46242a160 : SUCCESS in 5m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ba86161d2114b65b3622e3c92233f4b : SUCCESS in 5m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cfaa47081f6c47159833c8047eb7b212 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3322d08434744c19e059217c63ee44f : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/668f88bd1eb241a79df57cdac541201d : SUCCESS in 5m 24s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/3ff8122728744760a694b2eb60608321 : SUCCESS in 1h 03m 18s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/9cccee50407f42bcb87e837b14e64daa : SUCCESS in 40m 35s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/2b672725748f49988fecac034c1c8340 : SUCCESS in 2h 28m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/3d28e45368504d55947428e3f9156f41 : SUCCESS in 1h 23m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/b703e69953064994ae14c7c464687ca0 : SUCCESS in 2h 20m 35s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1564079e87434c7fa8842a8931fd5da5 : SUCCESS in 41m 03s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/715be6d176e74cb2840ee83b16239be9 : SUCCESS in 2h 14m 52s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b267ac6110604c07bccd6140684c1813 : SUCCESS in 1h 18m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/690bcdcd64e74c8090948335a02e37db : SUCCESS in 2h 15m 24s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/4ee6d4b3798846d6b22cd37209618e9e : SUCCESS in 1h 01m 45s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/491565b079c54114994095d21610eda2 : SUCCESS in 52m 25s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/d95dc90aff1b4c5ba377e3632b2c6eb1 : SUCCESS in 44m 22s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/54cf8bd7f38e4f77adab471ff0ddf565 : FAILURE in 1h 58m 28s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0afe7ed21bd746f29464a8d57f2547af : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":5},{"id":"1edabc9c7628a5ae0f399d29e8e82dcd7f756934","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-16 12:40:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a347d1488a29281cca034ad80b444d093e2ceddc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 12:58:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fc985a849df4167ba3befc9eba1c6dd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f93d3267afba4d2e9d40561199e66c6d : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a68eb35ed1984f23adf150e311b4c0c3 : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ca2eff2741e345a8878f91f9fa728add : SUCCESS in 14m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"93d51cacb36cae3d430cd46315ca938ece5694c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 15:16:45.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33b8652cba1843488c1a4f7500299e20\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8638a840edc4dedaaaf646920d0a213 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaa6c206a31a459e970dc27ae4d62e40 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1400520c6f724ee9b1b284fe5f9ed28e : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0c10b84d2724222969f3e195332b163 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df000581425e4a209c7e234b1d795efb : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d4387bf39048fb8f8c2adcc2adcc24 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9588649b9ab349b1aec24e6ec1c08f04 : SUCCESS in 4m 48s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/ad46842dd0784299abe40297a71c4f1c : SUCCESS in 49m 10s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/243744347d7947d085bb798db5f709f6 : SUCCESS in 51m 16s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/89ce3faee03a447ea6f1900d969ab8ef : SUCCESS in 1h 13m 08s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/10cc57dc11254950b52c11e045518644 : SUCCESS in 1h 51m 33s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/35ff48e417004eef851765361cd267e8 : SUCCESS in 1h 29m 31s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/51e7a43ff14c4541843c191b9409d0a3 : SUCCESS in 25m 30s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3f8050b648564f9ba853fedf76a34f71 : FAILURE in 31m 57s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/55ed6217065e49369bf5f6618fbd785c : SUCCESS in 46m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5be6db7c735f41ee9a0f2e6bd9b08918 : FAILURE in 1h 51m 02s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/3e3bc412a3494e74a9fc0fecb905167d : SUCCESS in 35m 07s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/df7a15557db542a08997dff7b65b1d45 : SUCCESS in 2h 04m 58s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/28fec226e9544dbaade011bee6270495 : SUCCESS in 46m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/ab35261b252d4d80b9972e0a93b9eaaa : SUCCESS in 2h 29m 35s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2f28a45233d041d6a52869cecc9b9c51 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":5},{"id":"3eccced54e5fb3f09d664a1f3062b8c9c13aac19","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-17 07:13:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0ed9c3c4a25786ae0d7a6234be83d27e9bcf8aa5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 07:25:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b044a7bcd9394a53af979ef71ca954a9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6099eab24d0242c6ab169f64d87f8384 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ec55dfa4fa544c5f8610ef9ac92fe0da : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e9fb9262603d4013bd1590c6422ef589 : SUCCESS in 9m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"be0875ac8c5b12983ca656a82dbb5462a1ed3cd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 09:36:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ed631c7b2bf476290fa3ff1f7e08395\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6adb39865f6f459fb35f2bbd25a53903 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d835a42f521143d0a2d068c2d8256bd9 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3c18a8dce1744ae8abeef798c894902 : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0da7809d5a34c2dabdc469cfafaefa3 : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/303ded6842b24a8d980d9eac38fe842b : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a5d457f1e62419e9bbcb875a72a48f6 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49c52032c9644c23b6a592466d04ae6f : SUCCESS in 2m 57s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/842bd006f6bc4a35a3db63d5292a291f : SUCCESS in 30m 39s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/b379b9e7eeb449fcae44f7bb4792f9fe : SUCCESS in 24m 16s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/3cf8a38bce384c5a944d62dd2145400a : FAILURE in 1h 21m 53s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/100e46be93cc47a6b2e206697d072238 : SUCCESS in 1h 56m 18s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/a4cbf590f4e84a87855e8e44619423d2 : SUCCESS in 2h 21m 47s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9f43b498077e4393bd1752e7046bff71 : SUCCESS in 55m 44s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6fa871f4006d4eb8899c51f80449c8af : SUCCESS in 2h 06m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b98480ac41f047919eaff5d066443e98 : FAILURE in 1h 22m 43s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8b019b8185fb46babfcae971ada4b206 : SUCCESS in 2h 01m 12s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/0c22c3d305b54e74ba710a23389f8952 : SUCCESS in 1h 02m 26s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a011c72f2d9744c39b39203544ed8b21 : SUCCESS in 2h 05m 58s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/321d4596ab3f4926b38555e0327279f6 : SUCCESS in 2h 04m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/3141b780ea51472c9d0fa0e77c3f175c : SUCCESS in 1h 54m 40s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/afc6f037c56e45708ad1e7a719c94a82 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":5},{"id":"ef3303a84ccb512715c3e0921896fc62c2c2de21","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-05-21 13:01:14.000000000","message":"Patch Set 5: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"97f59bd6422e6ec900e7c38fc0d38c370398ef6e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-22 06:54:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0c7350c28401f5e157105c07142b11434979e15e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-22 07:13:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5},{"id":"0f34c6d3ca51f5b9ecf95b41c2fd39e642fcbbd0","tag":"autogenerated:gerrit:restore","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-22 12:59:22.000000000","message":"Restored","accounts_in_message":[],"_revision_number":5},{"id":"aaedb20bb7823bf295a79eef9aea0d3f53379f9a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-22 13:03:54.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a844312e9a8f1637a9739bfc664ca0ff2de97e96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-22 13:28:09.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"c8f5705c4fd808e0a48e7dc74a38233464adf621","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-22 13:29:54.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"bc6c8fd422c603e23f5f1fec16016faac0b869a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-22 13:31:29.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"ca298c49b2e02f672ecc696dd093dd8b33b1d457","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-22 13:33:29.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"a80826f79494e75acfbc4ae231ab3fe858ef4318","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 14:03:45.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8095f0f719844abaaa8301b87dd030dc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a0155a29001a41f2a54d3aee8c3f23a3 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e5caeabc9c7543cf8b6facba7c12b016 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/57d272ef140349939b57e9a322686cee : SUCCESS in 15m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8fea85e65fe471e74643feca8ea6c5bb7fa5ff36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 14:32:07.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c38f137b72c4411c9d4e95530ed4d023\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7eaa6484c12c49a4b28554c2e4236b4a : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0e68ad1f63e43709ff73f397bea5aea : FAILURE in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/796f9c8a001d4a1e9b3f66a5b6e67336 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a20c7a65f12342599bbef9f19da8f175 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae4a3c159f534349b9c267034c7a27f1 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1c69a6f6156458c82722e8c3843f77e : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e780f4c6915f4b658eaa1ca001f8d9f0 : SUCCESS in 5m 47s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/24a9aa66fcd04456a043825f003fd406 : FAILURE in 37m 33s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6fb0cebe0a4f4ef9adc7d8901c5be02c : FAILURE in 32m 13s (non-voting)\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/9596bb5de4904ad7b124a45e0459d307 : FAILURE in 29m 56s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/30b12b7b93dc4f70ade661c611cd143f : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":8},{"id":"2054fc6c1531b087b0a87f452c3b83df30880ff6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-22 17:39:27.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"486bf33499d66d9ba23982f7ca39ec0252fe1a43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 17:56:12.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b36a6c6052bd408a883a800a66427989\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b803f5bc53b4d9baffbffb3bc6de9fe : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/38739d5dd24340efbd9564a539ae24d2 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bcee467ca0e54c44bf25a36e1bc6da37 : SUCCESS in 10m 47s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"95a02dc3b4ec1ecd0f064bc237a4c8e5a63b6a70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 18:13:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79d834e46af74da5a5cb73cc1646c517\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/649adeb3243c401e8641b905cea90cc0 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38a881271f814e508dfa8e18279930cd : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/251ce81234ec4f21805f4e56d2abad36 : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4afb3b6c9364b5d84c4429585e36093 : SUCCESS in 7m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb2a1229eb94415e8822a57c074fb230 : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/306a4ee35cf1437c9585488f829c6251 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e6a0a824f1a4126a4f18788d399c095 : SUCCESS in 5m 18s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/2c9258ea380b4f3fa6734394707504ef : FAILURE in 30m 05s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3ebc85883d8d40de8c8d8fc21172f35d : FAILURE in 17m 49s (non-voting)\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/d04076d18b254357b102b8463c05fcad : FAILURE in 32m 48s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/76f072865a9c485ba499804e9ee2e32d : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":9},{"id":"46533369b66d06dca52c23ed199de1e692e677a3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-23 06:53:07.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"77c497137d488ca05f9f6806aab6585b54fa58d6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-23 07:52:26.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"049646848ee5616af49852d7b12e58e597169430","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-05-27 13:51:17.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"f55f28aa27c6ebabd6d486c913179edd895aff21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 13:51:28.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"1d95c89519be65ae36002b48c8afe334b63da48c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 14:32:08.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/820b601d94294c86bc71d1850df9e868\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6edee02adfd149629053a15270e4fe37 : SUCCESS in 7m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf44841a108f4fdd953fedcda9ed592e : SUCCESS in 6m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7cde25c010524cee92e1f5e26d79f6f3 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80b837c6baf146ba8e868ff787014502 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4d71b66c8a84e3aace4b0837c1cc9e8 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":9},{"id":"3be5705a07a5bbcfc4ea696441ee41ff27acd929","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 14:32:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"d5a360a8e635a2707af32f61a5849373cc2ed8a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 14:33:28.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7005fa31ce24ed78910fe328aff4db7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8f9ac3f0230432cab7575369af7cee0 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf67b2ec2ed1426181a1d6ba0140d8ff : SUCCESS in 57s","accounts_in_message":[],"_revision_number":9},{"id":"bd7a13a5cff7cb2532cf22641eb8085947deea61","tag":"autogenerated:gerrit:revert","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-13 01:45:33.000000000","message":"Created a revert of this change as Ie0fea6e30c7944cfbad7de3da3f8f2e5272819cf","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"bcdec128e4f96b2d931e6a67cf6401eff5bb1cd4","revisions":{"6309b608b49ad41ec0c118af469f80f5dd4a6700":{"kind":"REWORK","_number":1,"created":"2025-05-15 16:28:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/949915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/949915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/949915/1"}}},"commit":{"parents":[{"commit":"76cd5a7bf941f22da8395b1cbc9e3ac7c0607758","subject":"Add build of modern Ubuntu distributions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76cd5a7bf941f22da8395b1cbc9e3ac7c0607758"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-15 16:27:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-15 16:27:41.000000000","tz":120},"subject":"Introduce mariadb_backup backup_strategy","message":"Introduce mariadb_backup backup_strategy\n\nWith mariabackup being renamed to mariadb-backup binary, to keep\nsome backwards compatability instead of renaming the binary\nto call this patch introduces the whole new strategy designed to\npretty much override only used binary name in\n\nChange-Id: I5336c8f98d9f1f33810113b9d7a84f1bef3deb66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6309b608b49ad41ec0c118af469f80f5dd4a6700"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6309b608b49ad41ec0c118af469f80f5dd4a6700"}]},"branch":"refs/heads/master"},"6a4bef08b449fa32d32fd4f213da939fdd12094c":{"kind":"REWORK","_number":2,"created":"2025-05-15 16:34:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/949915/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/949915/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/949915/2"}}},"commit":{"parents":[{"commit":"76cd5a7bf941f22da8395b1cbc9e3ac7c0607758","subject":"Add build of modern Ubuntu distributions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76cd5a7bf941f22da8395b1cbc9e3ac7c0607758"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-15 16:27:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-15 16:34:31.000000000","tz":120},"subject":"Introduce mariadb_backup backup_strategy","message":"Introduce mariadb_backup backup_strategy\n\nWith mariabackup being renamed to mariadb-backup binary, to keep\nsome backwards compatability instead of renaming the binary\nto call this patch introduces the whole new strategy designed to\npretty much override only used binary name in\n\nChange-Id: I5336c8f98d9f1f33810113b9d7a84f1bef3deb66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6a4bef08b449fa32d32fd4f213da939fdd12094c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6a4bef08b449fa32d32fd4f213da939fdd12094c"}]},"branch":"refs/heads/master"},"6786498007c4abbc8c33dda22796a112dfdfcc65":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-05-15 16:38:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/949915/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/949915/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/949915/3"}}},"commit":{"parents":[{"commit":"76cd5a7bf941f22da8395b1cbc9e3ac7c0607758","subject":"Add build of modern Ubuntu distributions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76cd5a7bf941f22da8395b1cbc9e3ac7c0607758"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-15 16:27:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-15 16:36:21.000000000","tz":120},"subject":"Introduce mariadb_backup backup_strategy","message":"Introduce mariadb_backup backup_strategy\n\nWith mariabackup being renamed to mariadb-backup binary, to keep\nsome backwards compatability instead of renaming the binary\nto call, this patch introduces the whole new strategy designed to\npretty much override only used binary name for modern MariaDB\nversions.\n\nThis new strategy is set as a default one in config. With that\noperator might override the strategy to `mariabackup` and keep\nusing EOL-ed MariaDB versions.\n\nThis patch is alternative to [1]\n\n[1] https://review.opendev.org/c/openstack/trove/+/948643\n\nChange-Id: I5336c8f98d9f1f33810113b9d7a84f1bef3deb66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6786498007c4abbc8c33dda22796a112dfdfcc65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6786498007c4abbc8c33dda22796a112dfdfcc65"}]},"branch":"refs/heads/master"},"2e37aa118a85f74a046c3953ade063710ab25cf8":{"kind":"REWORK","_number":4,"created":"2025-05-15 19:17:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/949915/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/949915/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/949915/4"}}},"commit":{"parents":[{"commit":"76cd5a7bf941f22da8395b1cbc9e3ac7c0607758","subject":"Add build of modern Ubuntu distributions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76cd5a7bf941f22da8395b1cbc9e3ac7c0607758"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-15 16:27:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-15 19:17:43.000000000","tz":120},"subject":"Introduce mariadb_backup backup_strategy","message":"Introduce mariadb_backup backup_strategy\n\nWith mariabackup being renamed to mariadb-backup binary, to keep\nsome backwards compatability instead of renaming the binary\nto call, this patch introduces the whole new strategy designed to\npretty much override only used binary name for modern MariaDB\nversions.\n\nThis new strategy is set as a default one in config. With that\noperator might override the strategy to `mariabackup` and keep\nusing EOL-ed MariaDB versions.\n\nThis patch is alternative to [1]\n\n[1] https://review.opendev.org/c/openstack/trove/+/948643\n\nChange-Id: I5336c8f98d9f1f33810113b9d7a84f1bef3deb66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2e37aa118a85f74a046c3953ade063710ab25cf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2e37aa118a85f74a046c3953ade063710ab25cf8"}]},"branch":"refs/heads/master"},"388d81c698104fec974124f7fd72e20c8fe6064c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-05-15 19:18:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/949915/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/949915/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/949915/5"}}},"commit":{"parents":[{"commit":"02c4f82729b180aaef03fbc30fc497917047fea9","subject":"Add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/02c4f82729b180aaef03fbc30fc497917047fea9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-15 16:27:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-15 19:18:59.000000000","tz":0},"subject":"Introduce mariadb_backup backup_strategy","message":"Introduce mariadb_backup backup_strategy\n\nWith mariabackup being renamed to mariadb-backup binary, to keep\nsome backwards compatability instead of renaming the binary\nto call, this patch introduces the whole new strategy designed to\npretty much override only used binary name for modern MariaDB\nversions.\n\nThis new strategy is set as a default one in config. With that\noperator might override the strategy to `mariabackup` and keep\nusing EOL-ed MariaDB versions.\n\nThis patch is alternative to [1]\n\n[1] https://review.opendev.org/c/openstack/trove/+/948643\n\nChange-Id: I5336c8f98d9f1f33810113b9d7a84f1bef3deb66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/388d81c698104fec974124f7fd72e20c8fe6064c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/388d81c698104fec974124f7fd72e20c8fe6064c"}]},"branch":"refs/heads/master","description":"Rebase"},"7a8c158016d0ba99959fd23e7749a3e78abea30e":{"kind":"REWORK","_number":6,"created":"2025-05-22 13:28:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/949915/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/949915/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/949915/6"}}},"commit":{"parents":[{"commit":"3e0b163d800afcb4d020dbf0c5f68886fa69bc23","subject":"Add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3e0b163d800afcb4d020dbf0c5f68886fa69bc23"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-15 16:27:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-22 13:27:09.000000000","tz":120},"subject":"Introduce mariadb_backup backup_strategy","message":"Introduce mariadb_backup backup_strategy\n\nWith mariabackup being renamed to mariadb-backup binary, to keep\nsome backwards compatability instead of renaming the binary\nto call, this patch introduces the whole new strategy designed to\npretty much override only used binary name for modern MariaDB\nversions.\n\nThis new strategy is selected whenever requested datastore version\nis equal or higher 10.4.\n\nThis patch is alternative to [1]\n\n[1] https://review.opendev.org/c/openstack/trove/+/948643\n\nChange-Id: I5336c8f98d9f1f33810113b9d7a84f1bef3deb66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a8c158016d0ba99959fd23e7749a3e78abea30e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a8c158016d0ba99959fd23e7749a3e78abea30e"}]},"branch":"refs/heads/master"},"6eb8cc2a38ab8be36097d072e5ed25123330f72d":{"kind":"REWORK","_number":7,"created":"2025-05-22 13:29:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/949915/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/949915/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/949915/7"}}},"commit":{"parents":[{"commit":"3e0b163d800afcb4d020dbf0c5f68886fa69bc23","subject":"Add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3e0b163d800afcb4d020dbf0c5f68886fa69bc23"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-15 16:27:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-22 13:29:53.000000000","tz":0},"subject":"Introduce mariadb_backup backup_strategy","message":"Introduce mariadb_backup backup_strategy\n\nWith mariabackup being renamed to mariadb-backup binary, to keep\nsome backwards compatability instead of renaming the binary\nto call, this patch introduces the whole new strategy designed to\npretty much override only used binary name for modern MariaDB\nversions.\n\nThis new strategy is selected whenever requested datastore version\nis equal or higher 10.4.\n\nThis patch is alternative to [1]\n\n[1] https://review.opendev.org/c/openstack/trove/+/948643\n\nChange-Id: I5336c8f98d9f1f33810113b9d7a84f1bef3deb66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6eb8cc2a38ab8be36097d072e5ed25123330f72d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6eb8cc2a38ab8be36097d072e5ed25123330f72d"}]},"branch":"refs/heads/master"},"d6a54ea9592eb4d4f850633f070c7dae64df295f":{"kind":"REWORK","_number":8,"created":"2025-05-22 13:31:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/949915/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/949915/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/949915/8"}}},"commit":{"parents":[{"commit":"3e0b163d800afcb4d020dbf0c5f68886fa69bc23","subject":"Add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3e0b163d800afcb4d020dbf0c5f68886fa69bc23"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-15 16:27:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-22 13:31:29.000000000","tz":0},"subject":"Introduce mariadb_backup backup_strategy","message":"Introduce mariadb_backup backup_strategy\n\nWith mariabackup being renamed to mariadb-backup binary, to keep\nsome backwards compatability instead of renaming the binary\nto call, this patch introduces the whole new strategy designed to\npretty much override only used binary name for modern MariaDB\nversions.\n\nThis new strategy is selected whenever requested datastore version\nis equal or higher 10.4.\n\nThis patch is alternative to [1]\n\n[1] https://review.opendev.org/c/openstack/trove/+/948643\n\nChange-Id: I5336c8f98d9f1f33810113b9d7a84f1bef3deb66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d6a54ea9592eb4d4f850633f070c7dae64df295f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d6a54ea9592eb4d4f850633f070c7dae64df295f"}]},"branch":"refs/heads/master"},"bcdec128e4f96b2d931e6a67cf6401eff5bb1cd4":{"kind":"REWORK","_number":9,"created":"2025-05-22 17:39:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/949915/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/949915/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/949915/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/949915/9"}}},"commit":{"parents":[{"commit":"3e0b163d800afcb4d020dbf0c5f68886fa69bc23","subject":"Add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3e0b163d800afcb4d020dbf0c5f68886fa69bc23"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-15 16:27:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-05-22 17:39:27.000000000","tz":0},"subject":"Introduce mariadb_backup backup_strategy","message":"Introduce mariadb_backup backup_strategy\n\nWith mariabackup being renamed to mariadb-backup binary, to keep\nsome backwards compatability instead of renaming the binary\nto call, this patch introduces the whole new strategy designed to\npretty much override only used binary name for modern MariaDB\nversions.\n\nThis new strategy is selected whenever requested datastore version\nis equal or higher 10.4.\n\nThis patch is alternative to [1]\n\n[1] https://review.opendev.org/c/openstack/trove/+/948643\n\nChange-Id: I5336c8f98d9f1f33810113b9d7a84f1bef3deb66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bcdec128e4f96b2d931e6a67cf6401eff5bb1cd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bcdec128e4f96b2d931e6a67cf6401eff5bb1cd4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
