)]}'
{"id":"openstack%2Ftrove~951005","triplet_id":"openstack%2Ftrove~master~Ifc88027096caeb46fabd84e4a952667e6b003d80","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-06-24 07:34:47.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-06-30 02:04:32.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-06-09 16:31:18.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ifc88027096caeb46fabd84e4a952667e6b003d80","subject":"Push datastore image to quay.io to support tests","status":"MERGED","created":"2025-05-27 14:45:55.000000000","updated":"2025-06-30 06:43:04.000000000","submitted":"2025-06-30 06:42:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951005","meta_rev_id":"239b05764a23d25ebf896516f154593f0ba7ee48","_number":951005,"virtual_id_number":951005,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-30 06:42:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"recommended":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-06-24 07:34:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-30 05:55:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2025-06-24 07:53:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-30 05:55:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-27 15:02:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-27 16:59:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-09 11:33:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-06-09 14:48:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-06-24 07:53:30.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"a0a64101d62f52c64d92e50d96bb5faa0b77573a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-05-27 14:45:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ce54356fdc2cdc503bf489d2d5537f9cecc8cf4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-05-27 14:47:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c39c46bb72c3ef0bc42eefa24ba3df9413e31ec4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 15:02:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cf02db2da184e90b203832e93713c00\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3f3a2fce592740da9f5d57445cf00464 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/afea488807eb4481a7bf339a9c721460 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4bd674438f04473791409821977ba21c : SUCCESS in 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4cb02ccccbbcbb0442d42c3a987edc73e3d7b16f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 16:59:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e439b4e88f704dc880d79004d95c69b5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72ef3bea9fca4c2497e8813375c124aa : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/383889fef0324b44893ca56f913e875f : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74942fe6c94f47c89bc243c66623ade3 : SUCCESS in 6m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/60727fe7198844aa945509e3f6ef557b : SUCCESS in 7m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac1965ef257d4ccdbb9d7f60c24f8464 : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1b2273708cd4cc09a9f495550ebf89e : SUCCESS in 7m 25s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/44116c9d8d91471ba3c14e6c3527f858 : SUCCESS in 52m 07s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/17fa1ad51f98408a95c450844bf72b0c : SUCCESS in 1h 02m 25s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/f76acaa869634ddbb510d39a74a35b29 : FAILURE in 42m 39s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/da89af64704043be82eea9a2b3dbc61a : FAILURE in 29m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/d60d213ae123483499857643d63f4885 : SUCCESS in 2h 12m 25s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3f30b3dd9f2b4c59a4ea8b074fd718cf : SUCCESS in 1h 00m 45s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/cb7d0c96e30b4773981ae3936790d5ca : FAILURE in 35m 47s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9e63b549f55b448c85508060b00e2e5a : FAILURE in 36m 24s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/945adbb8a5394f2f9618f3466abdbabb : SUCCESS in 1h 59m 56s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/7072b56ceb024710996e39d9b4eeb508 : SUCCESS in 1h 02m 18s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/c830e8db06a443acab1585cdc94dd3fa : FAILURE in 33m 52s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/4515d12b959946f080433bb7d2c0ebba : FAILURE in 38m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/d51cc0c7046f483cba04fb819fd6a101 : SUCCESS in 1h 37m 54s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/846b40520ce1418d968591f9b2339ef9 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"47defaaecf2dff5cf42714dae9a9d31102884318","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-09 11:33:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d13ad6e5835e1592a080b20dd11aaf80f035997","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-09 12:16:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6790c59aa7fffc69b45cda30b82cee5684fa60cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-09 14:47:51.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e2ca1f2886e1d6485ac4bfb8b8e1e221eae51dd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 15:18:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/126cdb0bd9274038acf4c93d29b3f152\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/867b86b2b4ed441cbb8bf2a7e2196415 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/49606daae8de4b3b9f2b410b82bed430 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3dee9fc107fa43498bfdc793e63b4a50 : SUCCESS in 11m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7be993e8a8eacc4d28d29a68b0cb83c2b205f00b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 16:31:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35e8dd33e9e4485eb4ed226178578d3f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47868e078545437385932444cff5b7ba : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2248f47c5ee04a659b2782f9b071c109 : SUCCESS in 6m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0df2d4d5b00a450891666790dc7412bd : SUCCESS in 5m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9914054416234d99bd4f2de7a7dd96a5 : SUCCESS in 5m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c6661ada1f7347799058ba164c38b48b : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4dafb5ee90b4b528cbd3f8071c98214 : SUCCESS in 5m 35s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/6debff102a5d4df0a943fe0dae4c94f7 : FAILURE in 1h 12m 42s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/53c9cc64b5f94f69942ad6decf439faa : FAILURE in 1h 01m 40s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/8ec702baa32c4494827e5db528b7eb05 : FAILURE in 39m 13s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/12ebd14db80543b680f5f8ded458d6e9 : FAILURE in 31m 53s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/2d193a76d347438488a4a185199d773d : FAILURE in 1h 39m 11s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fbff846f6df84991a62df8892a72b9ac : FAILURE in 49m 29s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a6e725d0debb4dfb9b7e4f7c643f2c6c : FAILURE in 24m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b6023c3971094a59be1f46ab6eb8c716 : FAILURE in 19m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/802b7800fba741e7b0f43a9cd4994d9f : FAILURE in 1h 31m 25s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/464a63d6d9484326ac31bf08061a3d47 : FAILURE in 49m 22s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/c2f2216fa5e84de288bcb77d9080950e : FAILURE in 23m 05s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/a9e1728cd88644088c395e2d644b2d36 : FAILURE in 19m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/354323dedd054f07b1cf64df8dbc1c86 : FAILURE in 1h 30m 40s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7cbd1ffdca8a43edbf9cbe2740fcdbb5 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":2},{"id":"b0b4799e99cddc573636d26ee0a060c0a4611849","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-19 02:47:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e3c0e95132768b26252625668f15dfde1ef854be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 03:39:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e8e9ca4fa894d4bb81a034db1df9772\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d61205181fec4596a1a046aae977418a : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9d234cc1bd06459b9c56abd2f59dc6e6 : SUCCESS in 12m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/611320e17ac942bea978aa3f0d0f01ff : SUCCESS in 16m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f9713d32dc68e676190e0fc6526800946e9ca077","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 05:55:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e56f929272e4e1485eb347786b726b9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea9f221810af44d7b8a8e99644cbd877 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cddaa97d9fdf4887b8eb70cc22fed6fa : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0f13c59163944f4b4ad0e9cb71b3d5e : SUCCESS in 6m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f6332fa8ee84ffea1078b92a2c96b0d : SUCCESS in 5m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0fb925fd701d40f5a4f8d2b413b17069 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/086c5c6466cd4182903a2075694135f7 : SUCCESS in 5m 12s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/ff4bc472f0104a9d8f41784a756ee510 : SUCCESS in 2h 27m 47s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/d0a6b5ec71bb4925b33fadada08b6e58 : FAILURE in 2h 08m 58s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/2c2d1bb469e041aca1f2491dbaf754fa : FAILURE in 1h 44m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/bf76277d405743cfa0faba9293a18e08 : FAILURE in 2h 35m 48s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2a6b612ad3ef4c26965f18b7773f66c7 : SUCCESS in 1h 18m 25s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0165a4d37db84e73b6ef18926290ea69 : FAILURE in 2h 02m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/558fc2415eb34728a487953f18bf8753 : FAILURE in 1h 55m 18s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c6b5bc6a5e8140188ed88b4e409a9e7a : TIMED_OUT in 3h 03m 51s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/3cab0730594441ecb49e044c46813861 : FAILURE in 52m 13s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/c14993fcf2b74b15a229e2fb3e6934f1 : FAILURE in 1h 23m 45s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/81e3892f38f84e8d967a2d79a15da676 : FAILURE in 1h 37m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/09f8c3245d194a3a9cd15eb43cf435b7 : FAILURE in 2h 43m 34s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/85258101ce7d408389be2ca00b6623da : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":3},{"id":"f046085fa517a6e7e2a48659840566cdc5516a33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-24 07:26:36.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"63678a26f8cb62e285b51eb721446193cb97e6b7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-24 07:34:47.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0ab282c1c48787e827e34732a7a9b516c2c6fe85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 07:46:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f82c81bbf554fd89bb563278d73c35f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/09b4f45bbc7042f49b92bedee5a40740 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cedb6ab6e7f8438c9233d698b4151de9 : SUCCESS in 13m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/76e6e1cfd5d948c2ba864f99f4358298 : SUCCESS in 17m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"653e07308ef9b0a6715b5072e4f963e27ad1078e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-24 07:53:30.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0282f8c8491f750f0c4f25a938888e35c7451022","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 09:57:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13caaf5f966f4603a080feb2084a8daf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf82d1287d54452ead63c785f986c4f3 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0f4c73cb3fb4a77892a773138011279 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9605c57e60b4374984cc195d18348cb : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c265142a18ef4fa6a691108fb69a6e61 : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/289a0e46b2d34a9b946b2e253a48a073 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a280ca7d94c945a8b1d1d8694b6c9bcc : SUCCESS in 5m 01s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/1501f1b22a734f3994e5d5f1589bacb7 : SUCCESS in 39m 54s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/224eaab4cd4049f999b3717d9f9e4b0d : FAILURE in 1h 24m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/c6d7a077bbc7418dbf75545531b21a37 : SUCCESS in 1h 05m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/c47313a8341d41e6ab5c606c766738e4 : FAILURE in 1h 37m 13s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/107674ff678746518e15fd514bdee6c4 : FAILURE in 35m 28s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e1d2dd7d95e3405e84d0df8148026ef2 : SUCCESS in 1h 04m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fcb9bb57fbd44ed79fb424a1dedaa4af : SUCCESS in 2h 00m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/36cce8396af14f4fb3ccb8653497eb65 : SUCCESS in 1h 08m 03s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/d9b5c1d861e34967ac2e78c826b4448a : FAILURE in 50m 33s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/10a2162b38d54fec9fcc47963dbecc3e : FAILURE in 1h 10m 56s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/aa9aedb9e85149c0b61e2cfba2c3c07b : FAILURE in 1h 38m 24s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/8733f7f2e57d456487c8755b59453ec0 : SUCCESS in 2h 28m 29s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cef1d5905b3e4b08ae3450bfa9f9d93b : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":4},{"id":"fd66607f0a79223018e318ad142c415941e56122","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-25 06:40:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"976e39027958570dc309f3c1191d8f042b45d91d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 07:30:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2459a9b91a6c441694072e1f3ba944fd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3b17f98f6b7e4dd69aaaa7ebd7aeada1 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bb46b62971bd452fb58c716401d03f94 : SUCCESS in 13m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3bd4c093ec4a4b20bf0a9a571373ae15 : SUCCESS in 11m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fa4fb4729b96064bb585a9cc5f2a0065380f9f65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 09:19:53.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/61f6049cfeee4c16a0726a177a8df0d2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65c9be526ec048688efccc77251a7669 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27de4ee1419542528d29a0994de06a6b : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3a738e78b374f0895565e902122b24f : SUCCESS in 5m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/469a3230e2054012a20e4b236d2c5297 : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6bdb6f63df9e45b7bc17988ec8dbcead : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9077a2a27a204499bab2a446d5c353d6 : SUCCESS in 3m 33s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/af98cac17d914c018925b5d20c7cf8f0 : SUCCESS in 1h 12m 32s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/e9d413a0f7af4d4d8cf800a67dad1bce : FAILURE in 37m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/28e2058d79784792bc59c88a3995b602 : SUCCESS in 2h 07m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/5966c564a8f2414a8a19d09476e113cd : SUCCESS in 2h 38m 44s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/331983dfb8c847938dd355a12f685cba : FAILURE in 30m 06s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8e1210d6dfac43de9d0749ee842a8303 : SUCCESS in 2h 07m 05s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7cda36e56e664a2fab9925a56931aa05 : FAILURE in 1h 02m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f0dd77f25f1f458ca62cebd4556f3f94 : SUCCESS in 2h 04m 52s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/91f36ec1833a4a0ba5072522cb3d6205 : SUCCESS in 41m 33s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/f68d4ffde73d45ba8c623571e2049b6d : SUCCESS in 2h 05m 54s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/7bab450438a14867b78344122517c213 : SUCCESS in 1h 26m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/f073d2a7939d40a28ce455ca81d1e418 : SUCCESS in 1h 49m 51s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a163abfae9374432a631787ec7bd8fde : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":4},{"id":"e47e3a294890794a2236023b305680719741176a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-26 02:13:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2becb29760234f44d917f4f548b3a0cf7cbf16c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 03:29:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64b21e2ad60249698113d641ea257aad\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/623e6b8d0e8449f895db94aa37172e72 : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/626c06596361438c9715ed21a52571fb : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/86a13fa5ec584e26bf029b0abaec8f6e : SUCCESS in 18m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f6ed84cf79c0a2f6c610ec7806c110b163b78efe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 04:47:29.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78f63495e54f4e6c86c7370f9a0b406e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94e2b00eda8744c99e45b3b988614b36 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad18c425194e4089be32f975548514e0 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e88862ad874944f8904e64d4766b5fa7 : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77f5c00d858344e5847e563fe7117c00 : SUCCESS in 5m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81da7b441da745fd8d37823be2b8c6bd : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd1a21d58b284b48ad7cdc03becc578d : SUCCESS in 5m 15s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/f99e91d645a34bc4828a03900b18847c : SUCCESS in 45m 18s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/5317d11cf60e4c43b314e9e373d01135 : FAILURE in 1h 54m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/801953df57bd4d04b2d4f80358a389cd : SUCCESS in 1h 53m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/b102e412d18c43b4af3ac0aa208b8019 : SUCCESS in 2h 20m 53s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fa922c957eff4d80b849bde985cb9b87 : FAILURE in 1h 18m 12s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/743270dfc843460fb13237960bacac0b : SUCCESS in 1h 46m 09s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4dd38ce472374076b69f053837db2993 : SUCCESS in 1h 06m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c9030c6b84bc48eb9447705d619c4afe : SUCCESS in 1h 37m 52s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/1068c294b35642778791b8c32cdf70d8 : SUCCESS in 38m 12s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/96614e41e0dd4d39bca360751f590547 : SUCCESS in 1h 50m 09s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/bfce6156c1584e19a27f31ccf94c0349 : SUCCESS in 1h 34m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/3e2f2855fe474db3a131a0bd253102a4 : SUCCESS in 1h 48m 22s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a4d2ce7b40564b1c82ca8fc138f487ca : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":4},{"id":"66857b48c4f5914d17c9a4212f20b249be5641df","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-30 02:04:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"be7764ecfc922611b2b97346cfa1f7ce20c76070","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 03:24:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e2bd7df5389415ab14fbebca15bdc21\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/17bad625dd5444999b1b1248e2110bd5 : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a02174a316b04c20921913e641395658 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d5ff6fafadef4ef08870ed6b774c6eb5 : SUCCESS in 18m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e612d1c2fdecc5dd1308c827d5b570ada3d889ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 04:29:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d288d44e366d45a6a9050f05f6dca242\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c8a96e2830a411dbcdde441051e4896 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25529f172f7c42b1ba9b68f10a39021d : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aeec93257a9b4e7c976b7898b5a888d3 : SUCCESS in 7m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c9e3e77f0204435be5cec4d3543f5f6 : SUCCESS in 5m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/690cce76889941ea8831577daf253d5b : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c31c4490fca4084b34bbe6617553d0c : SUCCESS in 5m 10s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/f8698c947d4a42e495246c46ba83d0f3 : SUCCESS in 46m 41s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/273cf136e8c5477590639ba201559ba2 : FAILURE in 1h 21m 44s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/c2e63693bbba4352af9249e32b05d436 : SUCCESS in 1h 20m 24s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/32bea425d6fb4b1d8b793d0a0e6d6210 : SUCCESS in 1h 36m 16s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/82463e798f3a416da20a3550a5fc4629 : SUCCESS in 41m 50s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9bf581e1bce0461fbf41578be5b40d93 : SUCCESS in 1h 20m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/451c77197d3c4d3889dfe96d752ef486 : SUCCESS in 1h 18m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e01879e55d254b8a80e16bb917db103e : SUCCESS in 1h 36m 40s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/2d7c6777c8624f72b5f980355176fae5 : SUCCESS in 59m 10s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/ba64aaed48ad458881b3a8f75f6421e1 : SUCCESS in 1h 55m 46s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/68cdf0ba389641579557ad0c660ccf0d : FAILURE in 53m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/aaa28215f2b34f59b6da1bfc57308c9b : SUCCESS in 1h 53m 02s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1c1e835fa2734ac3aec9ea968addc855 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":4},{"id":"72d8cc3168c6f948630be9d2557728e985e8271f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-30 05:55:25.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4107719426c212f1ec0d8e2823f614da17243f20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 05:55:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"97c8d48eb0395bfc7d79d27ee6e004a2b627b03c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 06:42:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fb0e6974877448e91f7fda67d6bbc24\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/627d86fae9d440819a1df181d4001199 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51bdd7fbbc0b407bb8f0b442156959c3 : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a191b2abf3944be929ca6dc8ec5b504 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fe5655110ee438a8943ff39f65b643f : SUCCESS in 5m 30s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/ced3a147a8ea4a5f9d8f9aa54fcb58c7 : SUCCESS in 46m 02s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d7a505fc99ae45129b207c1ce6a8832c : SUCCESS in 38m 51s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/8775d0107ed14d80a46254d642bda11d : SUCCESS in 41m 42s","accounts_in_message":[],"_revision_number":4},{"id":"4e52c01bffe4520dca4801689e0e7307360913be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 06:42:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"239b05764a23d25ebf896516f154593f0ba7ee48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 06:43:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef06174e3e45476f9e58f9df7aba1f4f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef76cfec31594622bb8e5b79c4eae91b : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d6f7e8e086b66ba2daa4cf4d0c915db4daf8b1f7","revisions":{"715e3205ef02445a6fd531642419ee095f5cc410":{"kind":"REWORK","_number":1,"created":"2025-05-27 14:45:55.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/05/951005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/951005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/951005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/951005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/951005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/951005/1"}}},"commit":{"parents":[{"commit":"b2d4bfd30807296f20e42401a5285f2bc6dfe991","subject":"Merge \"Fix cloud-init configuration file issue with duplicate headers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b2d4bfd30807296f20e42401a5285f2bc6dfe991"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-05-27 14:44:28.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-05-27 14:44:28.000000000","tz":480},"subject":"Push postgresql17 and mariadb 11.4 to quay.io to support tests","message":"Push postgresql17 and mariadb 11.4 to quay.io to support tests\n\nChange-Id: Ifc88027096caeb46fabd84e4a952667e6b003d80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/715e3205ef02445a6fd531642419ee095f5cc410"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/715e3205ef02445a6fd531642419ee095f5cc410"}]},"branch":"refs/heads/master"},"f4b5a8d6fa64078c1b06e0503c8056f73a84df06":{"kind":"REWORK","_number":2,"created":"2025-06-09 14:47:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/05/951005/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/951005/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/951005/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/951005/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/951005/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/951005/2"}}},"commit":{"parents":[{"commit":"b2d4bfd30807296f20e42401a5285f2bc6dfe991","subject":"Merge \"Fix cloud-init configuration file issue with duplicate headers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b2d4bfd30807296f20e42401a5285f2bc6dfe991"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-05-27 14:44:28.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-09 14:47:51.000000000","tz":0},"subject":"Push postgresql 17 and mariadb 11.8 to quay.io to support tests","message":"Push postgresql 17 and mariadb 11.8 to quay.io to support tests\n\nChange-Id: Ifc88027096caeb46fabd84e4a952667e6b003d80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4b5a8d6fa64078c1b06e0503c8056f73a84df06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f4b5a8d6fa64078c1b06e0503c8056f73a84df06"}]},"branch":"refs/heads/master"},"ad68f7a6fcea85aaf3f7041e3ee6b6198286b2fa":{"kind":"REWORK","_number":3,"created":"2025-06-19 02:47:50.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/05/951005/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/951005/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/951005/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/951005/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/951005/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/951005/3"}}},"commit":{"parents":[{"commit":"e9f62185701da253a66d3d4c16731ad1f3121134","subject":"Add mysql-8.0 tests and remove mysql-5.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9f62185701da253a66d3d4c16731ad1f3121134"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-05-27 14:44:28.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-19 02:46:57.000000000","tz":480},"subject":"Push postgresql17 and mariadb 11.4 to quay.io to support tests","message":"Push postgresql17 and mariadb 11.4 to quay.io to support tests\n\nChange-Id: Ifc88027096caeb46fabd84e4a952667e6b003d80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ad68f7a6fcea85aaf3f7041e3ee6b6198286b2fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ad68f7a6fcea85aaf3f7041e3ee6b6198286b2fa"}]},"branch":"refs/heads/master"},"d6f7e8e086b66ba2daa4cf4d0c915db4daf8b1f7":{"kind":"REWORK","_number":4,"created":"2025-06-24 07:26:36.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/05/951005/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/951005/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/951005/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/951005/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/951005/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/951005/4"}}},"commit":{"parents":[{"commit":"e9f62185701da253a66d3d4c16731ad1f3121134","subject":"Add mysql-8.0 tests and remove mysql-5.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9f62185701da253a66d3d4c16731ad1f3121134"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-05-27 14:44:28.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-24 07:25:25.000000000","tz":480},"subject":"Push datastore image to quay.io to support tests","message":"Push datastore image to quay.io to support tests\n\nthis commit push mysql8.4,mariadb11.4 and postgres 17 to\nquay.io\n\nChange-Id: Ifc88027096caeb46fabd84e4a952667e6b003d80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d6f7e8e086b66ba2daa4cf4d0c915db4daf8b1f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d6f7e8e086b66ba2daa4cf4d0c915db4daf8b1f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
