)]}'
{"id":"openstack%2Ftrove~951825","triplet_id":"openstack%2Ftrove~master~I67409071f9796121316c9f894c7a815846bc74ef","project":"openstack/trove","branch":"master","topic":"cinder-backup-fix","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-06-30 02:20:41.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I67409071f9796121316c9f894c7a815846bc74ef","subject":"Improve stop_slave logic for cinder backup","status":"MERGED","created":"2025-06-05 07:23:12.000000000","updated":"2025-06-30 03:35:05.000000000","submitted":"2025-06-30 03:34:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951825-cinder-backup-fix","meta_rev_id":"aab00663434ce825a9596d3e05dbab4670d2f628","_number":951825,"virtual_id_number":951825,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-30 03:34:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-30 02:20:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-30 02:20:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-05 07:56:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-05 09:32:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-12 07:09:26.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"CC"}],"messages":[{"id":"5215fc7e7273798764e911fa0f0bebf9d8301b8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-05 07:23:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fb301b508cbe60f10c9a6c015de313ba15e06fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 07:56:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea501baa79994ef39a76c3f294d706c0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/68f54cf5556940358855d0fb67899e9d : SUCCESS in 12m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e6d70f3b21a148ec9b82c2a03bd22eff : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/393097adf5194bdcb16330694bcbb98f : SUCCESS in 18m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c288af974eca7296d117bb4cacca3c998ef9748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 09:32:25.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c6307327fe14819ade982e9f5b0abee\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1298fb447bd24eb5824fee3e6bc3cfaf : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81057c5b34a340ceae59ea9b599d8a61 : FAILURE in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04c66d669ed7477d9b0b97e4f8e3af20 : SUCCESS in 5m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8aa3202a899f478c93172ff9ce6b367e : SUCCESS in 6m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5177a7be859c437994ab5edd19e1b878 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76f5d1f74b6e43f0bc5540aff5493192 : SUCCESS in 5m 15s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/698319a8ebc447208f5da6d8b82394e4 : SUCCESS in 27m 42s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/95369916908e4d5b8133a795193dc83f : SUCCESS in 1h 04m 31s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/41d35838482d4eb78c25e0cbb6ccdd5a : FAILURE in 22m 03s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/4669e8fade5b45c89f0c06c1c695f1e4 : FAILURE in 21m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/fbdce084f3504a67a04bfd82abad44ce : FAILURE in 2h 08m 17s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8d7bb40e72994551b07d83bdb7438bdb : SUCCESS in 1h 00m 26s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/cd96abdecc8047549329c75d41eeadef : FAILURE in 22m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/967600309d0d4b0dbb5fbb71f6d74581 : FAILURE in 23m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/46f2ef43fae8409a8d21f1ace0887437 : FAILURE in 1h 54m 32s\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/124a53b92b2343a0bdf736b99d6f0bec : SUCCESS in 1h 56m 28s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/49cedfc1d6444502a826439382f35b5d : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a9b4719b8c42cb9e4f2ad5a2fb167a2b3b716df7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-09 07:03:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"754e6ab72b3677eb2778922a5171395965239e10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 07:16:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccfdd37f0e334f11b1f7f51b36cf4af7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/028380fdf33b42109a827ca71a4d0421 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e7a8347630a94e57b2190a4df3bf42a8 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3713331019fe45559531f7178aab5d0b : SUCCESS in 11m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ccbd82b2d4af3e37f32d289b8d78b64faddbf323","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 08:44:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed8c6fa9d5414601b5e2f16350f46190\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72a43c94323f48399a0fd7cb6dd0d754 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10e68ced33df4ab9a520fcb9fb85ac30 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/009af6e2d0be4d38b598fba77fb016c0 : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/21d08f9747d1434db181563096e30411 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3aeda861f4b445e7927f66fc9c0e300e : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6122003430f24e13a22b1797c89230ba : SUCCESS in 4m 49s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/4759948a1256404ea4c0c93bfb21542a : FAILURE in 58m 37s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/36c1828b6f34415ea63eb1fccdf6d1a1 : FAILURE in 47m 25s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/b2138c008fa340cd9d828c788cecf48b : FAILURE in 28m 13s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/75f51ad6ada549499caa042327c857bc : FAILURE in 28m 33s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/aa1b008c842b4627aaabe25d39455f36 : FAILURE in 1h 29m 05s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/42699bd68d644952a31a37009dad9cc8 : FAILURE in 1h 01m 19s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d3e9d8fc4ac5484a987c673b2f35f44a : FAILURE in 22m 58s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f9a41554602a46abbc9719f9ef9e61c8 : FAILURE in 33m 09s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2b8d1f905d164fa2a253d162ea3662ca : FAILURE in 1h 28m 41s\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/b193c082a0624ce89cd63c5455e62783 : FAILURE in 1h 40m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9968bc3aed174294a4f0672a07ce76a2 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"e5f580c1e3806fa49ef76b0bee4855615c4696ff","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-10 05:05:40.000000000","message":"Topic fix-cinder-backup removed","accounts_in_message":[],"_revision_number":2},{"id":"2f93f6caed1047b2bcda9b48b112048e6c1f30e2","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2025-06-12 07:09:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84df04ab314cf55b98b7da633748bd755e7dceb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 07:21:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57d68249785a488b8ea9a9723779b5f4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1603b9cab4c445febb9b5c4d5144c59d : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba0abd71fdf547839b22a4723c85471c : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/da0a7288c4dc45558053b4a32f675db1 : SUCCESS in 9m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"50ab03a01b4ab65627882b530a26c232f0948d8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 09:15:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/163265518dee4479832cd9a42f0647d5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbd33aeb39514599849279f1d44b2c5e : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36100b54036045be9ef719641d588311 : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19de679e8208435db180c549dec90315 : SUCCESS in 5m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7898be255c0d422a9812df0a214cbc03 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/caa1e95b52c24e4e8cc4e7152656feed : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30915266a4a5495a89f37bf814ff3c7f : SUCCESS in 3m 29s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/0e633ec0e695423e8b6f7a2fed5bf119 : SUCCESS in 56m 05s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/1ec82661b69d4489a150289dd01cf8cd : SUCCESS in 56m 15s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/491bf03b0cf9471b91aa7cf78eb4a9af : FAILURE in 36m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/80334bb35b264784899b656b64136631 : FAILURE in 35m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/e678903f94024383a52bd1f38cec86f6 : SUCCESS in 1h 45m 14s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/38428fe3733540a48a34f1ec87331af5 : SUCCESS in 49m 12s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c1c4006e2bf54aaa8c1a90af9e70c823 : FAILURE in 38m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/f842624933324838b7759bc138d4b68c : FAILURE in 37m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/31af8bd49d5b4a39ab0d675364b93fdf : SUCCESS in 1h 37m 32s\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/8aa9589d60fa4f3bb36d00ea0ad2f237 : SUCCESS in 2h 04m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/06ef6bfdd48c44f88efb7046aaa1a433 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"964d455cc2a342c05ca36353c3836ef13e6d5e3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-19 02:40:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9995a21013297a4a5e86d1cc3be3eef8bba972ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 03:24:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bf620e7472942f5b417f79ba5e90b35\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/27a7748748de4f7d9517660a54744e40 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fb64ab692ce44dbea4ca3bb6b0c47de1 : SUCCESS in 11m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0356311eca4c4bd199f371150dcad740 : SUCCESS in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f092c6a238e9d09c9f9da2b3208a878ecb7fd096","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 05:38:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/894bf15a0d744447b4e11a8db962b698\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c1a84ff5ed245d7a8ca394c90b17440 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18e180a1f0554095993dab720629598a : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c57bf74199284a9fafabdbb02698b7a6 : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad681dc3b5ee486c90a602841185c09b : SUCCESS in 5m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dfcb5d75e35d477ba43629de61c99cc0 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b90804ca8d4452a1981ad2c90db56a : SUCCESS in 5m 30s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/1be85775a7684f6189658bbd95e38db4 : FAILURE in 1h 24m 32s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/acf673c1f08a43399086bdf621263225 : FAILURE in 2h 01m 03s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/9303552a82e94d44bb5efc124c409be0 : FAILURE in 2h 04m 33s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/1c4c17ad655c4fe489013af979d7f5a5 : FAILURE in 1h 57m 04s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1ebf8e9db97a4d8ab0c59a4378c4f82a : SUCCESS in 1h 10m 58s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/58390b185a9446ab87c42fd1bb33ca81 : FAILURE in 2h 53m 02s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/62763878b7184f4eba23faebcea89334 : FAILURE in 1h 49m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8fe54fce9938457189eeede0422c5990 : SUCCESS in 2h 40m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/5a340f37ed54482684b71f7a3ddebe34 : SUCCESS in 2h 52m 13s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7c65b1dcdfb24f129f7f9e2baf66b933 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":3},{"id":"9d3b1bb7e224695f323f3d29e4855e23f64b2bcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-22 02:16:28.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"8190d6a3333ab3051a310e8bba10e4d12d199d0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-22 03:16:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc1dec427a8249029c24a75587ad1d16\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1beeb584d51842a9916e081b83dddacf : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c66694b39b67494aad75d65cfa56bef5 : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c2ef191d6dbe4720a62cd97c28c2f130 : SUCCESS in 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"130ec9e7fbbb40d1627152a7475e6acfc611d2d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-22 04:56:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/590bbe037e3a49cea2f8fde5c82dfc07\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/980d813269db4879aad565881ebd7fd3 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0940ae995d524ee2895e3a22b463e9b5 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a14b1497c764f3ea3ff1a91d361f782 : SUCCESS in 7m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b89800900a4342629173741155c514bd : SUCCESS in 7m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e32f6fb3bf54df48055dd89802c99f1 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2677d39086114202865711935d6ba5c8 : SUCCESS in 8m 31s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/549af5b1256b457c89e141f6ad42d813 : FAILURE in 1h 17m 42s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/5e9475596d54407c90b74658ae40697f : FAILURE in 1h 23m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/6560cdb0b86d41babcab5f1ab6a55d20 : FAILURE in 1h 22m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/560d069116004438a55ce2f5146d684b : FAILURE in 1h 54m 42s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d5c32aa97a76409c9faeb38d1ac7af75 : SUCCESS in 48m 25s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/50ca3a944c9a4b6dab3cce286e91f1f9 : SUCCESS in 2h 05m 00s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7666983c74224238b9a2f3ece64df2a5 : SUCCESS in 1h 40m 44s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6ca9df2d09b1472c871f9a3614d55876 : SUCCESS in 2h 29m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/f69b0bc5fb214b87885d7a85b8a8aa71 : SUCCESS in 2h 15m 35s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5d80b2872cd546c2b8c8f5cdef364da2 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":4},{"id":"ecb660a49a8bb22ab030db14c4d36da13b1ccd35","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2025-06-27 07:29:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"247320c6ffe490518d7d5ce457b1cee50d5afc40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 07:57:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df7f5bb310f242e28dcd5e5587acb9de\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6ed68f38269a48bc88b0b8efdd30c7f9 : SUCCESS in 12m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7756f0b0781c420883ca836284689e6e : SUCCESS in 12m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/72accc16a190475aa682cadaa72872b5 : SUCCESS in 17m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"785986576af3e95172b2cc4e79e2f42a166f5a96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 10:16:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54e074d8cae94f42a9f20dbea5379cc8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/750e447173de4aabaaca20c39c6b766d : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2934eb127a74c1eb16c0792b22e902c : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c3967c745f24d069c1448c81b3184a0 : SUCCESS in 6m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97a65e9add0240158a4cb4d21c92e1e3 : SUCCESS in 5m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7231268b807b4335a1686b4130467ed7 : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d1a0a3ea12f46db9d1c031852d53311 : SUCCESS in 7m 46s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/2ac033899ae74e7ca79ba90c07781627 : SUCCESS in 1h 12m 18s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/d9b9be5fad1d44ef952acf3600bb5914 : FAILURE in 2h 04m 15s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/e719433610504739b07c06184b9ae516 : FAILURE in 1h 21m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/bf9cddd2e68e4570a3d536a88438cd58 : SUCCESS in 2h 45m 11s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4cae9ffb80d2439696d6372abb100de7 : SUCCESS in 56m 52s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/86228f1386f348158b88cf0b270ca232 : FAILURE in 1h 58m 44s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1f8761845131498388f738982f3972af : SUCCESS in 1h 07m 43s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/29140c4050894d829146a7804c1f7776 : SUCCESS in 2h 10m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/3fd8457a46754ed38931246e2f8429fd : FAILURE in 1h 50m 11s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/46baa36546414f69873c28488682f1cb : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":4},{"id":"fc575b9bca4c45f187f85ee9a04b0b5d842f7b5f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-30 02:20:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ee6a07e546a96920d81d68824bcb51c4aa5de838","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 02:21:05.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9349bf70322670c49ecb5e80d01e01367f87ab76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 03:34:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bf668a910c849579a78804dba1693e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/940bdbdafffa4174b7cd63de3442160a : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12768ed13a2049fa876baf7c80a84362 : SUCCESS in 6m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a7da06443154b22a3337285ee826a96 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7983b970ecc452e9a07857e40a3b28b : SUCCESS in 4m 53s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/25aaa38bbdce4b74a71f0fe8f44e3795 : SUCCESS in 1h 09m 42s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3023b540b35a4cc6967eb588ca3956ac : SUCCESS in 41m 25s","accounts_in_message":[],"_revision_number":4},{"id":"4fba3b2388cefc5b5aba3435d0fc3961c79a6c10","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 03:34:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"aab00663434ce825a9596d3e05dbab4670d2f628","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 03:35:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec8008c323dc4ae7b845e754e7482f29\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2526d030bf784b8280c43ab40722b033 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0c19d7c2ff8149e206ead008f95187e93edd16ec","revisions":{"9fabd37d61c430e12153ddb9791dc1959ae14c80":{"kind":"REWORK","_number":1,"created":"2025-06-05 07:23:12.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/25/951825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/951825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/951825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/951825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/951825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/951825/1"}}},"commit":{"parents":[{"commit":"00aa57453a5f0631d3ce052ef0f9206b5b73c2ae","subject":"Merge \"Introduce mariadb_backup backup_strategy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/00aa57453a5f0631d3ce052ef0f9206b5b73c2ae"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-05 07:09:10.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-05 07:09:10.000000000","tz":480},"subject":"Improve stop_slave logic for cinder backup","message":"Improve stop_slave logic for cinder backup\n\nAdd a check for result.first() to handle cases where the current backup\nis not a replication.\nFor better debugging, this patch ensures that any existing error is raised.\n\nChange-Id: I67409071f9796121316c9f894c7a815846bc74ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9fabd37d61c430e12153ddb9791dc1959ae14c80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9fabd37d61c430e12153ddb9791dc1959ae14c80"}]},"branch":"refs/heads/master"},"a251917d8b035b11f6cb38049a9a610b02b90cf8":{"kind":"REWORK","_number":2,"created":"2025-06-09 07:03:09.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/25/951825/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/951825/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/951825/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/951825/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/951825/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/951825/2"}}},"commit":{"parents":[{"commit":"00aa57453a5f0631d3ce052ef0f9206b5b73c2ae","subject":"Merge \"Introduce mariadb_backup backup_strategy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/00aa57453a5f0631d3ce052ef0f9206b5b73c2ae"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-05 07:09:10.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-09 07:02:47.000000000","tz":480},"subject":"Improve stop_slave logic for cinder backup","message":"Improve stop_slave logic for cinder backup\n\nAdd a check for result.first() to handle cases where the current backup\nis not a replication.\nFor better debugging, this patch ensures that any existing error is raised.\n\nChange-Id: I67409071f9796121316c9f894c7a815846bc74ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a251917d8b035b11f6cb38049a9a610b02b90cf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a251917d8b035b11f6cb38049a9a610b02b90cf8"}]},"branch":"refs/heads/master"},"1b228837df7781ba780d3d4a44dd4ff0391ca859":{"kind":"REWORK","_number":3,"created":"2025-06-19 02:40:20.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/25/951825/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/951825/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/951825/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/951825/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/951825/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/951825/3"}}},"commit":{"parents":[{"commit":"e9f62185701da253a66d3d4c16731ad1f3121134","subject":"Add mysql-8.0 tests and remove mysql-5.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9f62185701da253a66d3d4c16731ad1f3121134"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-05 07:09:10.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-19 02:39:09.000000000","tz":480},"subject":"Improve stop_slave logic for cinder backup","message":"Improve stop_slave logic for cinder backup\n\nAdd a check for result.first() to handle cases where the current backup\nis not a replication.\nFor better debugging, this patch ensures that any existing error is raised.\n\nChange-Id: I67409071f9796121316c9f894c7a815846bc74ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1b228837df7781ba780d3d4a44dd4ff0391ca859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1b228837df7781ba780d3d4a44dd4ff0391ca859"}]},"branch":"refs/heads/master"},"0c19d7c2ff8149e206ead008f95187e93edd16ec":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-06-22 02:16:28.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/25/951825/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/951825/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/951825/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/951825/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/951825/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/951825/4"}}},"commit":{"parents":[{"commit":"36105ddc3c3c5907bf32c1d30e8f185cb6fba134","subject":"Merge \"Revert \"Introduce mariadb_backup backup_strategy\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/36105ddc3c3c5907bf32c1d30e8f185cb6fba134"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-05 07:09:10.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-22 02:16:28.000000000","tz":0},"subject":"Improve stop_slave logic for cinder backup","message":"Improve stop_slave logic for cinder backup\n\nAdd a check for result.first() to handle cases where the current backup\nis not a replication.\nFor better debugging, this patch ensures that any existing error is raised.\n\nChange-Id: I67409071f9796121316c9f894c7a815846bc74ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0c19d7c2ff8149e206ead008f95187e93edd16ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0c19d7c2ff8149e206ead008f95187e93edd16ec"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
