)]}'
{"id":"openstack%2Ftrove~952033","triplet_id":"openstack%2Ftrove~master~Id6f12fcd207c6440dae103f6328beb93439a81b9","project":"openstack/trove","branch":"master","attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-01-08 18:55:46.000000000","reason":"Added by \u003cGERRIT_ACCOUNT_37598\u003e using the hovercard menu","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-19 09:55:01.000000000","reason":"removed on reply"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-23 08:00:37.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Id6f12fcd207c6440dae103f6328beb93439a81b9","subject":"Tenant replace to Project","status":"NEW","created":"2025-06-08 13:32:21.000000000","updated":"2026-01-23 08:00:37.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0448f750b8fa8364194d6752bde2207c3a06ec7c","_number":952033,"virtual_id_number":952033,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-01-12 09:27:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-01-19 09:55:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-08 13:34:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-06-08 13:50:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-08 15:08:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-06 21:08:00.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"43fdb9924bdcfa5247319b1d3195edc531b3931a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-08 13:32:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d776d91eb8c8fb8794d17b77756b381ed647ac9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-08 13:32:45.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1b181a66e2689f2c315908c64392b081310f0d13","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-08 13:33:56.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"274e595a8bdc49833aa3fa5620d89ab127988125","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-08 13:34:06.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"fdd46c4fe1d09a86cbc4197193246fe5bf27a998","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-08 13:50:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69932ca06e5d49cf9f94060efb58d05b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b2362430002546858db6e316c446f07e : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1207797440a6444286376856af9deea1 : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b253076c62e749a3bb04f192dcddb876 : SUCCESS in 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"44c4aceba9dbf2dc749d238feef6a50d7b75c7bb","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-08 14:47:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5144a19dd56f5ce1d235bd7e3d84100383dad24e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-08 15:00:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76b398a809f4463a82074ddc04d12cc8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d8b29c57042e4a1997e4fddc54594924 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07d59490237d4f37b81fd0e696c04560 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cecab20fe84946469566fbe1d12c0f47 : SUCCESS in 10m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6ebc6ed7d3cc57f5bcfeff1fff78303b6fb8aa15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-08 15:08:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88be7af6731d4b0486118822900adbd5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/24cc45d69b8a4c0f966f00561316f4dc : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23228b185718460b87b256c71892cf8c : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4011b5b0cfc49338f8fe4e789b9d144 : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75a36a24e280445b9a068c71012aadb9 : SUCCESS in 5m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52f7569ce9414e808c064bb35da8e26c : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bb93fb823c54fe7806abf1ab645dfd7 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/205c18b547e84426a9b0c21730950297 : SUCCESS in 3m 10s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/2106580b4bab4facac119a867a829026 : FAILURE in 59m 45s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/5707cfaacc6a42f290b9b1f26a265441 : FAILURE in 56m 05s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/85fc5c6044484a688f8d58b1f23e2359 : FAILURE in 31m 54s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/4cc725f9b68b46f4bb546534b2ea4df6 : FAILURE in 28m 36s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/bf38e8876777482aa0ac822ef54fcc5e : FAILURE in 1h 23m 31s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/867cc2967abf40929c7ddafa63b21da8 : FAILURE in 50m 35s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e361325b80be4aeb974653ddc3aee123 : FAILURE in 26m 27s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/dd16a842336b4732a00644569ddc888a : FAILURE in 26m 10s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/cbde135e3ef04d09a9a4db24918f842c : FAILURE in 1h 33m 45s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/40bb7337a5d943268f70752d8dcfaa33 : FAILURE in 59m 59s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/a34011d8eb3048f0b42e4c58022c1943 : FAILURE in 21m 28s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/53a089691014449ab312b5cf1721747e : FAILURE in 19m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/1b8e45c615244c9eb6fdcac6bda115d8 : FAILURE in 1h 33m 55s","accounts_in_message":[],"_revision_number":3},{"id":"e2b34bfb1e5fd2787172711a3439f84d7e4eaa69","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-08 15:09:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"60a8aac0d47832474bb9e8713924056d3f35f794","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-08 15:21:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff536193102e4624a3ef5a515e11738a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/81c59316082349ed8ebb053206515444 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/22971029bfd247bf849d0759325bcab7 : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0ddc588312c44a7ba2c77cefa3488552 : SUCCESS in 11m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f025e274bee24e72b038e90fb312c5be038398a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-08 16:48:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bc7ac73f22e4bd59f6a7a38c5974385\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/49ef18a908d74428ad7feaf2a459f998 : SUCCESS in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3daac4e023e454c8523d9f801269781 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9049d96b52647e0abf3126d0f0f8cae : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2cb85211690843f092ee1245c4d59bb5 : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e6b1a68835a54b7cbe815c221e60ff58 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec658dbcd9024a678f6163b2188cf8c2 : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd84b6bf85d94463a43a18ab16fff519 : SUCCESS in 6m 24s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/9d20a74a111f4e57a86573f1275454cb : FAILURE in 58m 09s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/df8008619c4c4253a8c95847c5048742 : FAILURE in 50m 17s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/e8d84373414a49d3901edb5109958bea : FAILURE in 18m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/7abd3a41394548009df38c67b7ffafdb : FAILURE in 43m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/652183e78bb54b6fae417fd34114ea29 : FAILURE in 1h 37m 34s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1440fffdc09a47459491da0336a49251 : FAILURE in 58m 02s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7f59cb3db201498da5f05f14d1077a55 : FAILURE in 18m 07s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/159d960984b84fde94914c6d981bb41e : FAILURE in 21m 18s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/abd61532e32740ed8a2f7ee8dabc6bf2 : FAILURE in 1h 33m 49s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/14f1bf9ebfb5452eb6ab1c2636301ff5 : FAILURE in 58m 40s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/0c08c339b4c94827b955a62d2da2051d : FAILURE in 42m 13s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/cf3014b59eaa4d84bb952398ba8fa409 : FAILURE in 18m 10s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/c3db47ebe7574b8685c9440c3e7311d2 : FAILURE in 1h 27m 14s","accounts_in_message":[],"_revision_number":3},{"id":"245917a4ffce37a7ebc4b188038fe7c4e1df3684","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-06 21:54:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3fe672591add06bca4110b76cc15c7b76edabd33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-06 22:11:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7042948fdc10422fa35f39ef61b63b58\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/312d40f332c54de4a98cd8375b2611b2 : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/63852d00d8084cfcb891cddd874dc73b : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0f562924ab664255b81c3152c10ba0a8 : SUCCESS in 15m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"23717df33025bafae21414f9d78c6e0ac339dee2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 00:05:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6922ad8e47b41c09e3fa9682793d014\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6291b90cb5944d47a717814f87ad7b7c : SUCCESS in 1m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef3af5953ec04d15a84fc81345363c43 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e4972a29ad1420996e53d87c04cce09 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/277737362fe744908cf1de2daa6ceda6 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/153c7d57bbad405786c68daaab6a24e1 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42ed9c8235f1488fb0d5954bf5db56fa : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81ddab3fcde42d1b528f1e7ad36063e : SUCCESS in 3m 22s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/d752c1f6a51c4c1d8c4ac4cd8a78cfe0 : SUCCESS in 48m 08s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/62acea5dc60d4c5ba1a4c1af6bfff6a3 : SUCCESS in 1h 33m 46s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/8bc68cf4d657440d80219b57cef53a8a : SUCCESS in 1h 24m 47s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/a3cd4d22f745451baf0d776f45f2ab6f : SUCCESS in 2h 10m 11s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/56b3048e98554b27a2f7998372e5e5c5 : SUCCESS in 36m 48s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/cba941cf87c34e0ea96515972fde4a46 : SUCCESS in 1h 31m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/f79f9fe527d54b4fb688b24b40c463aa : SUCCESS in 1h 09m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/bf36b5beda384144b12581fd65daabf8 : SUCCESS in 1h 30m 00s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/5d631fc51e3344fc8886688949d2ea31 : FAILURE in 21m 20s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/56ba87af360c43ebb7d2836a00913109 : SUCCESS in 1h 16m 39s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/dafe5aec8710424a9beff1b0d933a233 : SUCCESS in 34m 39s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/4d9e2691eff04c0b8304d844b9807c5e : SUCCESS in 1h 21m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9254951cfb8ebf2b4f29aff61eef59bdfbdf2f67","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-08 08:50:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b3eada3df878b060d81c69b5717ecc45925c7269","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 09:31:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66f5af1bb649475a9d1fb6b379df5267\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7e724d0e71a8489eb70980ca4499dad6 : SUCCESS in 15m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0d9026d25d364d0fadd1421081801b49 : SUCCESS in 15m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bf72465b221341f497e3256e674d1642 : SUCCESS in 15m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c1dca6a06175451f3dba60e9feea6a9e72e6e9c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 10:53:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/531cd5c42eeb4985bcac0c83e7b61871\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/26932622736b4f358e053c98600db8f8 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d8ad8e7ea1344329323363b534c4001 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d2b31473e7543c5a8e58a76fa465ef5 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/712198a056a7482b989415e2e0ab7fec : SUCCESS in 5m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c237f5ea50e46ea8ec5afcf5cef253f : SUCCESS in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abd0f3fad30444a6ba9bc506c2e63cf6 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d89fb6a486aa476eb2fda20cf459494b : SUCCESS in 5m 29s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/b6e7c00268e6400b92dbf4533e3a64cd : SUCCESS in 1h 03m 48s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/f02c446440744bdf9de261a3e0980911 : SUCCESS in 1h 55m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/acb4854c73bf44c5877f9e3fdc795903 : SUCCESS in 1h 33m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/3d00511c00a84e05bd0a6f6f72c0bda0 : FAILURE in 2h 00m 15s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/6a77e42a4a3c45ce92b8b2c025b91ca9 : SUCCESS in 45m 57s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/115432a295464928a18c64ba5fabb6fc : SUCCESS in 1h 37m 30s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/71d17a0bf8534befabadd69383e282b9 : SUCCESS in 1h 22m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/42501fd71dbb4eb89e7bba139525bd39 : FAILURE in 45m 43s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/2f1a0e6f393e44cc92652d30bbb71227 : SUCCESS in 57m 52s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/64b529ef93be4f7492135033b5fd0c9d : FAILURE in 26m 26s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/8d3cc079eeb348ac870b00586721b43e : SUCCESS in 1h 16m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/d69df1f25043496e82850681fa4f6965 : FAILURE in 49m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d72799c56851481213c323dd44e4306a4342539b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-05 23:10:20.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"828a4af2965cedb1e193dde510df66746e7ac72f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-05 23:10:21.000000000","message":"Topic license-classifier removed","accounts_in_message":[],"_revision_number":3},{"id":"1c69b81efa5b3158e5708d6e018ee3ac67b70994","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-05 23:10:39.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2635e308933358fc4437c8c6f09e5f48484a70f9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-05 23:11:38.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"e840da5b3f62976c714aa84dd8747d9364cd1cc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 23:23:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f97ab4ff352140baba95400378aeca61\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bfaa4713a0344c8396601dfbe6f83f8b : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bc277c32abe14e7ab4e2b81b70c229f3 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/10917cc4be3f4e3aa2b926689aa73f89 : SUCCESS in 9m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"82b56222bcef9d2d0890943beadb1b158fd9de8c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-05 23:25:42.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"5e9010fd70e6f354fd472a870ae99c068dbd8f46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 23:36:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb061805c3234fc1ae267ca7692d3c4b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ce506127a51b4ffdaaffa66f2ad31c92 : FAILURE in 5m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/daac7030fbae41818a67bf0bd266b317 : FAILURE in 6m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a3b85096569245f4a5c5d415e8df893e : FAILURE in 9m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8b910e7acc49e553bcc8fd2106c83b82f1d5776a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-05 23:37:15.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"1035b239323d62410ae2e2db9d13bfe73e541c6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 23:48:53.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61e634ee688e41e8abd440cb319a177a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc5c3451f0514e418933406dc5eeccb8 : FAILURE in 6m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0377e9c2268b4894ad7cc1f3777f627b : FAILURE in 6m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3e2b77afa2524747aae8b65442386625 : FAILURE in 8m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"388e733aa099fcbca4285cba9b2ddc1a2c55f8f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-06 00:00:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f19698f94f5046b298387def46e7965e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c58ef16fc9fe45f58118b1c7bc18f08d : FAILURE in 1m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e68fc60e3bb348afb644a620d1e2ba6c : FAILURE in 3m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8103c7134444220871493b5cf1825e9 : FAILURE in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1c9a9fe723a44e881ce6b7094b9c5cc : FAILURE in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7299a9c06b8e4fd4928a9cfac76fc913 : FAILURE in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2eec6b0524234dd392aa75b152e1de2f : FAILURE in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b61064d719d4888ad2dfb24aaed9b04 : FAILURE in 4m 03s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/112e453505ef407fa7fff21dca431488 : FAILURE in 10m 33s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/76d3b9a990f340cdbcb858fc4214c94e : FAILURE in 15m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/90dc414b7c1f4cfea72c06d420fbf302 : FAILURE in 7m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/8812e97acc9f40839335241ed0ce04d5 : FAILURE in 10m 42s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/116d6c2ab74e47729f86e8f8b851b847 : FAILURE in 9m 19s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/0f1fc39d0ffe425a82878d871c8784c7 : FAILURE in 10m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/95a97236b1304575aa1cda98fe9b168e : FAILURE in 7m 22s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/8b978480ecf34128bc887cf63bedc4a2 : FAILURE in 15m 29s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/612b6a51191e4ecfb65ee11cfe064ae4 : FAILURE in 10m 00s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/127c6639f62e47d8964f46afd82e2c9e : FAILURE in 9m 20s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/98d07a993c6d461f961672eef1a8e4dc : FAILURE in 10m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/928fe810ac744641ac1971ddf2bf4d40 : FAILURE in 15m 22s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a41bb91125e54df4a7a04e7457a69ca7 : FAILURE in 3m 57s","accounts_in_message":[],"_revision_number":7},{"id":"2649078835f048c482cb10a629df2c4be417d07a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-06 11:53:15.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"cef86fd0d1493c2a73947f7adb1a22d889938819","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-06 12:01:01.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"f8915da3d43b382593dbd8e8ddb570261e08cc32","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-06 12:06:33.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"a90e2b93990bd8b0e0a05e137abbc36ab536cb74","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-06 12:07:25.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"6077fcc99e1067266fa52777de01a45c8847f61b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-06 12:17:34.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"2526d910235b350187be8d2173605eb7ef2913b8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-06 12:26:08.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"85945bee58b87e6ac0581cfc05629bb377076d1c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-06 12:35:56.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"069e14c7e7692a44f7891c2b903fcc78d4b91740","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-06 12:49:57.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"de7d10ecd4d771a7c75fb6067d41f1c4cd3c3f38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-06 13:02:16.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03347856bafd4a139546903dc649c3e6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aef45e6c08fa4eec9b38df39c4b46dca : FAILURE in 10m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/50d224a60c054cf195c28efd58e3dc6e : FAILURE in 10m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/22d0005f0808445ab6d1586265b3c522 : FAILURE in 10m 11s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"037eebe90a4f47f525ee638416dba7876ca54eff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-06 13:27:35.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"6a03a9378758517d5a17565a704e3ead79c30589","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-06 13:37:47.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"f38f57abb5b47918ff7b602c5c68464c374cee53","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-06 13:48:48.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"853de9001f1a624a684c0309cd6624e749d54927","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-06 14:03:46.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acbaac9d660049e5b4acc95d4e11ea6e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/da656c1a3295460ab3ba840440919fa9 : FAILURE in 10m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/47856375192544eab16d64b385cbf7e2 : FAILURE in 8m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/12ae84090ea94a55ae1dbe90288c68ce : FAILURE in 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"497d699356993a8329006dec368b0c796caa6eb3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-06 14:04:46.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"099e2ed320bb4555eef4dcee6ad00a485035448b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-06 14:15:08.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"336aa03c42ad304432f9cab0c5edf6cd42a51561","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-06 14:23:07.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"5b8369ba778ae4c56fe02f014cbeb9c97d8c3d7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-06 14:39:10.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92466ca6b09e48b6af9f2b3822d25dce\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9216ef0b077440aba7c0d7b56eccf6dd : FAILURE in 9m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0330370914094c4da6e6afcf77d5d815 : FAILURE in 9m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f3e6ef307134497488e8ae509b504678 : FAILURE in 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"d94de4497244c3590174445057148624a77e35b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-06 14:42:07.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e15d247a2c6462891f15037dca7e8c1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6f7ef86813684e608a246423a309e8c9 : SUCCESS in 1m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/85f8271cf788417b896eb2ca5ff91e5d : SUCCESS in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d753ed19764c4a71a5026594e6fac402 : FAILURE in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9efb353de70f4e1b9cffa363efa1e851 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69f08554c1b646cb85455537a662e38f : FAILURE in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c659c09ef4d45b589dca148dcb4d3b6 : FAILURE in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ab24cc42ed463b998cc6521e2584e5 : SUCCESS in 4m 43s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/968d23728da94d2f88a839633041c13a : FAILURE in 15m 28s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/a424588a44c746ca918bf92af656fe7e : FAILURE in 16m 17s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/ed153dbe2afc426db63869db838be790 : FAILURE in 15m 20s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/ede8280af824452a90dee7fd5a36b0d8 : FAILURE in 18m 11s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/cb919ac498f84ab9b0bdfd25e6ab7c8c : FAILURE in 10m 21s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c51d98060ed8472ab941b894da318e35 : FAILURE in 15m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/f5d09d01ac224e688352ac9ad0c6efe9 : FAILURE in 11m 00s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/f096f0f6aa85450da025a094e42cc8e3 : FAILURE in 12m 30s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/475df379a54348ea9a37262c4d4b4309 : FAILURE in 16m 19s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/68415ea685644d95890c1c17fe1f9335 : FAILURE in 12m 32s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/7e3a9b44d571435da9b9849c680eef0f : FAILURE in 17m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/e94c440c258249bfb1d0f11f3d287b35 : FAILURE in 12m 32s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/932d8c0528444a06a9b806769685aaf9 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":21},{"id":"7be4e7df92a97e3447776554df32db50d926b132","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-06 14:48:40.000000000","message":"Patch Set 22: Published edit on patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"9b043fe8738d499608f401f8841fa2b5a0ba823e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-06 15:01:57.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da427778c0ce4a058365230dcd0ac41f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/820dc26b399940f1864aca404c213a60 : FAILURE in 11m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/87e8e3322296483a91217f1f7f178c02 : FAILURE in 10m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/880fa61423774b1fbccc8c27a1c90ec8 : FAILURE in 10m 11s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"9d71730607672abc51338b71ec33ab257e7b3a82","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-06 15:11:09.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"4e737450ead936e72d6b0b7f5d45e91c4b8374e4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-06 15:21:39.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"ad3a8c6d7eacb2e4037d404bc9aca285ea63f3a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-06 15:34:06.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88fec1fc8f044489ba2287e30da05923\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b31309c6bf64b34acab4c033fd20f4a : FAILURE in 6m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/74e1df652d6b4cdc9629739b7d8904af : FAILURE in 10m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f7f39f89404e4d9a9e68650b29737e8a : FAILURE in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"7e2f9c51329269ffc4a91038b7211deaadcb10eb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-12-06 15:36:55.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"89f273b3687f676dcca79f5ade159762b12ca924","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-06 15:51:26.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/118e9f637dc64e5ebb001acf03f6ac8f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c1cc87320db244c0af90f66ac8454330 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c11d62741d05414d81c6cae893a902db : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/66851c1974944c1698d85f2d75c52464 : SUCCESS in 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"9d3b9afed140c93d6fbf7c940dbefa9cd6091b12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-06 16:09:00.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05f18e3d68e24c6badb572fdfc431ac0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b8259991172a4885b6dfc012babe8335 : SUCCESS in 2m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d9a2f47b9e244c8a9b4248bf08f5aabf : SUCCESS in 3m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b78e1af8e9845dda45a5f82f7149da1 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27c4b750fa2e4055ab1577929866d806 : SUCCESS in 6m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dede697d934b4ee1a667625702222c76 : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d74e0538ad7440039d1027f461492b22 : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e6c049849aa46ae8861d925193323f5 : SUCCESS in 4m 42s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/b091fecdc68d4ff48484e4af3b60180f : FAILURE in 16m 14s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/3d0e2050419c4cc09e059699fa5ba562 : FAILURE in 17m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/60aeb39ec74746f781c436a586659d65 : FAILURE in 12m 26s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/1bf2810b60d34e9a9bfa5d718bf439cb : FAILURE in 18m 02s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c86494beb02042c8aae93cdc41738a4d : FAILURE in 14m 45s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/50948758a67a4f4088ababbba45a8040 : FAILURE in 16m 45s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c8583de3bd874d0e89081ad5e7b19dfa : FAILURE in 16m 50s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/58ad52c9792648379ee8ecbdc5bcd3e3 : FAILURE in 24m 38s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/da268ed4e0c6464fbe4a52ad87e196ff : FAILURE in 14m 13s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/91cd34996e2c494d89f39dc13b243976 : FAILURE in 24m 34s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/775bceb9003a431686936d17457e982f : FAILURE in 12m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/bfbdb35554b54ca4a1e3795955c4dc75 : FAILURE in 12m 36s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc79fecc067a47a4b07f9967ec80721a : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":25},{"id":"859e3db97354d769bec7418b136980fc6c03f4b6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-07 11:58:03.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"df327894936738096bed457fb38069a6f5d8252e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 12:16:10.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2228512677cd491ba1cb21def514e8e7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/72120eca80b940d586d54682056be39b : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/258ad6186ed84c1ba641338144bdd045 : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/053f5c01acff4f96b44d8d991bde9fe2 : SUCCESS in 16m 02s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"dd9ee33e2f93960e0c28d2c0af5dd1985f53a8df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 12:30:06.000000000","message":"Patch Set 25:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbae3a7347bb4fef96df1c4e58f1b861\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/528a38ccde41496f90d7795aeb6633be : SUCCESS in 1m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/66d058429dbf49fa90b5dcd0c4583cf4 : SUCCESS in 3m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee8ec97e05d545c2b9b0c755cf046e6d : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29c619353f0845a3aa0e4ec3d99c24b7 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a21e302194724e568ba1d0a2dd3863f0 : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7254c33383084f4eb4fffdeb8ffe0ec0 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b8955e6d3a24b8aba55cfac5ab8054f : SUCCESS in 4m 58s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/95ab063e6b3c489591d62fb5ade8e02d : FAILURE in 11m 54s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/2a2bd18202fd46a6b5d777211dd96c79 : FAILURE in 24m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/6882865da4d141158d067d6117e1a910 : FAILURE in 25m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/5a1a26e51d54407487e0e07af4186d5e : FAILURE in 15m 54s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1f9ae72bf74a4160b9f517bed5d0ead9 : FAILURE in 17m 24s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/ddefa7005d624b08a3633b157c4dd7a4 : FAILURE in 18m 25s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/43f99dda5bdb4e0da22dce3c4f39bcdc : FAILURE in 16m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/babf41c2e67a48a8b18174aa4a51161c : FAILURE in 15m 59s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/acfc54f8fabd4970948123ce2c97eeab : FAILURE in 17m 04s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/936197f1f1f9458ea3da5f6f4be8e243 : FAILURE in 15m 33s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/bcdeb60d1a414ece98b5d51684b567af : FAILURE in 17m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/4c7918730d9248709cdeba37a653ebf1 : FAILURE in 17m 50s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cd6de4fc6a9841a5ac87c28917f31b99 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":25},{"id":"32dea2ec918375109f2c9b57d47f1cbcb1847746","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-07 13:30:02.000000000","message":"Patch Set 26: Published edit on patch set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"dfb54f67af4e54b2aae5b11f2891c417b266f87b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 13:47:14.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29c9eea4a88d4740a383fe4572973bb4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8646272001e24cc8b36349ad2553eec8 : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3016040f16324798a78f53e952569a61 : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0dd475c4b710452a90d64f7aeb9088ed : SUCCESS in 15m 09s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"89cee93644660d122fbfec099c0b074e44e80ab8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 13:53:28.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1be5c38804a0460ca8f3ccd240ac92f2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/54bea73c4ddb4ff083e5f389ad37fc13 : SUCCESS in 1m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a41c53bca3bb45bf9b70e85594161c81 : SUCCESS in 1m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/798539648f2d416d80da20686be4a04f : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00a220958b0b4abbb8ebf5e56e4602b3 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/decafa365bcc46b4bcd0308faff9e34b : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8f3adfb2076466b979548b072b4c9dc : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e1e0a7004cd4b5b9de361c8f4a3df76 : SUCCESS in 4m 59s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/c01ad4b0ab4a4e1cb55f267c05537ead : FAILURE in 19m 18s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/b75001aa00404ecb832a27096311448e : FAILURE in 19m 43s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/a182973468be4437848c85a1e0143e6b : FAILURE in 19m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/87447a7526b342ad9828a6e0cc29fd0b : FAILURE in 19m 51s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/ae93db4ba1a24d4aac8a70f0978c479f : FAILURE in 18m 28s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/ac18323d84c149c99f9f16335c31835e : FAILURE in 19m 28s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/42c2da055e20471296ea23c86e58d644 : FAILURE in 19m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/12f1953b5f5d4c76acca0be7ed1b71fd : FAILURE in 22m 29s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/b064c56b11d34435b26d4f15ec149594 : FAILURE in 21m 02s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/7bb17f2ddb5a4b08b491835dcbf883e1 : FAILURE in 22m 20s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/e4a361b31c194ac9aeea38b2f5a949b5 : FAILURE in 22m 26s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/d2bb1f7bb3cf490789ea74da1b7f0e39 : FAILURE in 21m 16s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/65d7ef923db34a05a976d012741b2336 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":26},{"id":"27d543cb3ab99997788a848c9ab375b978f8d7a7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-09 13:31:13.000000000","message":"Patch Set 27: Published edit on patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"b89839b039bb7976bb442388e30be98a0b13fa86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 13:42:06.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e1a0ec68eaf460c9695a9f1dace3d4c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8e9c06308f5c433390493c41f477a392 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b2dad2143d364ff18484e5dc96d5c295 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e4c01022d4a74fac8da42bfe336800db : SUCCESS in 8m 59s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"a4bdc9006b1fbf496dd1fe91d59e3756fdc11b0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 13:52:03.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d9c5638ac2b4fc3a616543d03b22da8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1116e91ddf574261aff118d5e2150a8c : SUCCESS in 2m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8d1c5567e2a142b3adae0d366c207a64 : SUCCESS in 3m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a34840147e2d47e285bd4897fd9aedde : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4b4be196c16477799501ec98d4f646c : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d0742f03eb244f584ab1bcfdcb81ef2 : SUCCESS in 5m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b282294abacb4398a7bdf1533a19da9f : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7586db6f56d4a3c94b0de30e656fcf3 : SUCCESS in 5m 38s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/093bf1c3c6a84ae4aa40e87aefda0c62 : FAILURE in 17m 28s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/a181f2e771a74373b732cde37b7eeabb : FAILURE in 16m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/b075e3c7d1d6428b90528a4b7968f0ae : FAILURE in 19m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/7c3308dee5204a6f9412744ceda3b3b9 : FAILURE in 18m 09s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/d1d2337d93be495593a090abd6f553d1 : FAILURE in 15m 59s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/052c68acdcd0448985e14d54e98351fb : FAILURE in 19m 02s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b2ac0472688f4d26a889ffc32b4ea5ac : FAILURE in 18m 24s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/48e1ca2e98654476996fafd32662d39e : FAILURE in 19m 45s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/811cf3fe7e8043d89e84de89286e9790 : FAILURE in 17m 24s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/b127ef415bb14d599dce0f6f3a97f9c3 : FAILURE in 19m 14s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/1f735fb85e4540768900971f9e3f2f45 : FAILURE in 19m 24s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/a9700a39f7ef40bf90d240d40466250a : FAILURE in 19m 02s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff762bc11d2848d590dab9b8d44d7a3a : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":27},{"id":"14b1bc661e5e0092d787151531822f22607b5638","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 14:58:03.000000000","message":"Patch Set 28: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"3b03ac004d9510eaa8cee628d9b3c2e66f6506b7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 14:58:24.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"ff5cf933f43b9fdcc090bbda7abeaf23c9840f8e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 14:58:27.000000000","message":"Patch Set 30: Patch Set 29 was rebased","accounts_in_message":[],"_revision_number":30},{"id":"02c08affbe2d844c83c74d6dec6b2eafa2fe66bf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:03:51.000000000","message":"Patch Set 31: Published edit on patch set 30.","accounts_in_message":[],"_revision_number":31},{"id":"3db3d9cfe970c6ba6221e3c8b638c994b385b83c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:04:55.000000000","message":"Patch Set 32: Published edit on patch set 31.","accounts_in_message":[],"_revision_number":32},{"id":"9e3f0af7c2a3f81026553fa239dc29b1aefe3147","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:14:28.000000000","message":"Patch Set 33: Published edit on patch set 32.","accounts_in_message":[],"_revision_number":33},{"id":"1fec01eb8bcead71602f58d26450d16966443bbd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:15:29.000000000","message":"Patch Set 34: Published edit on patch set 33.","accounts_in_message":[],"_revision_number":34},{"id":"cb3f2f2c9c273ce5cca5aec0ade43dc8efca1ff1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:16:31.000000000","message":"Patch Set 35: Published edit on patch set 34.","accounts_in_message":[],"_revision_number":35},{"id":"d276a7733f9c8ff26fa367b819a2a7f414b73403","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:17:04.000000000","message":"Patch Set 36: Published edit on patch set 35.","accounts_in_message":[],"_revision_number":36},{"id":"d71ff1ac2b97c964782308b0d7be26404a805daa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:18:18.000000000","message":"Patch Set 37: Published edit on patch set 36.","accounts_in_message":[],"_revision_number":37},{"id":"8d2b5676584b94e732d51b182d0a52634943f637","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:19:04.000000000","message":"Patch Set 38: Published edit on patch set 37.","accounts_in_message":[],"_revision_number":38},{"id":"01e36985b1488330b22540c0a04a7f13f64f53d7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:28:46.000000000","message":"Patch Set 39: Published edit on patch set 38.","accounts_in_message":[],"_revision_number":39},{"id":"c6895235615c8502328eaf0884466635f31e9433","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:29:33.000000000","message":"Patch Set 40: Published edit on patch set 39.","accounts_in_message":[],"_revision_number":40},{"id":"21c90241be7c78e67fd9bd4d0246915df21a56b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:30:14.000000000","message":"Patch Set 41: Published edit on patch set 40.","accounts_in_message":[],"_revision_number":41},{"id":"8b348ebd3204857fb78e1bb68d29c1e570f40870","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:32:02.000000000","message":"Patch Set 42: Published edit on patch set 41.","accounts_in_message":[],"_revision_number":42},{"id":"3b6a9ea14a6fd243b0ff887556811a759c2ba643","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:32:47.000000000","message":"Patch Set 43: Published edit on patch set 42.","accounts_in_message":[],"_revision_number":43},{"id":"2be4969a102ccd99f7bbedb733f4a30248567ab0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:33:19.000000000","message":"Patch Set 44: Published edit on patch set 43.","accounts_in_message":[],"_revision_number":44},{"id":"9f8916c16dca171ac2d965e6b0fce7861b17e180","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:38:23.000000000","message":"Patch Set 45: Published edit on patch set 44.","accounts_in_message":[],"_revision_number":45},{"id":"6dd53112486f99ab01c46143333e545b376eceb2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:39:52.000000000","message":"Patch Set 46: Published edit on patch set 45.","accounts_in_message":[],"_revision_number":46},{"id":"d173392d1b8537fad33fc74a712598278b98acb2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:44:11.000000000","message":"Patch Set 47: Published edit on patch set 46.","accounts_in_message":[],"_revision_number":47},{"id":"248ea40f70c93c2df88341ed0483d3a7c1dea5e7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:44:43.000000000","message":"Patch Set 48: Published edit on patch set 47.","accounts_in_message":[],"_revision_number":48},{"id":"d1db2007f00297ac01552cb0d66f3d8fe6708e02","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:46:15.000000000","message":"Patch Set 49: Published edit on patch set 48.","accounts_in_message":[],"_revision_number":49},{"id":"228898ea44614fe37d3363bb787a1e081a417e44","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 15:59:42.000000000","message":"Patch Set 50: Published edit on patch set 49.","accounts_in_message":[],"_revision_number":50},{"id":"165fbf564a19257bb446423036795eebc061e2e4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 16:03:38.000000000","message":"Patch Set 51: Published edit on patch set 50.","accounts_in_message":[],"_revision_number":51},{"id":"c1e5d61e2a174262aeccf4cfcebf1c7d651cd891","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 16:04:27.000000000","message":"Patch Set 52: Published edit on patch set 51.","accounts_in_message":[],"_revision_number":52},{"id":"d0f6d7381e853df5599265144bc9ecb9abf7c54b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 16:18:48.000000000","message":"Patch Set 52:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67093af4e06741eaaab6ef9fdd1e2c78\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/179c5064ea7b46cc94988bf37f1ed4c8 : SUCCESS in 9m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cbd3a6f7e65c454182551eac951cdd6f : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1c0f7013eec34d689ffea67051c13a8d : SUCCESS in 9m 41s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"2aa0ffdbdab968857a549e514337873e6f883cec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 17:53:42.000000000","message":"Patch Set 52: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40722a0ef5964adaaeab495f127c43c8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c90144f8f93742928b6e16c6f3b5e060 : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/223d948310954c0889701739d3182865 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d8d87a9baaf4bda86fc4e6abf83d042 : FAILURE in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03835d156c024a2da13cf40a701a5df2 : SUCCESS in 5m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/93a09487bf8846e3964c918e31ad5deb : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba0eca1b83aa454ebe07915a4d68077b : SUCCESS in 4m 06s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/fd92a04066e14466875bf445c10ff17c : SUCCESS in 43m 55s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/fdf81f5f1aca457ca4c66947b5fc898f : SUCCESS in 1h 45m 25s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/8256920385f1472baef81ac00a36fa19 : SUCCESS in 1h 23m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/c25d4b566d0b49868eefefe15cdc3b9a : SUCCESS in 1h 36m 36s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/23e1490adddf4502b4826dce3514dfaa : SUCCESS in 42m 28s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/6d51c17f7c764dedb544f15efa27baea : SUCCESS in 1h 40m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/5c6a67371de54d42b6568848bcfd8522 : SUCCESS in 1h 18m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b57b7c75e3504828aa8b1c407dcb3d05 : SUCCESS in 1h 25m 19s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/4f014d1772d447ad81a354ea453a0863 : SUCCESS in 41m 30s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/89f4a7d27221487ebb761907edeb451b : SUCCESS in 1h 27m 33s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/efe7ba1dc1244a20a02bbdac85aa692d : SUCCESS in 1h 13m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/a3c361e70e90475bbd84363b3ce6e75c : SUCCESS in 1h 19m 04s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0c1c81eb2c144e0f8f2639023f1d6ae1 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":52},{"id":"3f6a36768c34c5562c15f2d8379167d0afd5f344","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 18:58:54.000000000","message":"Patch Set 53: Published edit on patch set 52.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":53},{"id":"abb9cb18c836addb0a28ca814f11864b0e9ccda6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 19:10:17.000000000","message":"Patch Set 53:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d26b5cb0edb4e568c483f5559e9dbed\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4e5d28da461c4556954353b7368fe4aa : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/441ade4c299c4ca0b9bbe4fe000aa634 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/99a1fe6d366e4c58b1e5fd0317622529 : SUCCESS in 9m 30s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"f670b03e61356bae4bd5cdaa0a200aea909bc7d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 20:54:05.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/945f9d8789d34b42b9cabfd4fd69a56c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/65eee3ddff6d4a29aa6999175907ec40 : SUCCESS in 4m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b28c87c17764da5bea416a35466249e : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72c17a839cbe4109b57391cb46f6bb54 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85f1e6399f3f47ca8f9901056651442b : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2385a3463f8c4fa0b370a36015055d18 : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/129e4510eec544d59f22a3303f2eb260 : SUCCESS in 5m 20s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/ccfd6b5bf92d4630933980d2503e2c13 : SUCCESS in 48m 41s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/958e25e4cfdd42058cb434a58dc35e55 : SUCCESS in 1h 54m 15s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/e9581acbb45742deb71d6381b6b37b7e : SUCCESS in 1h 28m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/9f2b0003b2e34db793c63b876c3babb4 : SUCCESS in 1h 51m 27s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/8e5dfdf959ff4a9b91ea3051f7921088 : SUCCESS in 44m 43s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b0a18326b23a4aa88326baecb7672a4a : FAILURE in 33m 54s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/649aa5785a524ed5af9b7e784308523b : FAILURE in 33m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/121810a08c154807a867142298ddde7f : SUCCESS in 1h 30m 30s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/30a68597165a4691b66aa4a7a880a5c1 : SUCCESS in 51m 29s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/101964ac45124d0ab46d31cf701fc184 : FAILURE in 19m 44s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/f89f38d9d6664871bf87ffe0461c8192 : SUCCESS in 1h 22m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/e8799e0f2bc54b15928f849494a3f11a : SUCCESS in 1h 22m 47s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e341d7bb96fe49c792e28d3666651da2 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":53},{"id":"a630178655ea50d8a143cc90a95d07c518d34f75","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 21:07:52.000000000","message":"Patch Set 54: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":54},{"id":"e9ed589dbd864c3f4e29330ef7100742ae564260","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 21:08:00.000000000","message":"Patch Set 54:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":54},{"id":"e76e03ae53e7f0a002c1055722b8cd50d4238da2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 21:23:55.000000000","message":"Patch Set 54:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66e945c35f0c46e593fa4aba80508043\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bafb374a6576495cb673721e15363978 : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2c6aa614f95d48fd980a4d6b15ba217b : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fe5c864c3db74ed1b8da734a6d4cd4d0 : SUCCESS in 14m 07s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"6b57e196eae7a9770f3fc56e5086cf7813e217ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 21:33:43.000000000","message":"Patch Set 55: Published edit on patch set 54.","accounts_in_message":[],"_revision_number":55},{"id":"e3e26e26605a64aeefbd681ea0e1f322ea134fa2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 21:50:57.000000000","message":"Patch Set 55:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c36c547bba446ac872e48682baf5c40\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4983f4fa84a94143ad7702859a3754df : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b45d488c378b4dbc89d335636a5b3c2b : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5e6c27b76206496c8093c949088d9aec : SUCCESS in 15m 16s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"22ecbbd742fefc37b8d63f40db0e0ee38717537a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 23:27:57.000000000","message":"Patch Set 55: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d24af87c90a44b6a90747b92707e794d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/61f9a7a5c33945888825c225828cce1d : SUCCESS in 4m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79b2738115bb4bae99ec01e753c5679d : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19866703a0f840ae9941ebba488af405 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b78f781b679643b8b631c34dc395518d : SUCCESS in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/69ffb04de2314a9db5445b271e525b3d : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/910ca3d4d49840f9b12ec052d9a52fac : SUCCESS in 5m 22s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/75d64b67937a44719a92aaa10d8a6122 : SUCCESS in 51m 48s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/494c52ee03104d21897adc7cb6bc5f72 : SUCCESS in 1h 53m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/e0928268f1b24996affe0e70b4addcea : SUCCESS in 1h 23m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/b4b5e1b966df4a6aac51c69f4a896347 : SUCCESS in 1h 47m 56s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/590752d1029545fd9f79c3ac017034c9 : SUCCESS in 48m 36s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/85f75244097f44ee80895241fc72d22a : SUCCESS in 1h 36m 25s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/2ae060e23f2d4633b76ac76d51d03c87 : SUCCESS in 1h 17m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b0f69538c8704eee9b211fbe41e05412 : SUCCESS in 1h 22m 53s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/3785e12a669041bbb3a0226b4fb1ffd8 : SUCCESS in 41m 12s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/f60d7ff8efbb427485dd1db52471e6c1 : SUCCESS in 1h 37m 03s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/5118049d02f24085bdaa8ed9f5d2a927 : SUCCESS in 1h 15m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/d405c3876a1e423583eb9366b241597b : SUCCESS in 1h 16m 10s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aebdb23c560c4dcf98745cac7a42376c : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":55},{"id":"ce229e98e488286aaab41b7cfee8fe8afcd36c15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 23:33:38.000000000","message":"Patch Set 56: Published edit on patch set 55.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":56},{"id":"fc9a14803bc81f3e4e989ce226d25c25596f2a29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 23:45:23.000000000","message":"Patch Set 56:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60cdbbd3fa884abe99da5ce2897b7911\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8a0de426d57b410885fbbfe773fff6f1 : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a8f792cb589649d795634454ad326d15 : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/290fb2696d8942e4a8d6fcd76e0bdcf3 : SUCCESS in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"a01ed7924ddeed46d1bf90b8496d334e2b9f64a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 02:07:49.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/887fc9018ea0450dbe24344d25543681\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dc18a596c8724e37bdabd6c606abf7cf : SUCCESS in 5m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42c512e77ef348e39b540f42b0240d93 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5551c9f1cfab47b1a4c4d9eb6098db8c : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f124a8c715c74b72bfe045c893846550 : SUCCESS in 5m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc083ec1c1ac413d9ff0c97f7ab532b2 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e4faa8c11b84d39933f31abecbf5144 : SUCCESS in 4m 54s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/dcf31ebca1c044a798431206d6da6a8b : SUCCESS in 31m 28s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/492a2196cb67436f8c5d6b420082eb56 : SUCCESS in 2h 23m 28s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/f694a482f6c544e58e9b95a384fa8957 : FAILURE in 2h 06m 17s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/aeaf910032184a9188c8fa984122e9bf : SUCCESS in 2h 31m 43s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/54b2703d3bd84f87aa16af7e2ed36d2b : SUCCESS in 58m 36s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/318722dc0e7f4779869ca8c95009d371 : SUCCESS in 1h 40m 25s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/8a5f3f93eeec4841a2b757d670af78c6 : SUCCESS in 43m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c85ecea03ad7438bab4cd56aeb609b5e : SUCCESS in 1h 25m 50s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/b62eefe576b84865a13e3b0827641313 : SUCCESS in 1h 07m 12s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/0b1095dc58154289923f22145e16b555 : FAILURE in 59m 47s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/9e7910ed4b0f4e109f980b165505992f : SUCCESS in 1h 11m 22s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/c5e7695d8ae7433c84e03e8deeccf580 : SUCCESS in 1h 20m 42s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a3b42ecdc3914b77b66b735eba1d87da : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":56},{"id":"00c62870c2c452368135e0712335f6f0c1a12bbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 06:29:10.000000000","message":"Patch Set 57: Published edit on patch set 56.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":57},{"id":"7a22896f7fd86bacde8a76b6fb303155997a43e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 06:30:04.000000000","message":"Patch Set 58: Published edit on patch set 57.","accounts_in_message":[],"_revision_number":58},{"id":"ebed8533ad8f6dd3a61496b53bed93622a094a0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 06:45:05.000000000","message":"Patch Set 58:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1ac69bcc34a42faa9cad646f8557f79\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2a40fff2b84f4e80ba0f6fa7e5d00c75 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f3197a3df1045af94e6906a9d7d1430 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c43750cbb3c7439383338e87bdcb4929 : SUCCESS in 13m 52s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"7a76a153c147fa475a7f36c736eaff86fa129cb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 07:24:48.000000000","message":"Patch Set 59: Published edit on patch set 58.","accounts_in_message":[],"_revision_number":59},{"id":"eec5cf5e4569e80734d9d94e3bb2027f8a02e58f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 07:24:52.000000000","message":"Patch Set 60: Patch Set 59 was rebased","accounts_in_message":[],"_revision_number":60},{"id":"ac5f79d5b8e6610608945ebb8fdf075f45b8b977","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 07:28:54.000000000","message":"Patch Set 61: Published edit on patch set 60.","accounts_in_message":[],"_revision_number":61},{"id":"7d1d7e27d04ee326a4afe4433038ba8473fb723d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 07:45:29.000000000","message":"Patch Set 61:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c68b19c86a14e978c2e7272fd61ca50\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7f1e3afd0038422fb72b4bde8af04d1d : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e4ee3643acd8499e8c54f5526c4723ba : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/68bd20cbe42f4808b32ff9fdd3ccbc38 : SUCCESS in 14m 24s (non-voting)","accounts_in_message":[],"_revision_number":61},{"id":"0092fdc6b6982215e3f2c483a96ffb45fae2a64f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 09:27:04.000000000","message":"Patch Set 61: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65f5877f44e34f6c8dbbed71f242682d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5e6fb9794b1240a1b66839a339ff5ee7 : SUCCESS in 5m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a3b4a6b747e40f5aeb4b0d9a4baaf05 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9e131eece2b4b5d8e143ef6b4744993 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd39367a0b2f4e30b1e3a419cdcdfc8d : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9622ca940ba0443ca5dcd5cc430e746b : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa68df52144d47099296adf2dd3440e3 : SUCCESS in 2m 53s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/8b93e2d59e94425d94b3c8a33c243a51 : SUCCESS in 44m 43s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/c9ab7351e4014bb1abb3a1b69b9cb30f : FAILURE in 1h 31m 12s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/780820921178415cb2f8a3a8fc42e044 : SUCCESS in 46m 40s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/6defd8aedddb45008c933eb6a0b87080 : SUCCESS in 1h 07m 21s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1683e35e0fa24d7393e5670bd17b8974 : SUCCESS in 38m 37s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/189f0b431dbb43fe8e008f9076baf71a : SUCCESS in 1h 31m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/03b0635a9df248ec9f4b904555548e60 : SUCCESS in 1h 16m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/502d4e184ae143d9a65290a95dc7b5aa : SUCCESS in 1h 56m 50s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/1fef6a75f00149d39b449a03c2953c8e : SUCCESS in 42m 41s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/216dbd9af15c41918c3d1f578a82d198 : SUCCESS in 1h 34m 45s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/821c4216d1cd481da8c412a93f8c11ee : SUCCESS in 1h 14m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/2fa3cb522a9c40819fbb8308a3ebe4cf : SUCCESS in 1h 32m 18s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/09c38a8f04f340d7abcf31edfc38ced3 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":61},{"id":"3225721d3ed1594e31ff007023ab9d75dc316038","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-19 09:55:01.000000000","message":"Patch Set 61: Code-Review+1","accounts_in_message":[],"_revision_number":61}],"current_revision_number":61,"current_revision":"228715592be56f641e302122b383c1cc6f82da3a","revisions":{"4a3bd39d8fb5286bc86657e1229ca8eec646add4":{"kind":"REWORK","_number":1,"created":"2025-06-08 13:32:21.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/1"}}},"commit":{"parents":[{"commit":"365dee5a8b86381c87ba84206180cdd5fc326fa0","subject":"Merge \"docs: add OpenStack-Ansible as an installation method \"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/365dee5a8b86381c87ba84206180cdd5fc326fa0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4a3bd39d8fb5286bc86657e1229ca8eec646add4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4a3bd39d8fb5286bc86657e1229ca8eec646add4"}]},"branch":"refs/heads/master"},"2c705e294f09029586a8aeaa7e11f2883008b892":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-06-08 13:32:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/2"}}},"commit":{"parents":[{"commit":"365dee5a8b86381c87ba84206180cdd5fc326fa0","subject":"Merge \"docs: add OpenStack-Ansible as an installation method \"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/365dee5a8b86381c87ba84206180cdd5fc326fa0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:45.000000000","tz":0},"subject":"Replace license classifier","message":"Replace license classifier\n\n... according to the following warning from setuptools.\n\nSetuptoolsDeprecationWarning: License classifiers are deprecated.\n\n********************************************************************************\nPlease consider removing the following classifiers in favor of a SPDX license expression:\n\nLicense :: OSI Approved :: Apache Software License\n\nSee https://packaging.python.org/en/latest/guides/writing-pyproject-toml/#license for details.\n********************************************************************************\n\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2c705e294f09029586a8aeaa7e11f2883008b892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2c705e294f09029586a8aeaa7e11f2883008b892"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6fbc8cff0da99d34ad0c677e85c798860ef829a2":{"kind":"REWORK","_number":3,"created":"2025-06-08 13:33:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/3"}}},"commit":{"parents":[{"commit":"365dee5a8b86381c87ba84206180cdd5fc326fa0","subject":"Merge \"docs: add OpenStack-Ansible as an installation method \"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/365dee5a8b86381c87ba84206180cdd5fc326fa0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:33:56.000000000","tz":0},"subject":"Replace license classifier","message":"Replace license classifier\n\n... according to the following warning from setuptools.\n\nSetuptoolsDeprecationWarning: License classifiers are deprecated.\n\n********************************************************************************\nPlease consider removing the following classifiers in favor of a SPDX license expression:\n\nLicense :: OSI Approved :: Apache Software License\n\nSee https://packaging.python.org/en/latest/guides/writing-pyproject-toml/#license for details.\n********************************************************************************\n\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6fbc8cff0da99d34ad0c677e85c798860ef829a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6fbc8cff0da99d34ad0c677e85c798860ef829a2"}]},"branch":"refs/heads/master"},"5094ce4bf7d18a06b2dec12fd8c4c4a9abe912b2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-12-05 23:10:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/4"}}},"commit":{"parents":[{"commit":"365dee5a8b86381c87ba84206180cdd5fc326fa0","subject":"Merge \"docs: add OpenStack-Ansible as an installation method \"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/365dee5a8b86381c87ba84206180cdd5fc326fa0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-05 23:10:39.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5094ce4bf7d18a06b2dec12fd8c4c4a9abe912b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5094ce4bf7d18a06b2dec12fd8c4c4a9abe912b2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e4340ced711e6452d449d25a2ee8c66c8ed19350":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-12-05 23:11:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/5"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-05 23:11:38.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e4340ced711e6452d449d25a2ee8c66c8ed19350"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e4340ced711e6452d449d25a2ee8c66c8ed19350"}]},"branch":"refs/heads/master","description":"Rebase"},"7987969532917a1ce354fed053978b8c3adaccf9":{"kind":"REWORK","_number":6,"created":"2025-12-05 23:25:42.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/6"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-05 23:25:41.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7987969532917a1ce354fed053978b8c3adaccf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7987969532917a1ce354fed053978b8c3adaccf9"}]},"branch":"refs/heads/master"},"380dd72208c34770d74562246e09af9300927981":{"kind":"REWORK","_number":7,"created":"2025-12-05 23:37:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/7"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-05 23:37:15.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/380dd72208c34770d74562246e09af9300927981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/380dd72208c34770d74562246e09af9300927981"}]},"branch":"refs/heads/master"},"d81cab65716e744f0d15f1d3cfabc98450d3f7e7":{"kind":"REWORK","_number":8,"created":"2025-12-06 11:53:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/8"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-06 11:53:15.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d81cab65716e744f0d15f1d3cfabc98450d3f7e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d81cab65716e744f0d15f1d3cfabc98450d3f7e7"}]},"branch":"refs/heads/master"},"cd7255d9cf4a2d9b639afbd6e9650b00e352c991":{"kind":"REWORK","_number":9,"created":"2025-12-06 12:01:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/9"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-06 12:01:01.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cd7255d9cf4a2d9b639afbd6e9650b00e352c991"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cd7255d9cf4a2d9b639afbd6e9650b00e352c991"}]},"branch":"refs/heads/master"},"52d0c3c6887f158f5674ca955a71640ffde09f77":{"kind":"REWORK","_number":10,"created":"2025-12-06 12:06:33.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/10"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-06 12:06:32.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/52d0c3c6887f158f5674ca955a71640ffde09f77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/52d0c3c6887f158f5674ca955a71640ffde09f77"}]},"branch":"refs/heads/master"},"42e784a73684e5db4cde88ebe52a87ecc580e87d":{"kind":"REWORK","_number":11,"created":"2025-12-06 12:07:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/11"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-06 12:07:25.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/42e784a73684e5db4cde88ebe52a87ecc580e87d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/42e784a73684e5db4cde88ebe52a87ecc580e87d"}]},"branch":"refs/heads/master"},"55cf65ec93689a5ec3173db3c5eaf7d472d836a4":{"kind":"REWORK","_number":12,"created":"2025-12-06 12:17:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/12"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-06 12:17:33.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/55cf65ec93689a5ec3173db3c5eaf7d472d836a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/55cf65ec93689a5ec3173db3c5eaf7d472d836a4"}]},"branch":"refs/heads/master"},"b9098df99afb5f864a6803402fa7941955962a9a":{"kind":"REWORK","_number":13,"created":"2025-12-06 12:26:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/13"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-06 12:26:07.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b9098df99afb5f864a6803402fa7941955962a9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b9098df99afb5f864a6803402fa7941955962a9a"}]},"branch":"refs/heads/master"},"16240547bb5fb5fb0fda11874a4e5b56ede66d1e":{"kind":"REWORK","_number":14,"created":"2025-12-06 12:35:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/14"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-06 12:35:56.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/16240547bb5fb5fb0fda11874a4e5b56ede66d1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/16240547bb5fb5fb0fda11874a4e5b56ede66d1e"}]},"branch":"refs/heads/master"},"348f351fc59a8feb38bdd0f0ef55275e4d46fc6e":{"kind":"REWORK","_number":15,"created":"2025-12-06 12:49:57.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/15"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-06 12:49:56.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/348f351fc59a8feb38bdd0f0ef55275e4d46fc6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/348f351fc59a8feb38bdd0f0ef55275e4d46fc6e"}]},"branch":"refs/heads/master"},"16ad6785d94acc0f8225d052399e9eb988a94f67":{"kind":"REWORK","_number":16,"created":"2025-12-06 13:27:35.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/33/952033/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/16"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-06 13:26:36.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/16ad6785d94acc0f8225d052399e9eb988a94f67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/16ad6785d94acc0f8225d052399e9eb988a94f67"}]},"branch":"refs/heads/master"},"7dcc48bb2a0516f9c1e2e727f019fec9d71b8574":{"kind":"REWORK","_number":17,"created":"2025-12-06 13:37:47.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/33/952033/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/17"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-06 13:37:44.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7dcc48bb2a0516f9c1e2e727f019fec9d71b8574"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7dcc48bb2a0516f9c1e2e727f019fec9d71b8574"}]},"branch":"refs/heads/master"},"5021e8788fc05bbc351dace7e6acb2acd49cabb2":{"kind":"REWORK","_number":18,"created":"2025-12-06 13:48:48.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/33/952033/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/18"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-06 13:47:56.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5021e8788fc05bbc351dace7e6acb2acd49cabb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5021e8788fc05bbc351dace7e6acb2acd49cabb2"}]},"branch":"refs/heads/master"},"be9223f5fd794ffb589d1d1ed7d46e2651e6ea4e":{"kind":"REWORK","_number":19,"created":"2025-12-06 14:04:46.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/33/952033/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/19"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-06 14:02:50.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9223f5fd794ffb589d1d1ed7d46e2651e6ea4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9223f5fd794ffb589d1d1ed7d46e2651e6ea4e"}]},"branch":"refs/heads/master"},"48d0d147d5c4c32a63351937ca5da0dcc7d4eb78":{"kind":"REWORK","_number":20,"created":"2025-12-06 14:15:08.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/33/952033/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/20"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-06 14:14:55.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/48d0d147d5c4c32a63351937ca5da0dcc7d4eb78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/48d0d147d5c4c32a63351937ca5da0dcc7d4eb78"}]},"branch":"refs/heads/master"},"6a088a42762c8c8bb9385bbb9dd0ffa0ee7061d1":{"kind":"REWORK","_number":21,"created":"2025-12-06 14:23:07.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/33/952033/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/21"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-06 14:22:49.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6a088a42762c8c8bb9385bbb9dd0ffa0ee7061d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6a088a42762c8c8bb9385bbb9dd0ffa0ee7061d1"}]},"branch":"refs/heads/master"},"722cc25459f6de5ae660c94329a9e4002dab6abc":{"kind":"REWORK","_number":22,"created":"2025-12-06 14:48:40.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/33/952033/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/22"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-06 14:48:36.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/722cc25459f6de5ae660c94329a9e4002dab6abc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/722cc25459f6de5ae660c94329a9e4002dab6abc"}]},"branch":"refs/heads/master"},"0d7a7fe5c01c401408b1dfc7fd94b291669e3bbd":{"kind":"REWORK","_number":23,"created":"2025-12-06 15:11:09.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/33/952033/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/23"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-06 15:11:09.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0d7a7fe5c01c401408b1dfc7fd94b291669e3bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0d7a7fe5c01c401408b1dfc7fd94b291669e3bbd"}]},"branch":"refs/heads/master"},"7c7f62e2c8124c8e650afd0744a6fec62c0dd311":{"kind":"REWORK","_number":24,"created":"2025-12-06 15:21:39.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/33/952033/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/24"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-06 15:21:06.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7c7f62e2c8124c8e650afd0744a6fec62c0dd311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7c7f62e2c8124c8e650afd0744a6fec62c0dd311"}]},"branch":"refs/heads/master"},"cddb4ce8a1ad72caf9c6d072c4c1910f3ef4845e":{"kind":"REWORK","_number":25,"created":"2025-12-06 15:36:55.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/33/952033/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/25"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-12-06 15:36:50.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cddb4ce8a1ad72caf9c6d072c4c1910f3ef4845e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cddb4ce8a1ad72caf9c6d072c4c1910f3ef4845e"}]},"branch":"refs/heads/master"},"324714f7fbcbd65071a864d564a1e483166879b9":{"kind":"REWORK","_number":26,"created":"2025-12-07 13:30:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/26"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-07 13:29:59.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/324714f7fbcbd65071a864d564a1e483166879b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/324714f7fbcbd65071a864d564a1e483166879b9"}]},"branch":"refs/heads/master"},"c8c073141f8b28425c2be55da8cfe205835aaf03":{"kind":"REWORK","_number":27,"created":"2025-12-09 13:31:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/27"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-09 13:31:13.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c8c073141f8b28425c2be55da8cfe205835aaf03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c8c073141f8b28425c2be55da8cfe205835aaf03"}]},"branch":"refs/heads/master"},"b4b6702255d7bd901c6e7b58a8eb7c6290fb49be":{"kind":"NO_CODE_CHANGE","_number":28,"created":"2026-01-06 14:58:03.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/28"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 14:58:03.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b4b6702255d7bd901c6e7b58a8eb7c6290fb49be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b4b6702255d7bd901c6e7b58a8eb7c6290fb49be"}]},"branch":"refs/heads/master","description":"Edit commit message"},"43e5a7463307473adc17067176e18d91b9edea8c":{"kind":"REWORK","_number":29,"created":"2026-01-06 14:58:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/29"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 14:58:22.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/43e5a7463307473adc17067176e18d91b9edea8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/43e5a7463307473adc17067176e18d91b9edea8c"}]},"branch":"refs/heads/master"},"e1d6dee10cab6d39e71d83e1c13e2a77ca15b513":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2026-01-06 14:58:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/30"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 14:58:27.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e1d6dee10cab6d39e71d83e1c13e2a77ca15b513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e1d6dee10cab6d39e71d83e1c13e2a77ca15b513"}]},"branch":"refs/heads/master","description":"Rebase"},"3dd612ff33f5956a323aa9d1d1a44d585f1aaf9b":{"kind":"REWORK","_number":31,"created":"2026-01-06 15:03:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/31"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:03:51.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3dd612ff33f5956a323aa9d1d1a44d585f1aaf9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3dd612ff33f5956a323aa9d1d1a44d585f1aaf9b"}]},"branch":"refs/heads/master"},"e214c41eb5b95d62ab3633153fac3a9f2ae31238":{"kind":"REWORK","_number":32,"created":"2026-01-06 15:04:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/32"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:04:55.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e214c41eb5b95d62ab3633153fac3a9f2ae31238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e214c41eb5b95d62ab3633153fac3a9f2ae31238"}]},"branch":"refs/heads/master"},"61c579fc6e532fe88eef39737fc8b5b2d4776f0c":{"kind":"REWORK","_number":33,"created":"2026-01-06 15:14:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/33"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:14:28.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61c579fc6e532fe88eef39737fc8b5b2d4776f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61c579fc6e532fe88eef39737fc8b5b2d4776f0c"}]},"branch":"refs/heads/master"},"ca6cd7aa4d04de99cb6b47e9e998b96d17617fa3":{"kind":"REWORK","_number":34,"created":"2026-01-06 15:15:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/34"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:15:29.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ca6cd7aa4d04de99cb6b47e9e998b96d17617fa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ca6cd7aa4d04de99cb6b47e9e998b96d17617fa3"}]},"branch":"refs/heads/master"},"b14c5390ca82a17e504c71de10c3122824eed6fd":{"kind":"REWORK","_number":35,"created":"2026-01-06 15:16:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/35"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:16:31.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b14c5390ca82a17e504c71de10c3122824eed6fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b14c5390ca82a17e504c71de10c3122824eed6fd"}]},"branch":"refs/heads/master"},"0876654dee3ba96ce0a1db8343476ed982449ccc":{"kind":"REWORK","_number":36,"created":"2026-01-06 15:17:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/36"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:17:03.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0876654dee3ba96ce0a1db8343476ed982449ccc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0876654dee3ba96ce0a1db8343476ed982449ccc"}]},"branch":"refs/heads/master"},"e71fc20f770677322275fb234138c7c736e612d9":{"kind":"REWORK","_number":37,"created":"2026-01-06 15:18:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/37"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:18:18.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e71fc20f770677322275fb234138c7c736e612d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e71fc20f770677322275fb234138c7c736e612d9"}]},"branch":"refs/heads/master"},"538b7b8aeeaf8a38e27d4dc6b5402b1cf455083c":{"kind":"REWORK","_number":38,"created":"2026-01-06 15:19:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/38"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:19:04.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/538b7b8aeeaf8a38e27d4dc6b5402b1cf455083c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/538b7b8aeeaf8a38e27d4dc6b5402b1cf455083c"}]},"branch":"refs/heads/master"},"e83072c250bff328e06b6e1ee4cb98cc3b18c8f2":{"kind":"REWORK","_number":39,"created":"2026-01-06 15:28:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/39"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:28:45.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e83072c250bff328e06b6e1ee4cb98cc3b18c8f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e83072c250bff328e06b6e1ee4cb98cc3b18c8f2"}]},"branch":"refs/heads/master"},"869a3c4c503a07c22a05d4dfebeb6e4c93e6b5a3":{"kind":"REWORK","_number":40,"created":"2026-01-06 15:29:33.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/40"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:29:33.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/869a3c4c503a07c22a05d4dfebeb6e4c93e6b5a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/869a3c4c503a07c22a05d4dfebeb6e4c93e6b5a3"}]},"branch":"refs/heads/master"},"ba7beb211cb687ace6679a1feac5127b2104f0d5":{"kind":"REWORK","_number":41,"created":"2026-01-06 15:30:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/41"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:30:14.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ba7beb211cb687ace6679a1feac5127b2104f0d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ba7beb211cb687ace6679a1feac5127b2104f0d5"}]},"branch":"refs/heads/master"},"a9cc14945338955512aca738bb8ca343a8aa8a13":{"kind":"REWORK","_number":42,"created":"2026-01-06 15:32:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/42"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:32:01.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a9cc14945338955512aca738bb8ca343a8aa8a13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a9cc14945338955512aca738bb8ca343a8aa8a13"}]},"branch":"refs/heads/master"},"6c743d49205279d3dc36c1d0a11df50fdef9d369":{"kind":"REWORK","_number":43,"created":"2026-01-06 15:32:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/43"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:32:47.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6c743d49205279d3dc36c1d0a11df50fdef9d369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6c743d49205279d3dc36c1d0a11df50fdef9d369"}]},"branch":"refs/heads/master"},"608269d6b48ed8f6e6a8da7f5ca5e6b000de2fac":{"kind":"REWORK","_number":44,"created":"2026-01-06 15:33:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/44"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:33:19.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/608269d6b48ed8f6e6a8da7f5ca5e6b000de2fac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/608269d6b48ed8f6e6a8da7f5ca5e6b000de2fac"}]},"branch":"refs/heads/master"},"88b63de14973430a1ef7b9c9865f848aaa0a6005":{"kind":"REWORK","_number":45,"created":"2026-01-06 15:38:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/45"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:38:23.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/88b63de14973430a1ef7b9c9865f848aaa0a6005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/88b63de14973430a1ef7b9c9865f848aaa0a6005"}]},"branch":"refs/heads/master"},"8e2a7f264ccbd30606d324113da29c2c0dbcf43b":{"kind":"REWORK","_number":46,"created":"2026-01-06 15:39:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/46"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:39:52.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8e2a7f264ccbd30606d324113da29c2c0dbcf43b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8e2a7f264ccbd30606d324113da29c2c0dbcf43b"}]},"branch":"refs/heads/master"},"3f34651a7a46bcf64869b94782c449504f10cdbb":{"kind":"REWORK","_number":47,"created":"2026-01-06 15:44:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/47"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:44:11.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f34651a7a46bcf64869b94782c449504f10cdbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f34651a7a46bcf64869b94782c449504f10cdbb"}]},"branch":"refs/heads/master"},"22073b33fa7fc6e086354a5c374e0510d5505b97":{"kind":"REWORK","_number":48,"created":"2026-01-06 15:44:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/48"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:44:43.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/22073b33fa7fc6e086354a5c374e0510d5505b97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/22073b33fa7fc6e086354a5c374e0510d5505b97"}]},"branch":"refs/heads/master"},"d48efeff1a9ba4eccca2f3c2c5eba0ed0ca217b0":{"kind":"REWORK","_number":49,"created":"2026-01-06 15:46:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/49"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:46:15.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d48efeff1a9ba4eccca2f3c2c5eba0ed0ca217b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d48efeff1a9ba4eccca2f3c2c5eba0ed0ca217b0"}]},"branch":"refs/heads/master"},"448f761bebe8194779304679d2a033a26f2b263d":{"kind":"REWORK","_number":50,"created":"2026-01-06 15:59:42.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/50"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 15:59:42.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/448f761bebe8194779304679d2a033a26f2b263d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/448f761bebe8194779304679d2a033a26f2b263d"}]},"branch":"refs/heads/master"},"de377f5d5b4e4ed383b36db78d1f5bf672ec9751":{"kind":"REWORK","_number":51,"created":"2026-01-06 16:03:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/51"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 16:03:38.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/de377f5d5b4e4ed383b36db78d1f5bf672ec9751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/de377f5d5b4e4ed383b36db78d1f5bf672ec9751"}]},"branch":"refs/heads/master"},"460745085e481317eeeb9976d48f922bb1bc604a":{"kind":"REWORK","_number":52,"created":"2026-01-06 16:04:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/52"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 16:04:27.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/460745085e481317eeeb9976d48f922bb1bc604a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/460745085e481317eeeb9976d48f922bb1bc604a"}]},"branch":"refs/heads/master"},"6b66544b1bb374035968ae788e708b497ff92803":{"kind":"REWORK","_number":53,"created":"2026-01-06 18:58:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/53"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 18:58:54.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6b66544b1bb374035968ae788e708b497ff92803"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6b66544b1bb374035968ae788e708b497ff92803"}]},"branch":"refs/heads/master"},"c22e18b0dcc03e4eee953e036c6259e3194bddd2":{"kind":"NO_CODE_CHANGE","_number":54,"created":"2026-01-06 21:07:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/54"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 21:07:52.000000000","tz":0},"subject":"Tenant replace to Project","message":"Tenant replace to Project\n\nWe now use Project in logs and etc.\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c22e18b0dcc03e4eee953e036c6259e3194bddd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c22e18b0dcc03e4eee953e036c6259e3194bddd2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"47dc1f5b2f3b1d31cd6f640aa5bad6657cffa9fb":{"kind":"REWORK","_number":55,"created":"2026-01-06 21:33:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/55"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 21:33:42.000000000","tz":0},"subject":"Tenant replace to Project","message":"Tenant replace to Project\n\nWe now use Project in logs and etc.\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/47dc1f5b2f3b1d31cd6f640aa5bad6657cffa9fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/47dc1f5b2f3b1d31cd6f640aa5bad6657cffa9fb"}]},"branch":"refs/heads/master"},"3f6fc9b50ed8710a847017197f8e354b4c5f88d0":{"kind":"REWORK","_number":56,"created":"2026-01-06 23:33:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/56"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-06 23:33:38.000000000","tz":0},"subject":"Tenant replace to Project","message":"Tenant replace to Project\n\nWe now use Project in logs and etc.\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f6fc9b50ed8710a847017197f8e354b4c5f88d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f6fc9b50ed8710a847017197f8e354b4c5f88d0"}]},"branch":"refs/heads/master"},"82725a124e4af39f9eb51009dd3ebabc02cbc479":{"kind":"REWORK","_number":57,"created":"2026-01-12 06:29:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/57"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 06:29:10.000000000","tz":0},"subject":"Tenant replace to Project","message":"Tenant replace to Project\n\nWe now use Project in logs and etc.\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/82725a124e4af39f9eb51009dd3ebabc02cbc479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/82725a124e4af39f9eb51009dd3ebabc02cbc479"}]},"branch":"refs/heads/master"},"af2713fc89c26f96390059c7ca43da5794ddbd5c":{"kind":"REWORK","_number":58,"created":"2026-01-12 06:30:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/58"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 06:30:04.000000000","tz":0},"subject":"Tenant replace to Project","message":"Tenant replace to Project\n\nWe now use Project in logs and etc.\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af2713fc89c26f96390059c7ca43da5794ddbd5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af2713fc89c26f96390059c7ca43da5794ddbd5c"}]},"branch":"refs/heads/master"},"18ada336f6e52bfb7ef76bc4f85f53668dfd8fef":{"kind":"REWORK","_number":59,"created":"2026-01-12 07:24:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/59"}}},"commit":{"parents":[{"commit":"0243ca14069331630d94a5dfd8a42f6eb64ea05e","subject":"Merge \"pg_hba.conf not re-creating after rebuild in PostgreSQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0243ca14069331630d94a5dfd8a42f6eb64ea05e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 07:24:47.000000000","tz":0},"subject":"Tenant replace to Project","message":"Tenant replace to Project\n\nWe now use Project in logs and etc.\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/18ada336f6e52bfb7ef76bc4f85f53668dfd8fef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/18ada336f6e52bfb7ef76bc4f85f53668dfd8fef"}]},"branch":"refs/heads/master"},"bfddbf0e85972168c86a122bb74280f402dcc58c":{"kind":"TRIVIAL_REBASE","_number":60,"created":"2026-01-12 07:24:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/60"}}},"commit":{"parents":[{"commit":"9d6895646ed7a6f4a1da348e7d668a48fd151d47","subject":"Merge \"Fix: Prevent config removal on instance update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9d6895646ed7a6f4a1da348e7d668a48fd151d47"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 07:24:52.000000000","tz":0},"subject":"Tenant replace to Project","message":"Tenant replace to Project\n\nWe now use Project in logs and etc.\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bfddbf0e85972168c86a122bb74280f402dcc58c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bfddbf0e85972168c86a122bb74280f402dcc58c"}]},"branch":"refs/heads/master","description":"Rebase"},"228715592be56f641e302122b383c1cc6f82da3a":{"kind":"REWORK","_number":61,"created":"2026-01-12 07:28:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/33/952033/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/33/952033/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/33/952033/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/33/952033/61"}}},"commit":{"parents":[{"commit":"9d6895646ed7a6f4a1da348e7d668a48fd151d47","subject":"Merge \"Fix: Prevent config removal on instance update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9d6895646ed7a6f4a1da348e7d668a48fd151d47"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:32:21.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 07:28:54.000000000","tz":0},"subject":"Tenant replace to Project","message":"Tenant replace to Project\n\nWe now use Project in logs and etc.\n\nChange-Id: Id6f12fcd207c6440dae103f6328beb93439a81b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/228715592be56f641e302122b383c1cc6f82da3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/228715592be56f641e302122b383c1cc6f82da3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
