)]}'
{"id":"openstack%2Ftrove~952537","triplet_id":"openstack%2Ftrove~master~Ie0fea6e30c7944cfbad7de3da3f8f2e5272819cf","project":"openstack/trove","branch":"master","topic":"mariadb-backup","attention_set":{},"removed_from_attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2025-06-19 03:01:25.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-06-19 03:01:25.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-06-19 03:01:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0fea6e30c7944cfbad7de3da3f8f2e5272819cf","subject":"Revert \"Introduce mariadb_backup backup_strategy\"","status":"MERGED","created":"2025-06-13 01:45:33.000000000","updated":"2025-06-19 03:03:11.000000000","submitted":"2025-06-19 03:01:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":949915,"submission_id":"952537-mariadb-backup","meta_rev_id":"557084fea7ebfd173f5a65cb3a5fa2c717f68087","_number":952537,"virtual_id_number":952537,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-19 03:01:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-19 02:40:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-19 02:40:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-13 01:45:33.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2025-06-13 01:45:33.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-06-13 01:45:33.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-06-13 01:45:33.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"efc6d40e18cdd32d83e322e97cb1ce595fd47a3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-13 01:45:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c1bcb3c5efb56e462e22083cbdf19a9ab0ea343","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 01:57:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e511aec7de554e93a86c830e85bd551f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7a043cdc4ea5465d8f8e21f91fbb23bd : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d2f397d33efe4a7cad0eaef5cd56ef12 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/26bfc6216fc544c69148b7ee736d5ded : SUCCESS in 9m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d07b95f05de95edd5343eb5c6fe5e5d5df5da5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 02:33:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a6716aacf18423799171fa0a578b4f3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6d5c8e26eca4ecb85f611ca76bb3457 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c9ca2bbdd174e078bf0dcd81281d87d : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4516ae2a799d4bdcb0fa1e901303c012 : SUCCESS in 7m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/669dcc49db324ddaab9d800ae3a5f7de : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f56e6e5392154ec68e16254124a7e7d1 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da0876e7ec6f4e46818dcef92e6ead1b : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10d4211ddcf64feeb694cd3b3f49d5cb : SUCCESS in 2m 40s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/23a5b038ce6d492896b70820f587a4c3 : FAILURE in 27m 31s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e4d027b22f0248f1b6b8356f745b93b6 : FAILURE in 44m 56s (non-voting)\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/b10bfea43a8f4282bac8e0f1a49daf9d : FAILURE in 35m 28s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9aa0d40ab01b4940b0daa93b20c7c08c : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"319047d3aca49af8ed580702b17bdffdec87fd07","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-19 02:40:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"651ef713b527c5a26288f07eb4f4a3c2fae6adc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 02:40:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3fa7fbab7ae6436ab656228e8368cd0536e89f8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 03:01:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cae3e98eba449178ad7f79e1f1f1989\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15dc36d49e95403f8691006e09e731f6 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f61d25941c0a489ab59b34be60531900 : SUCCESS in 7m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f3350bec1814711a4b7c3dd0213d452 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5adb1f0cb394cd2b65077b30f5b0ca4 : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ff47c5c9f81452e8c61a28b43886b84 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1b0462edd4e945e536cdc0f1154161bd402b007b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 03:01:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"557084fea7ebfd173f5a65cb3a5fa2c717f68087","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 03:03:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85072ac5311640f3ba193323f00bc198\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf647fe898634a5f8bd56b68bafe6222 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bec4c4fec7554183803e0e844c9e4a73 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34e4c52aabdb462377b775d015264d2bd934f293","revisions":{"34e4c52aabdb462377b775d015264d2bd934f293":{"kind":"REWORK","_number":1,"created":"2025-06-13 01:45:33.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/37/952537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/37/952537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/952537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/952537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/952537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/37/952537/1"}}},"commit":{"parents":[{"commit":"bcdec128e4f96b2d931e6a67cf6401eff5bb1cd4","subject":"Introduce mariadb_backup backup_strategy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bcdec128e4f96b2d931e6a67cf6401eff5bb1cd4"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-13 01:45:33.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-06-13 01:45:33.000000000","tz":0},"subject":"Revert \"Introduce mariadb_backup backup_strategy\"","message":"Revert \"Introduce mariadb_backup backup_strategy\"\n\nThis reverts commit bcdec128e4f96b2d931e6a67cf6401eff5bb1cd4.\n\nReason for revert: there three reasons:\n\n1. As the oldest version of Mariadb we supported is already 10.4, so we can use  `mariadb-backup` directly.\n2. Current patch has a bug where when performing restore action, the code didn\u0027t pass `swift_extra_metadata` to image, so we can\u0027t get the version result in an error\n3. I plan to remove 10.4 in this cycle and will add 11.4 instead\n\nChange-Id: Ie0fea6e30c7944cfbad7de3da3f8f2e5272819cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/34e4c52aabdb462377b775d015264d2bd934f293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/34e4c52aabdb462377b775d015264d2bd934f293"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
