)]}'
{"id":"openstack%2Ftrove~952625","triplet_id":"openstack%2Ftrove~master~Icee5ca1117e914c0abb18e135cd6b96c033d8ae0","project":"openstack/trove","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-05-29 04:03:54.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2026-05-29 04:49:25.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-03-25 02:34:44.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"Icee5ca1117e914c0abb18e135cd6b96c033d8ae0","subject":"Preparation for removing eventlet","status":"MERGED","created":"2025-06-13 23:49:41.000000000","updated":"2026-05-29 04:50:18.000000000","submitted":"2026-05-29 04:49:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":49,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952625-eventlet-removal","meta_rev_id":"0f6df86bacc4ba196d43f0cefeb22b6b5760eb63","_number":952625,"virtual_id_number":952625,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-29 04:49:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-05-29 04:03:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-29 02:58:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-05-29 04:03:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-13 23:59:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-14 01:43:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-22 00:10:18.000000000","updated_by":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-07-06 15:25:27.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-05-25 07:16:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"1e2fc76372bddeba315d2f11faed3aa9a0e476f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-06-13 23:49:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff331fe113ae659ed41748e7b64ca2348a5402e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 23:59:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b0d2b8705104918860a30611118a617\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8cd02c13b6154882b596bd0707ee68fa : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/97feafa8071a4266ac74bda72bafbe4c : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e78b8af6c1934adf91e40d2e48cf10f1 : SUCCESS in 9m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5977f7d20fcfa1be1dcadb1df0cb8598295f2967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-14 01:43:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/377df1f4a18c44a0af80351641f1505f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f588f8d9fab84bf9b85f9b7205ae1ee4 : SUCCESS in 1m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2ae696e47464d5cb529940f7d7990a7 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec9167acac3c412f8d27725d0b1a63d5 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed991b663bde4e989626130f4cd8c765 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28822e6ba0864ed4b5ab4cb5a48d3818 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c66205f1152f45af803eb127fc442ad2 : SUCCESS in 11m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d1e2c9c03d840c28aa815a0967c7052 : SUCCESS in 5m 18s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/d71083a467e640368e48b1eff531d6ee : FAILURE in 49m 22s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/88497dd83943491cb925a7b9e3ecfe5e : FAILURE in 42m 15s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/728188abd1c34f528d95bb540ebed217 : FAILURE in 36m 03s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/e9ff42528b154995ba21f4651423307b : FAILURE in 24m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/25626a581744474999e6e064eb60d34f : FAILURE in 1h 50m 33s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3e941a48e2494248850b8beffc5c5684 : FAILURE in 44m 59s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ae46e63ef43449e1bddf660f1c97bb35 : FAILURE in 24m 39s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/768083c435734cf8b968483c717e4cf7 : FAILURE in 17m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/85dcfbd29ae94675acb127e0930264d1 : SUCCESS in 1h 29m 56s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/431b75ae11844cadb834f7aa8f8fa532 : FAILURE in 42m 18s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/613fae955b1a43ef85f89b5e9120985e : FAILURE in 44m 40s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/4dc4265c1b0b421ba0af864474803452 : FAILURE in 24m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/c88a59effb51477293d4c675b4065f25 : SUCCESS in 1h 22m 52s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/980b4f5335204effaaaefa4b032a48b4 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3737b1c82a4468388dad3f5eed5c8eba01332513","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-06-14 10:16:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2e02144006174ba98da81babe5dd94b2194111ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-14 10:29:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ba1d0ccdcf443b0936c193fb790bc1d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9aa9692d065c405e96d0654d13180d67 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6291f614a2774ccc8bbd8ee8a165ff5f : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/678718d660b6404d8d6e1da4e05fa607 : SUCCESS in 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f8d8453116cb309fa75bdc4599dbcfdbbc5eb559","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-14 11:48:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72b6052ee4344a2b8cd06598f7ce17a7\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/74e79d1d805741baae8dfc4b41873a3e : SUCCESS in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf7bf4d0bf8c4d039873a8f0f026eca3 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbd698956b3a4e02a8f3801717c04380 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/590e77e0175a4d4e8237564274bebd62 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79e857aa098943d096f0d20d7d134073 : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28c4d6e394e144379c8d99ef1d95a0dd : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36fa6938ef28405ab795457a43cd73aa : SUCCESS in 4m 18s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/26051cf891d1467a830bc6f70606dafc : FAILURE in 41m 29s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/c290241956174157889d79b4380d1afa : FAILURE in 41m 12s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/4e5f9250f016432d81a845b6d2e45c13 : FAILURE in 27m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/3c1337ce1b1b4a4d94c6b0bff631c87e : FAILURE in 39m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/e2371691003c4ad48d789a9775ff7bc4 : FAILURE in 1h 23m 04s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b2e01191c69c41f584e353a969100f20 : FAILURE in 50m 29s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/952dedfaeb0f4ae6a69b3529677a70fa : FAILURE in 33m 16s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/46c603446acc497785f837f66db42713 : FAILURE in 37m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/22c72da1decd4d7d858f370d2c69d0e6 : SUCCESS in 1h 31m 13s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/017b721730c34e8095283d2408f0dfc7 : SUCCESS in 52m 59s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/c06063761ab74be4989c0ac9a6ea5a20 : FAILURE in 25m 03s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/e0b1bd98118c4fd78f460eabd70a8ee2 : FAILURE in 39m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/0b04f113ab17439d89b0e9aa7a9e1f84 : SUCCESS in 1h 04m 23s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d52cc2c351af4c9d8898ab67005a8bf9 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"786120829936c3095f8131b0f5253b8acb6e725a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-06-15 06:06:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"789029c26f8aa96ec10b6e466db4d83f6f7d7021","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-15 06:17:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ab18648919440f6b32d6ae1752a662a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb5f343341214ea182637471d9f63d44 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a5a932a6f803478e9c0bbf40800866fb : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c429a20ae25e4f01ad32c8f35beba947 : SUCCESS in 9m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d0dc9b969aa9a028a4f77f0844645240ca30452b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-06-15 06:53:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"528e428c2a02f8a22a59e67e1f5d4878becf2f08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-15 07:07:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c155aa992fdb482296ad08a1a016fcd6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8c1925564678471797aa5c10af7b3fd3 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/703bbeba38704be68ee5eeb147243cdf : SUCCESS in 11m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bf811839e5b34f799294a26753a21853 : SUCCESS in 9m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"182541564a184176dfcbe26018ee288b8227e78c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-06-15 07:17:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"47d9d4013ee8017dc77d876dd83e5db363adca18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-15 07:29:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02f095115cbf4533aa874e7a27da09b3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8c61be6ddded442abeb42f147e67a357 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f2569cdc606e4ee98df89f07786a4cd4 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a08cb64132314daf8ebc18541d2bc76f : SUCCESS in 10m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"863576a9f1bf5ff88d2d4dfac872190b4b2b2a1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-15 08:38:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37db80b7b98a44569187306e1837361e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1f422668fe614617b5c1348b410193df : SUCCESS in 3m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d046489af374baebcd9a85cccc379ed : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86ca6bdaecb54b8388533c0ce8b7db01 : SUCCESS in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64459a0a1b134d31a72429caf833870e : SUCCESS in 5m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b822a514be144fe8405f09f183aade1 : SUCCESS in 6m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a779e39b06b246cbbc47dd693c1361ac : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/328bf368e29141cd829a5398b9129b5e : SUCCESS in 4m 28s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/3b576be710b24b14be456c2cb0dc9931 : FAILURE in 24m 26s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/09732c8d82714012a90f7a99cdffc2d4 : FAILURE in 44m 13s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/e0e02e4111d9409bbc43b3880f0e99ae : FAILURE in 27m 59s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/fd4ac2e1cee5407c82a290215d0d7692 : FAILURE in 24m 50s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/f5d9ebaaa6cb4885819165192d1e0fee : FAILURE in 1h 20m 45s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/72f9d4d2d3e040b1a2d6f1d3513571fb : FAILURE in 23m 45s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1c18738401d24a51a598ff3fdda38539 : FAILURE in 27m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/4a4e8876ab36477f8c89d37e13e9a611 : FAILURE in 28m 29s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ffe3f7ac2c9742798c770f8f62557e3a : FAILURE in 1h 13m 13s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/e2e55e63140d42589bde1b291b509ebd : SUCCESS in 45m 00s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/5a8c21714982473aa610fd0117507298 : FAILURE in 28m 05s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/12bb9107f9134fe6954ca5bd69c9518a : FAILURE in 25m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/4a5904dd75ee421dace5d7ca7f80c3d8 : SUCCESS in 1h 06m 20s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/92048efa06c04265bd425f4391b3d35f : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":5},{"id":"1bb6c239fa7fe0900fb338c882335c101c687df2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-06-16 00:12:52.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3cbc4a055b671e60705385c1a111925ccf4c9cee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 00:25:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20216824a62249258738fe86a1cc4f89\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a33da934473043aca0f0bba87f1de6cf : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/01f4c76d859149b5913cb529db9b70bd : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6fca3c85d90c45929115594bf654fe56 : SUCCESS in 9m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9510ef9b0ea879affe3c4c2c84ad263a2ee854c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 02:00:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/01fa2391fe6a4bbdb591b727eed73535\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7b7c43c3d2b1403db1c49f37cf076352 : SUCCESS in 5m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a070f660590423fb0c042deb87530bf : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/462a4d750c294ab883136d8c0cb3051c : SUCCESS in 7m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac516558178d4435a3121fc34cd980b9 : SUCCESS in 8m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/397256b3e691495f8511ec32535fc835 : SUCCESS in 6m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d62bcedb0d34c848de1594c317bfdee : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dc0f88f629f4abe834860cc24d244f0 : SUCCESS in 6m 28s\n- trove-tempest-ubuntu-base-mysql5.7 https://zuul.opendev.org/t/openstack/build/c4b9420956f94dbe95554740a97a8ed9 : FAILURE in 43m 18s\n- trove-tempest-ubuntu-base-mysql5.7-jammy https://zuul.opendev.org/t/openstack/build/3707fd7604d64693853816074514fbfb : FAILURE in 29m 56s\n- trove-tempest-ubuntu-backup-mysql5.7 https://zuul.opendev.org/t/openstack/build/bf78e64f0f1f419e8182aa6925206b32 : FAILURE in 26m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mysql5.7 https://zuul.opendev.org/t/openstack/build/74f865b9162e4eddbea3e0192cb755d7 : FAILURE in 21m 40s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql5.7 https://zuul.opendev.org/t/openstack/build/dd19b902f6cb473d931954c4a077ac0d : SUCCESS in 1h 33m 23s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3038a18d3ea6495ab4bd6972840b2f66 : FAILURE in 42m 57s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fbe8ca9b6b214687bba44218aea998ea : FAILURE in 27m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b5482edda07a44589f4cf2de967aa0bd : FAILURE in 27m 39s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9e3689cd28ab4c3888459f86d44bd7af : SUCCESS in 1h 39m 51s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/d200f232a4c94429957eb8f8b8a13785 : SUCCESS in 45m 53s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/6456fff75b0c4f22b8d09ae190c19545 : FAILURE in 27m 10s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/336eda5fa9a146249d0c2245695729d2 : FAILURE in 28m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/0fde780449f142758313912d5e954057 : SUCCESS in 1h 24m 05s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/77c4ed470ee5471f86d38472957bba1b : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":6},{"id":"a745df9083141661f8f16dec6093b48d1339d742","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-06-21 08:24:48.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"94247ffe9885079f1fe654c6278df72fcca00917","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-21 08:39:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/296e3e3cf1274d2f8c2a050311ad23f7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0d3e0764824944d8ade9bc7427832fb0 : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d872ad743adf408db1dae60d881b5dea : SUCCESS in 11m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5505811fc98f42a388f9bc14a42a9f81 : SUCCESS in 11m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"20e0a762fe570bfce361291e0b8ce21bfff39ab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-21 10:16:46.000000000","message":"Patch Set 7: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc344c6026394f1cac9e75e4e8f889e4\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/16c2f517edb84160a473ecd840deb9d9 : SUCCESS in 4m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7c6183ab6e2469b9719559f6ad790a8 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71726f3f5f2b4717a183fdbca9bc7f75 : FAILURE in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f6d0ad3b1d3475798d8e5f81e59e73a : SUCCESS in 5m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c105d2821ffc43e693bf251df6e5afe0 : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa0105901a0f45f79dcf3d63e0555c34 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e09631bef824448a0eb40adc3c248a1 : SUCCESS in 5m 09s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/fe81275542dd4fc290aea5fafec6b90d : SUCCESS in 50m 33s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/c61353e72b2b495fb1d298c96ca74aec : FAILURE in 1h 24m 17s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/e71c88eb647e4fb58c7e3520cea5d72a : FAILURE in 1h 02m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/68eef3bb3e6243d18674ccd6bfe81fe1 : FAILURE in 1h 39m 09s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7bd5a6abc425477f9af70737c7f01b36 : SUCCESS in 41m 38s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/45321dd10b5047be8b61b2626755bfcc : SUCCESS in 1h 27m 16s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6c2cf839d12a4320b801877ba984a3f7 : SUCCESS in 1h 32m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a447ee4ae95f41aaa6be26e32e92d75e : SUCCESS in 1h 49m 22s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/8dbfdcb829564097b012a094ebae91b0 : SUCCESS in 47m 14s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/0b07188fc3604211bf37cdf4c9ded55f : FAILURE in 1h 13m 30s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/6c8caf481e074d9a978d21f41f249aef : SUCCESS in 1h 13m 15s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/3c08008f88d649fe99949be979d68219 : SUCCESS in 1h 31m 00s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bd3f69ea5f0d4475b6cf3bc0e882924b : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":7},{"id":"a82ef7c66791b30c28e17fd9e50c5e27144b8c9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-06-21 10:57:30.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"c8416ab1a2abbbb9ad9e46acad47e26802ed903e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-21 11:10:18.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffa4e89fe41b40fdb8167b3c39f8e395\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/64b4a6ac0aea4db49146cceb3f9694e5 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b51017c6ed364d13be77e9d0f8945808 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/01a7cee9d98a4db1bb77c4ca89274727 : SUCCESS in 10m 49s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9b25a5e648cce854a15e05a4c6d3f37aff76a3be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-21 13:00:17.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/acf25ae5856e4e12b1bf2f99df704dcb\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/16c1079baf554575bf71eb494ceacd2e : SUCCESS in 3m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18205d82309e4ae2a7f556ebf7b4ce16 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33a37ad5b2d84275bce346dd8d0a346f : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ae48a080165449899f37028cfa6594f : SUCCESS in 6m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/580468b7cba54a0098d4a8a7ef9d56cc : SUCCESS in 5m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3bf37508126148dabb67b8d2bbdf1351 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e1c9985749841b1822071ba678a7331 : SUCCESS in 4m 52s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/74dd025639d1488cb09f88e7bc3f93d2 : FAILURE in 1h 02m 37s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/37690ebcf56d44b9b2f52c25da07407a : FAILURE in 1h 15m 50s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/c27e6509241043a080ce29c58f07f3fc : FAILURE in 1h 13m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/1c51b1f66b70401ebdcec7681e88cc32 : FAILURE in 1h 37m 36s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0997d1f2874a4489b4fb1b73a1af6579 : SUCCESS in 45m 49s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bf3af30068734341800ca1a7882744cd : SUCCESS in 1h 30m 25s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9a4e16c343cc47fdbd28edc340336a02 : SUCCESS in 1h 34m 00s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d5b837a7649545b791f81445f4b30c41 : FAILURE in 1h 56m 31s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/43ac6b98fd2743768a82412b68039af1 : SUCCESS in 54m 55s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/eba5f02a04d04bed910442a87c6b6e30 : SUCCESS in 1h 51m 26s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/a321a3d70b8942cfbb5335e51331362d : SUCCESS in 1h 25m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/7adb415b401a4367b406ec156381208f : SUCCESS in 1h 45m 08s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cf1a8964ad644165bd3a39d9ebe308ba : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":8},{"id":"046543f8b20b7839ec2a8d58a41e6b512cdd9fa5","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-06-21 13:18:15.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4f52ac516a7e637ed0cc44850b1f68a024c34ddf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-21 13:31:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bc04e9b6a7142f3850ea250eff164a6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/11bd6d0fc7dc4b2a9c4da779ef4ab5b5 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/25638b1d32cd42ec9389a7d0de381f86 : SUCCESS in 9m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6d40ce7e29344642b892625e8ee3fc85 : SUCCESS in 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"31cf2abfd2453f1a256c7b2eb22d9ae2123a877d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-21 15:53:08.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/111c912bac2242d286fbbf3e143a597a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/10d36d26b08d4d689ba483df1c1e513c : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf32df81e8dc435f94a4df106a15c513 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3a19660592a4a21a897dce6a03b284d : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c011c8842eb4debab3538eedd41c621 : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da4f7860faba474f90b9adaf6f2e6576 : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81cfcb156b34437e8c45dfbf927a5dcf : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1e82c4b2061493ca87079cc0fd0f6d5 : SUCCESS in 5m 00s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/e826251542e64efbaefe068593888ec3 : FAILURE in 1h 20m 55s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/87936c8d63de44c2b7526a9ee0edad64 : FAILURE in 1h 30m 26s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/bc3b837eeeb94183959e97da756550d4 : FAILURE in 1h 31m 28s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/4db19420841e4c818624084ed417f39c : FAILURE in 2h 02m 16s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/3a46ec7a725941ff923a22149406986e : SUCCESS in 1h 08m 25s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/7adbafee5eb143d58d7e5e759527c37c : SUCCESS in 2h 31m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a8db7f2c8f91439cad15db998d050902 : SUCCESS in 2h 03m 22s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1ce68990a4634f1b827f5a921aef0ed1 : FAILURE in 2h 05m 34s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/3e194d062acf413cbb8478c6fdc492a8 : SUCCESS in 1h 06m 46s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/577dc21d076440e083d25a0eb55d568b : SUCCESS in 2h 23m 12s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/0728fc97199440128cd6064d8f6ecbf6 : FAILURE in 1h 26m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/837c4cd5f3c14096b8de2d8e9e3b0cae : SUCCESS in 1h 51m 44s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eb16ae47b134488cbab40e126bd6bcee : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":8},{"id":"0c3483a01979c0fe636c3f41d152024b9a51bb0f","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-06-21 17:16:31.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"de9970aba32db8b647e0d039f6da7563503a1dfe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-21 17:28:37.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cc47779c5744e598543e80a2055ef84\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f9d4f147550640d6935a83a203657126 : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8d84bae78fb64656abc9ef0d792dae28 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/152e7faafc0d4be5ba7eba9e582dda75 : SUCCESS in 10m 39s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9e7d57d05205385579cb7b1615b625f9f922f383","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-21 19:19:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e2949c0da93436583d836a7f19fa9ab\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8011fd36ef3340f69e0c626daec2b519 : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2479a1a5fee84143ab33d2eeb9eaa37d : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/786f3d263faa4b9984bc420cb8fd670e : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40b8a904807c42f5b731cd01f0f62ef3 : SUCCESS in 5m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3eab13becdb0435eb84e3f342f3feefc : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/855fa98becf2400685eaeba635e598c6 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba7b2ca99f544449a422283edb934252 : SUCCESS in 4m 33s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/9d2eafc7504d401d9962a17a93c0ec57 : SUCCESS in 55m 04s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/6bb190191d474e0298cda9935a55cd8f : FAILURE in 1h 05m 04s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/cbeec9bf69f140e1bfbfa93de18def55 : SUCCESS in 1h 32m 00s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/f2dbc5c0d4084033a1bda0b0f0bcc7dc : SUCCESS in 2h 01m 13s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/11d316cd36c442a990f7341967afe63e : SUCCESS in 46m 08s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/87d2097861b9462caf1534e52336b705 : SUCCESS in 1h 29m 58s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/dbcbba8b14d24bbb95f0f9dcc59331af : FAILURE in 50m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6265e20f89144876a606440196237724 : FAILURE in 1h 41m 39s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/6a9feeab6b544697be447d9309c0a080 : SUCCESS in 49m 42s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/904f623899f94153b351649b7b5dcc44 : SUCCESS in 1h 42m 01s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/f86474b9b9494d8d9f2c613a996fab64 : SUCCESS in 1h 12m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/dfe6b3b4bb56465dae659503c41eedc2 : SUCCESS in 1h 58m 17s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8857dc4036fb4cce92bcb8129e3e14b5 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":8},{"id":"bf038d990f505546d9588d7638dbdc408fb924f9","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-06-21 23:53:59.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a6d442a21683dbc72760afd846e34d79b0ad7347","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-06-22 00:10:18.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"242ed10947f6ac661f91488a8ec470cc65b60bed","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-22 02:15:44.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"897865f3d92242e0866abe059a407dc7b1e22808","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-06-22 06:24:06.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"26bd1089ef6972b97592bc43a59d5368ffb4ed10","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-08-05 03:05:34.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1e9e4043578fb967b292d2a23bf7b3bace2ebba6","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-08-05 05:39:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"163c3a9f284268d1116c348313a9d774aa6ca79e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-08-05 07:34:17.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1d1e3dcbccc00c6e5bf3a3fe9a5ec65f156d22fa","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-08-05 22:27:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7212bf2e266bbe31d8fccfc10795320afebfa747","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-08-06 01:16:30.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d8cc10a3f12b93d10532014b49524d4c82e80ac3","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-08-06 16:37:41.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1b5a78fcf543fedba714ae91412f8169060c2306","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-10-20 11:23:17.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":8},{"id":"188569e5c72bbcc3aaf543c08726037fb013d3ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-10-26 12:32:29.000000000","message":"Patch Set 9: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"cd2e642b0bd08596842f444893906d8da90540a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-10-26 12:32:48.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"cfa9cda51c08d57c863e52b380de4b786a3e51ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-26 12:49:34.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a377053d76f643a4bae82de256b64421\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a3e4e14b2a064cf5811cc254114ec8be : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e212e72a35b34af78053a80813038618 : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ddc9db78e3fa4d2e93aae5c879c27fb7 : SUCCESS in 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6cbe8fcf846eb62de846b4c6c8a1fdee5773829c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-26 14:19:54.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e841f59f0264530ba4be116fc8f74ff\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/776cfa50902e48db8b66b6dbd53f923a : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f810ab7f9b0740f1a1d0084ba08308eb : NODE_FAILURE Node(set) request 76f1f906d7ea438c83b10349aea23a6d failed in 0s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea275ffa48ed46f3991d51b9d808cf06 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1be47845bfc419fbe69a9322cfc6047 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42cb5815b89f4690a1c5fd2a7f5e28c7 : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa434dfcca64489be6dacff5e30803c : SUCCESS in 3m 07s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/41b4ebe95c1043658e5f5e97a666edde : SUCCESS in 50m 29s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/ec3070df75414f83b1848673efd071a6 : SUCCESS in 1h 31m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/05d2943d53674d96bf59708cc3be0151 : SUCCESS in 1h 18m 53s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/d163a895d50b40228582dfdfb16ab73c : SUCCESS in 1h 45m 50s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/e477971d437d44fcbc40299297fe21d4 : SUCCESS in 39m 06s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/0fb45013b87d4135b3b412a86d872b34 : SUCCESS in 52m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/ae8b5349299f409c9a8ae44aa5a02344 : SUCCESS in 1h 05m 19s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b174e6c6ae5c4407b1f23b76bdb0cd94 : SUCCESS in 1h 21m 30s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/864a5f88b9d54bf3adc00b514fbf7564 : SUCCESS in 49m 04s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/0f7505c0389343c5a3122fb1f3ab8768 : SUCCESS in 1h 17m 40s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/33c610d3736d41f28e630c0bd6e12d57 : SUCCESS in 1h 07m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/6acf9de03e1d40548eb15fbc5628c2cf : SUCCESS in 1h 14m 13s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0b19d56173f1489cbba8a57c912d9d84 : NODE_FAILURE Node(set) request 5c5acfbe46f54200a6c8a84a1305ef48 failed in 0s","accounts_in_message":[],"_revision_number":10},{"id":"77cdd582e146ec1158bc9d8f8702ffed3d008373","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-11-14 07:57:56.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"b6498778a8d6022a61ff6fa2aebcf92f2214eaf2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 08:13:56.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/658163eb61c74621958855f2c6897d4c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c1601167ccc4ec49b78b1ddb42a99cb : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c69d38537d51413eba74203979b116b7 : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d016fc70fc424536b0298991aca21925 : SUCCESS in 14m 05s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6e07c0da2cebc4be1e04bc2ea3980935ef2a7fb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 09:46:18.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a84938e5093148f5a0700f2c7f74f14c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3da2d74e2720451b8d2e9019b0f780ec : SUCCESS in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a56ef4e8e4fb4edea62635867f78e3d7 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/addd88d878c942c4b4cc5b3137611778 : SUCCESS in 7m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/867cbe0b83924068937a4158b20aeb4f : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/959d6e83d92f4b3d9dd735a738ab5503 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef7c80605ac348c99078ef7b7551f57b : SUCCESS in 3m 09s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/44a77c25ccdb487490eae18cfd99f991 : SUCCESS in 47m 30s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/b53cecff3d3c4d41a84c07b7d68a8d4d : SUCCESS in 1h 47m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/f38fa32e09a14567bf662bd9cc3972d5 : FAILURE in 1h 01m 03s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/010370ad392a47098f4ba78bb9d06b68 : FAILURE in 1h 39m 06s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a71b683f8ad74df2be8d66dcafad6c42 : SUCCESS in 46m 00s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/0ee7c94b29924a67a0bb097c142d2ba0 : FAILURE in 1h 43m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/56e8f19e8a03464fafccb58fd7711a6f : FAILURE in 57m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/740e7cb419a74c3c88bcff44aa500ca6 : FAILURE in 1h 15m 03s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/c9b48aef7599471f8d2015d1e52c9bf0 : FAILURE in 44m 13s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/ca28ac4167634246aab78204afe5e5e6 : SUCCESS in 1h 35m 04s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/7069eed44b174527bdd619dc38f05deb : FAILURE in 50m 00s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/3b21451122db4c83bbb334f1fc553d1f : FAILURE in 1h 12m 52s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/347ff32ac3b440e09849e8aca4daf13d : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":11},{"id":"402b6ce2ada981df4924a63adabb93af5fed02a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-12-10 23:22:37.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"70619663ddddc92c206a7b3ec09de8f2f34dc431","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 23:57:28.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee1952a7853e44c8b4e264f68941e4d6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4cb35cfa749844b9917000158fa7be4c : SUCCESS in 10m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/43af42db4b1c4ee9b073e0a0d6dd754b : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c393d87349984c7484dce71e0b41c1fd : SUCCESS in 9m 48s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4caeeda54f544ebc0cb19b18506bb1896851cf68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 01:20:15.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb4db9d119b343af9386c54160606122\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/908ea5dd4ad1477187eb934aee9c4429 : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2571783dcf1c41fca9e7420597621cc8 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77866cf4dc0a47e1a5c168b2bf59b5e4 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d453423a1e1946f09a96e34635947981 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92caff49078a4f22ab9594227bcf34c3 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4286ac11c5d4857a0635cefe9d7937f : SUCCESS in 6m 32s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/604610f4cfab4b5db1a4b423f8b811d1 : SUCCESS in 46m 58s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/5dcc596897b54d7d929351dda09725c3 : SUCCESS in 1h 45m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/4942da6442ed48a3948ac2d3a74b5b33 : SUCCESS in 1h 29m 48s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/5229174519534b6f969c252c021a9817 : SUCCESS in 1h 47m 55s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/de09e0baad4c4c19b9326e73271c5c2d : SUCCESS in 45m 17s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/ec22fec82fbb4300b08eef611c44a141 : SUCCESS in 1h 40m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/53a71f8c349847a2aebd914085c42eaa : SUCCESS in 1h 20m 56s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/2cac9c3c67164992a5e2dd77ee22002e : SUCCESS in 1h 32m 46s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/2976caf76d5c4319a67e9b06437ec856 : SUCCESS in 46m 07s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/cf7672271b1b42b890ac0f427203b499 : SUCCESS in 1h 41m 57s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/71422e96f32840d39c60814bca49a717 : SUCCESS in 1h 09m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/1d24445f01f8413a85b072bec93ebdbc : SUCCESS in 1h 30m 22s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c089b340570d4dfc87045b37905dd37d : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":12},{"id":"79e319dc41ebcebb6c7cba87f76015275c7affd1","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2025-12-15 09:53:51.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"2507334f8a17f1b486576ba2f4777321484edfe6","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-06 14:03:30.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"46add87f82f12fbd08d0bff79223a2d55b844356","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-03-17 08:08:13.000000000","message":"Patch Set 12: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":12},{"id":"68bdb272fa95146e695cd2ebe81f2dbff2c5fdff","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-03-17 08:20:07.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"87bb6e5a7e7157cfe10f4fede7fedc4f51fb92f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-18 12:06:18.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"60022036146ecd149f26ea27a21bd0a32f449588","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-18 12:24:25.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"520c05063aba833640200c9127b12d625d68a76f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-18 12:29:53.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"08516ddd9fc55da5e771701b0f91cccef570b43c","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-18 12:53:48.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"6837b5eb11e2afcc7d05d791c6075eefdbddcb0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 12:59:26.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c5a145cdd144ec09d64bc3c91c7f946\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9fb42739e4114f3194d7da7a9a182378 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d3dc72db378d449abc709f9d18db4add : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/72e3619ff7b347a9ad1d07689585f405 : SUCCESS in 14m 29s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"691dafa5a4565a580827ebb7ceaa79a28fedfe09","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-03-18 13:05:00.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b1dc0ae851acc2aa4600a8694f0bb572d29414c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-18 13:14:09.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"3c87aa1b2be59bb2372334cb6bf6e520110bfd54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-18 13:36:27.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"83b568f0fd71a7916493e92b8b19f06ec15e4074","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-18 13:49:54.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"5064ee00e73460d94a767b912c997a3f3d75e0c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-18 14:29:05.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"40c3236a7bd96e43269cc4e520fb42cf1db709d8","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-18 14:37:19.000000000","message":"Patch Set 19:\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"44cc3845019c2e6e15d720179c7a4f21a6ff5785","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 15:06:19.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4959ddc6d828471b8efeaf89ef9b993c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b42db00f9638441cab077086a8f8db8d : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/94139e4a48704148abe86da95c484e87 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d373cd45ab8d401196d62f76ec5b5bd3 : SUCCESS in 14m 22s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"fc7b2473c5e01258219f685070c2d6900a263fbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-18 15:59:15.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"5c0f52d91f3e137f9a1a90055bd74ab6ef318093","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-18 16:05:55.000000000","message":"Patch Set 20:\n\n(3 comments)","accounts_in_message":[],"_revision_number":20},{"id":"d98d6db67cec77417a73acffcf594265e254855b","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-18 16:17:12.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"8321fad0981c5b516e49150bee8492cf7b66d55e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-18 16:19:03.000000000","message":"Patch Set 21: Commit message was updated.","accounts_in_message":[],"_revision_number":21},{"id":"f9aaa247ea59bb7df8b24a131870742f46b9e413","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:37:52.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f44749120d542fe934761a821e0429e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b208d87f5dc04b3aa5017825a089a010 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bfa086c85e6841afb7a97612c94a6aa0 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2f40f0205a40465da28c34ea8b4465ae : SUCCESS in 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"bc7df7f56f5ef4bb0e521b21f48d4d76683b89a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 19:04:27.000000000","message":"Patch Set 21: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2db5d22571644af7bf8a99ddbc7f6604\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/87e938b461954bfe9692fabd20d2d337 : SUCCESS in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a94c94e12ea458f9fd723eb83992cfa : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9907ac1b85bc4cddadda15847c805076 : FAILURE in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca969b53afaa45b9bdca94a1fb1c9852 : SUCCESS in 6m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/424c7c2d9ebd40b4973a7f780deed962 : SUCCESS in 5m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/130d39ea6db24e6892e59e89dab4d815 : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d38e2dc0154cc19fe119834ddcb06a : SUCCESS in 6m 22s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/9a18ce3fd03240028880b642192d5b93 : SUCCESS in 1h 00m 57s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/cc4e1a135ce84e4995a8ac3bde6ea8ca : SUCCESS in 2h 16m 56s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/3da4b7e3948948399a1bd772cbf8dcd0 : SUCCESS in 1h 49m 24s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/b49b3c1d14cc4f44a2101d2cb3495771 : SUCCESS in 2h 14m 22s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/37ba0735a28d455ea29f1485c7b8a835 : FAILURE in 1h 09m 16s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/5f391309a5d24fec90ed499a2ad8169a : FAILURE in 49m 43s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1167e0f90b7d432ea5e6309ab996b9fd : SUCCESS in 2h 00m 19s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/154cfc7c7e954b77bd89059e98be04f6 : SUCCESS in 2h 22m 06s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/685044a06f38467aa6aa0a7b60879e94 : SUCCESS in 1h 08m 56s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/ef8def1ac8d342fc856e2ea3a03b01ab : SUCCESS in 2h 43m 49s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/c541efe74fad4fbeb1423a268efe849f : SUCCESS in 1h 37m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/fa061fde39b34342a25dd13efca3af31 : SUCCESS in 1h 51m 44s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b4a65c2c76314b5e868f613bc7b4d2f2 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":21},{"id":"095ac5a4478d20f377822a1f2f24202d62506080","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-18 23:27:12.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"722d65b1d1c25aa24f7bb61e93fb9fd06736ba3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 23:45:50.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec319184f4594833847f00ee0ccbb3fa\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fe3e7cbac02b422d9138dab0a3df4e6e : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/28f39f1f499049cea22283b2c2bae7e7 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4975db62a8644d1caa3ebb371a096f38 : SUCCESS in 10m 01s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"7d8c78ddd48ed12118fe47626a69b4e964350750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 01:14:41.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aab9120ce503423799b6f4e6950c3ff5\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/62236e4336be401fa80df4918b18a8b2 : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/289de029872e4e139fa2db5000b9727f : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d7c95aea7a54557b5ebfe9fd5d7e6bb : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12ffe6df5cef4cbcbb6a964d3ad2c845 : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a77298d10231410caf5aae771fec64af : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9962749a90be48ae921077d823af11e5 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d3e7ce4d394bb69d167761a70004ac : SUCCESS in 5m 17s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/b67c19bc77f8453ba04d64d4ace0f07d : FAILURE in 27m 02s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/f9526883c7664a11be69db7afc33c093 : SUCCESS in 1h 00m 05s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/90eaf7ffceed4dc1b1e932077f14bf99 : SUCCESS in 1h 22m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/c9a3886b64674928a6fbe73d6ae34e6f : SUCCESS in 1h 46m 35s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/c3be42bd53a340deb674b32ee9b118dc : SUCCESS in 47m 56s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/daadd781e02a4ed7ae0bbb43f561b7c0 : SUCCESS in 1h 40m 55s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/26ae4d19cfa047c285d1e0edbf7d2fe1 : SUCCESS in 1h 13m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/44f4993c625b4bf9bd49a479e4aa3fc2 : FAILURE in 1h 42m 18s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/ebc57eb65d3543cf869e5feffcbb0fd4 : SUCCESS in 46m 26s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/ad4759366c1f4d039fd95e70f3a40a5d : SUCCESS in 1h 39m 36s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/87f065a71cc945c591df1eb8535de36d : SUCCESS in 1h 09m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/7eaa3d008c2644bd9c26ae70b561e868 : SUCCESS in 1h 16m 46s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/76f59ea1c52f49068f5fd16ae506516b : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":22},{"id":"147146792c1c51c33839c25e6c083508d341a40c","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-20 18:06:56.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"02a7cfd9cdbe2f17902241199718040decab446b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 18:42:46.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b36271c15344336b6f72992afacfd1b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/54a067f256fc4a0e9d4592990f37a234 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/906f528475e840feb8bc7d4355fee611 : SUCCESS in 10m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d4e8696c13944c2e82e01daf64437d63 : SUCCESS in 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"319f67c735b285272ada79d3736a581b1843b316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 20:03:54.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/946291c3cd4341458adee11c776db949\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/df3658cbcf8048b099978da7cc4b148c : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18231f75417e463a81eea9bbee957f80 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c9cc7c37c3a4ac49426d7c672f259fb : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86f513685e854b0d8624ae6576300504 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b88315a618d4d3d8c4378604b1135ea : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8f8f9048cc6146bebca20723dfeb0c19 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c77b0c648ac48a881e9d1083e306166 : SUCCESS in 4m 45s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/855a4181b37f4787928dd0949812d3f1 : SUCCESS in 45m 40s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/b774a6213b0e4d2bbfcac4e6e9284b14 : FAILURE in 30m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/712af669ce4745baaf77d07dca1577fd : FAILURE in 41m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/a443ede608654b618767bfa4dfa45fee : SUCCESS in 1h 56m 00s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/341699dc402d4b60be307468cbf42189 : SUCCESS in 48m 44s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/23ed0780f5db4b04bd7efbb6b83308ad : FAILURE in 36m 13s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/88e8c6e2053c47659fc5a4d9e718809f : FAILURE in 31m 32s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/e042febe8b1947ff924020f855e5a7c8 : SUCCESS in 1h 53m 06s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/c0edeb5090f4483b8333c3c1b751b4f5 : SUCCESS in 57m 27s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/7265d7020a8a4671952a74e533641cf0 : FAILURE in 41m 57s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/9e78be5daa394f7aafb56c714adf33fc : FAILURE in 43m 33s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/ab01a57dd50f4495b8fc129154903912 : SUCCESS in 1h 32m 02s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e9000d1d9df465d9887c9041bef273c : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":22},{"id":"5ff93955ef318bc9ea867bffccb9ce8e9dffa7dc","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-20 21:27:42.000000000","message":"Patch Set 22: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"4d6f7aefcef278242c3f95778fe3dccc23178c92","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-03-23 12:19:42.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"6d60717f0cde263ae033d75b11ae23beb7f54070","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-23 23:43:35.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"34efb1ac3b3cefd2fa913da1141a7b5baa74d2a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-23 23:46:14.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"08b715392fa961efc4de7ff27202aecdeebad288","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-24 00:02:12.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"2b540ab5a42ec2f06117e92dee82edf9db74bbea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 00:20:09.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbba7ae6bef24d23afe4463b0af46e29\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7e278632932744818c13142a34045142 : FAILURE in 11m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/899c246797ba4e46971cf6915b275628 : FAILURE in 10m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ce33f3a4c3c84cbe88966de4d1cab623 : FAILURE in 14m 53s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"ab5db540156c791f9768e4e05bfa90e29ff3cdf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-24 00:45:29.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"1826aea0952d18856641a1c6530ff622ffadb472","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 01:06:27.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a5cb557ddb347359dd15e0e531899b2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c319f01d140e4b198f81a87373a22ce6 : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ff45f91c6c694bcd99c2ddd3793dbe10 : SUCCESS in 11m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/91b03223a4eb4e219dd20f5cba97e8a2 : SUCCESS in 17m 17s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"8103ed0e925ad48a5a3a38da05a472ac5b426f7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 02:29:12.000000000","message":"Patch Set 26: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a16c0374da543c0a4124110cf635efd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c66ce4b492c74dc69853ceaedf1565a1 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0248deae889b44f3a61724a669d6fbd0 : FAILURE in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/488ad3305ee74faaa758591059cb8188 : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d2cac443835d4b7f87dbfadc49d6a0ec : SUCCESS in 4m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e4eeee2b70e7430b980effb558bf3753 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55f7a42ec4bf49f292c8d377a2dd868d : SUCCESS in 6m 19s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/66b4f71a7c054ea3adeb594caf17f72f : SUCCESS in 32m 59s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/8f94086be07943a6bf4a701675783d3f : FAILURE in 57m 57s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/39f94e186d3c4a22b0f0967beb992d43 : FAILURE in 1h 42m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/ca38e33d6bb845b794784a64eefa69ae : FAILURE in 1h 22m 09s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/16422116cc34493e835c3fadb9288e00 : SUCCESS in 1h 03m 06s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/216d76b20fd04984aea7e114f9e52df0 : FAILURE in 56m 03s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/f5de802ac05c49f6897760c436e6b5b2 : FAILURE in 56m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/46e3c339b50d4d329b1d8f68d9f6a298 : SUCCESS in 1h 34m 07s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/e44bf090070e43d0a2100d0fe4ac5a7d : SUCCESS in 1h 07m 24s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/2328a8de952f4a85875d0eab8fa25350 : FAILURE in 57m 22s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/9299823f54f146c89db76344ec5a1dc0 : SUCCESS in 1h 17m 11s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/d29fefb913004da998ca6a2dcde42310 : SUCCESS in 1h 26m 32s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8f29dbe39c84469db02098b205e07cca : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":26},{"id":"46534ad86b4a276eea7ce6f7e37dee447daf1d05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-24 04:11:49.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"6d0b17ca2e2ade5fe1f85ded44b7e7dda55f117e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 05:15:27.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a913a2cc9af34e5aa186558cbe7d4155\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a9b8ed47505947e7b819eee017a4e735 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eba304e1eb7b40e0bd22c7d54a391815 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f07df95a953e4e36b6016498f7f82a75 : SUCCESS in 10m 05s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"e12748dbb7122eee9885f146feb8f4af76bb49f1","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-03-24 05:38:52.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"2d97a89e54c990d8705a075c90c4305c687503ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 06:20:35.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0586551522e4418db454e18a8d173b02\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45f481420268467695dfa1d2612a67ca : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e0262d156764f559bb37a21ebaf1582 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f929059adfb24ed1bbd6794760f975e4 : SUCCESS in 7m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cbd1c5a4506447f6ba780001e92cfad4 : SUCCESS in 7m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/38ba699734a14119955ed38ecaa5a208 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f476031232c4065afc97e7693408337 : SUCCESS in 3m 06s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/54fc3923f462486c834d4af0b6c7a71a : SUCCESS in 47m 26s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/ed25b2d148f74ad58948c63cf5c128f2 : FAILURE in 46m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/86c74b05d51b4e5ba012aca0320dce95 : SUCCESS in 1h 47m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/7df980bab0bb4831921c4245ed944477 : FAILURE in 29m 14s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/be927a8746674bf5b366c85de17e890a : SUCCESS in 56m 12s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/e422c4a3317840e98bd68ae0832cd20f : FAILURE in 26m 02s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/098acab23879496fb2a786597be3107b : FAILURE in 46m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/9dee53eb651a41b4a82a9c07bc4730bf : SUCCESS in 2h 07m 49s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/9bba45f14d1a478b814f9f412ddbd39c : SUCCESS in 59m 19s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/819908d86769480598b45f7f809b84f1 : FAILURE in 50m 00s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/7e7e30bd3edc4917a492d75c798dd5e9 : FAILURE in 46m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/cf914806c2264d00a2b44b38b0e75e15 : FAILURE in 26m 47s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cef1c0be29124a01aaa8c38eb47aa1ae : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":27},{"id":"c7183a3bc5796f14827ab805e79c241b8f61d162","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-24 07:05:00.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"b7aeecbc592cb848ae6b7f2db660dccaa6c6dafc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 08:57:56.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58bfb72efb1c4333bd547e591c4be034\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3144df17dc044e02b7336ddb7eb57e5d : SUCCESS in 15m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/986f5dff2357439fb75e6c4dc3ffc0bf : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/96be43c088654f57818e815b2896e49c : SUCCESS in 13m 12s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"587c87de9ea23228aeb39a0eb3fa4227d00cd5cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 10:10:07.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c5fa21ea0ff44b5b6dc4fd0645dfb06\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81e7b7ed1fd44da38e64bf47301fb077 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eac1aaffd984152897626dc2107e4d3 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c83433852d0d43338c38e0b94c6bf5c1 : SUCCESS in 6m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf261df91bcc4d98bf78c3cb1a708f48 : SUCCESS in 4m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/04b5783aaef54e54a83ffb38c0706af2 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd773023468b490db143138e83785774 : SUCCESS in 5m 03s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/4cc66010672f4c5a8843d1a43f3a5b22 : SUCCESS in 1h 48m 37s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/af63805662b04c658473595e30283cb5 : SUCCESS in 2h 17m 16s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/3791f945109c46c58dcbba99cb0b1b59 : TIMED_OUT in 3h 02m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/c3fe42b9a55b4bc6b5343cd2da44f7ad : SUCCESS in 2h 25m 14s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/8679bf1b387f4d49a828705d36c8e40c : SUCCESS in 1h 02m 51s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/2c2202fbab9d4a7f808e4e81f04f7423 : TIMED_OUT in 3h 03m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/649b133185bf443d922b8d16a2eae969 : SUCCESS in 2h 44m 47s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/50cb483e0b7346d59a6d81bfcd23f313 : SUCCESS in 1h 54m 35s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/0c3772badf3f4311a475cde4d42d2b80 : SUCCESS in 1h 16m 01s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/9fea02354c37449892db0787a72a2d6a : SUCCESS in 2h 23m 59s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/5fab73ca65224a44a333a59fa1889397 : SUCCESS in 1h 54m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/1bbdda231d3a4df982b15334037cb12a : SUCCESS in 3h 01m 06s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cc3d350d5d51428f911685c310f1babd : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":28},{"id":"ab58334f6b9fd58440397421d7826f6e3afd727f","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-24 11:26:24.000000000","message":"Patch Set 28:\n\n(4 comments)","accounts_in_message":[],"_revision_number":28},{"id":"5f1d8dc410ede2fba34f19d958a8496298cdabd3","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-03-24 19:25:52.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"7667fc65863797e999059df778effc9bc3c57f17","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-03-25 02:34:44.000000000","message":"Patch Set 28: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"03da47b878027f485a747f1c42e0d32792db0291","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-05-25 01:08:58.000000000","message":"Patch Set 29: Patch Set 28 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":29},{"id":"622c2a4fefee2b171c6dfd62fed657c38bf9b3ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 01:27:17.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41676ae7b27d4c78bd4e0433aca89177\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a43244d864884c1e8281296fea32fa3f : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fa7a14f7c7784a36a34f85d02ed2bbcc : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ae3231a6401149f8bc2e3f34a3b53c96 : SUCCESS in 15m 54s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"0239208b469f003228a0cca417fc0ed49976b7b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 04:07:32.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/328c395e2480491abcbe981f597dbbbd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8c073000fad48629adb7c094cf1f571 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b359e34c66654ca0a9cd672ee3ee0369 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f32eb0295ce943b98069ecfff2d99d79 : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/85837972bbbf42ec9887baa55da24462 : SUCCESS in 3m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/182ab93e3be1451fa2e0afa5ab907a43 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a4d65bf1644261b8e63b7ea22b69e5 : SUCCESS in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/164f84976cb24ce9a0045ddedaa51ce2 : SUCCESS in 7m 27s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/519dd3789c01480caa243ee6b5931007 : SUCCESS in 1h 14m 11s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/5670f13241234e96b6a4c1e6785e7c31 : SUCCESS in 2h 49m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/0ca34f81d6dd4f2eb4bbd1575e91d1f4 : SUCCESS in 1h 24m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/b686276e734646678ec6b636444adcff : SUCCESS in 1h 47m 40s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/933d7cf212b141c2b06f5addf5c60bf0 : SUCCESS in 29m 11s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a32fac819d6c42f9acc116689a2adb86 : SUCCESS in 1h 34m 01s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/682d2abbda484d08bc92f4c5d48debf4 : SUCCESS in 1h 14m 05s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1f98e441e7eb4279bd2a602047a586fa : FAILURE in 1h 41m 01s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/2ca1c93d4b2a4bc98ecd3efe087c9683 : SUCCESS in 43m 57s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/beac4017411d46db8aa865f2502c786c : SUCCESS in 52m 47s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/174feb8bff6e46c6ae8d5c41aa98d5f1 : SUCCESS in 1h 12m 05s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/da9a9148ef62486e8df0449c72671d61 : FAILURE in 1h 44m 20s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d99f376cd4f34942bd2714b38503d2b0 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":29},{"id":"1e41bdd0988483f22dd8e57f485f507be3f7e0f3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-05-26 01:58:00.000000000","message":"Patch Set 29: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"c0295daec5a78ade769b71caa9e42e3fac695d25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-05-27 22:26:43.000000000","message":"Patch Set 30: Patch Set 29 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":30},{"id":"7671ae68d57c6727f25acb79ef1b0b5ee5311c61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 23:08:49.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36a3f295757e49bbb802b76a7d884d87\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9bb6ed8be217496ab4f719127cdf51e4 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c88052748b9144f8b30e4622fd7d936b : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3f76720fac4d48e28cd24646d7118703 : SUCCESS in 17m 54s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"6af3a5ed29187432e5e4cf9ec31d7664ad633f79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 00:42:07.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93c12082cb9c4a7382925984bce925bb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8f848763f6b4e26af2acf4de05cb415 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cd98f6790e94e92b9c1e830eb674cde : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/653de5e16da34ed3a52b92859a6a2b2d : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/400f646320834adfabc8083ee9e81353 : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6eace2a6a2a94accae8b1f744248015e : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfbdd242dd0848f388915e4ac103322c : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/435ae48a4c0c47b1af8532539592d3a8 : SUCCESS in 4m 05s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/a5f50d74237f4060b0a1b3a5b219be4a : SUCCESS in 48m 26s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/7b0cf1c5e27d4d6db0c9c516f7948c17 : SUCCESS in 2h 13m 29s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/6982a7fb82194447910fd7a3d3e68bc4 : SUCCESS in 1h 39m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/87bda856c46645ba8b8da727b0bff114 : SUCCESS in 1h 49m 23s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/d6f432402db4467fb6971fb81bb2789f : SUCCESS in 43m 56s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/44ba408f97b0403f8c7959ac674213e0 : SUCCESS in 1h 39m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/761b334992454107b9ac575e5a4bf473 : SUCCESS in 1h 21m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/abd3f4453a8340ebae887e2f2b1ae48d : SUCCESS in 1h 30m 00s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/57083d1d28a3429a9ee752490c8c5971 : SUCCESS in 46m 33s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/522bf0156e53499ba4725867bb23df4b : FAILURE in 1h 02m 28s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/c0d97de9323e4dafa2638b04a12eb65e : SUCCESS in 1h 16m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/8cfb319383ef410c86ddd68cb53cb9d2 : SUCCESS in 1h 24m 20s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0f011aec8f5d43cea8d12f92fd206406 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":30},{"id":"ef52a2e8a8dc5a12cc2963dd06997c5a6f82208b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-05-28 12:45:19.000000000","message":"Patch Set 31: Patch Set 30 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":31},{"id":"e87ada5ec8ee8aa8338ead2d3f5fd0352fbddfb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 13:08:19.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42b7b64e8ada4d2e86ef33793512b2b0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9f57a496526641b985ee680e752d7ae3 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7178b451dbce4422a8a7a1cbe045788b : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/da17f39289b64667928719d19b0e6b16 : SUCCESS in 9m 53s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"e1288375a6a48e232f932d46d879b8fbf033f4e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 15:38:02.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9863c04a21d8460a98434e4f51a228fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0a76b35d46a4fcbab89954f1d1de95f : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bacb032c9c442b0be30d5fc87e85cde : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/80d6d7b4c31b4c40816e86ad558aefdf : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b0be1f34db7d4a34b4c5d9ba0c0955a6 : SUCCESS in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8f56da7db54a43a7b08c0e8839541b5e : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/410b333bc11648a2abcefd17551cb28b : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87cffeaa7fc14c5c959572ed3efd4868 : SUCCESS in 7m 15s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/0fa103071c3d4f14bf1dbe02e495f1c7 : SUCCESS in 48m 32s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/a30ca99ce4724a0ab423accc233e49ba : SUCCESS in 1h 28m 49s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/f7a91ddb671748f1b35d36d88cce2f09 : SUCCESS in 53m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/a517322e6d1b4bc6a63fa7a58658058c : SUCCESS in 2h 51m 13s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/0a39ca7680784291b57fc272749fa8ac : SUCCESS in 29m 02s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/8054b92c4528435cb0cc2da7a612a3e9 : FAILURE in 1h 03m 01s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/d5606c2753564b9e9e9cb37f07d1bb20 : FAILURE in 1h 08m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/8fa2e69d4359409a9bdc3d5b38329769 : FAILURE in 1h 30m 46s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/ad1647b0fef94e8797d1bee970bbc75f : SUCCESS in 51m 31s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/c8ad7de3cf6e4389a7c552324ef386c7 : SUCCESS in 1h 12m 21s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/40d546a15c2542519801d66122e22656 : SUCCESS in 46m 05s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/4f095f2ae49249c899b4e3aa62a71a16 : FAILURE in 2h 01m 36s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5be171181d374f78ab68d67c58f9846f : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":31},{"id":"531c8199b36d3494faf18a8609143f04d2d01549","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-28 16:02:01.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"4e2777d855b4fc315ac99b435ea5ca833d0862ba","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-05-29 02:58:23.000000000","message":"Patch Set 31: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"21306477bbb9f3e82401db53a79bef2d42b5afb9","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-05-29 04:03:54.000000000","message":"Patch Set 31: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"a46f40b81111b122195457ca1f9957f90dcf8dd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 04:04:18.000000000","message":"Patch Set 31: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":31},{"id":"a10f9c11208e6d202425f70970207bc8d895d49b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-05-29 04:38:31.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"2966732db075927a8baf09f5778d6c7b55ae9d9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 04:49:25.000000000","message":"Patch Set 31: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6452ce04920f4d619b3967c40dfd0799\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa4a59564d904a78abd6ce173edf4cc5 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f5fb4516fc94493a0207da3bd7b0335 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0581b2aa9df2486c849bf7b0da98a7c5 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d840aadf316d4157be27983c69855e17 : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eaf68482868045c08f73c78143e7d2f5 : SUCCESS in 3m 36s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/ead3283b57954e7dbf44e68190e0bbd1 : SUCCESS in 44m 37s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/7a8f19985402401e8dc5c4383fe8abcd : SUCCESS in 40m 26s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/291ae622091f4c5c8ca3beb497ceb508 : SUCCESS in 44m 21s","accounts_in_message":[],"_revision_number":31},{"id":"ec40b25dc21d659a808512a1f35ffeb0456ee7b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 04:49:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":31},{"id":"0f6df86bacc4ba196d43f0cefeb22b6b5760eb63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 04:50:18.000000000","message":"Patch Set 31:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7371f267457d41ec956ba2d543f4a003\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d3f6a88c2964af1bf611cb9d8fb0079 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":31}],"current_revision_number":31,"current_revision":"e816cd850ca3baad6bf3f675d40224a6eeb6dc42","revisions":{"486269e7afb0a69003bb3f4412dadd5a15ba2309":{"kind":"REWORK","_number":1,"created":"2025-06-13 23:49:41.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/1"}}},"commit":{"parents":[{"commit":"e9687dc07fc6c909c77e2a87ba24856950db4477","subject":"Merge \"docs: Add complete instructions for deploying Trove\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9687dc07fc6c909c77e2a87ba24856950db4477"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:49:24.000000000","tz":540},"subject":"Preparation for Removing Eventlet Usage","message":"Preparation for Removing Eventlet Usage\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/486269e7afb0a69003bb3f4412dadd5a15ba2309"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/486269e7afb0a69003bb3f4412dadd5a15ba2309"}]},"branch":"refs/heads/master"},"c72a690e52b8746fa0bd0272f74c14b545dfb10b":{"kind":"REWORK","_number":2,"created":"2025-06-14 10:16:30.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/2"}}},"commit":{"parents":[{"commit":"e9687dc07fc6c909c77e2a87ba24856950db4477","subject":"Merge \"docs: Add complete instructions for deploying Trove\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9687dc07fc6c909c77e2a87ba24856950db4477"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-14 10:15:44.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c72a690e52b8746fa0bd0272f74c14b545dfb10b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c72a690e52b8746fa0bd0272f74c14b545dfb10b"}]},"branch":"refs/heads/master"},"08cc1d5e1e4fabf0a8ab2a9c440c040c10725b2b":{"kind":"REWORK","_number":3,"created":"2025-06-15 06:06:01.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/3"}}},"commit":{"parents":[{"commit":"e9687dc07fc6c909c77e2a87ba24856950db4477","subject":"Merge \"docs: Add complete instructions for deploying Trove\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9687dc07fc6c909c77e2a87ba24856950db4477"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-15 06:05:38.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/08cc1d5e1e4fabf0a8ab2a9c440c040c10725b2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/08cc1d5e1e4fabf0a8ab2a9c440c040c10725b2b"}]},"branch":"refs/heads/master"},"d6989b3db14f01e76306408105f78cf302c5bdba":{"kind":"REWORK","_number":4,"created":"2025-06-15 06:53:30.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/4"}}},"commit":{"parents":[{"commit":"e9687dc07fc6c909c77e2a87ba24856950db4477","subject":"Merge \"docs: Add complete instructions for deploying Trove\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9687dc07fc6c909c77e2a87ba24856950db4477"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-15 06:53:07.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d6989b3db14f01e76306408105f78cf302c5bdba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d6989b3db14f01e76306408105f78cf302c5bdba"}]},"branch":"refs/heads/master"},"abb4721211c0cfb1422068373470621838fe14a4":{"kind":"REWORK","_number":5,"created":"2025-06-15 07:17:04.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/5"}}},"commit":{"parents":[{"commit":"e9687dc07fc6c909c77e2a87ba24856950db4477","subject":"Merge \"docs: Add complete instructions for deploying Trove\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9687dc07fc6c909c77e2a87ba24856950db4477"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-15 07:16:46.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/abb4721211c0cfb1422068373470621838fe14a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/abb4721211c0cfb1422068373470621838fe14a4"}]},"branch":"refs/heads/master"},"0b59f0ae73902d1bef86654fa3399ee1413d0d26":{"kind":"REWORK","_number":6,"created":"2025-06-16 00:12:52.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/6"}}},"commit":{"parents":[{"commit":"e9687dc07fc6c909c77e2a87ba24856950db4477","subject":"Merge \"docs: Add complete instructions for deploying Trove\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9687dc07fc6c909c77e2a87ba24856950db4477"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-16 00:12:31.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b59f0ae73902d1bef86654fa3399ee1413d0d26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0b59f0ae73902d1bef86654fa3399ee1413d0d26"}]},"branch":"refs/heads/master"},"ef99f100666d3bb2f88586bd5cc3ce76fd362158":{"kind":"REWORK","_number":7,"created":"2025-06-21 08:24:48.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/7"}}},"commit":{"parents":[{"commit":"e9687dc07fc6c909c77e2a87ba24856950db4477","subject":"Merge \"docs: Add complete instructions for deploying Trove\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9687dc07fc6c909c77e2a87ba24856950db4477"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-21 08:24:00.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ef99f100666d3bb2f88586bd5cc3ce76fd362158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ef99f100666d3bb2f88586bd5cc3ce76fd362158"}]},"branch":"refs/heads/master"},"5498a48e3cc528c9ad62e5ecc4dc503864fc8963":{"kind":"REWORK","_number":8,"created":"2025-06-21 10:57:30.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/8"}}},"commit":{"parents":[{"commit":"e9687dc07fc6c909c77e2a87ba24856950db4477","subject":"Merge \"docs: Add complete instructions for deploying Trove\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9687dc07fc6c909c77e2a87ba24856950db4477"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-21 10:57:10.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5498a48e3cc528c9ad62e5ecc4dc503864fc8963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5498a48e3cc528c9ad62e5ecc4dc503864fc8963"}]},"branch":"refs/heads/master"},"deed2f8cc02422713a0a7358d8979cfc8fcc90f7":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2025-10-26 12:32:29.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/9"}}},"commit":{"parents":[{"commit":"e9687dc07fc6c909c77e2a87ba24856950db4477","subject":"Merge \"docs: Add complete instructions for deploying Trove\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9687dc07fc6c909c77e2a87ba24856950db4477"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-10-26 12:32:29.000000000","tz":0},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/deed2f8cc02422713a0a7358d8979cfc8fcc90f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/deed2f8cc02422713a0a7358d8979cfc8fcc90f7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"26f1f208fc5689be28f7887c80b48903136aa498":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-10-26 12:32:48.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/10"}}},"commit":{"parents":[{"commit":"7ebcb39c9d948faa425040989f844ba712de6b2e","subject":"Merge \"Update master for stable/2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ebcb39c9d948faa425040989f844ba712de6b2e"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-10-26 12:32:48.000000000","tz":0},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26f1f208fc5689be28f7887c80b48903136aa498"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/26f1f208fc5689be28f7887c80b48903136aa498"}]},"branch":"refs/heads/master","description":"Rebase"},"1390dbac83fa57382910f171b8ce63a8a2d92a8d":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-11-14 07:57:56.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/11"}}},"commit":{"parents":[{"commit":"d932f98308fd39597a92b5fd88b773071699bbb5","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d932f98308fd39597a92b5fd88b773071699bbb5"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-11-14 07:57:56.000000000","tz":0},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1390dbac83fa57382910f171b8ce63a8a2d92a8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1390dbac83fa57382910f171b8ce63a8a2d92a8d"}]},"branch":"refs/heads/master","description":"Rebase"},"b1027702114c20c0303a763a4424078dba1c1b33":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-12-10 23:22:37.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/12"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-12-10 23:22:37.000000000","tz":0},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b1027702114c20c0303a763a4424078dba1c1b33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b1027702114c20c0303a763a4424078dba1c1b33"}]},"branch":"refs/heads/master","description":"Rebase"},"dca0ba46fb99713ac3e6b0f99bb5a84bf6203514":{"kind":"REWORK","_number":13,"created":"2026-03-18 12:06:18.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/13"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-18 12:06:14.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dca0ba46fb99713ac3e6b0f99bb5a84bf6203514"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dca0ba46fb99713ac3e6b0f99bb5a84bf6203514"}]},"branch":"refs/heads/master"},"a2b29bd9efe0f14eaddd8030262a6165a27e5c90":{"kind":"REWORK","_number":14,"created":"2026-03-18 12:24:25.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/14"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-18 12:22:13.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2b29bd9efe0f14eaddd8030262a6165a27e5c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a2b29bd9efe0f14eaddd8030262a6165a27e5c90"}]},"branch":"refs/heads/master"},"023f70e81a54d0cf54451f7be4799257fb6b749e":{"kind":"REWORK","_number":15,"created":"2026-03-18 12:29:53.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/15"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-18 12:29:36.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/023f70e81a54d0cf54451f7be4799257fb6b749e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/023f70e81a54d0cf54451f7be4799257fb6b749e"}]},"branch":"refs/heads/master"},"985cb909c3df49728e32d656504e5194954851fd":{"kind":"REWORK","_number":16,"created":"2026-03-18 13:14:09.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/16"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-18 13:13:52.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/985cb909c3df49728e32d656504e5194954851fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/985cb909c3df49728e32d656504e5194954851fd"}]},"branch":"refs/heads/master"},"9601d17b9ab9348a1ede5c6ff8b26196fe54a193":{"kind":"REWORK","_number":17,"created":"2026-03-18 13:36:27.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/17"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-18 13:36:07.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9601d17b9ab9348a1ede5c6ff8b26196fe54a193"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9601d17b9ab9348a1ede5c6ff8b26196fe54a193"}]},"branch":"refs/heads/master"},"5551860a93a0452b30110499df1c44a1f98b9d4e":{"kind":"REWORK","_number":18,"created":"2026-03-18 13:49:54.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/18"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-18 13:49:29.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5551860a93a0452b30110499df1c44a1f98b9d4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5551860a93a0452b30110499df1c44a1f98b9d4e"}]},"branch":"refs/heads/master"},"3cbb24f7298bcdb5eb9b5b24494410650908fbde":{"kind":"REWORK","_number":19,"created":"2026-03-18 14:29:05.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/19"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-18 14:28:35.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3cbb24f7298bcdb5eb9b5b24494410650908fbde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3cbb24f7298bcdb5eb9b5b24494410650908fbde"}]},"branch":"refs/heads/master"},"09721c3d84dc9c6eb2654a34076ee19a43ca2251":{"kind":"REWORK","_number":20,"created":"2026-03-18 15:59:15.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/20"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-18 15:58:51.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces. Key changes include:\n\n* Adding the cotyledon package to python-troveclient.\n* Adding an asynchronous interface to the GuestAgent API.\n* Replacing eventlet.Timeout with asyncio.timeout, limited to internal\n  Trove functions.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09721c3d84dc9c6eb2654a34076ee19a43ca2251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/09721c3d84dc9c6eb2654a34076ee19a43ca2251"}]},"branch":"refs/heads/master"},"2b165ee7daa5d5b0d22c146046587553137d9cef":{"kind":"NO_CODE_CHANGE","_number":21,"created":"2026-03-18 16:19:03.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/21"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-18 16:19:03.000000000","tz":0},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2b165ee7daa5d5b0d22c146046587553137d9cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2b165ee7daa5d5b0d22c146046587553137d9cef"}]},"branch":"refs/heads/master","description":"Edit commit message"},"487c18ae15d6854de9ec05148e7660586ff54774":{"kind":"REWORK","_number":22,"created":"2026-03-18 23:27:12.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/22"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-18 23:26:54.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/487c18ae15d6854de9ec05148e7660586ff54774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/487c18ae15d6854de9ec05148e7660586ff54774"}]},"branch":"refs/heads/master"},"d9b6f79d7df82c5093da3647e177aa3871dccb1e":{"kind":"REWORK","_number":23,"created":"2026-03-23 23:43:35.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/23"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-23 23:43:19.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d9b6f79d7df82c5093da3647e177aa3871dccb1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d9b6f79d7df82c5093da3647e177aa3871dccb1e"}]},"branch":"refs/heads/master"},"be9e4a00325b74b80a619be6d727d2983f595405":{"kind":"REWORK","_number":24,"created":"2026-03-23 23:46:14.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/24"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-23 23:45:58.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9e4a00325b74b80a619be6d727d2983f595405"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be9e4a00325b74b80a619be6d727d2983f595405"}]},"branch":"refs/heads/master"},"8060f2a64e5709c242cab04452d3b1095961ab29":{"kind":"REWORK","_number":25,"created":"2026-03-24 00:02:12.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/25"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-24 00:01:59.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8060f2a64e5709c242cab04452d3b1095961ab29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8060f2a64e5709c242cab04452d3b1095961ab29"}]},"branch":"refs/heads/master"},"7bbc175416472da7a569357e56c528e4207c9162":{"kind":"REWORK","_number":26,"created":"2026-03-24 00:45:29.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/26"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-24 00:43:16.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7bbc175416472da7a569357e56c528e4207c9162"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7bbc175416472da7a569357e56c528e4207c9162"}]},"branch":"refs/heads/master"},"3559458fc8c51567d2ed2cdbea1f097316e6e2a6":{"kind":"REWORK","_number":27,"created":"2026-03-24 04:11:49.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/27"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-24 04:11:02.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3559458fc8c51567d2ed2cdbea1f097316e6e2a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3559458fc8c51567d2ed2cdbea1f097316e6e2a6"}]},"branch":"refs/heads/master"},"59f6e3e0968ac302f7b139feb2c32e7d4f0893c2":{"kind":"REWORK","_number":28,"created":"2026-03-24 07:05:00.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/28"}}},"commit":{"parents":[{"commit":"d038f57505851d6681d437d12439942102a0f17c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d038f57505851d6681d437d12439942102a0f17c"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-03-24 07:04:41.000000000","tz":540},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/59f6e3e0968ac302f7b139feb2c32e7d4f0893c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/59f6e3e0968ac302f7b139feb2c32e7d4f0893c2"}]},"branch":"refs/heads/master"},"1dbd86728d1a9bffbfb3fc396ac6f7fd8c928f1d":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2026-05-25 01:08:58.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/29"}}},"commit":{"parents":[{"commit":"0994e6b1511359dedd20a4d1bf01608527f4c65a","subject":"Mark hashlib.md5 for non security purpose","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0994e6b1511359dedd20a4d1bf01608527f4c65a"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-05-25 01:08:58.000000000","tz":0},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1dbd86728d1a9bffbfb3fc396ac6f7fd8c928f1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1dbd86728d1a9bffbfb3fc396ac6f7fd8c928f1d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"59f6e3e0968ac302f7b139feb2c32e7d4f0893c2","theirs":"0994e6b1511359dedd20a4d1bf01608527f4c65a","contains_conflicts":false}},"bbd93753c5600163026fc479963d9e04246400fb":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2026-05-27 22:26:43.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/30"}}},"commit":{"parents":[{"commit":"2b75a7f30d1ef4ea374a342738c089be23946315","subject":"Merge \"replace .values()[0] with .scalar() in _migrate_legacy_database\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2b75a7f30d1ef4ea374a342738c089be23946315"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-05-27 22:26:43.000000000","tz":0},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bbd93753c5600163026fc479963d9e04246400fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bbd93753c5600163026fc479963d9e04246400fb"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"1dbd86728d1a9bffbfb3fc396ac6f7fd8c928f1d","theirs":"2b75a7f30d1ef4ea374a342738c089be23946315","contains_conflicts":false}},"e816cd850ca3baad6bf3f675d40224a6eeb6dc42":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2026-05-28 12:45:19.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/25/952625/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/952625/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/952625/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/952625/31"}}},"commit":{"parents":[{"commit":"b3ab18957a601524dd94e4085f959b16754cefdf","subject":"Fix checking for /dev/kvm device exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b3ab18957a601524dd94e4085f959b16754cefdf"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2025-06-13 23:46:14.000000000","tz":540},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2026-05-28 12:45:19.000000000","tz":0},"subject":"Preparation for removing eventlet","message":"Preparation for removing eventlet\n\nThis PR makes changes in preparation for removing eventlet usage while\nmaintaining compatibility with the current Trove API. The changes do\nnot impact the existing program interfaces.\n\nStory: 2011233\nTask: 52367\nChange-Id: Icee5ca1117e914c0abb18e135cd6b96c033d8ae0\nSigned-off-by: Hirotaka Wakabayashi \u003chiwkby@yahoo.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e816cd850ca3baad6bf3f675d40224a6eeb6dc42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e816cd850ca3baad6bf3f675d40224a6eeb6dc42"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"bbd93753c5600163026fc479963d9e04246400fb","theirs":"b3ab18957a601524dd94e4085f959b16754cefdf","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
