)]}'
{"id":"openstack%2Ftrove~953053","triplet_id":"openstack%2Ftrove~master~If7b6bcc4fa6c991f784884feeacb2b20dd2109e6","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-08 16:02:47.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-09-08 01:34:11.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-08-25 01:40:38.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"If7b6bcc4fa6c991f784884feeacb2b20dd2109e6","subject":"Replace LOG.warn","status":"MERGED","created":"2025-06-22 13:20:32.000000000","updated":"2025-09-08 16:04:45.000000000","submitted":"2025-09-08 16:02:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953053","meta_rev_id":"4e0f65386943c04f86c5b155b7e051eb3ea25cf1","_number":953053,"virtual_id_number":953053,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-08 16:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-08 01:34:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2025-08-25 01:40:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-08 01:34:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-22 13:51:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-22 16:02:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-23 08:34:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-06-23 08:34:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"11acf73ad27ec7264a31a237824f6aafe93aa424","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-22 13:20:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f3ca2f8396c58378cc29c7e92964952ce95ef79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-22 13:31:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b0cd3b11872b43eb8faa1678eda838bffc6044cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-22 13:37:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f33aec2f79e20b3f98185c1efa389587e427f303","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-22 13:51:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79edcc009ba64b9995c5430e394363e8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/157687e27d1e4a7086f92d57d0640319 : SUCCESS in 12m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5cb1671e7dab4636af0a928d19c2b14a : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5f6c66c83c0d483ba715abf06929ce09 : SUCCESS in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e01aedb4b4d2d236acec93a35da0758626886464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-22 16:02:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de01d7ecb0f44c168983ef09386a1dc7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4d55ac96f41149219cec7ae9e6139f6a : SUCCESS in 4m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/098301fb45ae4377b729da0186937477 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0f66cf23d944bf5af392f7a53a162e9 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40f83454906f4cc4b7fcd496ccd082d9 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7cf32845c00d433094e43a4a0dc4cd44 : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d64ac5ed3e9405a8586e34941f45b25 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/086d40ad1b07434188e6532e9fcf6d67 : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bc47414b9fe4cbb912b22b750ef7493 : SUCCESS in 5m 32s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/dfd8eba32787472e82aaba26898c4663 : SUCCESS in 1h 04m 26s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/b5d88467b666412084f4056fc386dbd6 : FAILURE in 1h 19m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/df5ce6a53daa495b9affaebe954b9eb9 : FAILURE in 1h 28m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/8f1837f27f3f4993b9318e20e8fd68ae : FAILURE in 1h 34m 33s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0d2a0100333148eeb1493bda31a7d086 : SUCCESS in 47m 04s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/819a832329b6497c9a3fcb504dd1599b : SUCCESS in 2h 02m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/edec064113894ae59f4e45636f4a87dc : SUCCESS in 1h 48m 17s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/424d65a1ed24425086e586b14fc3ca9c : SUCCESS in 2h 23m 19s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/a604bf89da6f4ec4bff9984b1139cdfe : SUCCESS in 1h 01m 46s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/0d921fdddb3f4792bfa60eecf2324392 : SUCCESS in 1h 56m 23s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/437f91bb4a63424cb82e2f1bf37c3c92 : SUCCESS in 1h 28m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/3c8ee5ff8ea44134b817bccbcf4824fa : SUCCESS in 1h 23m 30s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d7517edde0714268a0d9eef71a8f799f : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":3},{"id":"8fcf5c25f20ade5dc875478d921ed33f5ed91757","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-23 08:34:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"cfb942692589f6127b1b9204d59a5a6f792d270b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-08-05 03:08:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c723816071282e840e650de36b5fe450e936b0ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-19 13:22:05.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"023f2ff8548d3a9fdca7829e8ad7ccb189e76fa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-19 13:22:22.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"01b0fd90e880768b985b083984cc8d9b5a6f465b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 14:05:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f5f297bc8d9421c875f646d65ecda27\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f4bf7c2bebd8456ab3a84597d983b73f : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f023e61014e14e0f833eceedd595b068 : SUCCESS in 13m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/53a29b44fa1244e98bd8e8c4baddf0ba : SUCCESS in 17m 13s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7b31d03d1f773450bea549f4ba429718a52ce00e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 15:03:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/409d14c2cb8d4cc6aff9f8c0e87ee840\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bca378d3f77f4704875546f2cdd729b9 : SUCCESS in 4m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78951eeb3df642988af0cea36cb032e5 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6f262387dc142d08eeadd70588a1326 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebe19d12f13349969a4989f3661b0398 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e6494e298a3445bbd91a5cd55c3f567 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97c5d60f22e946f8852b6bda2f4d13a8 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39b6b63392aa45e7a0a2b053069470b2 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70f01d9d513e498a972db154270eb14b : SUCCESS in 3m 08s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/63d9f727b98f46ee9357067c9fd55502 : SUCCESS in 43m 58s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/71930b19dde047fbada391b429097a23 : SUCCESS in 1h 32m 53s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/814ca315745a4e5487193d0eb4e05e67 : SUCCESS in 1h 09m 49s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/fca7632ca3fb43069933172e1ff35fa1 : SUCCESS in 1h 39m 21s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/eb410215d89a4b56a53fb374d8f3d0db : SUCCESS in 48m 15s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/d80cf34d0a7647f8bd18820a39c48982 : SUCCESS in 1h 27m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/aa5acc82a1854c50a2840ed3d1af4191 : SUCCESS in 1h 10m 05s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/d6b3627e289a40b2972ddc676bb4387f : SUCCESS in 1h 28m 29s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/22ef61a1e8af461f84b7c0d00a0a218d : SUCCESS in 48m 06s (non-voting)\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/b63839222caa4619a8d94a23fa532a95 : SUCCESS in 1h 15m 00s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/4a0205d9b40045d29a22013ae8ca0b80 : FAILURE in 50m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/3ea1b92e2ec548a694059bb5c8867108 : SUCCESS in 1h 18m 15s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bc62f4b353d74c019b21756732b0d112 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":5},{"id":"73bff83a41a659365585e540281050d9ff7d9012","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-25 01:40:38.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"8c94a4e8a2388ba758aae3199a0dc98900077fdb","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-09-08 01:34:11.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"0c5896dcec8a58d23635b847a150ccbb80c862da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 01:34:32.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4fd43b07c52306a82505f3da8af3ab68cae24878","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 02:37:16.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa46f45c95724779878de4fb95d3f033\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/44da6e1e32a941f2aaca989b596ad788 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0450509499d1402faf43833ef93b06d5 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de98e1eaa7244a41ab7ef3517c9095c5 : SUCCESS in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb92096e325540dfa0261c57dcfc32b4 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c4e26b0b8bc45b4ae60828fbc078d52 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10ae2eb87f61484b82f85fc996b81482 : SUCCESS in 5m 24s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/801bfd98978d4fa38564a8eafa3306c7 : SUCCESS in 53m 13s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/89463c76562e4b7485760f1a99ecc41c : FAILURE in 16m 00s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/d5447bfd43e34bccb4395c787e0ab393 : FAILURE in 20m 52s","accounts_in_message":[],"_revision_number":5},{"id":"ca6e963c7aa461eaa55b7ad69de50f326ce08cc3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 03:09:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"42305b77c0ce7ccf1e190f8a945c9413e6280510","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 03:21:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/567f23ac1bf94c4f993311daa82ce920\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2bb807ea23ba47c5969f4db5f4d9ef9b : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/074af13931a8414a8e186672573543a0 : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1812b7ca137c40129aea053a2cfe4a6f : SUCCESS in 8m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"98da3f32a072ed65d328fb56e09b42af6eeda835","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 06:13:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2497c0b0f09493dbeb9a276dbca9683\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c9475aa17d6548caae99e7aa11a6120f : SUCCESS in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/257df2c45cb54d08a6a63c8bfdc0b04a : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/174ee9ba2ea34501bd017a72121979d9 : SUCCESS in 7m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c129670aa65b4cc4a2547d90aa96e380 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7663066384114cb59f28b3f6d13ed410 : SUCCESS in 6m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/831de0823e8d47ac8103badea336a28f : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2acac074ba164cfbb63232eef5aad20b : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d17035576dee4c79b3577e602a98bb3d : SUCCESS in 4m 59s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/c379a6c179864ca6885f50c3f8e0d04c : FAILURE in 20m 04s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/4a5cc14639a44613832c134384d371ed : SUCCESS in 1h 22m 49s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/49fca9e15eb14a6a9ff22d5fb5a180cc : FAILURE in 22m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/ca98458c4dfd4aa09aeefa8f25af959c : SUCCESS in 1h 38m 33s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/f4f286de08044d058dba29bcd58a6981 : FAILURE in 20m 42s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/3ffbcedfd9e7404eb3544b273205aebe : FAILURE in 45m 46s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/07c3b07f724449f9b31d82bfbdbb2aa7 : TIMED_OUT in 3h 02m 53s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/5ade267f3df84c0ea919d5a27b439561 : SUCCESS in 1h 15m 52s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/bfd40db364294ed397657fdfefd2ad74 : FAILURE in 19m 58s (non-voting)\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/1622400f118b47fd97ab3e5917f01cf3 : SUCCESS in 1h 15m 58s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/39a62cbddcb547d5b445ed90ac48b3a0 : SUCCESS in 1h 03m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/8878ceaa7e4f47e5a8a014656f1b56c4 : SUCCESS in 1h 18m 18s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/60a7cac8362c4b77b8fc23378bb1fc21 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":5},{"id":"fe477c0e55e45abbbc1407454c359e450c6fad0d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 13:04:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7914f26c8f03c7f854c575007857c58d15938b0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 13:44:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0531c6e9b6d5459d93ce477b1c9a7d8b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c415d12f94ac468c9e44a294b951966b : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bbd1b1e9e5cb4fc2adeaad890057bc5f : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5e95b7978d694b21a43c0c103f1e80ec : SUCCESS in 9m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9b1f08f5bc90d858ed06ec1f40c08c6a47258085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 15:09:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb8f645341244bbdb9135b7e19acdc20\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/abe409378b1e4db9b30f1406d9d3fe6c : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4a0152c349348efb59fd6a518876d04 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/688dd357ae4c4d50820de07ad9e9e328 : SUCCESS in 7m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e895a64ecd6454facd87b1f25752ae2 : SUCCESS in 5m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ff23965d4ca4fa688a75518d81ab5db : SUCCESS in 6m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02709c65e7044128a6468cf64d81a6d2 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d8e4a3be6d447318294237e5a061a79 : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19c66b8f14c743bfacb54d9d32104951 : SUCCESS in 6m 56s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/7a0366d870474bc4be1935af1100aec0 : SUCCESS in 46m 11s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/f8a62655cad34c2c80862a1e6f447c3b : SUCCESS in 1h 34m 55s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/66566a432bb44ac693d236521c449d50 : SUCCESS in 1h 21m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/ddbf72159ba3485382305792e181aa76 : SUCCESS in 2h 04m 11s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b93596d2990847c685885e584e8847cf : SUCCESS in 49m 59s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/350fb7afa0d74077ac64cb1ec0e6e2d3 : SUCCESS in 1h 21m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1302a05823854b66bf2aa4d3559a14ae : SUCCESS in 1h 10m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/abde4baa9cbb42eea01b8abff97a4726 : SUCCESS in 1h 26m 04s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/70f37283c0e34012a2cd399b1a2b63b2 : SUCCESS in 54m 17s (non-voting)\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/4a09dd0260f8405aa1aa2050fac7c855 : SUCCESS in 1h 19m 18s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/25305e01907c401ba97863a56abb12cf : SUCCESS in 1h 08m 53s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/de87dcd6c29e47c491366488035b2b41 : SUCCESS in 1h 21m 25s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cd9600fe27c24053a67c07df4b470359 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":5},{"id":"efedf0c22405f6f78689f10f01388d159e39f1e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 15:10:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8904f1a951a4c4ae5d603c239573260465824f6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 16:02:47.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cae07a1a3774a28a8f3ce33fa2d691d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4e8f220a098149d492274c47b113de1e : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d253ca34fc584d85aa8e81fae23a27bb : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2865e64709744d99905857ebc429c24b : SUCCESS in 7m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d53c9254f1a4eb59f6f222923abba5f : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc256a6da9444080a35d3d835590d966 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e41cc5d90b34d85a7017942e412f749 : SUCCESS in 5m 57s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/4015ec96442a4575901ed1b783896a60 : SUCCESS in 51m 35s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/545938e0be2740b39ffc24dae4182734 : SUCCESS in 42m 55s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/63a187a47760409f8e4f89afd4521edc : SUCCESS in 48m 09s","accounts_in_message":[],"_revision_number":5},{"id":"3c799c95dd91beb54fd407c937e2699b7ec0ecd3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 16:02:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"4e0f65386943c04f86c5b155b7e051eb3ea25cf1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 16:04:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebc341e1479d45d39d040819ae971d0b\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/9cbfff90fb3a4d8189b4ed3f96eb879c : SUCCESS in 1m 00s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2030ec6bbe884762816b69ecb061b4c3 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc2fb7fb14be4cde9e9e577b8a4bf96b : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"018bcb0900558832c1ead52200252e34631943b4","revisions":{"4e18fa8f5d7c039b29ca72a6525c4f5b4e39a105":{"kind":"REWORK","_number":1,"created":"2025-06-22 13:20:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/953053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/53/953053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/953053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/953053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/953053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/53/953053/1"}}},"commit":{"parents":[{"commit":"36105ddc3c3c5907bf32c1d30e8f185cb6fba134","subject":"Merge \"Revert \"Introduce mariadb_backup backup_strategy\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/36105ddc3c3c5907bf32c1d30e8f185cb6fba134"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-22 13:19:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-22 13:19:34.000000000","tz":540},"subject":"Replace LOG.warn","message":"Replace LOG.warn\n\nThe warn method is deprecated in favor of the warning method.\n\nChange-Id: If7b6bcc4fa6c991f784884feeacb2b20dd2109e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4e18fa8f5d7c039b29ca72a6525c4f5b4e39a105"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4e18fa8f5d7c039b29ca72a6525c4f5b4e39a105"}]},"branch":"refs/heads/master"},"6afdae7221158e8022b356d6a34b5bc4a9948a8c":{"kind":"REWORK","_number":2,"created":"2025-06-22 13:31:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/953053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/53/953053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/953053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/953053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/953053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/53/953053/2"}}},"commit":{"parents":[{"commit":"e1d572fdadcad38dbd8c28020d8693ca49f2433b","subject":"Use built-in hacking check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e1d572fdadcad38dbd8c28020d8693ca49f2433b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-22 13:19:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-22 13:31:31.000000000","tz":540},"subject":"Replace LOG.warn","message":"Replace LOG.warn\n\nThe warn method is deprecated in favor of the warning method.\n\nChange-Id: If7b6bcc4fa6c991f784884feeacb2b20dd2109e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6afdae7221158e8022b356d6a34b5bc4a9948a8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6afdae7221158e8022b356d6a34b5bc4a9948a8c"}]},"branch":"refs/heads/master"},"cb7288e399e4b23ec1cd9c08517906c0c21a492f":{"kind":"REWORK","_number":3,"created":"2025-06-22 13:37:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/953053/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/53/953053/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/953053/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/953053/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/953053/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/53/953053/3"}}},"commit":{"parents":[{"commit":"e1d572fdadcad38dbd8c28020d8693ca49f2433b","subject":"Use built-in hacking check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e1d572fdadcad38dbd8c28020d8693ca49f2433b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-22 13:19:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-22 13:37:27.000000000","tz":540},"subject":"Replace LOG.warn","message":"Replace LOG.warn\n\nThe warn method is deprecated in favor of the warning method.\n\nChange-Id: If7b6bcc4fa6c991f784884feeacb2b20dd2109e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cb7288e399e4b23ec1cd9c08517906c0c21a492f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cb7288e399e4b23ec1cd9c08517906c0c21a492f"}]},"branch":"refs/heads/master"},"4f56e9d20a13cfc82eee4c9d5229a89ce7c2e80b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-08-19 13:22:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/953053/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/53/953053/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/953053/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/953053/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/953053/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/53/953053/4"}}},"commit":{"parents":[{"commit":"14acd73dd06045badab644f218f865630cf8d03e","subject":"Use built-in hacking check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/14acd73dd06045badab644f218f865630cf8d03e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-22 13:19:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 13:21:54.000000000","tz":540},"subject":"Replace LOG.warn","message":"Replace LOG.warn\n\nThe warn method is deprecated in favor of the warning method.\n\nChange-Id: If7b6bcc4fa6c991f784884feeacb2b20dd2109e6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4f56e9d20a13cfc82eee4c9d5229a89ce7c2e80b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4f56e9d20a13cfc82eee4c9d5229a89ce7c2e80b"}]},"branch":"refs/heads/master"},"018bcb0900558832c1ead52200252e34631943b4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-08-19 13:22:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/953053/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/53/953053/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/953053/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/953053/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/953053/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/53/953053/5"}}},"commit":{"parents":[{"commit":"bf116bae67bb66b227c1ca2f7acc29b7722a683a","subject":"Use built-in hacking check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bf116bae67bb66b227c1ca2f7acc29b7722a683a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-22 13:19:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 13:22:22.000000000","tz":0},"subject":"Replace LOG.warn","message":"Replace LOG.warn\n\nThe warn method is deprecated in favor of the warning method.\n\nChange-Id: If7b6bcc4fa6c991f784884feeacb2b20dd2109e6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/018bcb0900558832c1ead52200252e34631943b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/018bcb0900558832c1ead52200252e34631943b4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
