)]}'
{"id":"openstack%2Ftrove~953217","triplet_id":"openstack%2Ftrove~master~I510adc33116760ec5af54a9158b76c9a628ef981","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-07-02 06:00:20.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I510adc33116760ec5af54a9158b76c9a628ef981","subject":"Fix trove-imager-loader.service not started","status":"MERGED","created":"2025-06-24 15:10:53.000000000","updated":"2025-07-02 09:01:04.000000000","submitted":"2025-07-02 09:00:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953217","meta_rev_id":"5ea93454130c3f63a02edbec6debf093e57bfca6","_number":953217,"virtual_id_number":953217,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-02 09:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-02 03:25:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-02 03:25:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-24 15:36:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-24 17:10:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d105a261fc0b74eab68135a55b7290f478966190","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-24 15:10:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65667b51118c080a6b1749fe58c923934d749e6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 15:36:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d18d7c6a57c411f9d169da88562cbbf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6288d0290d2a41ad9c1fff2541a39e27 : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f511adfe728f4807a46cc8280bd93ac5 : SUCCESS in 11m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bdedef99a1324554b64d774e94e6591f : SUCCESS in 16m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"36c5e39387ec925cf4de6feb3d56c3971d09a7a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 17:10:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1d23649037b433f8e25d1ff25ab9353\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba0f9f74cecd4405a03c6fe1eac87015 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf69a8b17b374fafbef8aef28bc2041a : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa3330b9ae624649bdf402e45d270cb9 : SUCCESS in 5m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/860d6dc106d04c04acd0a58688c305e2 : SUCCESS in 5m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dff0192a449c4439ac3bfecba7613b5c : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef796d20dea04c55955e778c7681341f : SUCCESS in 6m 20s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/4196c8d482524933b32a2902ded8026a : FAILURE in 1h 13m 24s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/bcc72e947cb746e4a9cdf486b277ed45 : FAILURE in 1h 01m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/ecebd3074cf74bde99588ef7e0fbc434 : FAILURE in 1h 13m 36s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/1a63a95da4984e5a82fcd272efca4652 : FAILURE in 1h 32m 26s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/cfc6aee3f60949c58845796c83531114 : FAILURE in 55m 17s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1b3d1b71adbc4469a0b9bdd77adf215b : FAILURE in 1h 04m 15s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/fc8c82a9054b44bb9ce19554b24bd490 : FAILURE in 56m 18s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6f8ce0f2856e4e589ebf577caefd014e : SUCCESS in 1h 55m 49s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/7d1309a32bca4df98b9b129ce1d75095 : SUCCESS in 1h 05m 12s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/858df26a6747440dae9df756613c8ce0 : SUCCESS in 1h 34m 13s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/7cade74cc0264a23af685e45390f33bf : FAILURE in 1h 13m 18s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/6fca150736a94644b5f248d1d3a7eec6 : SUCCESS in 1h 44m 00s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bdeca880e72246b58189f877be57f0d9 : SUCCESS in 5m 56s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/e7348d00c3714c308ea4c9a2ec12f978 : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2e95d3f1a3bfd317a0b67fb4aaae7893dd8a5d76","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-06-25 06:39:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"efd87da91584d0fdf970059cd352189aa994266a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 07:18:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8946652990b4dabbe43a9eb747de9bf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/018f26af97f54644ad3ce99aeed978f6 : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1005727a72034c8eba35d44ad5307eb3 : SUCCESS in 13m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e77f632a055547378eb7d37418f7fa3f : SUCCESS in 16m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07910a5330ac8f496a9f7b2aa9ca570a9eec322b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 08:54:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7faf6e0be63441ef942853661a369ba4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/907816900a49490c80e4706156da4c26 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2152165f943401b99382ab983c7f303 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0e69ba31db34a98a71e0fbe76d6f3e6 : SUCCESS in 5m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0cc67edeeb1e4c5aaa6983a7dd1594fc : SUCCESS in 6m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b2a0328ce7744e6ebfb0436a67f393a7 : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8be04fdbf9943fb87135f3f04c64603 : SUCCESS in 5m 59s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/a2bace5ae3df4d03acdb5ee15bf2fafe : FAILURE in 49m 20s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/8194b405ae2848ef90790d3d2558d284 : FAILURE in 37m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/0aa88ddded7f435399a875af0dc942ff : SUCCESS in 41m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/28aaef29cecb418c9f1ebb36e6808c71 : FAILURE in 57m 56s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/1f681a2a55ba4dc18e4528bcba64a4ff : FAILURE in 29m 05s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/728f6c68bec545538d388fb901a9419f : FAILURE in 36m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/db90a0a3c2ea45979a5d582d72b8356a : FAILURE in 36m 33s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/d15744e188e14abda93a623ca18617a3 : FAILURE in 57m 27s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/1e26a9ddb6574eb28440d40b265a0ea0 : SUCCESS in 1h 12m 15s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/29ef5629781f4bd1a2741c921258e4f3 : FAILURE in 36m 49s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/5fd0424deb2c4dbfa99e50b2e5383fba : FAILURE in 36m 42s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/394e450ee4b94d8ab0015ba0db1b7ad0 : FAILURE in 2h 13m 22s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8fddd37ffd774c2ea0ea25615e64efe9 : SUCCESS in 4m 53s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/f0d844f95a1248e99dc5cfb7966e1fca : SUCCESS in 8m 45s","accounts_in_message":[],"_revision_number":1},{"id":"97635bb02b636b26fc0b925122e966adc77f52ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-01 02:57:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"95eff301c9f48ca782f2f66c28c1f759f58fddeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 03:39:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ee373596ba64b0f8f68736950e6c881\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ef6f0e6f84ef4b7e8efdc74e27959ac5 : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2792a787c611471a99b96a5635edd00e : SUCCESS in 13m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e9a74e9a6c73424eb08b92f6aeb99523 : SUCCESS in 17m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"11bcf1c0dcf72aa3f33b0d6c9658395e873a83bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 04:58:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fae55453fd1c46988aff15e4f86033ba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c57fb3bcba604c33a8236f6f473e5ea6 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b380ffbed6f04e0883bc7f8d4fa11beb : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c1c7838dd644e0e902862d8d477cfee : SUCCESS in 6m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e90c46c821d7440184f5fadb5acd4628 : SUCCESS in 6m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43dd1acef15b4fc7932aef3ff30c0bc8 : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d5513fb8a5241f5a68b848c9baaec98 : SUCCESS in 6m 36s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/500f4fedf07c431790f8a8c2f6e72952 : SUCCESS in 28m 45s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/977da06de5ac45d680ade7b71daee865 : FAILURE in 1h 22m 05s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/7091c6bd594d4a2a821b424a007fe416 : SUCCESS in 1h 46m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/9f6ccde733474305833259a977513085 : FAILURE in 1h 34m 06s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e6c4a8e00cba4b819c37d8b7e6f987a3 : FAILURE in 25m 36s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/db040d3bbe7249519373c24127ccf4d8 : SUCCESS in 1h 53m 27s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/704498c04f5b4935b628b28628db30ee : SUCCESS in 36m 47s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a22867aa02784cdeb77406a9208997b7 : SUCCESS in 1h 47m 52s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/aec08b9570fd46288e64b2746501c1be : SUCCESS in 1h 02m 06s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/9717b6d96e2a4ecebc1ec53b72e3ed62 : SUCCESS in 1h 25m 41s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/7ce6eaf396b94ffda18e5431e3285fb6 : SUCCESS in 35m 24s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/1b7904bf354841cb8716c3e1c4c5cac3 : FAILURE in 1h 21m 37s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7591442ad01c46da95c4b46620dcf763 : SUCCESS in 5m 37s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/2703c02d8f15481d91b2d68f91a8ef2b : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":2},{"id":"cb5529da02c748417c065f531f27aa8e2b973653","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-01 05:34:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5eb65184d807d08baa8edfe21ba0814fda86cbfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 07:19:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c57586ed57c48048174635ee80e8ef3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26182181669f40eb92e998b2aefd6736 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/330a20af445b47bf83449ff8824a02cf : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e0f551746714ed2b5a0ba57821d0d75 : SUCCESS in 5m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a39fb00fdb246a09aa697518e4de856 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a00985bbbe81458fa317c7d8765b3304 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34ca14dd928b4ad08504eba30a97204c : SUCCESS in 3m 54s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/4395379091a740ef978434c7d897ff6e : SUCCESS in 43m 21s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/e6547e03221845e08468936fbbe6c628 : FAILURE in 43m 45s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/4253133f30154d47b2ca959c7bb26bb7 : SUCCESS in 45m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/ce41506b0afc4643978d3393cf4413ed : SUCCESS in 1h 09m 19s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/138975a2cab14884a6afe7cdc2e29640 : SUCCESS in 28m 28s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/372db95d3f2443b5a8057e06cded0229 : SUCCESS in 46m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/6954945730964710b2f8b91d325ba1c2 : FAILURE in 30m 39s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/70f4b28e413944048c11ad183e1413ee : SUCCESS in 1h 44m 15s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/3665ce3dc234417ab555f8db25de1dfe : FAILURE in 25m 29s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/e9552cbd137344f5adc72271866c6e2b : SUCCESS in 45m 50s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/fba6e188e3be436c941705dacba24016 : SUCCESS in 35m 58s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/4183e132e1024e5bbdb01fc02e9a0f7a : FAILURE in 48m 34s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6167c22c87074cb5b97d9d615648b722 : SUCCESS in 4m 45s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/81dd26156a554678959b835b2977ae2c : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"3e697a80c7e8291b625fdb0a2cbb521e90fa5af1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 08:37:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d743efe9e4ff4d9893d69d9c321e6121\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c072597046c045da81ee1454c7a35f62 : SUCCESS in 10m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/be728dcf70a24ceb82ee69651bd41f15 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2321b3e4d15b4b4eb0c6093c3b811589 : SUCCESS in 12m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f80e7676b06d2741bbc69e263f642d5fea73a19c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-02 01:13:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ee14f2335c49014cee05130e251f54d7934b3a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 01:30:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b952347bd6c14cf5b5146efd62636519\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/beefb34ccd34455d87337d0cc1489b39 : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9c57a9c1df804304ae4f76ef1696580a : SUCCESS in 13m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/54d8f57443744219a30ef0cad2b1ab0b : SUCCESS in 15m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f6946257b978cecee234b587bddffa77a407522a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 03:08:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87f49ea332bc426b9613958bc50c3f26\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f467d68aa464204b0a37227764168bc : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2599e82552940ba81eed6c2ffcfa682 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5dfc7ffa17f64d0ea51a9e85a2a31767 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4d7e3eeab2c4047ac99e66c3af6b5f1 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ad24486f6704ba2bd5b9d642470d4e9 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3045dc68d47469989c7ef03f6427bc7 : SUCCESS in 5m 07s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/99271ca7182b4d5d917d3b14ebcf2607 : SUCCESS in 52m 27s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/a2dd0dcf1a594c45b7c9baf9862b84c0 : SUCCESS in 1h 54m 25s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/aaf0d8da8ba14d9eb1a41f8ac5f3f792 : SUCCESS in 1h 25m 12s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/e314efc3124340e9ba7558d97c174fe1 : SUCCESS in 1h 06m 20s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/05c48ab528d24d5986a8d973d328e2f4 : SUCCESS in 41m 58s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a8cfa644247243c580f405716770b9bc : FAILURE in 27m 15s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ac7ab360fb1f4d358d03ec916853a4f4 : SUCCESS in 1h 09m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/58c744cc471647368a1175744dcc1f24 : SUCCESS in 52m 53s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/944a88e08a324a2aaa914c7a19972260 : SUCCESS in 47m 10s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/273832a95b284eeb8f02d9f52ecef239 : FAILURE in 27m 29s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/117f00901daf40f981ba764f97377c85 : SUCCESS in 1h 03m 46s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/3088a6f3d25949809cc1824a98a4a8cd : SUCCESS in 1h 17m 26s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/475241e129824ff6abfa55889a62e834 : SUCCESS in 3m 58s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/c87e41b22be94804ab1e80a8dc87ec1c : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":2},{"id":"a5ed1ddfe97634847cbe3962e75089a3bac972e2","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-02 03:25:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c74d73acd2e2b34d0895f53c305c2d130759683b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 03:26:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1b5800ba220cee45a2ee032a4f393001629fab47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 04:48:41.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc8c2869ee0941c8aca5884a202b8877\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83b9add362a04e6caeadb64411ecf7d8 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da97ecc567cd4f958d5ae7c3d7c0e36c : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec4681659d53481b96255313722ee377 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eda35e2807f14bbe8cf25ed8f511e4e7 : SUCCESS in 4m 19s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/0589ee152d464e609fd5ac6d5a81f990 : SUCCESS in 26m 12s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/e76eacf79e484dc4a01d875ffa4a1ebd : FAILURE in 24m 51s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/368c2f498a544022a8a34f335de706e7 : SUCCESS in 1h 14m 02s","accounts_in_message":[],"_revision_number":2},{"id":"8facc57d4df784b71ee9fe096b332cb23d84380b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-02 06:00:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a512fe128c388d24f120e8d32717a091a2b85b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 06:14:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3463dea3f7e2470482b62fb31c342bec\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8dedc6c7de874a8f8c0bfe35228ffca1 : SUCCESS in 12m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/943950926c284610a9dc23dff68c5761 : SUCCESS in 12m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ce3c5c9612c04b4bba9d4810ab53e7d3 : SUCCESS in 11m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"201f4dc26375f6f8011d11ea7edf14e7fffaf1e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 07:54:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1834c93917547da9a0b659559bc58c9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca318cbe21ec484ebb0c35c6b34785d2 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23ab8a077c334702b7bc262d9c10b624 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3715e794c6a4823803ba470eac47788 : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79e8338193994dc4bc81c74b928ac6d0 : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07de05bf18154bbf8ab85b5382fb1729 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f088a0ff3fb54b1b909348bb6da743cc : SUCCESS in 4m 57s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/c16bcbda25de4fdeb32a64f1a2a045bb : SUCCESS in 47m 07s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/013c035e2f4e49fbb04e2f105fd29f56 : SUCCESS in 1h 53m 15s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/51ea5a3627e5434ab890a133a7cd14fb : FAILURE in 1h 02m 09s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/6a3a1fca982d4110b1e30822b9a5ccc6 : FAILURE in 1h 48m 15s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/9d016f5565a042e49b34a29504554811 : SUCCESS in 26m 31s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/8a3d9b740e454b89b21cd5c98a2b043e : SUCCESS in 1h 29m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/2882d95ea10f4ee99884bb33356bd2aa : FAILURE in 27m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/bf9ad64ad23b4e29acf9d9c1a9814624 : SUCCESS in 1h 32m 56s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/59b3b72e45e54848b60c9c154e56235e : SUCCESS in 28m 22s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/ccdae75d56de4a9a8cbc9f1bbb1c07d2 : SUCCESS in 1h 11m 42s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/1c85c259486245fb950f4a0e267475b5 : SUCCESS in 35m 37s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/fa8b9ba105f640d992f33574074776b1 : SUCCESS in 51m 30s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a9246d1f417b4abfa937d64c02fb2239 : SUCCESS in 5m 24s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/c7c249aa1f3e48428d74aea6d3196ec7 : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":2},{"id":"881055f99a3b8855cca74082fcda60b861e8646f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 07:54:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c929f9104925e50ca21e55b6d2e27766669a5618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 09:00:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cdd9756c91a48088948f11e752281ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c484c11886954741ab58196f5bc4eacd : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddf86666a5a648c2822ceeb206b5d9a0 : SUCCESS in 5m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/268f3bfdb10444cc953b7824a3b5cf3b : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9498a5b22ea34b108f18157ceee3072b : SUCCESS in 7m 09s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/85693da2cd60403faa4c0044e69c8a3c : SUCCESS in 29m 26s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/36288474cbeb43b8bb17f2c00bc7e824 : SUCCESS in 51m 13s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/8cd97ad8b023470eb0cc7ecfb75de0e1 : SUCCESS in 1h 01m 05s","accounts_in_message":[],"_revision_number":2},{"id":"2ef6c1ab4bbafb7cd2eb8fff07dedca5135c8e00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 09:00:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5ea93454130c3f63a02edbec6debf093e57bfca6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 09:01:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3d51a1e593f4e4faba29b387b1c6112\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/325b9c2b0de74d2aa1663a1bcd79025e : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f6cd2198f03b911cfc2fd81dec83abd577b8df88","revisions":{"daa1301a66404c4a8d074f7758c06e79627cc127":{"kind":"REWORK","_number":1,"created":"2025-06-24 15:10:53.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/17/953217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/17/953217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/953217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/953217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/953217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/17/953217/1"}}},"commit":{"parents":[{"commit":"e9f62185701da253a66d3d4c16731ad1f3121134","subject":"Add mysql-8.0 tests and remove mysql-5.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e9f62185701da253a66d3d4c16731ad1f3121134"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-24 15:08:51.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-24 15:10:51.000000000","tz":480},"subject":"Fix trove-imager-loader.service not started","message":"Fix trove-imager-loader.service not started\n\nThis service will pull database images after the system started.\n\nChange-Id: I510adc33116760ec5af54a9158b76c9a628ef981\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/daa1301a66404c4a8d074f7758c06e79627cc127"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/daa1301a66404c4a8d074f7758c06e79627cc127"}]},"branch":"refs/heads/master"},"f6cd2198f03b911cfc2fd81dec83abd577b8df88":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-01 02:57:51.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/17/953217/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/17/953217/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/953217/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/953217/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/953217/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/17/953217/2"}}},"commit":{"parents":[{"commit":"f8bece7cce0823c2a026d315fb3ee2af682a886c","subject":"Merge \"Add swift upload retry mechanism\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f8bece7cce0823c2a026d315fb3ee2af682a886c"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-24 15:08:51.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-07-01 02:57:51.000000000","tz":0},"subject":"Fix trove-imager-loader.service not started","message":"Fix trove-imager-loader.service not started\n\nThis service will pull database images after the system started.\n\nChange-Id: I510adc33116760ec5af54a9158b76c9a628ef981\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f6cd2198f03b911cfc2fd81dec83abd577b8df88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f6cd2198f03b911cfc2fd81dec83abd577b8df88"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
