)]}'
{"id":"openstack%2Ftrove~954015","triplet_id":"openstack%2Ftrove~master~I1046e5113b3d86284d4d06e1d046bff422695648","project":"openstack/trove","branch":"master","topic":"trove-push-mariadb11.4-11.8","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-07-04 00:57:23.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I1046e5113b3d86284d4d06e1d046bff422695648","subject":"Change default timeout for mariadb 11.4 \u0026 11.8","status":"MERGED","created":"2025-07-03 06:47:24.000000000","updated":"2025-07-06 12:42:25.000000000","submitted":"2025-07-04 01:45:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954015-trove-push-mariadb11.4-11.8","meta_rev_id":"4aae1966807de35a82c6192083826084523bf5a6","_number":954015,"virtual_id_number":954015,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-04 01:45:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"recommended":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-07-03 17:39:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-04 00:57:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2025-07-06 12:42:25.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-04 00:57:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-03 06:59:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-03 09:57:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-03 17:39:40.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-07-06 12:42:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"ed01176584dd981149b1adc501019c7ac959befe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-03 06:47:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee2f55a9f9fdc7dfadbdea9ef041267139087f19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 06:59:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc37941583c34ebc84e0a82b1d7ff6c4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/518b7a341e9e4379b9a56bc341143868 : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a9148fead824e0ba77b42cd8252bf5c : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ac5bb15098bf442598875abcf28b3ab8 : SUCCESS in 8m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fe4862d87b4b9a92cb7f7c54a7a72c659d602e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 09:57:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/96a4649c3e434fb39c67a0fba427208b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a266aab73c3e4b6c8938d6151301b5ec : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e744af7e676844839f4e65e3e28011ee : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5514271a390144b08970ad3c93acf48d : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fed2e5727f1b4eb691d8d8ed1329f156 : SUCCESS in 5m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56142f6dfd744de490150c0f68fddacd : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a74df2ecdc24a5ebbbbc1fba1428d9c : SUCCESS in 5m 23s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/b41476f16b9e4b51bf47a44633a30458 : FAILURE in 1h 23m 56s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/08eb9cc5aea24e619264a87bade97a30 : SUCCESS in 2h 56m 18s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/f662c6674a3942219556db52758e0e59 : SUCCESS in 2h 16m 13s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/5cd0242ea18a4733b7f4377b4f2d07e4 : FAILURE in 2h 10m 57s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/06b9d28bb71647428beac8ae99ac4228 : SUCCESS in 1h 15m 54s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/402eea5c3eee44c88bfd6fe8f2069bd0 : TIMED_OUT in 3h 03m 41s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/ba65908c10ba49ffb4737becb57d37c3 : SUCCESS in 1h 22m 14s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c999076fc78f4f5b891390157148b3e1 : SUCCESS in 2h 23m 04s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/bbfa71125824433c9eec471d7af94959 : SUCCESS in 35m 51s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/2e809e29532c48b4a830b04a73ccbc18 : SUCCESS in 2h 11m 05s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/6dd33b65e3f84b6996096a8321b76958 : SUCCESS in 1h 06m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/75802e7b357847b99a896c1a3c8267e7 : SUCCESS in 2h 36m 19s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8a890fadeb6343559268668574d57bc0 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7b7d418514705642c7a7e6de34bd9072c8e4783d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-03 14:01:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"659171e1ac7b3d7599491f06b4a75eb885d6ec63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 15:38:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8a7d72dee91447aa4a2524e3a90e06d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3dcc41c0abfb4cd8bdb26846e5f0e2a0 : SUCCESS in 10m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6505555e55d3409b88211676ca1cb1ac : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7c401d7b444042f1b498efd424a9829e : SUCCESS in 14m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84145ce3368a1969117f729f7b3fee6ff0077f9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 16:01:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/327760bb9b25419e82e53400a2150635\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/287befd1a1814337b23bccbb4c3c1e0b : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b30a72f7ab284203bfb38c824bf0e97a : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46e8b6ab93154dd98c88f62993ee4a04 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/12405ad8d6fc4b3aa92982581f3cd484 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd5ee54b1e244876acb087ab67fc757e : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/113d4b9015a14f36b5c72fbf9ea2b7d0 : SUCCESS in 3m 55s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/02c0a56176f549358ae9fb2646388339 : SUCCESS in 33m 09s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/0f552e34f9064d21926b3853975c39dc : SUCCESS in 1h 23m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/c5bf50e233324e40aab30611776e19a9 : SUCCESS in 1h 21m 38s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/b0af9a38f4ec48b49ea184149a06b9d6 : SUCCESS in 1h 59m 20s (non-voting)\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/0a0b615bf352409e9a22ae17fa61661d : SUCCESS in 40m 25s\n- trove-tempest-ubuntu-backup-mariadb10.4 https://zuul.opendev.org/t/openstack/build/c46c0eaddae04bf5b1c96fd5befc00f5 : SUCCESS in 1h 10m 19s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb10.4 https://zuul.opendev.org/t/openstack/build/a4af6799c018475082a8cd9db2171088 : FAILURE in 52m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb10.4 https://zuul.opendev.org/t/openstack/build/b2a0d26197dd4829941b1f3ac1defb58 : SUCCESS in 1h 32m 51s (non-voting)\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/4f9a3bc737bc420591c47af22e194eb6 : SUCCESS in 49m 59s\n- trove-tempest-ubuntu-backup-postgresql12 https://zuul.opendev.org/t/openstack/build/4c2c011abafe4d458db4800dc6cd8593 : FAILURE in 28m 17s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql12 https://zuul.opendev.org/t/openstack/build/c08a574e64884d068b7975d9c427f518 : SUCCESS in 1h 17m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql12 https://zuul.opendev.org/t/openstack/build/c7566b9ac4dc487b836ee9f55b01ef15 : SUCCESS in 47m 34s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4416c62266134bf180bb1c8abbfdaffb : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"aa37b5c66d635499661bac7827a2a82248cda0d6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-07-03 17:39:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b461026078e9dfdf0f019155843030ea3f399386","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-04 00:57:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"25bda2c47d1cf4be22d6e3c93cc19599b6bf40f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 00:57:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a92070a0554d0f757390d6510ed5f64d636de3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 01:45:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73366c7c147441129ab5d4854b0afcaf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2ba38ff22a24a5a95ad42890b171701 : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17b0263f7f5e42f3b4aeb4daba78f65e : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/748b45072c0e460991015895ab14e478 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/514dd76e432444b9b218389072f0b46a : SUCCESS in 5m 01s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/b105db7f0af742368b5567c7a989878e : SUCCESS in 46m 46s\n- trove-tempest-ubuntu-base-mariadb10.4 https://zuul.opendev.org/t/openstack/build/5895617b8e7647da8c15fad55822aef5 : SUCCESS in 36m 15s\n- trove-tempest-ubuntu-base-postgresql12 https://zuul.opendev.org/t/openstack/build/d47b654e71404c2d9dd598be9ccc813d : SUCCESS in 44m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0c5e7b6835f8031ac3f464598db60ea96faa9d80","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 01:45:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eef73e0b3ad3027095c735a6c2077e6e6f99a5a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 01:47:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4a93fd3d4d74f4e804db192fc3cd05a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0601a6888c846e0b4c56f16dae9b856 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4aae1966807de35a82c6192083826084523bf5a6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-06 12:42:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e92f23471a93a55fe6b53113b636eb0613f5fd26","revisions":{"e92f23471a93a55fe6b53113b636eb0613f5fd26":{"kind":"REWORK","_number":1,"created":"2025-07-03 06:47:24.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/15/954015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/954015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/954015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/954015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/954015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/954015/1"}}},"commit":{"parents":[{"commit":"58bc2aee27a886b7367c8ce97cb08528248ac17b","subject":"Merge \"Follow up: Ibd8dbb8ad8fe895ba9d14e00038ed08e99fbfe0c\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/58bc2aee27a886b7367c8ce97cb08528248ac17b"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-07-03 06:46:03.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-07-03 06:47:18.000000000","tz":480},"subject":"Change default timeout for mariadb 11.4 \u0026 11.8","message":"Change default timeout for mariadb 11.4 \u0026 11.8\n\nThis change sets default timeout from 30s to 120s and push to quay.io\n\nChange-Id: I1046e5113b3d86284d4d06e1d046bff422695648\nSigned-off-by: wu.chunyang \u003cwchy1001@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e92f23471a93a55fe6b53113b636eb0613f5fd26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e92f23471a93a55fe6b53113b636eb0613f5fd26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
