)]}'
{"id":"openstack%2Ftrove~954181","triplet_id":"openstack%2Ftrove~master~I110007516323e3193414e24ed6799e40b7870bd8","project":"openstack/trove","branch":"master","topic":"postgresql-17","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-07-11 02:26:24.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-07-11 02:26:24.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-07-10 06:10:38.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-07-11 02:26:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I110007516323e3193414e24ed6799e40b7870bd8","subject":"Add support for PostgreSQL 16 \u0026 17","status":"MERGED","created":"2025-07-05 12:52:34.000000000","updated":"2025-07-11 02:28:36.000000000","submitted":"2025-07-11 02:26:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"954181-postgresql-17","meta_rev_id":"1507568c870a54826472cfa98d8d83ff41b21767","_number":954181,"virtual_id_number":954181,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-11 02:26:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-11 01:15:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-11 01:15:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-05 13:12:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-05 14:27:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-06 12:30:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-07-06 12:30:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-07-06 12:31:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"f8fa2181a49b72135f850c30d6172fb53e85314b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-05 12:52:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e8afa9f656bdd1500a563490b9f82411d22a81f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-05 13:12:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf9d2382ecd048acbbf5ca4087e33664\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bedf824197354c2b8065cbe9e51c915d : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3daa6164ab684cf790c543a48bad49d9 : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/71fd639204124059b1b313969ff88228 : SUCCESS in 17m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0fcd8fec752ad4c74f4513b87059c53881e6c2e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-05 14:27:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e99d6fe57bb846ba966e6157c11f699d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/505daffb6cf44864842d2699776c0973 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37914d7f1eaf451296a3dd7594f8ef7e : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e58b192eb0f7495abe026b929c820a52 : SUCCESS in 5m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/14e0f21bc5134f14b159bde7a954c84e : SUCCESS in 5m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ea8745c208e4099a45a589de628fe2c : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e373cb6213e42c8a57b1560850c0a61 : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62c7870c45b64e2991435664df9e70af : SUCCESS in 5m 43s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/9a2b1bf2a77f47a4858dd5d179b1623c : SUCCESS in 50m 20s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/4e11070c138242fd8df2ecce23a2cbc2 : SUCCESS in 1h 33m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/f659838fea314705bead440a0a28e8b0 : SUCCESS in 1h 16m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/081791b44b8642aba464d03c1c88a2e8 : SUCCESS in 1h 08m 35s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/bfff4bb5231d41c38e44328fd2826de8 : SUCCESS in 36m 26s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/0a9ef1e62a9d4b419cbd06a6bc79217b : SUCCESS in 44m 33s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/814644ce49324760a16d6f8e2f1b49be : SUCCESS in 44m 52s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/d3c94f0050ad4d8ba8e4f9617118d3f3 : SUCCESS in 1h 27m 22s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/2f7f375b7ffc48e49be56b3883388cb0 : SUCCESS in 48m 18s (non-voting)\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/8271c9ffef264338984dd0f544b48a68 : SUCCESS in 1h 21m 37s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/42838bb64bb64f2999a1dcfac912ce4f : SUCCESS in 1h 06m 24s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/cd2a7a845cfe40629fa5eb1a5d8f01d9 : SUCCESS in 1h 22m 44s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/90d28652b4ac4ecba2a64f3e2fad70c2 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"62e5b0685ec17b758c4924654111ed2b67046bbd","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-06 12:30:54.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5b0b5af68d2763a3ad57324ae031181791e14f9b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-06 12:40:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"410923e039db76c707d68d856cd172180fca7be3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-07-08 09:23:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df6f7d843c186f26e96e3c36ff357cb10534176f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-07-08 09:24:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b4535b99a4dcf2e70679c46643bfe92af8a86ff7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-07-08 09:25:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bcfe373826670cea8fc7d67dae51ad9460ea186d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-08 10:34:11.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"55d3c0af3f511eb88e2da76a1cbbbe725b6cc050","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-08 10:36:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68ab7b4e8aa0251872c940446618304d16ced0da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-08 10:36:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1cf8538db6d3649d28987781b47d3a0c7a396537","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-08 10:38:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"53852e395b41a1cce32b556dc2bf9ad0348dfd1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 10:49:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f28e17ea0e9e4d92bcb511e918e43d93\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/265b60e868ae4a47bb4e743a54d8836f : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/51a9972f09fe493eb758471a0bb258f6 : SUCCESS in 10m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e262aa48788f4ed199a2b32c4a13d87b : SUCCESS in 9m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"02abb701bd001db41fdee47342a5848180878b8e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-07-08 10:57:49.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"adcb3894d2d785547e85ec389ebee102b4f1c483","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-08 11:25:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5e1069a35aef06029ece3e0207d2100e3116ded2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 12:08:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b61642e98a014109aa3fbd9506946e50\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cda5d1e3de14752ac04f34d0489cf0f : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dbad20cdea44043aaf0732ee68da73e : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fc2dc381d0445a4b105275b0b74c3ac : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f20412629e3a4d55a9e732af105f37a2 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59aa9dd259204d208aa39891dba8240a : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc3b208725e498cb4d40ea78d2b1745 : SUCCESS in 2m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a687888e2d154bdf81b3e4686b45622e : SUCCESS in 5m 46s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/76a9510f1d054896b412f8c7fbbc2623 : SUCCESS in 28m 04s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/debbb26531d74a5db97973a08ef93a5d : FAILURE in 29m 42s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/2a7365031b494702aaab7473a548b776 : SUCCESS in 42m 01s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/35f77e893f784f33bf9142944cc1dc9d : SUCCESS in 1h 05m 02s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/99bb25b4979b4a83bb3dd4d789833ccf : SUCCESS in 40m 32s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/42fab30195c14c57a644cba79a9094f3 : FAILURE in 27m 06s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/d337dace150d4914999319ccfc4b891c : SUCCESS in 1h 17m 57s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/60c379e2b4ef49b2a1f938df9c08726c : SUCCESS in 1h 30m 51s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/a89cf22e0ec84e0aa4c9fc418f11c8ee : SUCCESS in 42m 19s (non-voting)\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/b5a8d2dd2d634d768e8109a5927deb21 : SUCCESS in 1h 14m 01s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/324ae77313fe453f932d66629022b7c1 : SUCCESS in 59m 19s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/54574ab310094143b5ddef6541f30a27 : SUCCESS in 1h 13m 53s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e8a6801311054ceab13fd6ac6c18be96 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":2},{"id":"88ee9f0ef8b202be71672288b6004ddf4b7edbb1","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-09 04:27:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"80ee65aa7a0d6479e6e70dbfe8eff699241f465e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-09 04:33:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"583f7eda782c83f41d8151ed9fb1f21157088ef8","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-09 04:34:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5f2e414521168dde4b8b21d8e8b62a001a14047b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-09 04:34:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1f2c7d75c8a5dffffd0b2b3719ef6d4d3000e4db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 04:50:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fc20679fa7c4debb07d9fd5fb972dc5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/32726223023145f8a7b4f620e323ba63 : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/759c804d21a848b5a59898d7004572a3 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/56dc72baefbc439c88f3bce86a71ae94 : SUCCESS in 13m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"05817cef47b7950109f38829e64e5f161bbc46c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 06:28:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/955aa473113f409f9b15dfed55325d46\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03b0b870743d40a798bc6e4adb9e210a : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2cac892848146d19c530bdbfc59be08 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3435ab75c87f408cb67d7d221f19f77f : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b5037bf1f2c4efeb66f2068966dc8d6 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8dcc3d23dc4440ce94edb953253c952b : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae2be10946394e39b5691fb8a5271585 : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71da8f74bb1147899de64e7417e60307 : SUCCESS in 4m 26s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/950a5d92c4d04041bb3fac06e17861bd : SUCCESS in 50m 56s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/adbe8b6923b24c6c9878dfe94bc74abb : SUCCESS in 1h 27m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/75e5a8169c0d47c9bac7ab40e5727a6b : SUCCESS in 1h 12m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/4d91c40266b54ebc8d64ad380ea0d709 : SUCCESS in 1h 52m 01s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1c2e2b6f35c34f0ab579b857be2f845f : SUCCESS in 36m 09s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/83a35ff77bc84b85b1e1c5135e61b2b3 : SUCCESS in 1h 18m 53s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/187ab50aca0f4343b5e18fb36d3b210d : SUCCESS in 1h 06m 44s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b46a547b03b24410889f0dcb24f4e0ce : SUCCESS in 1h 13m 39s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/f7abdb2ebbbd4b94ad83a055c4d8e41c : SUCCESS in 42m 06s (non-voting)\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/5c008c8406984320b28b0a75dd32c819 : SUCCESS in 1h 18m 23s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/3c087f3f608e43f4a6e1efe4d6e68f59 : SUCCESS in 1h 00m 55s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/c782c829d5ed47358f624d87cf441c9f : SUCCESS in 1h 20m 27s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0d6fd959b8f848f0ad0d6d72809a6d45 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":4},{"id":"aedc875dfeb4c942e2e4341e9ed33b250a923fb9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-07-09 07:31:27.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5413395715b8238554a8f56abefef7d6207db28d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-10 06:10:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f189d94ef1f12e6ba9001ee0b1a00e5c24b99df2","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-11 01:15:46.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"776327ecd8086a10ccc9d8a70a14f3ba489cba13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 01:16:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"84f8d967b778e3aa069789898828b8ba30f1d138","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 02:26:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd03439e755c42cbb088ab3305915a20\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d939ace16e40a89c10fda9506acfa8 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1baf22a946b748009f4d94aa5a890d60 : SUCCESS in 5m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6fd88df5d78a45e68b1b9f2ca827e88b : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec3d45e89c34d22a2fd7c4d972f352d : SUCCESS in 2m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0db65105b315469a92041e4ac99f2f07 : SUCCESS in 5m 24s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/387de8fc74f746db97a6a3a8228d2c92 : SUCCESS in 33m 11s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/0fd0b7e0273f4c9299c8e849cc0466be : SUCCESS in 39m 55s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/7175e57d4d6e4229ab8fb80733642401 : SUCCESS in 1h 05m 04s","accounts_in_message":[],"_revision_number":4},{"id":"64ce351aa26654b171452eba414f6a9a106e8161","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 02:26:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"1507568c870a54826472cfa98d8d83ff41b21767","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 02:28:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9782beeec7694bb3bdd9ea5b10cc5e93\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1567be6579c44c98a5ee6e5e3110a00 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/634763601b4e4841b06e03248757e7cd : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bacf37e68926c46ed33e9a17afaa0d41923b441d","revisions":{"110369c9e88bcd69c06798bf274599054a9e3450":{"kind":"REWORK","_number":1,"created":"2025-07-05 12:52:34.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/81/954181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/954181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/954181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/954181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/954181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/954181/1"}}},"commit":{"parents":[{"commit":"9e46762ae28566e90353ebe42984e9c83e415623","subject":"Add support for mariadb 11.4 \u0026 11.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9e46762ae28566e90353ebe42984e9c83e415623"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-13 06:36:03.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-07-05 12:52:23.000000000","tz":480},"subject":"Add support for PostgreSQL 17","message":"Add support for PostgreSQL 17\n\nThis commit adds healthcheck for postgresql to check health status.\n\nChange-Id: I110007516323e3193414e24ed6799e40b7870bd8\nSigned-off-by: wu.chunyang \u003cwchy1001@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/110369c9e88bcd69c06798bf274599054a9e3450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/110369c9e88bcd69c06798bf274599054a9e3450"}]},"branch":"refs/heads/master"},"77a94a3963af9fabb7fcde25d65056da3f364715":{"kind":"REWORK","_number":2,"created":"2025-07-08 10:36:43.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/81/954181/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/954181/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/954181/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/954181/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/954181/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/954181/2"}}},"commit":{"parents":[{"commit":"9e46762ae28566e90353ebe42984e9c83e415623","subject":"Add support for mariadb 11.4 \u0026 11.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9e46762ae28566e90353ebe42984e9c83e415623"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-13 06:36:03.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-07-08 10:36:20.000000000","tz":480},"subject":"Add support for PostgreSQL 16 \u0026 17","message":"Add support for PostgreSQL 16 \u0026 17\n\nThis commit adds healthcheck for postgresql to check health status.\n\nChange-Id: I110007516323e3193414e24ed6799e40b7870bd8\nSigned-off-by: wu.chunyang \u003cwchy1001@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77a94a3963af9fabb7fcde25d65056da3f364715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77a94a3963af9fabb7fcde25d65056da3f364715"}]},"branch":"refs/heads/master"},"dc702c850ed0d9cd78ecb3c11ac246879a771174":{"kind":"REWORK","_number":3,"created":"2025-07-09 04:33:03.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/81/954181/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/954181/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/954181/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/954181/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/954181/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/954181/3"}}},"commit":{"parents":[{"commit":"9e46762ae28566e90353ebe42984e9c83e415623","subject":"Add support for mariadb 11.4 \u0026 11.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9e46762ae28566e90353ebe42984e9c83e415623"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-13 06:36:03.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-07-09 04:32:56.000000000","tz":480},"subject":"Add support for PostgreSQL 16 \u0026 17","message":"Add support for PostgreSQL 16 \u0026 17\n\nThis commit adds healthcheck for postgresql to check health status.\n\nChange-Id: I110007516323e3193414e24ed6799e40b7870bd8\nSigned-off-by: wu.chunyang \u003cwchy1001@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dc702c850ed0d9cd78ecb3c11ac246879a771174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dc702c850ed0d9cd78ecb3c11ac246879a771174"}]},"branch":"refs/heads/master"},"bacf37e68926c46ed33e9a17afaa0d41923b441d":{"kind":"REWORK","_number":4,"created":"2025-07-09 04:34:58.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/81/954181/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/81/954181/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/954181/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/954181/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/81/954181/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/81/954181/4"}}},"commit":{"parents":[{"commit":"9e46762ae28566e90353ebe42984e9c83e415623","subject":"Add support for mariadb 11.4 \u0026 11.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9e46762ae28566e90353ebe42984e9c83e415623"}]}],"author":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-06-13 06:36:03.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2025-07-09 04:34:52.000000000","tz":480},"subject":"Add support for PostgreSQL 16 \u0026 17","message":"Add support for PostgreSQL 16 \u0026 17\n\nThis commit adds healthcheck for postgresql to check health status.\n\nChange-Id: I110007516323e3193414e24ed6799e40b7870bd8\nSigned-off-by: wu.chunyang \u003cwchy1001@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bacf37e68926c46ed33e9a17afaa0d41923b441d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bacf37e68926c46ed33e9a17afaa0d41923b441d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
