)]}'
{"id":"openstack%2Ftrove~955248","triplet_id":"openstack%2Ftrove~master~I78f4c93d4d196e5d870a39256cb90dcbec41da45","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-07-22 11:49:31.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-07-22 11:57:01.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I78f4c93d4d196e5d870a39256cb90dcbec41da45","subject":"Replaced usage outdate egrep to grep","status":"MERGED","created":"2025-07-16 21:59:34.000000000","updated":"2025-07-22 12:40:27.000000000","submitted":"2025-07-22 12:38:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955248","meta_rev_id":"0b8f5bcc6a72642cb91ba45741e1693ccadbeb1b","_number":955248,"virtual_id_number":955248,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-22 12:38:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-22 11:49:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-22 11:49:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-16 22:45:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-16 23:55:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-17 08:41:29.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"a6f9d0ed66bd997a00e4c3c015c9b1f780063f66","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-16 21:59:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be080b48eb239801eccdda9cec003878fb42b825","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-16 22:02:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8e93965773e16564567b4c911c8f9dca42ac90c5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-16 22:03:50.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e14317f1b6cdb62358a64730d748fd95266ab8cc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-16 22:03:55.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"95ac461fa7088eaafbf45144e0f50b99c8d1f690","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 22:45:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b6d2310dbf24696bdd69881ff2eef3c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5fa82f32f0854aa3ad54340988a97f0c : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/454892cbac6b4ad080e4d2f4de67262c : SUCCESS in 12m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cf2ba47e05314e83a6e3de3b34386e72 : SUCCESS in 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9c5979437b10f73e434b3ca7c6b68bbed80b1a04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 23:55:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d4b4588cc7148c092ae5d6cfdd0120b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f8f6041883e4de98edbc550f5fd7621 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b3ebb85e0e44549869dfcca51e2a361 : SUCCESS in 7m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0063cbf97cb04cdaae28da71b49a847e : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fc7b57cabe5c42869997fce76fa9c693 : SUCCESS in 6m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a750275187114c638301fe160290bcc3 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e00eefd302154196b7fab6cd8bc9028b : SUCCESS in 2m 43s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/9bd49100372245bda47dba36a4b060da : SUCCESS in 46m 32s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/5531533fff2d40e7839690f9691c0e4f : SUCCESS in 1h 28m 30s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/975b3f835e13435a9a00e1622c311ab7 : SUCCESS in 1h 12m 26s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/3d70d96fb50a45c28320ca556c6e2dab : SUCCESS in 1h 47m 57s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/4a69cc6ea1a843138b8cd6dced36c48e : SUCCESS in 37m 14s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/9befd99925f1474a8ff0b638d84436b5 : SUCCESS in 1h 17m 38s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/bddd7960ad0344658dc95563eafde764 : SUCCESS in 1h 02m 02s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a84d2458f2ba41b2b3d1161b85816e4e : SUCCESS in 1h 18m 17s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/b6b807d43a264ca4b0323f8816280d58 : SUCCESS in 41m 02s (non-voting)\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/cc8ad8e4364a49a8a745660a1cabc5d3 : SUCCESS in 1h 11m 46s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/e5c214264a4f4ed5b462855705bd6727 : SUCCESS in 1h 02m 04s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/24adcb43ffab41578a1ece3cd486afb4 : SUCCESS in 1h 15m 41s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d8e91d1c7d4d422799f85e9691b6a2d7 : SUCCESS in 6m 55s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/d860bb472c054e15b6f807188ce55b46 : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":3},{"id":"4a4f4306048d469350f2d9b548761b9df4ad91b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-18 12:59:06.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"35540528a300e10803c087faf199550893c5e9d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 13:34:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b91948fe75a410e9540c8653b02bfed\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/17b268a570f24aa486630fd976061d4a : SUCCESS in 13m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b62d2af08419496d98b55921ffd6fd74 : SUCCESS in 11m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8721c83389e246b6ba891360545d94eb : SUCCESS in 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e778b7b59f5e66ea2286c0dfa4eb7ed4243e2017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 14:48:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9b1fcdd615b4ae78844271c88dea7e0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a0b4ed72d2c44978e5a305bcedbc044 : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff03f042177342638d8eadfe7720fe27 : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/193847e549534e79b4226c925571850d : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/562d0e0b5d6d4127817955240a40a5cf : SUCCESS in 6m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee8759165cd54af1b8dc051598082bc2 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49d2bde57206402a90ebd4b5d5bd5131 : SUCCESS in 5m 10s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/043113fc547e4457be5d32a3c38d9841 : SUCCESS in 47m 33s\n- trove-tempest-ubuntu-backup-mysql8.0 https://zuul.opendev.org/t/openstack/build/9135676fb5c34f6985d5c9e333009d9e : SUCCESS in 1h 32m 10s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.0 https://zuul.opendev.org/t/openstack/build/f7b577ae68b047a9880d4c78957f4a95 : SUCCESS in 1h 15m 52s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.0 https://zuul.opendev.org/t/openstack/build/db5020c0c5f64650bede5715bcd2ecdd : SUCCESS in 1h 48m 13s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/ca71991b5e9a4a1eb49ae3befc11e403 : SUCCESS in 41m 00s (non-voting)\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/52e89d1376e2433b8a3fa286a6efbc09 : SUCCESS in 1h 18m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/007c5d2aa35c423787812f162c48f162 : SUCCESS in 1h 11m 51s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/e2a3ccff269e4d01be3695a1373919a7 : SUCCESS in 1h 21m 34s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/fd49a437ec944e85aa7b147b466b4e31 : SUCCESS in 49m 31s (non-voting)\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/87e1308006434545a1aa5997a9b174a9 : SUCCESS in 1h 16m 40s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/14b39b4e47ec43d89b86f42faa0dc14d : SUCCESS in 1h 07m 10s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/8ce6718d91d048608575501d9c6b96d7 : SUCCESS in 1h 18m 47s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f5bfc2b0cbd44fd58ab4481b0b1a48ca : SUCCESS in 4m 59s\n- trove-ubuntu-guest-image-build https://zuul.opendev.org/t/openstack/build/bf72ac4ec44d423abfafb056f5e05eb1 : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":4},{"id":"76c501ac5f0ea060b28ce78e8f70c8a5acced276","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-07-22 11:49:31.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"68c48a91ddb11f8db1870a5f29906506c2824086","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 11:49:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"da40ba851ad10604d86ea593bb26c396f4cecc7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 12:38:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecbd542653dd488093aa453edff46a0a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf4825aa71d84c23adbfa9f2c59073d1 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5dbdd1f650724c558a9bc99492dfd003 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c1fe94f35834df8bd398c083a9ff7d7 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4760bc8fb0994ab791d6018d1aa0cf94 : SUCCESS in 2m 44s\n- trove-tempest-ubuntu-base-mysql8.0 https://zuul.opendev.org/t/openstack/build/87d8419c4d36461288af0640c4a10aa1 : SUCCESS in 48m 27s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/41bbfee565524b7591203e39ed0ba3d1 : SUCCESS in 40m 35s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/049acc6aec2841f1bfc95c5639b6eceb : SUCCESS in 43m 22s","accounts_in_message":[],"_revision_number":4},{"id":"6b81bdc4683f524cdae3c525075fbbb3828ff055","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 12:38:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"0b8f5bcc6a72642cb91ba45741e1693ccadbeb1b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 12:40:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2ce9d9e354b452d93ca11b27828d78c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8977ac037414c2dae9a1756b3b4b70f : SUCCESS in 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5937eb6cb546d49b4ccb67aed670ab7a254eda64","revisions":{"586012b8dc465cca989f6bade420c73c8d1477a5":{"kind":"REWORK","_number":1,"created":"2025-07-16 21:59:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/48/955248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/955248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/955248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/955248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/955248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/955248/1"}}},"commit":{"parents":[{"commit":"8420adb0a79fe1f65759d4fbea3fc37cf34c45ac","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8420adb0a79fe1f65759d4fbea3fc37cf34c45ac"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-16 21:59:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-16 21:59:34.000000000","tz":0},"subject":"Replaced usage outdate egrep to grep","message":"Replaced usage outdate egrep to grep\n\nChange-Id: I78f4c93d4d196e5d870a39256cb90dcbec41da45\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/586012b8dc465cca989f6bade420c73c8d1477a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/586012b8dc465cca989f6bade420c73c8d1477a5"}]},"branch":"refs/heads/master"},"0ec42a984a7c8e573712b72c3b44e68801f50e2e":{"kind":"REWORK","_number":2,"created":"2025-07-16 22:02:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/48/955248/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/955248/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/955248/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/955248/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/955248/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/955248/2"}}},"commit":{"parents":[{"commit":"8420adb0a79fe1f65759d4fbea3fc37cf34c45ac","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8420adb0a79fe1f65759d4fbea3fc37cf34c45ac"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-16 21:59:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-16 22:02:45.000000000","tz":0},"subject":"Replaced usage outdate egrep to grep","message":"Replaced usage outdate egrep to grep\n\nChange-Id: I78f4c93d4d196e5d870a39256cb90dcbec41da45\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0ec42a984a7c8e573712b72c3b44e68801f50e2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0ec42a984a7c8e573712b72c3b44e68801f50e2e"}]},"branch":"refs/heads/master"},"04ce1ac2fe697c6e23f659cb5eb468fd57ef1ad7":{"kind":"REWORK","_number":3,"created":"2025-07-16 22:03:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/48/955248/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/955248/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/955248/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/955248/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/955248/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/955248/3"}}},"commit":{"parents":[{"commit":"8420adb0a79fe1f65759d4fbea3fc37cf34c45ac","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8420adb0a79fe1f65759d4fbea3fc37cf34c45ac"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-16 21:59:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-16 22:03:50.000000000","tz":0},"subject":"Replaced usage outdate egrep to grep","message":"Replaced usage outdate egrep to grep\n\nChange-Id: I78f4c93d4d196e5d870a39256cb90dcbec41da45\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/04ce1ac2fe697c6e23f659cb5eb468fd57ef1ad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/04ce1ac2fe697c6e23f659cb5eb468fd57ef1ad7"}]},"branch":"refs/heads/master"},"5937eb6cb546d49b4ccb67aed670ab7a254eda64":{"kind":"REWORK","_number":4,"created":"2025-07-18 12:59:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/48/955248/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/48/955248/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/955248/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/955248/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/48/955248/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/48/955248/4"}}},"commit":{"parents":[{"commit":"8420adb0a79fe1f65759d4fbea3fc37cf34c45ac","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8420adb0a79fe1f65759d4fbea3fc37cf34c45ac"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-16 21:59:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-18 12:59:05.000000000","tz":0},"subject":"Replaced usage outdate egrep to grep","message":"Replaced usage outdate egrep to grep\n\nChange-Id: I78f4c93d4d196e5d870a39256cb90dcbec41da45\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5937eb6cb546d49b4ccb67aed670ab7a254eda64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5937eb6cb546d49b4ccb67aed670ab7a254eda64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
