)]}'
{"id":"openstack%2Ftrove~962552","triplet_id":"openstack%2Ftrove~master~I909c628b96173a5633572909524926a24bc1596d","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-12-05 09:27:47.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-12-05 08:39:03.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-12-05 09:27:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I909c628b96173a5633572909524926a24bc1596d","subject":"Push datastore image to quay.io to support tests","status":"MERGED","created":"2025-09-29 21:55:18.000000000","updated":"2025-12-05 09:28:35.000000000","submitted":"2025-12-05 09:27:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962552","meta_rev_id":"f40985fded0dbad235a5ed4c856d6887e48a418f","_number":962552,"virtual_id_number":962552,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-05 09:27:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-05 08:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-05 08:39:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-29 22:19:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-29 23:43:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-09 18:10:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-12-02 07:28:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"06ce12e78a48b549192ea1d452b45eca17a31a9f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-29 21:55:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12fb39b32b54fa04eb1505f16ab12515f5681a60","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-29 22:04:44.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f540b602f1101e0c5a8521a78404e337eaf87def","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-29 22:05:58.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"6aa689244adbf0471c36653ccef7d6e315777f4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 22:19:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b61f4fb21fe437684976c2a5e5503f8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f0fd9acb3a444a78b3a93509e3c8669b : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/78eb358dc9d94df9b104c0b1ac2bae6a : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0b4b6ff02c364b4cb0d18f8470665c82 : SUCCESS in 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0e3e531b27501745be356dc55441f4fb26a22fda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 23:43:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69617c1f0b1d422c8566a22a92ee3e1e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed94f3171bcb4a9bba91d74fe8333c02 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fd8a6e1ba1d4281b680abed1d534d36 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60ff623083e7491f804e1612bd163e94 : SUCCESS in 7m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3bcf1cdf252c4b3ab496431d05fba4f3 : SUCCESS in 7m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a62ccc43a19e427bbd3b315df1fc9b41 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3260cdd75144df8807bf0766f3590bf : SUCCESS in 4m 42s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/e059f5ade7544f81977d9c0dda4738e9 : SUCCESS in 50m 05s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/4e411f09fa664d809f873550ee42161e : SUCCESS in 1h 29m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/580b6f184ada4d4cbccad98fbf0d4528 : SUCCESS in 1h 08m 53s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/62434eef5fa94b5ea902b64c051dc086 : SUCCESS in 1h 36m 34s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/d05e92e82a7941eda48d972bba8ea0b1 : SUCCESS in 37m 53s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/26de96b5d39a4a82a72665bc56d03b39 : SUCCESS in 1h 16m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/5365af8356fb47b8be652e88c2679aa9 : SUCCESS in 1h 03m 40s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/7bf42509e6e24d8c87e301b8f451b9ab : SUCCESS in 1h 20m 08s (non-voting)\n- trove-tempest-ubuntu-base-postgresql18 https://zuul.opendev.org/t/openstack/build/4e514828f1db4f35bae3f1d709e5a3c3 : FAILURE in 38m 23s\n- trove-tempest-ubuntu-backup-postgresql18 https://zuul.opendev.org/t/openstack/build/c96137aea8774410825f46036fbead83 : FAILURE in 49m 11s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql18 https://zuul.opendev.org/t/openstack/build/b5cac2789de343f78c0dec5856aae372 : FAILURE in 49m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql18 https://zuul.opendev.org/t/openstack/build/60e95b8fc3fb4cf492cb43c30c9d8a4d : FAILURE in 41m 12s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ca8afd14ad8d4926808c84050f0b1986 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":3},{"id":"50c8d03f0738b7f6dcbf93b3e666cab3f93201ff","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-29 23:57:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3ca502c4609dfdff04e853468fb23cda63b72e27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 00:13:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9b84a3bbde94825bc3891bcaedc45ea\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/80c96c78aef94f99a5f55ca7cca40250 : SUCCESS in 13m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a11fbe18a6294a27be6a3e9703a249d4 : SUCCESS in 12m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ea91a86bf97243d783183c957773419e : SUCCESS in 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8424c794bd56fc9d57238be4231e948640528f19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 01:41:11.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/393a02e452494ed59011ce9aed732a7e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ebd7a59abec4ea6bd45c096dc391e42 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/292d7e105d944f6d89b00dd54de050a0 : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/963b4251d0a74ad9a8f618570f530971 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35db420ab3094712bbdb4727c75a6835 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9885a6906232464e93de75e945d8f640 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a03828f8ff96459f81fb1868ed70d532 : SUCCESS in 6m 33s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/e6866f426b7047b59153dfdcfe59db25 : SUCCESS in 46m 00s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/376baf8686514d7b83d0ba53bca66f0a : SUCCESS in 1h 24m 01s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/5e80f0f59df2415396ba2091f442bc74 : SUCCESS in 1h 17m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/c153778d8f98406398ec3711ecdfd206 : SUCCESS in 1h 42m 38s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/9f43296482724df8b3dc126dd156e5b0 : SUCCESS in 40m 17s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/d64fc8bb0fe8427e8d7a1d8cbce067d7 : SUCCESS in 1h 24m 51s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/ddc4c877081d44458fe1263935cfccf8 : SUCCESS in 1h 09m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/bb0160f50f5c463a88249ac07564d2d2 : SUCCESS in 1h 19m 48s (non-voting)\n- trove-tempest-ubuntu-base-postgresql18 https://zuul.opendev.org/t/openstack/build/2eded04174f24beb94dcb58cbc74406a : FAILURE in 38m 29s\n- trove-tempest-ubuntu-backup-postgresql18 https://zuul.opendev.org/t/openstack/build/b6b566809a6f4e80a3e1dc8b62d94e6c : FAILURE in 44m 36s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql18 https://zuul.opendev.org/t/openstack/build/22fcc97372044c749de4f6deeeb6bccc : FAILURE in 46m 30s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql18 https://zuul.opendev.org/t/openstack/build/96d728f20a7c47a7b0d48ad2827f7cf1 : FAILURE in 48m 03s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/13b29c412bd54e818b32cfddf325aab8 : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":3},{"id":"d47fcdc8d8bcc58b9afd3a44f43d257d0c7f5a8c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-10-08 19:46:27.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"990c2f7d13a029473e652ba814ff80af7672e1ab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-10-08 20:32:15.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"f082d0a3a8dbdc0c23a6c38a9a0a62b6b7ab5119","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 21:01:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49db8dbdd42048438f615471e6cc137f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/820183d3808745c3956d6179b71d56d9 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9fb8aab4f113499aa37d93f1f4a79675 : SUCCESS in 14m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c3c0a97a16754f808a2d53a63a69596a : SUCCESS in 10m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d44a3e21112940dacfafe73f86d5eb2b3179adae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 22:19:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d70d0d169f34733aba27af89fb09e34\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e23bd5c1a39e4e578f69ba1f4f8e09d7 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/043a0a10777840438da491b6ad7cd145 : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1efa86c4267441e8131e4082cdc801f : SUCCESS in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed959f31e49e4d4bad491d0126a0ce9b : SUCCESS in 6m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b02b3ca9641e4baf8f96cb920c07491e : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22db4e10ba0e4d369d00fa5e3cfaad66 : SUCCESS in 6m 26s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/4bb733e7d1514315b0d3710ce9de2fb2 : SUCCESS in 43m 37s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/1654fdd88cda4c7f94923785b740c48d : SUCCESS in 1h 36m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/9cae056bf3d840cb85de4e5a3353e3ce : SUCCESS in 1h 20m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/8fafb06fcdc544d7adda4ed1f4251674 : SUCCESS in 1h 46m 26s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/fee9ab66fa224ea78b66edea79dece3b : SUCCESS in 46m 54s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/7e9c2017df51462cbb73381bc7c074ce : SUCCESS in 1h 17m 08s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a2df74df21784aafbc08cd0496b2c806 : SUCCESS in 1h 17m 21s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/7b0c0403525e44e899c7417c8a950b76 : SUCCESS in 1h 16m 09s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/e7f2f56a296e4a6587679e261d7fbc6c : SUCCESS in 43m 54s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/b20faceabaf54178aa9c43fda2279e46 : SUCCESS in 1h 28m 39s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/cf8884aabd0f4c8ba9d4feed0610cc92 : SUCCESS in 1h 00m 09s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/714d7758bef645f698cd9f459d053ac4 : SUCCESS in 1h 27m 49s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fe079f3ba5b04a3391e2aff457e9749b : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":5},{"id":"441eeb4011eed5a958e89087a4db396011a3b3fd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-10-08 22:48:29.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"884a9fd1f0f359d3a5717aa40f6a3d2e3463e280","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-10-08 22:48:32.000000000","message":"Patch Set 6:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6},{"id":"99db1371cb39ed7b423f9c55cc8815968275c4ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 23:09:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc021a94ecb1480b961992dcd0578e54\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8be6531ae1534f5e953c14c31662bdea : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/64305a3c051a4e658a50c9f411c1c5cb : SUCCESS in 14m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/22a57ba2074f401da9a7c0d39b5df2c3 : SUCCESS in 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9e727726300590d3b5844f4bcf8920557a5c3fdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 00:29:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e21d43aad5c4559a18395d39573f62f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0253a07e13ca49b5b2ddc18cbb82208e : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d078a11e1fe2454ebfc4b23d2643d55b : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f046dc2ba66476ead3f0487045cb518 : SUCCESS in 5m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a933144840fe40888d7bfa1e1a62e510 : SUCCESS in 6m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/933b68de86334aa4b53dd27450e0fbbc : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/755bd3ce99954c3aac31ff274df28b3b : SUCCESS in 5m 53s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/6feacf9a4eda489c9d3ec4cf27479dcf : SUCCESS in 46m 29s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/7039ad395e8e4e2b8edd12aee1f4fa66 : SUCCESS in 1h 26m 39s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/51af26d9a72f46efa5b54c1e70d3b15d : SUCCESS in 1h 13m 55s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/a385b7a888a948d7b836c8e1ab87b5c7 : SUCCESS in 1h 39m 35s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/eaac160bfaf44cdeb0a65dd7648c5b09 : SUCCESS in 42m 56s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/47a4adf4e7d34702a8f565c81446372d : SUCCESS in 1h 18m 17s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/ab0c4618c68c4af9b4b7c61b2aeee4ef : SUCCESS in 1h 06m 38s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/d2a815e136fc4b61b57ed82bdf7751ae : SUCCESS in 1h 21m 01s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/a79e77bcf8464015853fb4a3721bd1db : SUCCESS in 42m 51s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/6c7aa68bcabc49bf848f1c312201c566 : SUCCESS in 1h 22m 24s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/73e92f6ceaa44ab8b031b4400f03aace : SUCCESS in 1h 06m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/13eec74eb09c4aebb71a06a98cd6542b : SUCCESS in 1h 24m 46s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9d27e2b9b8a8480a8bd66cf2d6fa1327 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":6},{"id":"3ab44d9a0c4300090c298988c366e23bd7062f1a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-12-02 06:38:15.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"99be63da15c37e2e3a4e2daa7ca4761c206f35c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-02 07:27:35.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"85115334cac14878796ba23ed0753de086adced1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-02 07:27:54.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a38f22e58c2e0e5a51f8c1aec111412a0ff713e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 07:45:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/623a4bbe281e4a4c947cc625e6935acc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b98240bf53e34c1f984488cbe4ed3a2f : SUCCESS in 11m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/705366d4fc074f1f8872a44a32b681cb : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aee2418f3b244794b8ab70654c759115 : SUCCESS in 15m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ec4a0348eeb8a6dbc05650a3e078505776fe903b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 09:09:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/794d861c15264576abfab055102d8667\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/152e572c84cf4cba99a722290d6ba769 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bca07afa7d54a7292a3dca08f56ec1c : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9eba0ddb49964c0688ec64161f33ee5b : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/514f6a6199784267add84eae04d66392 : SUCCESS in 9m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e0dee14409c414dbe87029539e9cb44 : SUCCESS in 6m 50s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/c9f5863e93e1472bb0cf7883eaee2337 : SUCCESS in 44m 50s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/e3730fe3216245a9bc87d3b52b4a2fe4 : SUCCESS in 1h 40m 50s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/a60fc7e411dc4c8a92984dbcdf2c56a0 : SUCCESS in 1h 28m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/e55707536b3a4116a9b861ead6852cb8 : FAILURE in 1h 33m 13s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/5a96456b12264ad489ec2739e21a9ee8 : SUCCESS in 43m 51s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/2ce0a620066c4de99c6d245b28bad02d : SUCCESS in 1h 32m 40s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/2e203b8a3acb4e6c8056def6d6a51893 : SUCCESS in 1h 17m 25s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/4778dc0ab0a042e88e2de7387a2571bd : SUCCESS in 1h 25m 35s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/3a6bd9bbd11243fc9db47d4d8a1207dd : SUCCESS in 46m 22s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/a42a7220d0b041ca9193dcfe87c222c8 : SUCCESS in 1h 32m 12s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/a58b758c53234107afa24c4a4a1d321c : SUCCESS in 1h 14m 10s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/a1a019ded46c4d36a61397c25bfcba59 : SUCCESS in 1h 18m 13s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1d7bb28ded20444b9e6400f99cb3d992 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":7},{"id":"4ac65b0a9b4ca66ba0c3377f6fffa476a4c2e8a9","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2025-12-05 08:39:03.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"d53aa8f1d1ae86f9865790e914a824079330d94d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 08:39:25.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"09a89436dce3a09c519ecda6b5c7ae9523425531","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 09:27:46.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b840478076a448f48ed77e8af5b0e2ea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1451c6c04b3747fda743b6a58e52dbfa : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45578cd2bc334b678413174bb885a55e : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6277b6b23e6c48bf869fb6f34b4a2472 : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efaf78264cbc47008a0e167c1ab7cdbc : SUCCESS in 5m 14s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/423f3cc0c3a04be684ed54b12abdbb28 : SUCCESS in 45m 38s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/f6698254bc9f41cca83afe4d19c5e30e : SUCCESS in 40m 13s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/b44fb558c0824961ae11428e07027c09 : SUCCESS in 42m 59s","accounts_in_message":[],"_revision_number":7},{"id":"88a09a5e51f951ef47e262e1997cac28f8c0dd0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 09:27:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"f40985fded0dbad235a5ed4c856d6887e48a418f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 09:28:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f34eaa0ed948420590d697c10cabafa7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6f20c9dd5cc4feaa20a96f55077b054 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"5c8be98cfc1799059239c3ce4b87be432e09dd8d","revisions":{"217627c3c27fedbf08e27d291a07155109545b2f":{"kind":"REWORK","_number":1,"created":"2025-09-29 21:55:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/52/962552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/962552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/962552/1"}}},"commit":{"parents":[{"commit":"7ebcb39c9d948faa425040989f844ba712de6b2e","subject":"Merge \"Update master for stable/2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ebcb39c9d948faa425040989f844ba712de6b2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-29 21:55:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-29 21:55:18.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I909c628b96173a5633572909524926a24bc1596d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/217627c3c27fedbf08e27d291a07155109545b2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/217627c3c27fedbf08e27d291a07155109545b2f"}]},"branch":"refs/heads/master"},"d1dddf1087a92cf9271513217abb8dc2b5e9d8a4":{"kind":"REWORK","_number":2,"created":"2025-09-29 22:04:44.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/52/962552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/962552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/962552/2"}}},"commit":{"parents":[{"commit":"7ebcb39c9d948faa425040989f844ba712de6b2e","subject":"Merge \"Update master for stable/2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ebcb39c9d948faa425040989f844ba712de6b2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-29 21:55:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-29 22:04:44.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I909c628b96173a5633572909524926a24bc1596d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d1dddf1087a92cf9271513217abb8dc2b5e9d8a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d1dddf1087a92cf9271513217abb8dc2b5e9d8a4"}]},"branch":"refs/heads/master"},"b17ec8beb1791b8b49c7cdcc4b3a8f3ed9c56e34":{"kind":"REWORK","_number":3,"created":"2025-09-29 22:05:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/52/962552/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/962552/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/962552/3"}}},"commit":{"parents":[{"commit":"7ebcb39c9d948faa425040989f844ba712de6b2e","subject":"Merge \"Update master for stable/2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ebcb39c9d948faa425040989f844ba712de6b2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-29 21:55:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-29 22:05:58.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I909c628b96173a5633572909524926a24bc1596d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b17ec8beb1791b8b49c7cdcc4b3a8f3ed9c56e34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b17ec8beb1791b8b49c7cdcc4b3a8f3ed9c56e34"}]},"branch":"refs/heads/master"},"f9303bc58b5521bcf23373c2376540bd537acf23":{"kind":"REWORK","_number":4,"created":"2025-10-08 19:46:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/52/962552/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/962552/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/962552/4"}}},"commit":{"parents":[{"commit":"7ebcb39c9d948faa425040989f844ba712de6b2e","subject":"Merge \"Update master for stable/2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ebcb39c9d948faa425040989f844ba712de6b2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-29 21:55:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-10-08 19:45:26.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I909c628b96173a5633572909524926a24bc1596d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f9303bc58b5521bcf23373c2376540bd537acf23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f9303bc58b5521bcf23373c2376540bd537acf23"}]},"branch":"refs/heads/master"},"68585985e1d9d6885397c4a134554fb69441043f":{"kind":"REWORK","_number":5,"created":"2025-10-08 20:32:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/52/962552/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/962552/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/962552/5"}}},"commit":{"parents":[{"commit":"7ebcb39c9d948faa425040989f844ba712de6b2e","subject":"Merge \"Update master for stable/2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ebcb39c9d948faa425040989f844ba712de6b2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-29 21:55:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-10-08 20:32:11.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I909c628b96173a5633572909524926a24bc1596d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/68585985e1d9d6885397c4a134554fb69441043f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/68585985e1d9d6885397c4a134554fb69441043f"}]},"branch":"refs/heads/master"},"5545be8deec480016a7b270f124218d6c2ad11ce":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-10-08 22:48:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/52/962552/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/962552/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/962552/6"}}},"commit":{"parents":[{"commit":"7ebcb39c9d948faa425040989f844ba712de6b2e","subject":"Merge \"Update master for stable/2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ebcb39c9d948faa425040989f844ba712de6b2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-29 21:55:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-10-08 22:48:29.000000000","tz":0},"subject":"Push datastore image to quay.io to support tests","message":"Push datastore image to quay.io to support tests\n\nFor PostgreSQL 18\n\nChange-Id: I909c628b96173a5633572909524926a24bc1596d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5545be8deec480016a7b270f124218d6c2ad11ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5545be8deec480016a7b270f124218d6c2ad11ce"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5c8be98cfc1799059239c3ce4b87be432e09dd8d":{"kind":"REWORK","_number":7,"created":"2025-12-02 07:27:35.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/52/962552/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/962552/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/962552/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/962552/7"}}},"commit":{"parents":[{"commit":"7ebcb39c9d948faa425040989f844ba712de6b2e","subject":"Merge \"Update master for stable/2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ebcb39c9d948faa425040989f844ba712de6b2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-29 21:55:18.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-02 07:27:35.000000000","tz":0},"subject":"Push datastore image to quay.io to support tests","message":"Push datastore image to quay.io to support tests\n\nFor PostgreSQL 18\n\nChange-Id: I909c628b96173a5633572909524926a24bc1596d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5c8be98cfc1799059239c3ce4b87be432e09dd8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5c8be98cfc1799059239c3ce4b87be432e09dd8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
