)]}'
{"id":"openstack%2Ftrove~969925","triplet_id":"openstack%2Ftrove~master~I1d56d0ba1ca735c7c9514092cb753650c4a96378","project":"openstack/trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-12 17:48:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1d56d0ba1ca735c7c9514092cb753650c4a96378","subject":"Delay string interpolations at logging calls","status":"ABANDONED","created":"2025-12-05 13:52:49.000000000","updated":"2026-07-12 17:48:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7c6611307c5da48b124e191abe1ff64323130020","_number":969925,"virtual_id_number":969925,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-12-05 14:37:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-05 14:06:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-05 14:37:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4735e397223001921a906dc7d765205e9989db9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-05 13:52:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47212e7a2962743e4fa2d72d155aa58e15923fe1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 14:06:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9543c4783c3b4051aa6126bc29b76700\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1191e7cef4de4998952f5a49edca7a3f : FAILURE in 8m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bd011570cb9c43848fda4036e5a1eb5e : FAILURE in 7m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/85f1d55567bb4717a6f0aaf70ca9f6a9 : FAILURE in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ace2a355d6e4df9b404b9ac904e4469f00c96eb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 14:37:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9302ea8263594155935d5db334d5cbf4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0e52c07692954810a2660320955e0422 : SUCCESS in 4m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcab0ac006ca4f409348acb17baff17c : FAILURE in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5c3ab5d50d7433295573090aefa7c47 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8243a5d5b7a349678082568956e1ae02 : FAILURE in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f00abd4fb33c49a5b4687216ad4e289f : FAILURE in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e72582cd597c4624a7e7f80d3d8ef1e4 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6aba8c4e3414a559a6020176c06f2e6 : SUCCESS in 6m 11s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/8b16e0cb0b8f40a9a2cb1872ab5641b2 : FAILURE in 29m 34s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/d4d4cbca2ef44495b10747a130d4925b : FAILURE in 36m 09s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/80197f3b7c0b477bb44e52f7a1928013 : FAILURE in 35m 28s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/77bcb96b4710447686398eb4ebbe1795 : FAILURE in 30m 36s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/b2ce10dc453a4448af435f1029fdd95a : FAILURE in 31m 17s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/5c279e42b1464a4caf3522bdd593c2c9 : FAILURE in 40m 39s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/f908554166bc4b3382ba46bdd1614c47 : FAILURE in 39m 26s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a88953f73e1044eeb650f573c13d6379 : FAILURE in 29m 55s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/101eb692fc9c462691c7a73fdec338ee : FAILURE in 29m 39s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/b17565e5147144b3b2806401cbe0b8b6 : FAILURE in 39m 37s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/4dfaf8b7f7314dd79a685a678abbc19c : FAILURE in 39m 23s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/5475ce7d00a54a338b964f0d7ca0153b : FAILURE in 28m 38s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9f217b835684473587437cae290c28fd : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b08a6f2a45a76c43ab0a5a4e579c28c466bcaf33","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-12 17:47:57.000000000","message":"Topic H904 removed","accounts_in_message":[],"_revision_number":1},{"id":"7c6611307c5da48b124e191abe1ff64323130020","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-12 17:48:30.000000000","message":"Abandoned\n\nreplaced by https://review.opendev.org/c/openstack/trove/+/996948","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"db326196649a8fd26a58a8a8e590547677df9e7b","revisions":{"db326196649a8fd26a58a8a8e590547677df9e7b":{"kind":"REWORK","_number":1,"created":"2025-12-05 13:52:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/969925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/25/969925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/969925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/969925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/25/969925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/25/969925/1"}}},"commit":{"parents":[{"commit":"d932f98308fd39597a92b5fd88b773071699bbb5","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d932f98308fd39597a92b5fd88b773071699bbb5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-05 13:52:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-05 13:52:35.000000000","tz":540},"subject":"Delay string interpolations at logging calls","message":"Delay string interpolations at logging calls\n\nThis is a general best practise which brings (slight) performance\nimprovements and also structured logging.\n\nEnable the hacking check to enforce it.\n\nChange-Id: I1d56d0ba1ca735c7c9514092cb753650c4a96378\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/db326196649a8fd26a58a8a8e590547677df9e7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/db326196649a8fd26a58a8a8e590547677df9e7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
