)]}'
{"id":"openstack%2Ftrove~970789","triplet_id":"openstack%2Ftrove~master~I09d3f44e436ba543777551ac66e0be2d619c6377","project":"openstack/trove","branch":"master","topic":"bug/dont-remove-configuration-on-instance-update","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-01-12 06:22:40.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2026-01-12 07:18:10.000000000","reason":"Change was submitted"},"31737":{"account":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"last_update":"2026-01-12 07:18:10.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-01-08 13:14:32.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-08 13:14:32.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I09d3f44e436ba543777551ac66e0be2d619c6377","subject":"Fix: Prevent config removal on instance update","status":"MERGED","created":"2025-12-12 05:15:08.000000000","updated":"2026-01-12 07:19:08.000000000","submitted":"2026-01-12 07:18:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"970789-bug/dont-remove-configuration-on-instance-update","meta_rev_id":"cb0c61c94122ab2707aaed9b4e49aaf09447e917","_number":970789,"virtual_id_number":970789,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-12 07:18:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-12 01:51:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2026-01-12 04:47:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-12 01:51:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-12 05:39:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-12 06:58:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-12 12:01:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-12-12 12:01:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"state":"REVIEWER"},{"updated":"2025-12-12 12:01:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-12-18 07:38:21.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"}],"messages":[{"id":"3de633f34b7011898a068f1cec223ebae33ba296","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2025-12-12 05:15:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d85869d326c45c90900fa6d1daf4957157f7acde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 05:39:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4eafd424dcbe4a59b970b9fa590cf7b2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3c5186f8725d4d888dc06917fddc8f0b : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/89151ba854c54c2a843377cd924c93fd : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ab53d115c3f8441ca1ea5823ee0e8234 : SUCCESS in 10m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"052d432a753dee31a381ba2676ed71a8b8eb5735","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 06:58:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74f814c662324920ab85fcffed2b87ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f922a83837524248bd82f9121d6e2de6 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9f061b95ddb426aa44177a040364b35 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80a485848052445f8c658180fdde8b77 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c81308e2e8274e1da56a981e5a5ba286 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73f3a4a52949499f888b783048705515 : SUCCESS in 2m 49s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/89166915aa634fb397a4c0a2a1376e7f : SUCCESS in 48m 09s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/bed13d4453814741afce7541732a3ac6 : SUCCESS in 1h 38m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/f43bb6397f71448ca36d2e2c8f57a4f1 : SUCCESS in 1h 27m 48s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/9b1abcee96b6415ba48d135ae841f0b0 : SUCCESS in 1h 40m 16s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/fc477223fc5a4a39920bce78bcaa5472 : SUCCESS in 43m 30s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/fc48f3890a044965bf184a0909cc779b : SUCCESS in 1h 36m 11s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/e89a83613129497bac101528f6301585 : SUCCESS in 1h 15m 07s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/4695f45dde75400ab42f8490308a4674 : SUCCESS in 1h 29m 43s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/01836069f8d14c9db670c7214a845127 : SUCCESS in 42m 57s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/90e604760d2f49d2a4bfb8123d1d2f16 : SUCCESS in 1h 29m 41s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/735747043a324f89afae423ef53a5584 : SUCCESS in 1h 13m 17s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/03fff735eadc448e9049f7e46019e8bf : SUCCESS in 1h 19m 26s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2949fa6f4e9547e68af3a0804bc8671f : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5fd5c2da9ed88a5d33dedf2809728c47d64fb3f8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-12 12:01:19.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"346ff174db523a1871db07d0b1a24a7b19a7e87b","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2025-12-18 07:38:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30dfe09e9c84e8ce1c11ecd1c18c3da2769a2291","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 13:55:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"414c574723dcf421bd158684b0e33f9a039c9416","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-06 14:03:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"801801446429f6cd814c664bda90228edaae1b03","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-07 14:12:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14ab8a09d836381efac5cea0d904964e1bf599e2","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-07 14:31:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00a965f30a8b00f6d5d1b78e0cc0e6f22776040d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-07 15:19:25.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a571dbb1621999398cab6f2f89e7795c50ccbfdb","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-07 15:21:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"374f6f06f4ce8795de5fc54ef740b144a2b51ec6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 15:38:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74b4e6da50e74a59829186020183d07c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8902d93719194acc9f50e519fbda970c : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f73893197443475f9dd31ea85649c4fb : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f05e1547974f40baa896257810beec7a : SUCCESS in 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a3530229c4474a330282f020e50eee281040efaa","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-01-07 15:46:57.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"36e566f1a953fff55b0b0488c882993f2770fc68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 17:11:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d83a1cdec25e46a1889dda4c9c5f70b4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7097b2295c454f7c852896c05a1c2156 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0647fd4a02a243eb9648120120683d51 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/864381704d594297be4471c51297a688 : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd802f5d4ebd427db5f97daa00eaa68a : SUCCESS in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e322a06eb6f9457fb6a45feaca8647b7 : SUCCESS in 5m 02s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/cb3853ec61454e299dbd7e874fd2e85c : SUCCESS in 43m 05s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/946871048c264501a256626271d99b0b : SUCCESS in 1h 37m 37s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/611d76aac78846f6bc788125cbcc2dd5 : SUCCESS in 1h 40m 54s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/825ee07ea70245ac9d20399dc9cf9c7c : SUCCESS in 1h 40m 47s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/64ef1613e867473cb775642f81417ed6 : SUCCESS in 47m 39s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/1ca893a2ea5346c89228e8543a7e0f11 : SUCCESS in 1h 38m 16s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a20d62d266db42c994113b6314b2c4d2 : SUCCESS in 1h 22m 00s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/5d52edb65be641bcb613551f92000e0d : SUCCESS in 1h 21m 25s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/239c596db7d14c8e9c856ee96cea5af4 : SUCCESS in 48m 31s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/f3314b61fca649e2b446e7fab1552c74 : SUCCESS in 1h 40m 40s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/67846f1bbd7742d58f401618c4fb7af9 : SUCCESS in 1h 18m 47s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/b76657d5981c4f6fa0f97fd85d8d3fc5 : SUCCESS in 1h 26m 40s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e5acf83e0e7548c68ee39b3fc74cc125 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":2},{"id":"3041371fe9aabb810ba0943fdf367c60df87806b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-08 01:38:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b567724a55cdc823c7f45bea06c99559b82b901","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-01-08 03:27:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"68963efb5ff7d340891919637bd1af7fc07b5901","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-08 05:52:31.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"86de1a6da265934b1276443af66e2aeff66a10d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-01-08 10:24:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e4ed860da00173f4bcdebca9b8d4610c59a5c392","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 10:35:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b09cf1703b554fae842bf3e99fcffe68\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42177471af11481bb0ac41c9fcc0589b : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3f28a0631b394491ab674590d40215b3 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/03fb5c5e48a44599972d243ba9df3a67 : SUCCESS in 8m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aa850e3291486f8980e5a197e06265eabba81a3e","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-01-08 10:37:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a816eb2cc1a842d57b435b6b4c2fc5bdf55d6e26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 12:06:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/abce505c3a214dd489b921a5963a8310\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c115f9fceee5460d92c266c1e2696ab3 : SUCCESS in 4m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03a01bb25a3b481fb25223aac4bd86f4 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1ceb68ca28a4451a2bc11e81d59eb2d : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a1fecbd913d4a3da1f26f71131bb89e : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a9818c25f21497e8e799c5f138b2fbd : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd781226640d45b8a8df3579cad57759 : SUCCESS in 5m 57s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/a3a6716c9071403d9aedaaefa7f551ec : SUCCESS in 47m 54s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/6a7df9243be94665baf1916101040972 : FAILURE in 26m 20s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/c2af175a2e364ba583d6a52734df65bf : FAILURE in 27m 19s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/1de6ffe994344b7e92753d116248e030 : SUCCESS in 1h 37m 52s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/20924c4969c44d75a13a8f9b73fff04e : FAILURE in 20m 59s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/fd2ae1611d984ff99955bb4cd309fa07 : FAILURE in 26m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/91682ee2e05f43bf85da4fa2f4a9f64b : FAILURE in 19m 52s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/edda43827dc14140b88ed6cf69f57a5f : SUCCESS in 1h 32m 57s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/19af4ecf9c5541d1bf1c839ce73b3683 : FAILURE in 17m 42s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/1153a6237e564ec7b26ca2d3371e45d3 : SUCCESS in 1h 40m 36s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/03db260eb5ed475fa37b8162229a72d3 : SUCCESS in 1h 18m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/1311e1b96f2944c398ab5baad5f6862e : SUCCESS in 1h 24m 53s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fcf0e561fde94aeaa9d02ee7da597e16 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":3},{"id":"95ae461a209dc2234510002d666da613cd525eff","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-08 13:14:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"595bafc5a59e2246711b66fc47edbec52ac06285","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-08 13:14:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"141e95cd34ad08e3eef7382e9a1a0a0a53b225c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 13:27:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d62ce5d8086e4baca7d84a905378066e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4d227feb306c40bdb9669fc556168a13 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5833db217c564ae18f9a091906f7938c : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5a613ba5c15943a48119a4fcda64ddbb : SUCCESS in 8m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0a106bbf7cca8e4d907cb27117ba205f7f6d01e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 15:10:25.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f16f62b1bdb4eaf80470045d37ed171\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eea276f1cf0c4323921c6a6b2d7b4cf3 : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a09b132e6e88449eae668aff0981adcc : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6a6e1a2781f4de2ac7f30cbe13d2ffa : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a64b4dc42e7469d9fb75d724bcd9dac : SUCCESS in 5m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b23c8ead90cf40b286ee0d57371c2bb3 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab62188a19284af0ab15f1327c3cb7f7 : SUCCESS in 5m 09s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/7250722fbd4f41d180c8334cf72d93d2 : SUCCESS in 51m 24s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/449c0f17fc424554b6c1d068c8a58d5b : SUCCESS in 1h 52m 50s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/337ecb7143c8461480700a0ccdc48c34 : SUCCESS in 1h 39m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/5ff44a7fcae24663aa68286d6b841647 : SUCCESS in 1h 48m 07s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/a5909895d4a74a66a1f0ad99f6806bf7 : FAILURE in 25m 26s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/131609c657224dc5b5c6254aaa744edd : FAILURE in 51m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/3af074f2aa0240f9a12d3d03f60e9769 : SUCCESS in 1h 24m 48s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/63c3533763b247f8bd07b594d2b7aa00 : SUCCESS in 1h 36m 46s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/3871ab7f521e45fdafcfa126973b9f42 : SUCCESS in 48m 52s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/d8eefc53e2c849ab9dd15012724add2e : FAILURE in 24m 35s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/632dc35af0054889b87c262b4ea84b51 : SUCCESS in 1h 21m 05s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/d1e1c5f4cce64b189f74b8a76436c560 : SUCCESS in 1h 18m 39s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2390e18b00ba4f4cbf41594ad68dddfa : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":3},{"id":"8ecb2d0b6d88bbdac15519ce7062d20c4853d508","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-11 11:28:51.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_36080\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}],"_revision_number":4},{"id":"75317c4c9d81e1c993e2cffde19136b352564314","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 11:46:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6e1608f179a4833bd902d49f0704597\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4b2e4671d67450bac8103c2dd016141 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4d1373ceb74c445bbca656cfabe96bdb : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/88033aabc97c45f1bac9ac0f0e2bfe6d : SUCCESS in 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"42b61eba3df6ac88f01b5ce5de5f52c7a49c8228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-11 14:15:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed53f4f664e7413e946bc52c26ef7242\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/33632e8654a3452da6d6a9b54b499308 : SUCCESS in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddf8e7858bc8460a9a445fe92ec33cbf : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/674c339177684f7e85f487b1910e7e81 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4914a66b8564e4386918841ba3ef051 : SUCCESS in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7dcd31b255334a8381c10e711b0b6c13 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b34ad02db8a4f8aa905dd5bac1ec98c : SUCCESS in 5m 29s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/7f49d002ce35438da5d4d25b66ab0aee : SUCCESS in 1h 06m 17s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/ad6f020a8b2a44bd8849b177b3bd3f0d : SUCCESS in 2h 45m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/0759d104b40d42dd8692d6f9820d1ff7 : FAILURE in 2h 04m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/4e3afe1a76934a599d6e6cc86f894597 : FAILURE in 50m 59s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/320af7ee7f2d45d995ddc61385cf5080 : SUCCESS in 40m 46s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/f262d9a585b0410ebaa1e28d7296986a : SUCCESS in 1h 40m 22s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/acebc8c092da4774b67d687977eb2103 : SUCCESS in 1h 16m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/9fc86780d05f438e928bda5005d502fa : SUCCESS in 1h 28m 25s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/ba072320cf9b45d890d90d60f97d3610 : SUCCESS in 45m 35s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/776eb95ed5054121a9c2e7cb8838cc46 : SUCCESS in 1h 26m 26s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/4065dd15cd16403993f9da046ea84709 : SUCCESS in 40m 31s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/04fb15482fd34161b794d01c3ee94c34 : SUCCESS in 1h 17m 21s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6d52b5572ea248b58066bd65c8d96bb0 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":4},{"id":"10a78468714ed7a5e09c2656957264acbab89b50","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-01-12 01:51:46.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"79ee17d66d3f9af3e47d60b8a0dadd661c22d43f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 01:52:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d4c500ae294aa4a48c9ffe39d8162a5c86e1b72f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 02:51:08.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/989863ba034a43ec933a5173f6cabf49\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/58e3f8de7956425689cdc6c4086af2ce : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f15f2c05c504f638dd969680b0f73ac : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fdb1682950464e97b840df5699f4fc8f : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f4bac32db9041759331d6629d6459f0 : SUCCESS in 12m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe1a961a23eb4fd5a1376b7f46b4be70 : SUCCESS in 4m 59s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/636d16f786b24b34931be87936a2b93e : SUCCESS in 32m 53s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/f79dbb1370d1427690d42fbfdcae70b7 : SUCCESS in 41m 44s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/11a755afd28f49dea88377a0427933df : FAILURE in 54m 05s","accounts_in_message":[],"_revision_number":4},{"id":"2c7b5902382ce568511048fed39041da7ef3293a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 04:47:21.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4aa9d5fa1ef06404ad1e9be9a70e69dbc776ec31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 05:03:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e4de2608295456783bdc6be9f4e8fb9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e9a63933f9f44e53a4644751ac3cf9d6 : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dc5ae1350abc4f88bdd163e14324fa91 : SUCCESS in 11m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/60061f9ada0641fca1392255525f27a9 : SUCCESS in 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"86cf84f5225cfc78c7e2d8ba72e472c5e969a3c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 06:33:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e851dde2d93642138e509945883a7b1e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5fc0b39dd7534e139a4a2e5c08e2317e : SUCCESS in 4m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bf3836e58b44f8d91b0464304e566b3 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcd4f4d92f114b8db9b9b626cf3fd6f4 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ddbf2ce0bef4e5e9a07974c610aaba4 : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/54794d82cb714fcd899bb101a711479c : SUCCESS in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f79b1668b5aa48009a91c2ea9332b7d0 : SUCCESS in 4m 22s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/0191457112d4419bb73a12e313febc35 : SUCCESS in 53m 10s\n- trove-tempest-ubuntu-backup-mysql8.4 https://zuul.opendev.org/t/openstack/build/473307ed83a84d57b14d63e86038baa9 : SUCCESS in 1h 45m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mysql8.4 https://zuul.opendev.org/t/openstack/build/761b1c2ca6ea4e1281911001a71de651 : SUCCESS in 1h 27m 35s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql8.4 https://zuul.opendev.org/t/openstack/build/a1c13d389b4348d9a3f6c32d70c65070 : SUCCESS in 1h 42m 12s (non-voting)\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/725e0ae18d3f4d5c8cabf38755d66281 : SUCCESS in 43m 49s\n- trove-tempest-ubuntu-backup-mariadb11.4 https://zuul.opendev.org/t/openstack/build/bb2ebb16de834dc3816fb91c113926f1 : SUCCESS in 1h 39m 57s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb11.4 https://zuul.opendev.org/t/openstack/build/e730be23b1154cb3a264791624f8615c : SUCCESS in 1h 12m 36s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb11.4 https://zuul.opendev.org/t/openstack/build/f50ceee39c0d4705addacf5551843dd7 : SUCCESS in 1h 29m 18s (non-voting)\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/fdc1bec32caa459280a66e3c0d2df85d : SUCCESS in 44m 19s\n- trove-tempest-ubuntu-backup-postgresql17 https://zuul.opendev.org/t/openstack/build/8e0a9c00c21f418bba020e3007f5df64 : SUCCESS in 1h 36m 02s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql17 https://zuul.opendev.org/t/openstack/build/37866f987a3e48469da115ce9f58b541 : SUCCESS in 1h 10m 45s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql17 https://zuul.opendev.org/t/openstack/build/e4782794df774e4dacdabff0aa55afe9 : SUCCESS in 1h 19m 23s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eee0edeeb9484c13a7bf26a81e4fef97 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":4},{"id":"fcb0c98e9684df4329722175a995858c541153a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 06:34:05.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"87cd6eb284a6d218f57ba0659cb2a7883a770d07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 07:18:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19daadacb7e54b1e921a39f827f09099\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/62623ea84a054d5da64cc7225cedc250 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/103e8d1442ea43388ec6eb829243336c : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b934f2e06f8843ad98625c43fa7b47eb : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f047d54e58714a208cfcba54cd62cfe7 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bee88e21dfbb41c989232ebb0cd38b54 : SUCCESS in 6m 23s\n- trove-tempest-ubuntu-base-mysql8.4 https://zuul.opendev.org/t/openstack/build/a93ccde1878c4ed38a9d012d155a5567 : SUCCESS in 43m 38s\n- trove-tempest-ubuntu-base-mariadb11.4 https://zuul.opendev.org/t/openstack/build/2d33fc6fcca845a5a7a9f2b817de4694 : SUCCESS in 39m 57s\n- trove-tempest-ubuntu-base-postgresql17 https://zuul.opendev.org/t/openstack/build/d340c0bb360b4849884f752d80772a0a : SUCCESS in 40m 48s","accounts_in_message":[],"_revision_number":4},{"id":"dc9e97d83fa1f3ff6f55431e1ae77bdb21c0c269","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 07:18:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"cb0c61c94122ab2707aaed9b4e49aaf09447e917","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 07:19:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4eec160e53b94755a0ec57b3ee5b098b\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/2a1c65e7919e48fe9c714d50ead53468 : SUCCESS in 46s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c412e0371e7d44bea6ead9bc335e325c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fd7186bdcc194b8965fd16a6c7e32eb93bb91061","revisions":{"9f627a5ef1f257527e393b2dd5b8443626a9f1d6":{"kind":"REWORK","_number":1,"created":"2025-12-12 05:15:08.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/89/970789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/970789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/970789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/970789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/970789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/970789/1"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 05:14:27.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 05:14:27.000000000","tz":300},"subject":"Fix: Prevent config removal on instance update","message":"Fix: Prevent config removal on instance update\n\nAn unexpected side effect was causing configuration to be removed\nwhen an instance method was updated (e.g., changing access or name).\nThis patch resolves the issue by ensuring configuration persistence\nduring instance updates, preserving expected behavior.\n\nUnit tests have been added to validate that instance operations\nno longer inadvertently modify or delete existing configuration.\n\nChange-Id: I09d3f44e436ba543777551ac66e0be2d619c6377\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9f627a5ef1f257527e393b2dd5b8443626a9f1d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9f627a5ef1f257527e393b2dd5b8443626a9f1d6"}]},"branch":"refs/heads/master"},"5f49cc6c62d517c7d794017a9402c431177ab3ab":{"kind":"REWORK","_number":2,"created":"2026-01-07 15:19:25.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/89/970789/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/970789/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/970789/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/970789/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/970789/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/970789/2"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 05:14:27.000000000","tz":300},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2026-01-07 15:19:25.000000000","tz":0},"subject":"Fix: Prevent config removal on instance update","message":"Fix: Prevent config removal on instance update\n\nAn unexpected side effect was causing configuration to be removed\nwhen an instance method was updated (e.g., changing access or name).\nThis patch resolves the issue by ensuring configuration persistence\nduring instance updates, preserving expected behavior.\n\nUnit tests have been added to validate that instance operations\nno longer inadvertently modify or delete existing configuration.\n\nChange-Id: I09d3f44e436ba543777551ac66e0be2d619c6377\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5f49cc6c62d517c7d794017a9402c431177ab3ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5f49cc6c62d517c7d794017a9402c431177ab3ab"}]},"branch":"refs/heads/master"},"2509581457d9a13c8aa25346e634bbc3fe6e4572":{"kind":"REWORK","_number":3,"created":"2026-01-08 10:24:41.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/89/970789/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/970789/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/970789/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/970789/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/970789/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/970789/3"}}},"commit":{"parents":[{"commit":"3ec0c5f8dd634525e2606ec1945a070cb69690fd","subject":"Merge \"Return X-OpenStack-Request-Id header in response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ec0c5f8dd634525e2606ec1945a070cb69690fd"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 05:14:27.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-08 10:11:42.000000000","tz":300},"subject":"Fix: Prevent config removal on instance update","message":"Fix: Prevent config removal on instance update\n\nAn unexpected side effect was causing configuration to be removed\nwhen an instance method was updated (e.g., changing access or name).\nThis patch resolves the issue by ensuring configuration persistence\nduring instance updates, preserving expected behavior.\n\nUnit tests have been added to validate that instance operations\nno longer inadvertently modify or delete existing configuration.\n\nChange-Id: I09d3f44e436ba543777551ac66e0be2d619c6377\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2509581457d9a13c8aa25346e634bbc3fe6e4572"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2509581457d9a13c8aa25346e634bbc3fe6e4572"}]},"branch":"refs/heads/master"},"fd7186bdcc194b8965fd16a6c7e32eb93bb91061":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-11 11:28:51.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/89/970789/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/89/970789/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/970789/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/970789/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/89/970789/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/89/970789/4"}}},"commit":{"parents":[{"commit":"35b336b43a8bf082288150e9d23a8a2be2a4a83d","subject":"Merge \"Fix: Allow float values in KeyValueCodec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/35b336b43a8bf082288150e9d23a8a2be2a4a83d"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2025-12-12 05:14:27.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-01-11 11:28:51.000000000","tz":0},"subject":"Fix: Prevent config removal on instance update","message":"Fix: Prevent config removal on instance update\n\nAn unexpected side effect was causing configuration to be removed\nwhen an instance method was updated (e.g., changing access or name).\nThis patch resolves the issue by ensuring configuration persistence\nduring instance updates, preserving expected behavior.\n\nUnit tests have been added to validate that instance operations\nno longer inadvertently modify or delete existing configuration.\n\nChange-Id: I09d3f44e436ba543777551ac66e0be2d619c6377\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd7186bdcc194b8965fd16a6c7e32eb93bb91061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fd7186bdcc194b8965fd16a6c7e32eb93bb91061"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
